)]}'
{"id":"openstack%2Fkolla~668286","triplet_id":"openstack%2Fkolla~stable%2Frocky~I98405e3fb32cbec2d9863817a913ed71c5597e15","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I98405e3fb32cbec2d9863817a913ed71c5597e15","subject":"Fixes ceph-osd deploys with loop devices.","status":"MERGED","created":"2019-06-29 07:38:47.000000000","updated":"2019-07-26 18:44:29.000000000","submitted":"2019-07-26 18:44:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668286-1564166669525-e397d3ca","meta_rev_id":"589ebc84b5080118ae24b1bd619406a855f7e947","_number":668286,"virtual_id_number":668286,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-26 18:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-26 11:40:24.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-26 17:48:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-07-26 17:48:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 10:59:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-25 10:58:44.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-07-26 11:40:24.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-26 17:48:25.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-26 18:44:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8757361714acb00dd2a29433561635475b9354c4","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-06-29 07:38:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fc78e3b11d32902d01736c561f7f34dd395e0164","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-29 08:17:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/86/668286/1/openstack-check/tripleo-build-containers-fedora-28-master/c7d5624/ : RETRY_LIMIT in 2m 39s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/86/668286/1/openstack-check/tripleo-build-containers-fedora-28-stein/3ee33e2/ : SUCCESS in 37m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e649158e78ed10a127b3ff9e1acdc81d4a0c7486","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 09:09:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/86/668286/1/check/kolla-build-centos-binary/fddc359/ : SUCCESS in 42m 48s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/86/668286/1/check/kolla-build-centos-source/b52d9a8/ : SUCCESS in 45m 23s\n- kolla-ansible-centos-source http://logs.openstack.org/86/668286/1/check/kolla-ansible-centos-source/dbfe9ac/ : SUCCESS in 1h 22m 43s\n- kolla-ansible-centos-binary http://logs.openstack.org/86/668286/1/check/kolla-ansible-centos-binary/1869817/ : SUCCESS in 1h 06m 20s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/86/668286/1/check/kolla-ansible-centos-source-ceph/5e8729e/ : POST_FAILURE in 3m 29s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/86/668286/1/check/kolla-build-oraclelinux-binary/4c0078d/ : SUCCESS in 47m 35s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/86/668286/1/check/kolla-build-oraclelinux-source/2eb8fe9/ : SUCCESS in 49m 23s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/86/668286/1/check/kolla-ansible-oraclelinux-source/0e41bc7/ : SUCCESS in 1h 07m 10s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/86/668286/1/check/kolla-ansible-oraclelinux-binary/43ba828/ : SUCCESS in 59m 27s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/86/668286/1/check/kolla-ansible-oraclelinux-source-ceph/58e1eff/ : POST_FAILURE in 1m 58s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/86/668286/1/check/openstack-tox-lower-constraints/902b143/ : SUCCESS in 3m 04s\n- openstack-tox-pep8 http://logs.openstack.org/86/668286/1/check/openstack-tox-pep8/d5caf3e/ : SUCCESS in 4m 23s\n- openstack-tox-py27 http://logs.openstack.org/86/668286/1/check/openstack-tox-py27/98f6390/ : SUCCESS in 2m 49s\n- openstack-tox-py35 http://logs.openstack.org/86/668286/1/check/openstack-tox-py35/f0ff490/ : SUCCESS in 3m 54s\n- build-openstack-sphinx-docs http://logs.openstack.org/86/668286/1/check/build-openstack-sphinx-docs/26e9023/html/ : SUCCESS in 3m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/86/668286/1/check/kolla-build-ubuntu-binary/7b67d85/ : SUCCESS in 51m 57s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/86/668286/1/check/kolla-build-ubuntu-source/6f899e5/ : SUCCESS in 28m 26s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/86/668286/1/check/kolla-ansible-ubuntu-source/35d7135/ : SUCCESS in 56m 15s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/86/668286/1/check/kolla-ansible-ubuntu-binary/52af433/ : SUCCESS in 1h 20m 37s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/86/668286/1/check/kolla-ansible-ubuntu-source-ceph/47d2c13/ : POST_FAILURE in 2m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"afb43a763c327078bdaf83288973ac89bde22ee7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-29 13:55:25.000000000","message":"Patch Set 1:\n\nFirst get change merged in master, then propose to backport to stable (starting with stein today).","accounts_in_message":[],"_revision_number":1},{"id":"f9e971fbdebc71e6b71d4a91303e3bd985e083b5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-29 13:59:36.000000000","message":"Patch Set 1:\n\nCI: rocky ceph is broken","accounts_in_message":[],"_revision_number":1},{"id":"31499105ef6d76732575b0201c24d23cee70154f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-29 13:59:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6c04fbb6f3abddcd5d43bda185ebf2adad6149ed","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-30 18:36:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d8b394ce55b08e095b361ca823280ccc9b4a4f4d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-30 18:55:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e0f374645632087e540376190af11c1e970de7be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 18:58:52.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"76da0fe9b84a6fc6b90dcbec1ba6e580092d90b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-30 18:58:52.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"3ffa72f38633ec0c9ff0b2feefb566d06f87181b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-30 19:30:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4c7da56a85f253a1ae1e539dcca8a7712d4d2172","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-30 20:13:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/86/668286/2/openstack-check/tripleo-build-containers-fedora-28-master/8716280/ : RETRY_LIMIT in 2m 31s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/86/668286/2/openstack-check/tripleo-build-containers-fedora-28-stein/f25495f/ : SUCCESS in 41m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3152c9b82151c5a85fe451a1efd962ef61b39ecd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-30 21:40:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/86/668286/2/check/kolla-build-centos-binary/31e37a0/ : SUCCESS in 36m 56s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/86/668286/2/check/kolla-build-centos-source/ac12486/ : SUCCESS in 34m 19s\n- kolla-ansible-centos-source http://logs.openstack.org/86/668286/2/check/kolla-ansible-centos-source/b910ae1/ : SUCCESS in 1h 15m 50s\n- kolla-ansible-centos-binary http://logs.openstack.org/86/668286/2/check/kolla-ansible-centos-binary/bc50654/ : SUCCESS in 1h 12m 47s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/86/668286/2/check/kolla-ansible-centos-source-ceph/1f144fd/ : SUCCESS in 1h 50m 50s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/86/668286/2/check/kolla-build-oraclelinux-binary/1d4f492/ : SUCCESS in 33m 35s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/86/668286/2/check/kolla-build-oraclelinux-source/528bf26/ : FAILURE in 3m 41s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/86/668286/2/check/kolla-ansible-oraclelinux-source/c5eac7c/ : SUCCESS in 1h 16m 26s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/86/668286/2/check/kolla-ansible-oraclelinux-binary/7e89bf4/ : SUCCESS in 1h 17m 26s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/86/668286/2/check/kolla-ansible-oraclelinux-source-ceph/447785a/ : SUCCESS in 2h 00m 31s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/86/668286/2/check/openstack-tox-lower-constraints/3b2c537/ : SUCCESS in 3m 36s\n- openstack-tox-pep8 http://logs.openstack.org/86/668286/2/check/openstack-tox-pep8/a8f3eef/ : SUCCESS in 3m 34s\n- openstack-tox-py27 http://logs.openstack.org/86/668286/2/check/openstack-tox-py27/5ea37c1/ : SUCCESS in 4m 13s\n- openstack-tox-py35 http://logs.openstack.org/86/668286/2/check/openstack-tox-py35/8c531f7/ : SUCCESS in 4m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/86/668286/2/check/build-openstack-sphinx-docs/b69b0d2/html/ : SUCCESS in 5m 28s\n- kolla-build-ubuntu-binary http://logs.openstack.org/86/668286/2/check/kolla-build-ubuntu-binary/2cb0363/ : SUCCESS in 1h 28m 58s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/86/668286/2/check/kolla-build-ubuntu-source/df7f36e/ : SUCCESS in 40m 20s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/86/668286/2/check/kolla-ansible-ubuntu-source/51a2d00/ : SUCCESS in 57m 27s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/86/668286/2/check/kolla-ansible-ubuntu-binary/8d488a9/ : SUCCESS in 1h 52m 24s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/86/668286/2/check/kolla-ansible-ubuntu-source-ceph/2b26d24/ : SUCCESS in 1h 18m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bb32f85ac8803e040da0a99c46e371da34fef338","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-01 05:00:47.000000000","message":"Patch Set 2:\n\nrecheck\n\nDocker repo failure","accounts_in_message":[],"_revision_number":2},{"id":"c5b95e21c60a19db412d8fb7d41a12a8bc2feb66","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-01 05:38:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/86/668286/2/openstack-check/tripleo-build-containers-fedora-28-master/580c01b/ : RETRY_LIMIT in 3m 02s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/86/668286/2/openstack-check/tripleo-build-containers-fedora-28-stein/99f9971/ : SUCCESS in 35m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e2a1dbef2e24c168a1c4958b98a962778c3a910d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 07:07:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/86/668286/2/check/kolla-build-centos-binary/bb2e1d0/ : SUCCESS in 21m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/86/668286/2/check/kolla-build-centos-source/0d535ee/ : SUCCESS in 56m 40s\n- kolla-ansible-centos-source http://logs.openstack.org/86/668286/2/check/kolla-ansible-centos-source/838414f/ : SUCCESS in 1h 15m 47s\n- kolla-ansible-centos-binary http://logs.openstack.org/86/668286/2/check/kolla-ansible-centos-binary/a6da8ea/ : SUCCESS in 1h 10m 46s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/86/668286/2/check/kolla-ansible-centos-source-ceph/bd6c820/ : SUCCESS in 1h 50m 19s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/86/668286/2/check/kolla-build-oraclelinux-binary/a2aeee5/ : SUCCESS in 25m 32s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/86/668286/2/check/kolla-build-oraclelinux-source/e2419e7/ : SUCCESS in 57m 47s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/86/668286/2/check/kolla-ansible-oraclelinux-source/58fa491/ : SUCCESS in 1h 11m 51s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/86/668286/2/check/kolla-ansible-oraclelinux-binary/1be4fbe/ : SUCCESS in 1h 04m 09s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/86/668286/2/check/kolla-ansible-oraclelinux-source-ceph/b1f0624/ : SUCCESS in 1h 49m 20s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/86/668286/2/check/openstack-tox-lower-constraints/0d173e0/ : SUCCESS in 3m 27s\n- openstack-tox-pep8 http://logs.openstack.org/86/668286/2/check/openstack-tox-pep8/6f5af22/ : SUCCESS in 3m 41s\n- openstack-tox-py27 http://logs.openstack.org/86/668286/2/check/openstack-tox-py27/0c8582b/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/86/668286/2/check/openstack-tox-py35/835e2eb/ : SUCCESS in 4m 50s\n- build-openstack-sphinx-docs http://logs.openstack.org/86/668286/2/check/build-openstack-sphinx-docs/fb4c84e/html/ : SUCCESS in 3m 34s\n- kolla-build-ubuntu-binary http://logs.openstack.org/86/668286/2/check/kolla-build-ubuntu-binary/e6c40ff/ : SUCCESS in 56m 33s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/86/668286/2/check/kolla-build-ubuntu-source/1788d8e/ : SUCCESS in 28m 22s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/86/668286/2/check/kolla-ansible-ubuntu-source/89cadeb/ : SUCCESS in 1h 03m 16s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/86/668286/2/check/kolla-ansible-ubuntu-binary/c4c0fd7/ : SUCCESS in 1h 25m 29s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/86/668286/2/check/kolla-ansible-ubuntu-source-ceph/9af4b4d/ : FAILURE in 37m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"156c237236d1b13f51bbe9f4daca67531ea2cb2b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-01 07:23:39.000000000","message":"Patch Set 2:\n\nCI issue irrelevant","accounts_in_message":[],"_revision_number":2},{"id":"70fe12c381120ad6545b9f750e7a942f0221dcf3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 11:17:42.000000000","message":"Patch Set 2: Code-Review-1\n\nLet\u0027s wait for master to merge.","accounts_in_message":[],"_revision_number":2},{"id":"2c9ea8085779fef0874e36610ce84d8e927eaae8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 08:35:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cbb37d49d78afb577348abd8ff5fc914a7bd1de5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 08:37:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ac4227148dad3719e7dbf7e0542581b3099235b8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-04 08:38:01.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ab32bb50dee097a4358f187cfc44538181279753","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 09:13:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/86/668286/3/openstack-check/tripleo-build-containers-fedora-28-master/57c7400/ : RETRY_LIMIT in 3m 41s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/86/668286/3/openstack-check/tripleo-build-containers-fedora-28-stein/04cf7ca/ : SUCCESS in 36m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"088d9e2f16c9d8671b2131e9a04dbd80953da8f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:49:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/86/668286/3/check/kolla-build-centos-binary/db06693/ : SUCCESS in 45m 43s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/86/668286/3/check/kolla-build-centos-source/ca6cccf/ : SUCCESS in 48m 02s\n- kolla-ansible-centos-source http://logs.openstack.org/86/668286/3/check/kolla-ansible-centos-source/02b4a94/ : FAILURE in 54m 08s\n- kolla-ansible-centos-binary http://logs.openstack.org/86/668286/3/check/kolla-ansible-centos-binary/2c59598/ : SUCCESS in 1h 09m 00s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/86/668286/3/check/kolla-ansible-centos-source-ceph/919dff7/ : POST_FAILURE in 3m 59s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/86/668286/3/check/kolla-build-oraclelinux-binary/200c40c/ : SUCCESS in 42m 10s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/86/668286/3/check/kolla-build-oraclelinux-source/e8b6b85/ : SUCCESS in 40m 56s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/86/668286/3/check/kolla-ansible-oraclelinux-source/7130e66/ : SUCCESS in 1h 11m 18s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/86/668286/3/check/kolla-ansible-oraclelinux-binary/63fcb52/ : SUCCESS in 1h 05m 24s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/86/668286/3/check/kolla-ansible-oraclelinux-source-ceph/6e2395c/ : POST_FAILURE in 3m 26s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/86/668286/3/check/openstack-tox-lower-constraints/c449c5e/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/86/668286/3/check/openstack-tox-pep8/1f8288c/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/86/668286/3/check/openstack-tox-py27/a094155/ : SUCCESS in 3m 19s\n- openstack-tox-py35 http://logs.openstack.org/86/668286/3/check/openstack-tox-py35/d6b4ab5/ : SUCCESS in 3m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/86/668286/3/check/build-openstack-sphinx-docs/65c46cb/html/ : SUCCESS in 3m 41s\n- kolla-build-ubuntu-binary http://logs.openstack.org/86/668286/3/check/kolla-build-ubuntu-binary/8990a5c/ : SUCCESS in 1h 27m 17s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/86/668286/3/check/kolla-build-ubuntu-source/d2d7d8b/ : SUCCESS in 31m 19s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/86/668286/3/check/kolla-ansible-ubuntu-source/7a9161d/ : SUCCESS in 1h 06m 55s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/86/668286/3/check/kolla-ansible-ubuntu-binary/570cfb3/ : SUCCESS in 1h 55m 41s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/86/668286/3/check/kolla-ansible-ubuntu-source-ceph/622ed24/ : POST_FAILURE in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"240dea3f15811d024d4548f256c07361634c6b65","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 10:59:48.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"dc1e8bc499715ea5c699e940d8f7daea0a3bc7c6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-04 11:06:42.000000000","message":"Patch Set 3: -Code-Review\n\n(unrelated) FAILURE -\u003e nova : Waiting for nova-compute service up (never made it)\n\n(ceph) POST_FAILURE -\u003e copy default ansible kolla-ansible inventory","accounts_in_message":[],"_revision_number":3},{"id":"ac98ca06e4b272e6d560574870c0fdb7b7cb7fa3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-17 12:20:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"32b70b50a4fac7aa3c62bb5fb44da38e623c5495","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 13:15:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/86/668286/3/openstack-check/tripleo-build-containers-fedora-28-master/c01eb67/ : RETRY_LIMIT in 2m 59s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/86/668286/3/openstack-check/tripleo-build-containers-fedora-28-stein/8d9b78b/ : SUCCESS in 52m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d508212c2859f875e7dbe3417a04296e6b39fe83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 14:34:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/86/668286/3/check/kolla-build-centos-binary/1ca62e0/ : SUCCESS in 38m 42s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/86/668286/3/check/kolla-build-centos-source/d9cd8fe/ : SUCCESS in 55m 52s\n- kolla-ansible-centos-source http://logs.openstack.org/86/668286/3/check/kolla-ansible-centos-source/97e12c1/ : SUCCESS in 1h 18m 53s\n- kolla-ansible-centos-binary http://logs.openstack.org/86/668286/3/check/kolla-ansible-centos-binary/3eab48b/ : SUCCESS in 1h 03m 50s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/86/668286/3/check/kolla-ansible-centos-source-ceph/6bbc358/ : POST_FAILURE in 4m 12s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/86/668286/3/check/kolla-build-oraclelinux-binary/991132f/ : SUCCESS in 32m 09s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/86/668286/3/check/kolla-build-oraclelinux-source/e250203/ : SUCCESS in 34m 41s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/86/668286/3/check/kolla-ansible-oraclelinux-source/eff672b/ : SUCCESS in 1h 13m 30s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/86/668286/3/check/kolla-ansible-oraclelinux-binary/3e5fb17/ : SUCCESS in 1h 06m 20s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/86/668286/3/check/kolla-ansible-oraclelinux-source-ceph/7cb9a57/ : POST_FAILURE in 4m 02s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/86/668286/3/check/openstack-tox-lower-constraints/808d79b/ : SUCCESS in 10m 12s\n- openstack-tox-pep8 http://logs.openstack.org/86/668286/3/check/openstack-tox-pep8/5495859/ : SUCCESS in 9m 55s\n- openstack-tox-py27 http://logs.openstack.org/86/668286/3/check/openstack-tox-py27/f938faf/ : SUCCESS in 4m 13s\n- openstack-tox-py35 http://logs.openstack.org/86/668286/3/check/openstack-tox-py35/51dcafa/ : SUCCESS in 3m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/86/668286/3/check/build-openstack-sphinx-docs/010a891/html/ : SUCCESS in 6m 39s\n- kolla-build-ubuntu-binary http://logs.openstack.org/86/668286/3/check/kolla-build-ubuntu-binary/72a05a6/ : SUCCESS in 1h 35m 02s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/86/668286/3/check/kolla-build-ubuntu-source/20a3756/ : SUCCESS in 33m 51s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/86/668286/3/check/kolla-ansible-ubuntu-source/19893c5/ : SUCCESS in 1h 07m 21s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/86/668286/3/check/kolla-ansible-ubuntu-binary/0dacefe/ : TIMED_OUT in 2h 01m 26s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/86/668286/3/check/kolla-ansible-ubuntu-source-ceph/52abc0e/ : POST_FAILURE in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09a33a4553a0808c35bee59d7069d45f7d69285b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-25 10:58:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d25b240c57fec3c8b07236a04f732d331dde875b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 11:49:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/86/668286/3/openstack-check/tripleo-build-containers-fedora-28-master/04cc853/ : RETRY_LIMIT in 3m 06s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/86/668286/3/openstack-check/tripleo-build-containers-fedora-28-stein/a705f30/ : SUCCESS in 49m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f8fe8ebf16cd1854fb48352fdd3a91177590ce3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 13:01:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/86/668286/3/check/kolla-build-centos-binary/62c475c/ : SUCCESS in 43m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/86/668286/3/check/kolla-build-centos-source/9da4b9e/ : SUCCESS in 53m 59s\n- kolla-ansible-centos-source http://logs.openstack.org/86/668286/3/check/kolla-ansible-centos-source/72ef794/ : SUCCESS in 1h 14m 48s\n- kolla-ansible-centos-binary http://logs.openstack.org/86/668286/3/check/kolla-ansible-centos-binary/f4d706c/ : SUCCESS in 1h 11m 29s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/86/668286/3/check/kolla-ansible-centos-source-ceph/4c90e43/ : POST_FAILURE in 4m 04s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/86/668286/3/check/kolla-build-oraclelinux-binary/5dc6153/ : SUCCESS in 40m 56s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/86/668286/3/check/kolla-build-oraclelinux-source/5ceb66a/ : SUCCESS in 43m 14s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/86/668286/3/check/kolla-ansible-oraclelinux-source/046ebb6/ : SUCCESS in 1h 14m 09s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/86/668286/3/check/kolla-ansible-oraclelinux-binary/050dea7/ : SUCCESS in 1h 02m 07s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/86/668286/3/check/kolla-ansible-oraclelinux-source-ceph/52f0ee2/ : POST_FAILURE in 3m 37s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/86/668286/3/check/openstack-tox-lower-constraints/eac7517/ : SUCCESS in 5m 34s\n- openstack-tox-pep8 http://logs.openstack.org/86/668286/3/check/openstack-tox-pep8/1afe01c/ : SUCCESS in 4m 33s\n- openstack-tox-py27 http://logs.openstack.org/86/668286/3/check/openstack-tox-py27/003d9df/ : SUCCESS in 3m 53s\n- openstack-tox-py35 http://logs.openstack.org/86/668286/3/check/openstack-tox-py35/f8c9fd7/ : SUCCESS in 6m 55s\n- build-openstack-sphinx-docs http://logs.openstack.org/86/668286/3/check/build-openstack-sphinx-docs/6fea98a/html/ : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary http://logs.openstack.org/86/668286/3/check/kolla-build-ubuntu-binary/61592da/ : SUCCESS in 59m 44s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/86/668286/3/check/kolla-build-ubuntu-source/b580f40/ : SUCCESS in 29m 33s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/86/668286/3/check/kolla-ansible-ubuntu-source/569c82a/ : SUCCESS in 1h 00m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/86/668286/3/check/kolla-ansible-ubuntu-binary/c6a794a/ : SUCCESS in 1h 56m 41s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/86/668286/3/check/kolla-ansible-ubuntu-source-ceph/b36a606/ : POST_FAILURE in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c544a129746a7908e5209892db29fa64f20c3d8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-25 14:41:20.000000000","message":"Patch Set 3:\n\nlet\u0027s first fix CI: https://review.opendev.org/672749","accounts_in_message":[],"_revision_number":3},{"id":"746719711a3380a98d91cb2456da7654f967142f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-25 16:17:30.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"34061a04f2bbe4fd302f2c1278fe153673feb68d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 17:08:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/86/668286/4/openstack-check/tripleo-build-containers-fedora-28-master/3efb41d/ : RETRY_LIMIT in 3m 35s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/86/668286/4/openstack-check/tripleo-build-containers-fedora-28-stein/d733db1/ : SUCCESS in 49m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8bdd4d59d450501e14f4cb689f632b218eb41d12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 18:25:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/86/668286/4/check/kolla-build-centos-binary/4f92c90/ : SUCCESS in 42m 08s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/86/668286/4/check/kolla-build-centos-source/5b28228/ : SUCCESS in 47m 19s\n- kolla-ansible-centos-source http://logs.openstack.org/86/668286/4/check/kolla-ansible-centos-source/3079d8a/ : SUCCESS in 1h 41m 37s\n- kolla-ansible-centos-binary http://logs.openstack.org/86/668286/4/check/kolla-ansible-centos-binary/5cfc547/ : SUCCESS in 1h 15m 41s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/86/668286/4/check/kolla-ansible-centos-source-ceph/8d268c1/ : FAILURE in 1h 03m 15s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/86/668286/4/check/kolla-build-oraclelinux-binary/0d3fc67/ : FAILURE in 29m 25s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/86/668286/4/check/kolla-build-oraclelinux-source/1ea1696/ : FAILURE in 55m 10s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/86/668286/4/check/kolla-ansible-oraclelinux-source/be683c6/ : SUCCESS in 1h 14m 41s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/86/668286/4/check/kolla-ansible-oraclelinux-binary/a512186/ : SUCCESS in 1h 07m 25s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/86/668286/4/check/kolla-ansible-oraclelinux-source-ceph/bf5ecac/ : SUCCESS in 1h 57m 04s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/86/668286/4/check/openstack-tox-lower-constraints/46db61b/ : SUCCESS in 4m 33s\n- openstack-tox-pep8 http://logs.openstack.org/86/668286/4/check/openstack-tox-pep8/8251d73/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/86/668286/4/check/openstack-tox-py27/237ea6a/ : SUCCESS in 6m 06s\n- openstack-tox-py35 http://logs.openstack.org/86/668286/4/check/openstack-tox-py35/72d39d4/ : SUCCESS in 5m 28s\n- build-openstack-sphinx-docs http://logs.openstack.org/86/668286/4/check/build-openstack-sphinx-docs/5ac3ab9/html/ : SUCCESS in 5m 10s\n- kolla-build-ubuntu-binary http://logs.openstack.org/86/668286/4/check/kolla-build-ubuntu-binary/050cb77/ : SUCCESS in 57m 46s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/86/668286/4/check/kolla-build-ubuntu-source/d869daa/ : SUCCESS in 32m 16s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/86/668286/4/check/kolla-ansible-ubuntu-source/c6eaace/ : SUCCESS in 1h 00m 06s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/86/668286/4/check/kolla-ansible-ubuntu-binary/cd689c6/ : TIMED_OUT in 2h 01m 33s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/86/668286/4/check/kolla-ansible-ubuntu-source-ceph/54fd9ae/ : SUCCESS in 1h 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5d5945259fc7e73ff752b2d893e6958504f03047","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-25 19:04:51.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d40df39992504d8aa7783a0191b2bbbdbec90e20","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 20:05:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/86/668286/4/openstack-check/tripleo-build-containers-fedora-28-master/95571f8/ : RETRY_LIMIT in 3m 07s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/86/668286/4/openstack-check/tripleo-build-containers-fedora-28-stein/e883bfc/ : SUCCESS in 54m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3848e9b9a3827962fa8aa4e4307f57cc97f8ef30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 21:08:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/86/668286/4/check/kolla-build-centos-binary/2d687e1/ : SUCCESS in 33m 16s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/86/668286/4/check/kolla-build-centos-source/fe17c82/ : SUCCESS in 41m 31s\n- kolla-ansible-centos-source http://logs.openstack.org/86/668286/4/check/kolla-ansible-centos-source/2cde41d/ : SUCCESS in 1h 47m 01s\n- kolla-ansible-centos-binary http://logs.openstack.org/86/668286/4/check/kolla-ansible-centos-binary/99f09af/ : SUCCESS in 1h 45m 26s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/86/668286/4/check/kolla-ansible-centos-source-ceph/4840dec/ : TIMED_OUT in 2h 01m 50s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/86/668286/4/check/kolla-build-oraclelinux-binary/2a86cda/ : SUCCESS in 34m 57s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/86/668286/4/check/kolla-build-oraclelinux-source/35c1e26/ : SUCCESS in 49m 28s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/86/668286/4/check/kolla-ansible-oraclelinux-source/94be912/ : SUCCESS in 1h 09m 17s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/86/668286/4/check/kolla-ansible-oraclelinux-binary/6c75c38/ : SUCCESS in 1h 06m 15s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/86/668286/4/check/kolla-ansible-oraclelinux-source-ceph/4bb0274/ : FAILURE in 1h 25m 44s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/86/668286/4/check/openstack-tox-lower-constraints/55feb03/ : SUCCESS in 5m 38s\n- openstack-tox-pep8 http://logs.openstack.org/86/668286/4/check/openstack-tox-pep8/281f2e2/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/86/668286/4/check/openstack-tox-py27/c1e0ab0/ : SUCCESS in 4m 48s\n- openstack-tox-py35 http://logs.openstack.org/86/668286/4/check/openstack-tox-py35/b8e3ce9/ : SUCCESS in 5m 13s\n- build-openstack-sphinx-docs http://logs.openstack.org/86/668286/4/check/build-openstack-sphinx-docs/2372de5/html/ : SUCCESS in 5m 37s\n- kolla-build-ubuntu-binary http://logs.openstack.org/86/668286/4/check/kolla-build-ubuntu-binary/d9ec086/ : SUCCESS in 1h 02m 16s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/86/668286/4/check/kolla-build-ubuntu-source/8bce040/ : SUCCESS in 33m 52s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/86/668286/4/check/kolla-ansible-ubuntu-source/a7d02de/ : SUCCESS in 1h 05m 35s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/86/668286/4/check/kolla-ansible-ubuntu-binary/cefa1be/ : SUCCESS in 1h 35m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/86/668286/4/check/kolla-ansible-ubuntu-source-ceph/26ad77d/ : SUCCESS in 1h 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5da70c694ba5471dbb0115ff714817aafaf51007","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-26 10:49:20.000000000","message":"Patch Set 4:\n\nrecheck\n\nOL \"failed successfully\" bootstrapping ceph loop osds","accounts_in_message":[],"_revision_number":4},{"id":"182894e6d3ab0754434cf26565a25051c5fc12d8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-26 10:49:36.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"5de3fddd3029b18dc3543a06f0cdc7ed6deff9ab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-26 11:40:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/86/668286/4/openstack-check/tripleo-build-containers-fedora-28-master/863f4ea/ : RETRY_LIMIT in 3m 08s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/86/668286/4/openstack-check/tripleo-build-containers-fedora-28-stein/4e0abbc/ : SUCCESS in 49m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"422843d478053d5fe30f4a651b1411cdb429b4d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 13:14:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/86/668286/4/check/kolla-build-centos-binary/1d2ab2f/ : SUCCESS in 38m 47s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/86/668286/4/check/kolla-build-centos-source/a0c9866/ : SUCCESS in 1h 00m 48s\n- kolla-ansible-centos-source http://logs.openstack.org/86/668286/4/check/kolla-ansible-centos-source/95010cc/ : SUCCESS in 1h 13m 40s\n- kolla-ansible-centos-binary http://logs.openstack.org/86/668286/4/check/kolla-ansible-centos-binary/b1207c8/ : SUCCESS in 1h 07m 44s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/86/668286/4/check/kolla-ansible-centos-source-ceph/b8acc47/ : SUCCESS in 1h 49m 26s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/86/668286/4/check/kolla-build-oraclelinux-binary/d03ddee/ : SUCCESS in 45m 50s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/86/668286/4/check/kolla-build-oraclelinux-source/9ce8f62/ : SUCCESS in 48m 01s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/86/668286/4/check/kolla-ansible-oraclelinux-source/7dfceed/ : SUCCESS in 1h 10m 46s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/86/668286/4/check/kolla-ansible-oraclelinux-binary/f221b39/ : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/86/668286/4/check/kolla-ansible-oraclelinux-source-ceph/3d258b4/ : SUCCESS in 1h 39m 56s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/86/668286/4/check/openstack-tox-lower-constraints/bb22682/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/86/668286/4/check/openstack-tox-pep8/c99598b/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/86/668286/4/check/openstack-tox-py27/a045858/ : SUCCESS in 4m 25s\n- openstack-tox-py35 http://logs.openstack.org/86/668286/4/check/openstack-tox-py35/1fa8e74/ : SUCCESS in 3m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/86/668286/4/check/build-openstack-sphinx-docs/0665a2b/html/ : SUCCESS in 3m 44s\n- kolla-build-ubuntu-binary http://logs.openstack.org/86/668286/4/check/kolla-build-ubuntu-binary/596d24a/ : SUCCESS in 1h 25m 12s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/86/668286/4/check/kolla-build-ubuntu-source/f29ec01/ : SUCCESS in 31m 15s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/86/668286/4/check/kolla-ansible-ubuntu-source/bf2803b/ : SUCCESS in 1h 01m 33s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/86/668286/4/check/kolla-ansible-ubuntu-binary/b85954b/ : SUCCESS in 1h 58m 06s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/86/668286/4/check/kolla-ansible-ubuntu-source-ceph/4aeafc1/ : SUCCESS in 1h 31m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bc69a35081f0a3e067c45371b24f59c164921966","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-26 17:48:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nbackport (whatever happens in rocky CI, this should not make it worse)","accounts_in_message":[],"_revision_number":4},{"id":"6094e3af7b302d5d0a25e8753aeeb34e04d01b83","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 17:48:39.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"33024de3245eec2cb2ee1d3d4d1f1d0b6952a8fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 18:44:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/86/668286/4/gate/kolla-build-centos-source/f8fafda/ : SUCCESS in 46m 08s\n- kolla-build-oraclelinux-source http://logs.openstack.org/86/668286/4/gate/kolla-build-oraclelinux-source/4ba885e/ : SUCCESS in 50m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/86/668286/4/gate/openstack-tox-lower-constraints/568a20b/ : SUCCESS in 4m 03s\n- openstack-tox-pep8 http://logs.openstack.org/86/668286/4/gate/openstack-tox-pep8/6135a92/ : SUCCESS in 4m 52s\n- openstack-tox-py27 http://logs.openstack.org/86/668286/4/gate/openstack-tox-py27/6a061a5/ : SUCCESS in 4m 13s\n- openstack-tox-py35 http://logs.openstack.org/86/668286/4/gate/openstack-tox-py35/98b8e29/ : SUCCESS in 4m 06s\n- build-openstack-sphinx-docs http://logs.openstack.org/86/668286/4/gate/build-openstack-sphinx-docs/e040032/html/ : SUCCESS in 3m 52s\n- kolla-build-ubuntu-source http://logs.openstack.org/86/668286/4/gate/kolla-build-ubuntu-source/fee2b3b/ : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":4},{"id":"589ebc84b5080118ae24b1bd619406a855f7e947","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 18:44:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b9d53dacad1bcf39a84d73a32be5db620e7691cb","revisions":{"e3c1e3fe5178810ca9bd273896823e24788838de":{"kind":"REWORK","_number":1,"created":"2019-06-29 07:38:47.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/86/668286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/668286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/668286/1"}}},"commit":{"parents":[{"commit":"b68632ff7e1e782712d94af6005c6348f6b1c640","subject":"Run nova api_db sync before db sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b68632ff7e1e782712d94af6005c6348f6b1c640"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-06-28 18:00:18.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-06-29 07:38:47.000000000","tz":0},"subject":"Fixes ceph-osd deploys with loop devices.","message":"Fixes ceph-osd deploys with loop devices.\n\nChange-Id: I98405e3fb32cbec2d9863817a913ed71c5597e15\nCloses-Bug: #1834678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3c1e3fe5178810ca9bd273896823e24788838de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e3c1e3fe5178810ca9bd273896823e24788838de"}]},"branch":"refs/heads/stable/rocky"},"a8e6ab42498e613b3e706d660f7de04c2a9ce0c2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-30 18:55:38.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/86/668286/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/668286/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/668286/2"}}},"commit":{"parents":[{"commit":"b68632ff7e1e782712d94af6005c6348f6b1c640","subject":"Run nova api_db sync before db sync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b68632ff7e1e782712d94af6005c6348f6b1c640"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-06-28 18:00:18.000000000","tz":180},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-30 18:55:35.000000000","tz":0},"subject":"Fixes ceph-osd deploys with loop devices.","message":"Fixes ceph-osd deploys with loop devices.\n\nDepends-On: https://review.opendev.org/668335\nChange-Id: I98405e3fb32cbec2d9863817a913ed71c5597e15\nCloses-Bug: #1834678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8e6ab42498e613b3e706d660f7de04c2a9ce0c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8e6ab42498e613b3e706d660f7de04c2a9ce0c2"}]},"branch":"refs/heads/stable/rocky"},"dbc136a95fabd92334d0c76eb6ce5c1686229dab":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2019-07-04 08:35:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/668286/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/668286/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/668286/3"}}},"commit":{"parents":[{"commit":"ffdd9faf5c631980b017deefd509af67bcb6d257","subject":"Merge \"Fix deployment of ceph-nfs container failing due to wrong USER\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffdd9faf5c631980b017deefd509af67bcb6d257"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-06-28 18:00:18.000000000","tz":180},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-04 08:35:22.000000000","tz":0},"subject":"Fixes ceph-osd deploys with loop devices.","message":"Fixes ceph-osd deploys with loop devices.\n\nChange-Id: I98405e3fb32cbec2d9863817a913ed71c5597e15\nCloses-Bug: #1834678\n(cherry picked from commit f5ac80e332c2a057ee68d95983492bc7e1156b9a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbc136a95fabd92334d0c76eb6ce5c1686229dab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbc136a95fabd92334d0c76eb6ce5c1686229dab"}]},"branch":"refs/heads/stable/rocky"},"b9d53dacad1bcf39a84d73a32be5db620e7691cb":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-07-25 16:17:30.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/86/668286/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/668286/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/668286/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/668286/4"}}},"commit":{"parents":[{"commit":"ffdd9faf5c631980b017deefd509af67bcb6d257","subject":"Merge \"Fix deployment of ceph-nfs container failing due to wrong USER\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffdd9faf5c631980b017deefd509af67bcb6d257"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-06-28 18:00:18.000000000","tz":180},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-07-25 16:17:27.000000000","tz":0},"subject":"Fixes ceph-osd deploys with loop devices.","message":"Fixes ceph-osd deploys with loop devices.\n\nDepends-On: https://review.opendev.org/672775\nChange-Id: I98405e3fb32cbec2d9863817a913ed71c5597e15\nCloses-Bug: #1834678\n(cherry picked from commit f5ac80e332c2a057ee68d95983492bc7e1156b9a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9d53dacad1bcf39a84d73a32be5db620e7691cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9d53dacad1bcf39a84d73a32be5db620e7691cb"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
