)]}'
{"id":"openstack%2Fkolla~668291","triplet_id":"openstack%2Fkolla~stable%2Fstein~I98405e3fb32cbec2d9863817a913ed71c5597e15","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I98405e3fb32cbec2d9863817a913ed71c5597e15","subject":"Fixes ceph-osd deploys with loop devices.","status":"MERGED","created":"2019-06-29 13:58:41.000000000","updated":"2019-07-05 10:31:30.000000000","submitted":"2019-07-05 10:31:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668291-1562322690800-b16e9fe7","meta_rev_id":"7a34af77553220a440d0063057edd3b0a9949644","_number":668291,"virtual_id_number":668291,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-05 10:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-04 09:12:22.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":2,"date":"2019-07-05 09:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-07-04 08:36:38.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-07-04 08:37:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":1,"date":"2019-07-05 09:19:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-30 02:00:26.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-07-04 08:36:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-04 09:12:22.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-05 09:19:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-05 10:31:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ef45ebe38f0aa60f529430d7724507ecc7a9f6f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-29 13:58:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b53b62b526613858c73995879341469abc90c5f2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-29 13:58:59.000000000","message":"Patch Set 1:\n\nlooks good and has already been proposed back to rocky","accounts_in_message":[],"_revision_number":1},{"id":"68de3f010861fa362e574469443be1f0b048bcf0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-06-29 13:59:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7c96f92cb9ae0f61a94e0c7f5d4da7e43c943e5d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-29 14:40:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/91/668291/1/openstack-check/tripleo-build-containers-fedora-28-master/39fc546/ : FAILURE in 5m 45s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/91/668291/1/openstack-check/tripleo-build-containers-fedora-28-stein/9dffdc2/ : SUCCESS in 36m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34f6af23b8e0d4508d81164f1909aaa3ee900057","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-29 15:31:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/91/668291/1/check/kolla-build-centos-binary/76fbce3/ : SUCCESS in 21m 55s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/91/668291/1/check/kolla-build-centos-source/03320f5/ : SUCCESS in 31m 01s\n- kolla-ansible-centos-source http://logs.openstack.org/91/668291/1/check/kolla-ansible-centos-source/f50d27b/ : SUCCESS in 1h 07m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/91/668291/1/check/kolla-ansible-centos-binary/22abedc/ : SUCCESS in 1h 00m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/91/668291/1/check/kolla-ansible-centos-source-upgrade/6565ca0/ : SUCCESS in 1h 09m 34s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/91/668291/1/check/tripleo-build-containers-centos-7/f32fa70/ : SUCCESS in 51m 12s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/91/668291/1/check/kolla-ansible-centos-source-ceph/c821325/ : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/91/668291/1/check/kolla-build-oraclelinux-binary/e72a7cf/ : SUCCESS in 49m 18s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/91/668291/1/check/kolla-build-oraclelinux-source/d7e51d6/ : SUCCESS in 41m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/668291/1/check/openstack-tox-lower-constraints/c33fca4/ : SUCCESS in 3m 49s\n- openstack-tox-pep8 http://logs.openstack.org/91/668291/1/check/openstack-tox-pep8/07e2256/ : SUCCESS in 3m 50s\n- openstack-tox-py27 http://logs.openstack.org/91/668291/1/check/openstack-tox-py27/d05dcbb/ : SUCCESS in 4m 09s\n- openstack-tox-py36 http://logs.openstack.org/91/668291/1/check/openstack-tox-py36/19a2abf/ : SUCCESS in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/91/668291/1/check/openstack-tox-py37/2093f3e/ : SUCCESS in 4m 36s\n- openstack-tox-docs http://logs.openstack.org/91/668291/1/check/openstack-tox-docs/e6d64f3/html/ : SUCCESS in 3m 08s\n- kolla-build-ubuntu-binary http://logs.openstack.org/91/668291/1/check/kolla-build-ubuntu-binary/65a1927/ : SUCCESS in 36m 58s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/91/668291/1/check/kolla-build-ubuntu-source/5e7b23c/ : SUCCESS in 26m 40s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/91/668291/1/check/kolla-ansible-ubuntu-source/fd64341/ : SUCCESS in 54m 42s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/91/668291/1/check/kolla-ansible-ubuntu-binary/390835a/ : SUCCESS in 44m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/91/668291/1/check/kolla-ansible-ubuntu-source-upgrade/713c653/ : SUCCESS in 48m 53s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/91/668291/1/check/kolla-ansible-ubuntu-source-ceph/c276c16/ : SUCCESS in 1h 13m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e933e7edc013c241002e095d66bb7be0f5885331","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-06-29 20:45:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"db88b2d8f5009581bde975a2a7e7bf0037b97c16","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-06-29 20:50:52.000000000","message":"Removed reviewer Dincer Celik with the following votes:\n\n* Code-Review+1 by Dincer Celik \u003chello@dincercelik.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"4917173f6209a4659719de507566fdde1f2a1a87","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-06-30 02:00:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"96d5e517899652f5602c0f0ba6aff6e856071937","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-01 11:17:12.000000000","message":"Patch Set 1: Code-Review-1\n\nLet\u0027s wait for master to merge then add \u0027cherry picked from\u0027 message.","accounts_in_message":[],"_revision_number":1},{"id":"f7e7466edd766a4ce8b90c698426afd04f6fef86","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-04 08:27:30.000000000","message":"Patch Set 1:\n\nmaster merged","accounts_in_message":[],"_revision_number":1},{"id":"fcb46de44677fe7b9a7bf58c0ba8bcc51cc19aa3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 08:35:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9192bbcb34affe6476502f3bdc50df05b1c15d1e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 08:36:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6489f3d57b1057e74d11679db14e12870655c03d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-04 08:37:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"773a0d06d47fc68dd2330151988685b194fd083e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-04 09:12:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/91/668291/2/openstack-check/tripleo-build-containers-fedora-28-master/9d909e2/ : FAILURE in 5m 59s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/91/668291/2/openstack-check/tripleo-build-containers-fedora-28-stein/19802a5/ : SUCCESS in 36m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f4c3f2993374ff25ce1a7933a8a6565118042532","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 10:07:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/91/668291/2/check/kolla-build-centos-binary/e89c7ef/ : FAILURE in 37m 55s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/91/668291/2/check/kolla-build-centos-source/8574fe5/ : SUCCESS in 43m 44s\n- kolla-ansible-centos-source http://logs.openstack.org/91/668291/2/check/kolla-ansible-centos-source/b856c6a/ : SUCCESS in 1h 12m 05s\n- kolla-ansible-centos-binary http://logs.openstack.org/91/668291/2/check/kolla-ansible-centos-binary/8f3d4d4/ : SUCCESS in 59m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/91/668291/2/check/kolla-ansible-centos-source-upgrade/9137d0b/ : SUCCESS in 1h 11m 07s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/91/668291/2/check/tripleo-build-containers-centos-7/3e37a39/ : SUCCESS in 44m 37s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/91/668291/2/check/kolla-ansible-centos-source-ceph/f5f686a/ : SUCCESS in 1h 14m 57s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/91/668291/2/check/kolla-build-oraclelinux-binary/79ad3a2/ : SUCCESS in 41m 10s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/91/668291/2/check/kolla-build-oraclelinux-source/49dc849/ : SUCCESS in 32m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/668291/2/check/openstack-tox-lower-constraints/ff5d7a5/ : SUCCESS in 3m 57s\n- openstack-tox-pep8 http://logs.openstack.org/91/668291/2/check/openstack-tox-pep8/d1a7e2e/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/91/668291/2/check/openstack-tox-py27/c5cef30/ : SUCCESS in 3m 12s\n- openstack-tox-py36 http://logs.openstack.org/91/668291/2/check/openstack-tox-py36/322103c/ : SUCCESS in 3m 07s\n- openstack-tox-py37 http://logs.openstack.org/91/668291/2/check/openstack-tox-py37/81fce67/ : SUCCESS in 3m 32s\n- openstack-tox-docs http://logs.openstack.org/91/668291/2/check/openstack-tox-docs/fe49ac8/html/ : SUCCESS in 3m 59s\n- kolla-build-ubuntu-binary http://logs.openstack.org/91/668291/2/check/kolla-build-ubuntu-binary/810b6a0/ : SUCCESS in 22m 56s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/91/668291/2/check/kolla-build-ubuntu-source/b355dd2/ : SUCCESS in 30m 33s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/91/668291/2/check/kolla-ansible-ubuntu-source/093cf22/ : SUCCESS in 58m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/91/668291/2/check/kolla-ansible-ubuntu-binary/7316c89/ : SUCCESS in 47m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/91/668291/2/check/kolla-ansible-ubuntu-source-upgrade/f8cfeb7/ : SUCCESS in 1h 06m 49s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/91/668291/2/check/kolla-ansible-ubuntu-source-ceph/9d107d8/ : SUCCESS in 1h 13m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"185051da19dfb3af716abd4f5e471a64236a9d31","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-05 09:19:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"44a46428bf41396212ee9e8634d1c702282f9e65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 09:19:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e66a0a3a09fd4136163dc9c0d94756e905c3d3eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 10:31:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7a34af77553220a440d0063057edd3b0a9949644","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 10:31:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/91/668291/2/gate/kolla-build-centos-source/d2384e4/ : SUCCESS in 43m 44s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/91/668291/2/gate/tripleo-build-containers-centos-7/cf8ac6e/ : SUCCESS in 46m 04s\n- kolla-build-oraclelinux-source http://logs.openstack.org/91/668291/2/gate/kolla-build-oraclelinux-source/9ac4f10/ : SUCCESS in 58m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/91/668291/2/gate/openstack-tox-lower-constraints/4303fd1/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/91/668291/2/gate/openstack-tox-pep8/7e28920/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/91/668291/2/gate/openstack-tox-py27/2458996/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/91/668291/2/gate/openstack-tox-py36/d8c2b37/ : SUCCESS in 3m 46s\n- openstack-tox-py37 http://logs.openstack.org/91/668291/2/gate/openstack-tox-py37/5e8697b/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/91/668291/2/gate/openstack-tox-docs/afbf8fb/html/ : SUCCESS in 3m 49s\n- kolla-build-ubuntu-source http://logs.openstack.org/91/668291/2/gate/kolla-build-ubuntu-source/a49c992/ : SUCCESS in 31m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e930fb35ac603a148eb8fabd8684052d9c63e5f6","revisions":{"5e8c54615fe6c3a380f945eb1879b607d2bc96a9":{"kind":"REWORK","_number":1,"created":"2019-06-29 13:58:41.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/91/668291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/668291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/668291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/668291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/668291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/668291/1"}}},"commit":{"parents":[{"commit":"c75e469f6c2951c349ba36fd250c16a5998d0181","subject":"Merge \"Clean up test process\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c75e469f6c2951c349ba36fd250c16a5998d0181"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-06-28 18:00:18.000000000","tz":180},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-29 13:58:41.000000000","tz":0},"subject":"Fixes ceph-osd deploys with loop devices.","message":"Fixes ceph-osd deploys with loop devices.\n\nChange-Id: I98405e3fb32cbec2d9863817a913ed71c5597e15\nCloses-Bug: #1834678\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e8c54615fe6c3a380f945eb1879b607d2bc96a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e8c54615fe6c3a380f945eb1879b607d2bc96a9"}]},"branch":"refs/heads/stable/stein"},"e930fb35ac603a148eb8fabd8684052d9c63e5f6":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2019-07-04 08:35:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/91/668291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/91/668291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/668291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/668291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/91/668291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/91/668291/2"}}},"commit":{"parents":[{"commit":"6fd4b02178653b94ea87e085c1e4148d38e28494","subject":"Update versions: ./tools/version-check.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fd4b02178653b94ea87e085c1e4148d38e28494"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-06-28 18:00:18.000000000","tz":180},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-07-04 08:35:14.000000000","tz":0},"subject":"Fixes ceph-osd deploys with loop devices.","message":"Fixes ceph-osd deploys with loop devices.\n\nChange-Id: I98405e3fb32cbec2d9863817a913ed71c5597e15\nCloses-Bug: #1834678\n(cherry picked from commit f5ac80e332c2a057ee68d95983492bc7e1156b9a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e930fb35ac603a148eb8fabd8684052d9c63e5f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e930fb35ac603a148eb8fabd8684052d9c63e5f6"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
