)]}'
{"id":"openstack%2Fkolla~668625","triplet_id":"openstack%2Fkolla~stable%2Fpike~I9828c23e4bcec7553a69aa5f20e990e0ef0538ea","project":"openstack/kolla","branch":"stable/pike","topic":"pike-em","hashtags":[],"change_id":"I9828c23e4bcec7553a69aa5f20e990e0ef0538ea","subject":"Fix bandit failure B105 in keystone fernet","status":"MERGED","created":"2019-07-02 07:41:30.000000000","updated":"2019-07-03 07:37:50.000000000","submitted":"2019-07-03 07:37:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668625-1562139470840-1ef4935a","meta_rev_id":"f6d8aa3a47828533b384b59b1b30886ee5e007ce","_number":668625,"virtual_id_number":668625,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-03 07:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-02 08:20:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-07-03 06:03:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-07-02 09:28:39.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-07-03 06:03:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 08:20:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-02 09:28:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-03 06:03:46.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-03 07:37:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34b14e50228e9115846966879f623e354e0b18ed","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-07-02 07:41:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f01e745357671999fd2e639e55b3d2b4aefa7112","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-02 08:20:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/25/668625/1/openstack-check/tripleo-build-containers-fedora-28-master/3664be0/ : RETRY_LIMIT in 2m 57s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/25/668625/1/openstack-check/tripleo-build-containers-fedora-28-stein/056c691/ : SUCCESS in 37m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ea2d21e98087c995e80dd3225ee27a97e0e6963","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 08:54:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/25/668625/1/check/openstack-tox-pep8/19ca95c/ : SUCCESS in 5m 25s\n- openstack-tox-py27 http://logs.openstack.org/25/668625/1/check/openstack-tox-py27/40174c5/ : SUCCESS in 5m 32s\n- openstack-tox-py35 http://logs.openstack.org/25/668625/1/check/openstack-tox-py35/f92040d/ : SUCCESS in 4m 21s\n- build-openstack-sphinx-docs http://logs.openstack.org/25/668625/1/check/build-openstack-sphinx-docs/75440de/html/ : SUCCESS in 5m 19s\n- kolla-build-centos-binary http://logs.openstack.org/25/668625/1/check/kolla-build-centos-binary/18be59f/ : SUCCESS in 40m 36s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/25/668625/1/check/kolla-build-centos-source/c208215/ : SUCCESS in 55m 23s\n- kolla-build-ubuntu-binary http://logs.openstack.org/25/668625/1/check/kolla-build-ubuntu-binary/807722f/ : SUCCESS in 27m 14s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/25/668625/1/check/kolla-build-ubuntu-source/e3b253d/ : SUCCESS in 27m 05s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/25/668625/1/check/kolla-build-oraclelinux-binary/cbc2e11/ : SUCCESS in 38m 18s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/25/668625/1/check/kolla-build-oraclelinux-source/8ef72f7/ : SUCCESS in 57m 06s\n- kolla-ansible-centos-source http://logs.openstack.org/25/668625/1/check/kolla-ansible-centos-source/3a3c631/ : SUCCESS in 58m 11s (non-voting)\n- kolla-ansible-centos-binary http://logs.openstack.org/25/668625/1/check/kolla-ansible-centos-binary/aa98f5c/ : SUCCESS in 53m 51s (non-voting)\n- kolla-ansible-ubuntu-source http://logs.openstack.org/25/668625/1/check/kolla-ansible-ubuntu-source/6fce8ca/ : SUCCESS in 47m 39s (non-voting)\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/25/668625/1/check/kolla-ansible-ubuntu-binary/5229328/ : SUCCESS in 47m 58s (non-voting)\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/25/668625/1/check/kolla-ansible-oraclelinux-source/43e2571/ : SUCCESS in 1h 01m 47s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/25/668625/1/check/kolla-ansible-oraclelinux-binary/8fa9ce6/ : SUCCESS in 49m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"400150593cd1488c4df03c58987ef87ed8754737","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-02 09:28:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4c734b0d22f50bc692bdc77f24cf0a38dd0c3dbc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-03 06:03:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7659e2af33705755db209b3fb85843cc01077e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 06:03:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7940ca3cf76b0c437e634f8870cca9280dd59c27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:37:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"041279cbedf3af22420f4ddd3d3ce801a073ab75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 07:37:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/25/668625/1/gate/openstack-tox-pep8/a4c416d/ : SUCCESS in 4m 50s\n- openstack-tox-py27 http://logs.openstack.org/25/668625/1/gate/openstack-tox-py27/7d90f74/ : SUCCESS in 5m 26s\n- openstack-tox-py35 http://logs.openstack.org/25/668625/1/gate/openstack-tox-py35/afcb467/ : SUCCESS in 5m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/25/668625/1/gate/build-openstack-sphinx-docs/9a49d4b/html/ : SUCCESS in 5m 18s\n- kolla-build-centos-source http://logs.openstack.org/25/668625/1/gate/kolla-build-centos-source/a821709/ : SUCCESS in 54m 35s\n- kolla-build-ubuntu-source http://logs.openstack.org/25/668625/1/gate/kolla-build-ubuntu-source/8370307/ : SUCCESS in 30m 28s\n- kolla-build-oraclelinux-source http://logs.openstack.org/25/668625/1/gate/kolla-build-oraclelinux-source/26eefe3/ : SUCCESS in 45m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"658090a3af8756f625ea9db08ea0826953b9264a","revisions":{"658090a3af8756f625ea9db08ea0826953b9264a":{"kind":"REWORK","_number":1,"created":"2019-07-02 07:41:30.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/25/668625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/668625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/668625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/668625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/668625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/668625/1"}}},"commit":{"parents":[{"commit":"e212bf5d9d287a07422f36832407901cf6c61b30","subject":"Merge \"Bump service versions for Pike EM\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e212bf5d9d287a07422f36832407901cf6c61b30"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-09 13:28:21.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2019-07-02 07:40:05.000000000","tz":120},"subject":"Fix bandit failure B105 in keystone fernet","message":"Fix bandit failure B105 in keystone fernet\n\n\u003e\u003e Issue: [B105:hardcoded_password_string] Possible hardcoded password: \u0027/etc/keystone/fernet-keys\u0027\n   Severity: Low Confidence: Medium\n      Location: docker/keystone/keystone-fernet/fetch_fernet_tokens.py:27\n         More Info: https://bandit.readthedocs.io/en/latest/plugins/b105_hardcoded_password_string.html\n\nIgnore bandit failures on this line.\n\nChange-Id: I9828c23e4bcec7553a69aa5f20e990e0ef0538ea\nCloses-Bug: #1828416\n(cherry picked from commit e6387634c6a7170e96606b9a8ff680f08d81e801)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/658090a3af8756f625ea9db08ea0826953b9264a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/658090a3af8756f625ea9db08ea0826953b9264a"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
