)]}'
{"id":"openstack%2Fkolla~668630","triplet_id":"openstack%2Fkolla~master~I8927ce98d669480866d6fcddaf7b08ef3cbbc37e","project":"openstack/kolla","branch":"master","topic":"bug/1697570","hashtags":[],"change_id":"I8927ce98d669480866d6fcddaf7b08ef3cbbc37e","subject":"Fix ovs-appctl by adding --pidfile to ovsdb command.","status":"MERGED","created":"2019-07-02 08:15:10.000000000","updated":"2019-08-20 04:49:48.000000000","submitted":"2019-07-05 09:17:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668630-1562318237597-b16e9fe7","meta_rev_id":"ff8f6ef57f952cd04efa3be5bc43ded93e62d863","_number":668630,"virtual_id_number":668630,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-05 09:17:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-02 08:54:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"all":[{"value":1,"date":"2019-07-03 03:31:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-07-02 09:49:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-05 08:19:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-07-05 08:19:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"date":"2019-07-03 06:26:47.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-02 08:54:58.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-02 09:49:34.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-03 03:31:59.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-07-03 06:26:47.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-07-05 08:19:31.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-05 09:17:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5a13d63c177c0cd45561aaf696fe15c64b886054","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-02 08:15:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa2c4399451b8ed9951b6d7e685ec2dc7da5b851","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-02 08:26:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"02e0a282dad4578dab3cf942ec359b17b35db3de","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-02 08:54:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/30/668630/1/openstack-check/tripleo-build-containers-fedora-28-master/e854345/ : FAILURE in 5m 53s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/30/668630/1/openstack-check/tripleo-build-containers-fedora-28-stein/1df30f9/ : SUCCESS in 38m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e6eda5d57c414f676b1f1724660beebdfccecea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-02 09:44:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/30/668630/1/check/kolla-build-centos-binary/7ae3ba4/ : SUCCESS in 53m 45s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/30/668630/1/check/kolla-build-centos-source/351f7dd/ : SUCCESS in 46m 40s\n- kolla-ansible-centos-source http://logs.openstack.org/30/668630/1/check/kolla-ansible-centos-source/055b921/ : SUCCESS in 1h 08m 45s\n- kolla-ansible-centos-binary http://logs.openstack.org/30/668630/1/check/kolla-ansible-centos-binary/f72f5f3/ : SUCCESS in 1h 11m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/30/668630/1/check/kolla-ansible-centos-source-upgrade/3a37727/ : SUCCESS in 1h 20m 24s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/30/668630/1/check/tripleo-build-containers-centos-7/5394eb3/ : SUCCESS in 53m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/668630/1/check/openstack-tox-lower-constraints/a7d12a0/ : SUCCESS in 4m 48s\n- openstack-tox-pep8 http://logs.openstack.org/30/668630/1/check/openstack-tox-pep8/cd08b5c/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/30/668630/1/check/openstack-tox-py27/92d79c9/ : SUCCESS in 2m 50s\n- openstack-tox-py36 http://logs.openstack.org/30/668630/1/check/openstack-tox-py36/80e876d/ : SUCCESS in 3m 11s\n- openstack-tox-py37 http://logs.openstack.org/30/668630/1/check/openstack-tox-py37/15d260b/ : SUCCESS in 4m 50s\n- openstack-tox-docs http://logs.openstack.org/30/668630/1/check/openstack-tox-docs/38fe26f/html/ : SUCCESS in 4m 06s\n- kolla-build-ubuntu-binary http://logs.openstack.org/30/668630/1/check/kolla-build-ubuntu-binary/ee1f12b/ : SUCCESS in 25m 39s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/30/668630/1/check/kolla-build-ubuntu-source/a3d0e39/ : SUCCESS in 32m 58s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/30/668630/1/check/kolla-ansible-ubuntu-source/f4f04c8/ : SUCCESS in 46m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/30/668630/1/check/kolla-ansible-ubuntu-binary/7c9a296/ : SUCCESS in 49m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/30/668630/1/check/kolla-ansible-ubuntu-source-upgrade/0c17990/ : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":1},{"id":"452a31abb8adc3ae00ee00bb15d4336eca6ee44e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-02 09:49:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8f946dfae1155b42906253867f45852e7f5bc03","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-07-03 03:31:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cfca3b4528b609270dbe9f09570d8d8d4c2860ff","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-05 08:19:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0b97e4a5e0b8ddca0b3997a87c2bee047ddc71d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 08:19:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1cc54592219e716deba2bf3c36379d7d733b5d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 09:17:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"076ec17cb6b24d432ab5ef0ebea7ce3a717e7fa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 09:17:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/30/668630/1/gate/kolla-build-centos-source/e0a0db2/ : SUCCESS in 36m 29s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/30/668630/1/gate/tripleo-build-containers-centos-7/bf6ce76/ : SUCCESS in 49m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/30/668630/1/gate/openstack-tox-lower-constraints/929f71e/ : SUCCESS in 3m 44s\n- openstack-tox-pep8 http://logs.openstack.org/30/668630/1/gate/openstack-tox-pep8/10b0469/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/30/668630/1/gate/openstack-tox-py27/fe83b4e/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/30/668630/1/gate/openstack-tox-py36/a4a89c7/ : SUCCESS in 3m 31s\n- openstack-tox-py37 http://logs.openstack.org/30/668630/1/gate/openstack-tox-py37/74dccfc/ : SUCCESS in 4m 27s\n- openstack-tox-docs http://logs.openstack.org/30/668630/1/gate/openstack-tox-docs/fc07b74/html/ : SUCCESS in 3m 52s\n- kolla-build-ubuntu-source http://logs.openstack.org/30/668630/1/gate/kolla-build-ubuntu-source/2b86237/ : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a8a789969aae22a182a5cb27e32e4e1ad58c6c5f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 07:41:56.000000000","message":"Patch Set 1:\n\nDoes this need to be backported?","accounts_in_message":[],"_revision_number":1},{"id":"04fb8f28a032fba74f0aa6386f913dcd255b1157","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-20 04:49:07.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 0e78cbb84c895eb14d8d6e63989d6cfc88474b98","accounts_in_message":[],"_revision_number":1},{"id":"1e5a03d8596a97fc33231c457e88337dde801ddc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-20 04:49:25.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit fc806d33a8601059455790527782989dbb2f56e9","accounts_in_message":[],"_revision_number":1},{"id":"c95a97ab371a2dd2db34ee768202399b4dce38d2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-20 04:49:48.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 13ad3974560dfb9cc89b0f966a61f7a592efee30","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ec7017f7a358553158dc927285ab9f4af1a641a","revisions":{"0ec7017f7a358553158dc927285ab9f4af1a641a":{"kind":"REWORK","_number":1,"created":"2019-07-02 08:15:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/30/668630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/668630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/668630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/668630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/668630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/668630/1"}}},"commit":{"parents":[{"commit":"5ecb9107ea286bd0c1a45d4d95d064c764d33296","subject":"Merge \"[prometheus/openstack-exporter] Changed repository location and version.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ecb9107ea286bd0c1a45d4d95d064c764d33296"}]}],"author":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-07-02 08:11:02.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"michal.nasiadka@nokia.com","date":"2019-07-02 08:15:08.000000000","tz":120},"subject":"Fix ovs-appctl by adding --pidfile to ovsdb command.","message":"Fix ovs-appctl by adding --pidfile to ovsdb command.\n\nChange-Id: I8927ce98d669480866d6fcddaf7b08ef3cbbc37e\nCloses-Bug: #1697570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ec7017f7a358553158dc927285ab9f4af1a641a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ec7017f7a358553158dc927285ab9f4af1a641a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
