)]}'
{"id":"openstack%2Fkolla~668815","triplet_id":"openstack%2Fkolla~stable%2Fstein~If58f15e18e420aca790496132352f1c1dcdb2640","project":"openstack/kolla","branch":"stable/stein","topic":"extract-ci-vars_stable/stein","hashtags":[],"change_id":"If58f15e18e420aca790496132352f1c1dcdb2640","subject":"Extract CI vars for use in kolla-ansible jobs","status":"MERGED","created":"2019-07-03 07:51:54.000000000","updated":"2019-07-04 11:56:46.000000000","submitted":"2019-07-04 11:56:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668815-1562241406931-0934339c","meta_rev_id":"6b5de4ef8d7a823c60a38b7a60e8c78516718504","_number":668815,"virtual_id_number":668815,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-07-04 09:16:07.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-04 11:56:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-07-04 08:16:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-07-03 15:48:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-03 22:58:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-07-04 08:16:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 15:48:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-03 22:58:48.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-07-04 08:16:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-04 11:56:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6262c824885b21cf4fdcd1bca326cd82ad666f5d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-03 07:51:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64dd56b31f721f8dc5bda79dac9c5cb08b87421a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 09:19:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/15/668815/1/check/kolla-build-centos-binary/5539849/ : SUCCESS in 42m 09s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/668815/1/check/kolla-build-centos-source/99d8e05/ : SUCCESS in 34m 05s\n- kolla-ansible-centos-source http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-source/756113e/ : FAILURE in 28m 48s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-binary/7790df8/ : SUCCESS in 1h 02m 42s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-source-upgrade/7a23f25/ : SUCCESS in 1h 21m 13s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/668815/1/check/kolla-build-oraclelinux-binary/b9a3da7/ : SUCCESS in 43m 57s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/668815/1/check/kolla-build-oraclelinux-source/1af5757/ : SUCCESS in 1h 01m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/668815/1/check/openstack-tox-lower-constraints/79d5e93/ : SUCCESS in 4m 15s\n- openstack-tox-pep8 http://logs.openstack.org/15/668815/1/check/openstack-tox-pep8/1144808/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/15/668815/1/check/openstack-tox-py27/c30735b/ : SUCCESS in 3m 08s\n- openstack-tox-py36 http://logs.openstack.org/15/668815/1/check/openstack-tox-py36/ab75104/ : SUCCESS in 3m 17s\n- openstack-tox-py37 http://logs.openstack.org/15/668815/1/check/openstack-tox-py37/bcab98e/ : SUCCESS in 2m 59s\n- openstack-tox-docs http://logs.openstack.org/15/668815/1/check/openstack-tox-docs/7322178/html/ : SUCCESS in 2m 33s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/668815/1/check/kolla-build-ubuntu-binary/a248513/ : SUCCESS in 22m 09s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/668815/1/check/kolla-build-ubuntu-source/96728a0/ : SUCCESS in 33m 20s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-source/d9a30c8/ : SUCCESS in 57m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-binary/dd7ccbb/ : SUCCESS in 38m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-source-upgrade/c209505/ : SUCCESS in 55m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2fdcfa62195464ee164aede3331541ec7ed3a9d9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-03 09:23:10.000000000","message":"Patch Set 1:\n\nrecheck\n\nZuul issue","accounts_in_message":[],"_revision_number":1},{"id":"3d56fd216486d939d96e88f3780bd428c745b737","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 10:41:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/15/668815/1/check/kolla-build-centos-binary/622998f/ : SUCCESS in 47m 07s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/668815/1/check/kolla-build-centos-source/369c159/ : RETRY_LIMIT in 4m 03s\n- kolla-ansible-centos-source http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-source/508ee95/ : RETRY_LIMIT in 4m 20s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-binary/174d83e/ : SUCCESS in 1h 07m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-source-upgrade/54e4ace/ : SUCCESS in 1h 10m 51s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/668815/1/check/kolla-build-oraclelinux-binary/c0568d8/ : SUCCESS in 50m 37s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/668815/1/check/kolla-build-oraclelinux-source/610a169/ : RETRY_LIMIT in 4m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/668815/1/check/openstack-tox-lower-constraints/e040c73/ : SUCCESS in 3m 37s\n- openstack-tox-pep8 http://logs.openstack.org/15/668815/1/check/openstack-tox-pep8/23f3296/ : SUCCESS in 3m 22s\n- openstack-tox-py27 http://logs.openstack.org/15/668815/1/check/openstack-tox-py27/e864d9e/ : SUCCESS in 2m 51s\n- openstack-tox-py36 http://logs.openstack.org/15/668815/1/check/openstack-tox-py36/cc7267e/ : SUCCESS in 3m 40s\n- openstack-tox-py37 http://logs.openstack.org/15/668815/1/check/openstack-tox-py37/bb47631/ : SUCCESS in 3m 53s\n- openstack-tox-docs http://logs.openstack.org/15/668815/1/check/openstack-tox-docs/d201501/html/ : SUCCESS in 4m 21s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/668815/1/check/kolla-build-ubuntu-binary/f483aa5/ : SUCCESS in 30m 14s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/668815/1/check/kolla-build-ubuntu-source/eb1ad7f/ : SUCCESS in 29m 37s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-source/da71ebc/ : SUCCESS in 52m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-binary/e310a94/ : SUCCESS in 46m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-source-upgrade/fdd2cea/ : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3a10aa6f6db194dce0cc878f58e142278ab18912","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-03 10:54:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2e97c101dcf10f930f5e086d07806ee7fa28cdb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 12:17:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/15/668815/1/check/kolla-build-centos-binary/7d7f642/ : SUCCESS in 48m 57s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/668815/1/check/kolla-build-centos-source/16b44a1/ : SUCCESS in 37m 25s\n- kolla-ansible-centos-source http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-source/2b0dfc2/ : SUCCESS in 1h 08m 20s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-binary/113442e/ : SUCCESS in 1h 10m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-source-upgrade/d638624/ : SUCCESS in 1h 17m 00s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/668815/1/check/kolla-build-oraclelinux-binary/eed4767/ : SUCCESS in 48m 35s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/668815/1/check/kolla-build-oraclelinux-source/8f97e4d/ : SUCCESS in 41m 37s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/668815/1/check/openstack-tox-lower-constraints/a82f3bf/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 http://logs.openstack.org/15/668815/1/check/openstack-tox-pep8/cd02508/ : SUCCESS in 3m 09s\n- openstack-tox-py27 http://logs.openstack.org/15/668815/1/check/openstack-tox-py27/12b3331/ : SUCCESS in 3m 21s\n- openstack-tox-py36 http://logs.openstack.org/15/668815/1/check/openstack-tox-py36/c3c5e09/ : SUCCESS in 3m 41s\n- openstack-tox-py37 http://logs.openstack.org/15/668815/1/check/openstack-tox-py37/ea028ab/ : SUCCESS in 3m 26s\n- openstack-tox-docs http://logs.openstack.org/15/668815/1/check/openstack-tox-docs/aa74b45/html/ : SUCCESS in 2m 54s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/668815/1/check/kolla-build-ubuntu-binary/e4a2bee/ : SUCCESS in 24m 53s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/668815/1/check/kolla-build-ubuntu-source/2597455/ : SUCCESS in 27m 44s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-source/01ddea8/ : SUCCESS in 51m 35s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-binary/24ddd9f/ : SUCCESS in 44m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-source-upgrade/b28be4a/ : SUCCESS in 57m 44s","accounts_in_message":[],"_revision_number":1},{"id":"330d73b9d970584954b147aeebb8e951e3dc10df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-03 15:48:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44e9007ba3bc7432b58cceffff8b5b643f8108a5","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-07-03 22:58:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1c19d576a9cd07383718a5c1a9e8a7b9711790dc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-04 08:16:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9c8c8e5f6f87b9c834bf39d1cbc3fcb0a09c18b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 08:16:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"515060efa4924a8ec32021053e3e1807095a3075","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 09:10:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source http://logs.openstack.org/15/668815/1/gate/kolla-build-centos-source/9c34de0/ : FAILURE in 47m 49s\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/668815/1/gate/kolla-build-oraclelinux-source/d2c4e3d/ : SUCCESS in 37m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/668815/1/gate/openstack-tox-lower-constraints/1754bc8/ : SUCCESS in 3m 25s\n- openstack-tox-pep8 http://logs.openstack.org/15/668815/1/gate/openstack-tox-pep8/a399d46/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/15/668815/1/gate/openstack-tox-py27/2d6333a/ : SUCCESS in 4m 31s\n- openstack-tox-py36 http://logs.openstack.org/15/668815/1/gate/openstack-tox-py36/562ec32/ : SUCCESS in 3m 53s\n- openstack-tox-py37 http://logs.openstack.org/15/668815/1/gate/openstack-tox-py37/8de7893/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/15/668815/1/gate/openstack-tox-docs/e1390d5/html/ : SUCCESS in 3m 17s\n- kolla-build-ubuntu-source http://logs.openstack.org/15/668815/1/gate/kolla-build-ubuntu-source/de44af5/ : SUCCESS in 32m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8a35af1cbcf9594c431375ac7fe50420ada4a5ab","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-04 09:16:07.000000000","message":"Patch Set 1:\n\nrecheck\n\niad.rax","accounts_in_message":[],"_revision_number":1},{"id":"ba8732e561da502316287da18613ae9e858a7856","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 11:04:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/15/668815/1/check/kolla-build-centos-binary/4cc7f7a/ : SUCCESS in 33m 06s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/15/668815/1/check/kolla-build-centos-source/a258dc5/ : SUCCESS in 1h 01m 12s\n- kolla-ansible-centos-source http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-source/15229bd/ : SUCCESS in 1h 11m 17s\n- kolla-ansible-centos-binary http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-binary/b7ff8d0/ : SUCCESS in 1h 09m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/15/668815/1/check/kolla-ansible-centos-source-upgrade/b1ddb3a/ : SUCCESS in 1h 24m 04s\n- kolla-build-oraclelinux-binary http://logs.openstack.org/15/668815/1/check/kolla-build-oraclelinux-binary/2bf1439/ : SUCCESS in 31m 01s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/668815/1/check/kolla-build-oraclelinux-source/aa3b0ff/ : SUCCESS in 43m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/668815/1/check/openstack-tox-lower-constraints/2c904a2/ : SUCCESS in 3m 52s\n- openstack-tox-pep8 http://logs.openstack.org/15/668815/1/check/openstack-tox-pep8/4bcaf16/ : SUCCESS in 6m 12s\n- openstack-tox-py27 http://logs.openstack.org/15/668815/1/check/openstack-tox-py27/b2accb6/ : SUCCESS in 6m 18s\n- openstack-tox-py36 http://logs.openstack.org/15/668815/1/check/openstack-tox-py36/9ca4b8f/ : SUCCESS in 4m 32s\n- openstack-tox-py37 http://logs.openstack.org/15/668815/1/check/openstack-tox-py37/bafb34a/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/15/668815/1/check/openstack-tox-docs/9146ab8/html/ : SUCCESS in 3m 10s\n- kolla-build-ubuntu-binary http://logs.openstack.org/15/668815/1/check/kolla-build-ubuntu-binary/5ddb088/ : SUCCESS in 21m 55s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/15/668815/1/check/kolla-build-ubuntu-source/c573586/ : SUCCESS in 29m 28s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-source/750a278/ : SUCCESS in 56m 03s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-binary/3383e41/ : SUCCESS in 43m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/15/668815/1/check/kolla-ansible-ubuntu-source-upgrade/913d1d0/ : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e37413e7581d8ed37c30797197931916a340f3d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 11:04:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"549b0e6b6050159ece417dc54af6d9784d61667a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 11:56:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"39c84d17de3e14cff9f012bfcc0cdf70954a3141","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 11:56:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/15/668815/1/gate/kolla-build-centos-source/2e69245/ : SUCCESS in 44m 01s\n- kolla-build-oraclelinux-source http://logs.openstack.org/15/668815/1/gate/kolla-build-oraclelinux-source/3d769f9/ : SUCCESS in 45m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/15/668815/1/gate/openstack-tox-lower-constraints/24f7403/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/15/668815/1/gate/openstack-tox-pep8/c3317de/ : SUCCESS in 3m 59s\n- openstack-tox-py27 http://logs.openstack.org/15/668815/1/gate/openstack-tox-py27/95d65f9/ : SUCCESS in 3m 22s\n- openstack-tox-py36 http://logs.openstack.org/15/668815/1/gate/openstack-tox-py36/ab40505/ : SUCCESS in 3m 32s\n- openstack-tox-py37 http://logs.openstack.org/15/668815/1/gate/openstack-tox-py37/e3cfa3d/ : SUCCESS in 4m 05s\n- openstack-tox-docs http://logs.openstack.org/15/668815/1/gate/openstack-tox-docs/7aa5279/html/ : SUCCESS in 3m 25s\n- kolla-build-ubuntu-source http://logs.openstack.org/15/668815/1/gate/kolla-build-ubuntu-source/c070654/ : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac66f586981b5333448545403242c2a5ebc90371","revisions":{"ac66f586981b5333448545403242c2a5ebc90371":{"kind":"REWORK","_number":1,"created":"2019-07-03 07:51:54.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/15/668815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/668815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/668815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/668815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/668815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/668815/1"}}},"commit":{"parents":[{"commit":"6fd4b02178653b94ea87e085c1e4148d38e28494","subject":"Update versions: ./tools/version-check.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fd4b02178653b94ea87e085c1e4148d38e28494"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-06-28 18:18:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-07-03 07:49:11.000000000","tz":120},"subject":"Extract CI vars for use in kolla-ansible jobs","message":"Extract CI vars for use in kolla-ansible jobs\n\nAdapted for stable/stein (oraclelinux still tested)\n\nChange-Id: If58f15e18e420aca790496132352f1c1dcdb2640\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit 8270ae20f9a8d528d5bea9e275a84a3ef4cf23aa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac66f586981b5333448545403242c2a5ebc90371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac66f586981b5333448545403242c2a5ebc90371"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
