)]}'
{"id":"openstack%2Fkolla~668971","triplet_id":"openstack%2Fkolla~master~I1815c3f30c3208b31501d1b7c2bd6be99db437e5","project":"openstack/kolla","branch":"master","topic":"ci-gate-queue","hashtags":[],"change_id":"I1815c3f30c3208b31501d1b7c2bd6be99db437e5","subject":"CI: set the same gate queue for kolla and kolla-ansible","status":"MERGED","created":"2019-07-03 18:20:47.000000000","updated":"2019-07-04 08:41:17.000000000","submitted":"2019-07-04 07:39:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668971-1562225995696-5004bf0a","meta_rev_id":"a0149ab5ab3f51dab355487a8b24241e471dfd4d","_number":668971,"virtual_id_number":668971,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-07-03 19:57:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-04 07:39:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-07-04 06:42:38.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-07-03 18:30:11.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":1,"date":"2019-07-04 06:42:38.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-03 18:30:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-04 06:42:38.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-07-04 07:39:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3f59db6aba6b6d9309d5262ad3c05b43f2125ef1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-03 18:20:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69cba7c49967c5a7abd9d358d72dd02e8c3f7a16","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-03 18:30:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a7df7289561364a720c81b752c1dfba9cef037a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 19:56:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/71/668971/1/check/kolla-build-centos-binary/9a6b8b0/ : SUCCESS in 46m 33s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/668971/1/check/kolla-build-centos-source/883f048/ : SUCCESS in 49m 41s\n- kolla-ansible-centos-source http://logs.openstack.org/71/668971/1/check/kolla-ansible-centos-source/cf79bda/ : SUCCESS in 1h 08m 55s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/668971/1/check/kolla-ansible-centos-binary/029f789/ : SUCCESS in 1h 05m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/71/668971/1/check/kolla-ansible-centos-source-upgrade/e47e2bd/ : SUCCESS in 1h 26m 55s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/71/668971/1/check/tripleo-build-containers-centos-7/73c79ad/ : SUCCESS in 40m 27s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/668971/1/check/openstack-tox-lower-constraints/5568df2/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/71/668971/1/check/openstack-tox-pep8/3833a3e/ : SUCCESS in 3m 24s\n- openstack-tox-py27 http://logs.openstack.org/71/668971/1/check/openstack-tox-py27/bb80b00/ : SUCCESS in 3m 18s\n- openstack-tox-py36 http://logs.openstack.org/71/668971/1/check/openstack-tox-py36/2f5fbbb/ : SUCCESS in 3m 30s\n- openstack-tox-py37 http://logs.openstack.org/71/668971/1/check/openstack-tox-py37/570c48d/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/71/668971/1/check/openstack-tox-docs/3e0fe63/html/ : SUCCESS in 3m 00s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/668971/1/check/kolla-build-ubuntu-binary/ea53122/ : FAILURE in 22m 45s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/668971/1/check/kolla-build-ubuntu-source/3601def/ : FAILURE in 26m 20s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/668971/1/check/kolla-ansible-ubuntu-source/d2d3521/ : SUCCESS in 46m 25s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/668971/1/check/kolla-ansible-ubuntu-binary/bd1ee23/ : SUCCESS in 48m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/71/668971/1/check/kolla-ansible-ubuntu-source-upgrade/533491c/ : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7773dc2d1cb53b3f9d0dd7c8c513af0e092353b9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-03 19:57:54.000000000","message":"Patch Set 1:\n\nrecheck\n\niad.rax","accounts_in_message":[],"_revision_number":1},{"id":"b81649282a627047e39fde47572d805049b424ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 21:30:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/71/668971/1/check/kolla-build-centos-binary/a6c16f2/ : SUCCESS in 32m 38s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/668971/1/check/kolla-build-centos-source/66dea9d/ : SUCCESS in 54m 08s\n- kolla-ansible-centos-source http://logs.openstack.org/71/668971/1/check/kolla-ansible-centos-source/1ddb045/ : SUCCESS in 1h 12m 50s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/668971/1/check/kolla-ansible-centos-binary/9ced503/ : SUCCESS in 1h 04m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/71/668971/1/check/kolla-ansible-centos-source-upgrade/b21cf48/ : SUCCESS in 1h 26m 29s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/71/668971/1/check/tripleo-build-containers-centos-7/5ebe4e1/ : SUCCESS in 49m 45s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/668971/1/check/openstack-tox-lower-constraints/97cedbb/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/71/668971/1/check/openstack-tox-pep8/51fe769/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/71/668971/1/check/openstack-tox-py27/751e7ac/ : SUCCESS in 3m 29s\n- openstack-tox-py36 http://logs.openstack.org/71/668971/1/check/openstack-tox-py36/44c9b7b/ : SUCCESS in 3m 26s\n- openstack-tox-py37 http://logs.openstack.org/71/668971/1/check/openstack-tox-py37/5c24908/ : SUCCESS in 3m 32s\n- openstack-tox-docs http://logs.openstack.org/71/668971/1/check/openstack-tox-docs/46111d7/html/ : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/668971/1/check/kolla-build-ubuntu-binary/0a4520b/ : SUCCESS in 26m 12s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/668971/1/check/kolla-build-ubuntu-source/785e070/ : SUCCESS in 33m 30s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/668971/1/check/kolla-ansible-ubuntu-source/6bfa26e/ : SUCCESS in 53m 08s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/668971/1/check/kolla-ansible-ubuntu-binary/819b49c/ : SUCCESS in 41m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/71/668971/1/check/kolla-ansible-ubuntu-source-upgrade/ebae502/ : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":1},{"id":"dfcf77e8d25da5739d746497a39d3efa8ac4e5c1","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-07-04 06:42:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d71be3f798ccaa0aa004755d470b4069552cd20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 06:42:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8cbd1eb106419c14ac0b8e61b4ede184c40a27dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 07:39:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/71/668971/1/gate/kolla-build-centos-source/c432841/ : SUCCESS in 54m 37s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/71/668971/1/gate/tripleo-build-containers-centos-7/5859962/ : SUCCESS in 47m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/71/668971/1/gate/openstack-tox-lower-constraints/2357d4b/ : SUCCESS in 3m 53s\n- openstack-tox-pep8 http://logs.openstack.org/71/668971/1/gate/openstack-tox-pep8/b8de39f/ : SUCCESS in 3m 33s\n- openstack-tox-py27 http://logs.openstack.org/71/668971/1/gate/openstack-tox-py27/e5ce343/ : SUCCESS in 4m 02s\n- openstack-tox-py36 http://logs.openstack.org/71/668971/1/gate/openstack-tox-py36/9541ad9/ : SUCCESS in 4m 22s\n- openstack-tox-py37 http://logs.openstack.org/71/668971/1/gate/openstack-tox-py37/3904074/ : SUCCESS in 5m 15s\n- openstack-tox-docs http://logs.openstack.org/71/668971/1/gate/openstack-tox-docs/81e8321/html/ : SUCCESS in 3m 06s\n- kolla-build-ubuntu-source http://logs.openstack.org/71/668971/1/gate/kolla-build-ubuntu-source/b6ac867/ : SUCCESS in 26m 20s","accounts_in_message":[],"_revision_number":1},{"id":"923c8c4641aa27f5f10a6a3b4767c795627ab325","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 07:39:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f812f014098e8ed09f19d2b87d4890947fd4c834","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-04 08:40:58.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit ae53af6922b2fbbbe7f344eec4c04d94eb5cdfa6","accounts_in_message":[],"_revision_number":1},{"id":"da67782f04ec14feb30b48fe21b59d2358610f8c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-04 08:41:10.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit e2175e66d12417e0ae359b5d5728a95ded58ea01","accounts_in_message":[],"_revision_number":1},{"id":"f49f1de9fd768e7841c7e5e1526cd677dce7bf02","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-04 08:41:17.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit b01e581a86d046db2b827779b76ba79045463596","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4884203a9ed8e1a2e67d3877e195359d4e3ad726","revisions":{"4884203a9ed8e1a2e67d3877e195359d4e3ad726":{"kind":"REWORK","_number":1,"created":"2019-07-03 18:20:47.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/71/668971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/668971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/668971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/668971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/668971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/668971/1"}}},"commit":{"parents":[{"commit":"0d6fbd872e9f304bb3d17a094f1fd19c86b25ece","subject":"Merge \"Blacklist sphinx 2.1.0 (autodoc bug)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d6fbd872e9f304bb3d17a094f1fd19c86b25ece"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-07-03 18:18:54.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-07-03 18:18:54.000000000","tz":120},"subject":"CI: set the same gate queue for kolla and kolla-ansible","message":"CI: set the same gate queue for kolla and kolla-ansible\n\nThis is to ensure that any Depends-On does not cause Zuul not to pick up\nthe change for gating due to no notifications between queues.\nPreviously W+1-ing a change which depended on non-merged change from\nthe other project caused it to remain in the same state.\n\nChange-Id: I1815c3f30c3208b31501d1b7c2bd6be99db437e5\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4884203a9ed8e1a2e67d3877e195359d4e3ad726"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4884203a9ed8e1a2e67d3877e195359d4e3ad726"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
