)]}'
{"id":"openstack%2Fkolla~669104","triplet_id":"openstack%2Fkolla~stable%2Frocky~I1815c3f30c3208b31501d1b7c2bd6be99db437e5","project":"openstack/kolla","branch":"stable/rocky","topic":"ci-gate-queue-stable/rocky","hashtags":[],"change_id":"I1815c3f30c3208b31501d1b7c2bd6be99db437e5","subject":"CI: set the same gate queue for kolla and kolla-ansible","status":"MERGED","created":"2019-07-04 08:41:10.000000000","updated":"2019-07-05 10:40:23.000000000","submitted":"2019-07-05 10:40:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669104-1562323223667-769ffea3","meta_rev_id":"fe9fbe0738bf3f7ffc09b928d904170a65dc5467","_number":669104,"virtual_id_number":669104,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-05 10:40:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-07-05 09:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-07-04 08:53:03.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-07-05 09:19:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 08:53:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-05 09:19:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-05 10:40:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a69eb7d1e7bee9ef69680805c24c6c4929f7dbd1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-04 08:41:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e3bb9048d96975a45a73460ede849ce4c7e4bac9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 08:53:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1220a319ef8c1bdfe05ef07b3f569ced1fd5be21","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-04 08:53:08.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"dfb11534719f94fc15db0c394577e9ba64979b79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 11:23:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/04/669104/1/check/kolla-build-centos-binary/da141c6/ : SUCCESS in 39m 29s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/04/669104/1/check/kolla-build-centos-source/c75f137/ : SUCCESS in 31m 06s\n- kolla-ansible-centos-source http://logs.openstack.org/04/669104/1/check/kolla-ansible-centos-source/999d937/ : SUCCESS in 1h 16m 41s\n- kolla-ansible-centos-binary http://logs.openstack.org/04/669104/1/check/kolla-ansible-centos-binary/5f437a2/ : SUCCESS in 1h 14m 39s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/04/669104/1/check/kolla-build-oraclelinux-binary/7207985/ : SUCCESS in 39m 22s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/04/669104/1/check/kolla-build-oraclelinux-source/cddb49a/ : SUCCESS in 49m 48s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/04/669104/1/check/kolla-ansible-oraclelinux-source/8ee21ff/ : SUCCESS in 1h 16m 31s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/04/669104/1/check/kolla-ansible-oraclelinux-binary/8848f91/ : SUCCESS in 1h 09m 04s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/04/669104/1/check/openstack-tox-lower-constraints/e9e6ee3/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/04/669104/1/check/openstack-tox-pep8/713346c/ : SUCCESS in 3m 44s\n- openstack-tox-py27 http://logs.openstack.org/04/669104/1/check/openstack-tox-py27/b122d20/ : SUCCESS in 3m 43s\n- openstack-tox-py35 http://logs.openstack.org/04/669104/1/check/openstack-tox-py35/67c17ad/ : SUCCESS in 6m 27s\n- build-openstack-sphinx-docs http://logs.openstack.org/04/669104/1/check/build-openstack-sphinx-docs/2273296/html/ : SUCCESS in 3m 25s\n- kolla-build-ubuntu-binary http://logs.openstack.org/04/669104/1/check/kolla-build-ubuntu-binary/ace2618/ : SUCCESS in 1h 24m 48s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/04/669104/1/check/kolla-build-ubuntu-source/76b5aaa/ : SUCCESS in 35m 04s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/04/669104/1/check/kolla-ansible-ubuntu-source/3567a39/ : SUCCESS in 58m 21s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/04/669104/1/check/kolla-ansible-ubuntu-binary/143e536/ : SUCCESS in 1h 54m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68341b4d271043b6c4960518f0ea6835cb7d2c6b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-05 09:19:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"65409384c8a3872e9ba304c8a0f73db828513159","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 09:20:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5923fd9c967f363351a4d04eb373a353e8a85a72","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 10:40:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bf13ceb005c0b9da8ccae93a1c33c78a0aad32bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 10:40:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/04/669104/1/gate/kolla-build-centos-source/e4696b9/ : SUCCESS in 49m 00s\n- kolla-build-oraclelinux-source http://logs.openstack.org/04/669104/1/gate/kolla-build-oraclelinux-source/4c08d73/ : SUCCESS in 41m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/04/669104/1/gate/openstack-tox-lower-constraints/764488a/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/04/669104/1/gate/openstack-tox-pep8/fbe4b7a/ : SUCCESS in 4m 08s\n- openstack-tox-py27 http://logs.openstack.org/04/669104/1/gate/openstack-tox-py27/f9196aa/ : SUCCESS in 3m 09s\n- openstack-tox-py35 http://logs.openstack.org/04/669104/1/gate/openstack-tox-py35/b07ad17/ : SUCCESS in 3m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/04/669104/1/gate/build-openstack-sphinx-docs/a37b6db/html/ : SUCCESS in 4m 23s\n- kolla-build-ubuntu-source http://logs.openstack.org/04/669104/1/gate/kolla-build-ubuntu-source/b532f7e/ : SUCCESS in 31m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e2175e66d12417e0ae359b5d5728a95ded58ea01","revisions":{"e2175e66d12417e0ae359b5d5728a95ded58ea01":{"kind":"REWORK","_number":1,"created":"2019-07-04 08:41:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/04/669104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/669104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/669104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/669104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/669104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/669104/1"}}},"commit":{"parents":[{"commit":"ffdd9faf5c631980b017deefd509af67bcb6d257","subject":"Merge \"Fix deployment of ceph-nfs container failing due to wrong USER\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ffdd9faf5c631980b017deefd509af67bcb6d257"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-07-03 18:18:54.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-07-04 08:41:10.000000000","tz":0},"subject":"CI: set the same gate queue for kolla and kolla-ansible","message":"CI: set the same gate queue for kolla and kolla-ansible\n\nThis is to ensure that any Depends-On does not cause Zuul not to pick up\nthe change for gating due to no notifications between queues.\nPreviously W+1-ing a change which depended on non-merged change from\nthe other project caused it to remain in the same state.\n\nChange-Id: I1815c3f30c3208b31501d1b7c2bd6be99db437e5\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit 4884203a9ed8e1a2e67d3877e195359d4e3ad726)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2175e66d12417e0ae359b5d5728a95ded58ea01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2175e66d12417e0ae359b5d5728a95ded58ea01"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
