)]}'
{"id":"openstack%2Fkolla~669375","triplet_id":"openstack%2Fkolla~master~I3cb7d2c482cf6d57e01892b6ba0f76d9580ff53c","project":"openstack/kolla","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I3cb7d2c482cf6d57e01892b6ba0f76d9580ff53c","subject":"Add Python 3 Train unit tests","status":"MERGED","created":"2019-07-05 15:43:44.000000000","updated":"2019-07-08 17:00:12.000000000","submitted":"2019-07-08 17:00:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669375-1562605212757-eaa21e4c","meta_rev_id":"e27fbf55dd82b631160c23974bee934907345bcb","_number":669375,"virtual_id_number":669375,"owner":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-08 17:00:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-08 15:53:38.000000000","_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-07-08 08:32:28.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-07-08 16:01:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-07-08 16:01:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-08 08:32:28.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-07-08 16:01:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-08 17:00:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d909f006a54a6113511d2f1ebe5b7970e248b05","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-05 15:43:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a9c1377ef01b494ffa6bcbf15b45b256e88e86a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 17:18:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/75/669375/1/check/kolla-build-centos-binary/81c79f2/ : SUCCESS in 40m 48s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/75/669375/1/check/kolla-build-centos-source/b438673/ : SUCCESS in 52m 11s\n- kolla-ansible-centos-source http://logs.openstack.org/75/669375/1/check/kolla-ansible-centos-source/8c11213/ : SUCCESS in 1h 11m 15s\n- kolla-ansible-centos-binary http://logs.openstack.org/75/669375/1/check/kolla-ansible-centos-binary/295552b/ : SUCCESS in 1h 06m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/75/669375/1/check/kolla-ansible-centos-source-upgrade/8eaaada/ : SUCCESS in 1h 28m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/669375/1/check/openstack-tox-lower-constraints/f7ea955/ : SUCCESS in 3m 13s\n- openstack-tox-pep8 http://logs.openstack.org/75/669375/1/check/openstack-tox-pep8/8fd2f82/ : SUCCESS in 2m 40s\n- openstack-tox-py27 http://logs.openstack.org/75/669375/1/check/openstack-tox-py27/d4937fc/ : SUCCESS in 3m 01s\n- openstack-tox-py36 http://logs.openstack.org/75/669375/1/check/openstack-tox-py36/dce9baf/ : SUCCESS in 3m 02s\n- openstack-tox-py37 http://logs.openstack.org/75/669375/1/check/openstack-tox-py37/04b0712/ : SUCCESS in 3m 17s\n- openstack-tox-docs http://logs.openstack.org/75/669375/1/check/openstack-tox-docs/bbc23d7/html/ : SUCCESS in 3m 11s\n- build-openstack-releasenotes http://logs.openstack.org/75/669375/1/check/build-openstack-releasenotes/b96250c/html/ : SUCCESS in 4m 02s\n- kolla-build-ubuntu-binary http://logs.openstack.org/75/669375/1/check/kolla-build-ubuntu-binary/c1f00c9/ : SUCCESS in 24m 56s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/75/669375/1/check/kolla-build-ubuntu-source/5b86b7d/ : SUCCESS in 25m 39s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/75/669375/1/check/kolla-ansible-ubuntu-source/0790092/ : SUCCESS in 48m 13s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/75/669375/1/check/kolla-ansible-ubuntu-binary/c864dff/ : SUCCESS in 51m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/75/669375/1/check/kolla-ansible-ubuntu-source-upgrade/80b1b93/ : SUCCESS in 1h 05m 39s","accounts_in_message":[],"_revision_number":1},{"id":"dd5445e3f2efd276f930bab557078f03e46d4d8b","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-07-08 08:32:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a07c532ad98f0d9c32c12cf4a030e3888e3e2606","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 09:14:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66d727298caefa49e12df8e51b12dd32b2777ffe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 09:16:42.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"77474e87ffb21c11a3f22057764d3b7aa96f0d14","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2019-07-08 15:53:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88b0e1ab1ebb5df5d79dae39aa8828d956780063","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-08 16:01:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a01bc53680c8669dc19f7641d3e7d08858f3e58c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 16:01:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a67fa792eb655211d000e84efb5a7895dfddc967","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 17:00:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/75/669375/1/gate/kolla-build-centos-source/d8d8513/ : SUCCESS in 52m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/75/669375/1/gate/openstack-tox-lower-constraints/367e4af/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/75/669375/1/gate/openstack-tox-pep8/875b111/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/75/669375/1/gate/openstack-tox-py27/c8b2f92/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/75/669375/1/gate/openstack-tox-py36/dc7a748/ : SUCCESS in 3m 25s\n- openstack-tox-py37 http://logs.openstack.org/75/669375/1/gate/openstack-tox-py37/9ae3091/ : SUCCESS in 3m 18s\n- openstack-tox-docs http://logs.openstack.org/75/669375/1/gate/openstack-tox-docs/1adebbb/html/ : SUCCESS in 2m 49s\n- build-openstack-releasenotes http://logs.openstack.org/75/669375/1/gate/build-openstack-releasenotes/9f27118/html/ : SUCCESS in 4m 14s\n- kolla-build-ubuntu-source http://logs.openstack.org/75/669375/1/gate/kolla-build-ubuntu-source/3fbd16a/ : SUCCESS in 30m 44s","accounts_in_message":[],"_revision_number":1},{"id":"766d66d1027973efb95d4fbb5774c316ad76d3ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-08 17:00:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"008aa957f27b26caff3be8b18110e71cdaf70772","revisions":{"008aa957f27b26caff3be8b18110e71cdaf70772":{"kind":"REWORK","_number":1,"created":"2019-07-05 15:43:44.000000000","uploader":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"ref":"refs/changes/75/669375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/669375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/669375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/669375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/669375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/669375/1"}}},"commit":{"parents":[{"commit":"c6b88c581dd7e856dcc241c4b46210e50face7dd","subject":"Merge \"CI: Clean up Zuul repo mirrors for CentOS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6b88c581dd7e856dcc241c4b46210e50face7dd"}]}],"author":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:38:37.000000000","tz":-240},"committer":{"name":"Corey Bryant","email":"corey.bryant@canonical.com","date":"2019-07-05 15:43:43.000000000","tz":-240},"subject":"Add Python 3 Train unit tests","message":"Add Python 3 Train unit tests\n\nThis is a mechanically generated patch to ensure unit testing is in place\nfor all of the Tested Runtimes for Train.\n\nSee the Train python3-updates goal document for details:\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I3cb7d2c482cf6d57e01892b6ba0f76d9580ff53c\nStory: #2005924\nTask: #34216\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/008aa957f27b26caff3be8b18110e71cdaf70772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/008aa957f27b26caff3be8b18110e71cdaf70772"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
