)]}'
{"id":"openstack%2Fkolla~669471","triplet_id":"openstack%2Fkolla~master~I3bcef4fbba1f7c59b564446552532ea949bc34c3","project":"openstack/kolla","branch":"master","topic":"rhel8-containers-buildupdate","hashtags":[],"change_id":"I3bcef4fbba1f7c59b564446552532ea949bc34c3","subject":"DNM: Add rhel8 support for early CentOS8 testing","status":"ABANDONED","created":"2019-07-05 19:19:47.000000000","updated":"2020-02-07 16:28:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5095239a79a86de11cd9d67eebe7c98112cc2380","_number":669471,"virtual_id_number":669471,"owner":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"date":"2020-02-06 10:12:04.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-31 19:54:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-01-31 17:59:32.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"date":"2020-02-07 16:28:53.000000000","_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 08:17:34.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-11-25 12:12:07.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-01-31 17:59:32.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-31 19:54:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-06 10:12:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"4bb48affd3e70719593fd1cb5dc139daf5456529","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-05 19:19:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8ae72cd280cce3969131a2ae5f3258fa6391a40","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-05 19:55:48.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0244f022148fba8e9590850d27f674840e5eca11","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-05 20:31:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/71/669471/2/openstack-check/tripleo-build-containers-fedora-28-master/0471646/ : FAILURE in 5m 41s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/71/669471/2/openstack-check/tripleo-build-containers-fedora-28-stein/c5878fa/ : SUCCESS in 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"707318a44eac74040b95ac87d745363cf3f6ad61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 21:51:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/71/669471/2/check/kolla-build-centos-binary/89af369/ : SUCCESS in 42m 59s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/669471/2/check/kolla-build-centos-source/8bf34fe/ : SUCCESS in 48m 16s\n- kolla-ansible-centos-source http://logs.openstack.org/71/669471/2/check/kolla-ansible-centos-source/68f71c8/ : SUCCESS in 1h 23m 28s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/669471/2/check/kolla-ansible-centos-binary/6a83f11/ : SUCCESS in 1h 05m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/71/669471/2/check/kolla-ansible-centos-source-upgrade/1077153/ : SUCCESS in 1h 16m 23s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/71/669471/2/check/tripleo-build-containers-centos-7/e435c60/ : SUCCESS in 46m 08s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/71/669471/2/check/kolla-ansible-bifrost-centos-source/6947f1c/ : SUCCESS in 47m 06s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/71/669471/2/check/kolla-ansible-centos-source-ceph/9c8c908/ : SUCCESS in 1h 27m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/71/669471/2/check/kolla-ansible-centos-source-upgrade-ceph/d8afe26/ : SUCCESS in 1h 41m 29s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm http://logs.openstack.org/71/669471/2/check/kolla-ansible-centos-source-cinder-lvm/ac5b891/ : SUCCESS in 1h 23m 10s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv http://logs.openstack.org/71/669471/2/check/kolla-ansible-centos-source-scenario-nfv/e2793af/ : SUCCESS in 1h 25m 06s (non-voting)\n- kolla-ansible-centos-source-ironic http://logs.openstack.org/71/669471/2/check/kolla-ansible-centos-source-ironic/09c07eb/ : SUCCESS in 1h 12m 59s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/669471/2/check/openstack-tox-lower-constraints/b9f174f/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/71/669471/2/check/openstack-tox-pep8/649fde7/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/71/669471/2/check/openstack-tox-py27/2f28b52/ : SUCCESS in 3m 16s\n- openstack-tox-py36 http://logs.openstack.org/71/669471/2/check/openstack-tox-py36/f42a246/ : SUCCESS in 3m 50s\n- openstack-tox-py37 http://logs.openstack.org/71/669471/2/check/openstack-tox-py37/61882e7/ : SUCCESS in 5m 27s\n- openstack-tox-docs http://logs.openstack.org/71/669471/2/check/openstack-tox-docs/49945c1/html/ : SUCCESS in 3m 20s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/669471/2/check/kolla-build-ubuntu-binary/940324f/ : SUCCESS in 26m 33s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/669471/2/check/kolla-build-ubuntu-source/c5f006a/ : SUCCESS in 28m 24s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/669471/2/check/kolla-ansible-ubuntu-source/cb4cc2e/ : SUCCESS in 54m 34s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/669471/2/check/kolla-ansible-ubuntu-binary/68866f9/ : SUCCESS in 46m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/71/669471/2/check/kolla-ansible-ubuntu-source-upgrade/827988a/ : SUCCESS in 48m 47s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/71/669471/2/check/kolla-ansible-ubuntu-source-zun/3727e2f/ : SUCCESS in 46m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/71/669471/2/check/kolla-ansible-ubuntu-source-ceph/2c454cd/ : SUCCESS in 1h 00m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/71/669471/2/check/kolla-ansible-ubuntu-source-upgrade-ceph/0b80efc/ : SUCCESS in 1h 15m 18s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm http://logs.openstack.org/71/669471/2/check/kolla-ansible-ubuntu-source-cinder-lvm/1047cb4/ : SUCCESS in 1h 02m 34s (non-voting)\n- kolla-ansible-ubuntu-source-ironic http://logs.openstack.org/71/669471/2/check/kolla-ansible-ubuntu-source-ironic/ce88e3b/ : SUCCESS in 59m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d94b0226522ff97abb6ed187444a7bb633ac56b4","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-07-11 22:33:25.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"fc7c560625db442679a98ea61ca7cea34e638b17","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 23:12:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/71/669471/3/openstack-check/tripleo-build-containers-fedora-28-master/b4878f6/ : FAILURE in 5m 25s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/71/669471/3/openstack-check/tripleo-build-containers-fedora-28-stein/fd0c19c/ : SUCCESS in 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d4d4c48c1660098c7ef8a2a17d15fdcb9260d31f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 00:26:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/71/669471/3/check/kolla-build-centos-binary/119b776/ : SUCCESS in 41m 53s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/669471/3/check/kolla-build-centos-source/3491db4/ : SUCCESS in 54m 29s\n- kolla-ansible-centos-source http://logs.openstack.org/71/669471/3/check/kolla-ansible-centos-source/39dfc49/ : SUCCESS in 1h 08m 52s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/669471/3/check/kolla-ansible-centos-binary/acf3f42/ : SUCCESS in 1h 01m 39s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/71/669471/3/check/kolla-ansible-centos-source-upgrade/c14292f/ : SUCCESS in 1h 20m 24s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/71/669471/3/check/tripleo-build-containers-centos-7/36ddf66/ : SUCCESS in 46m 55s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/71/669471/3/check/kolla-ansible-bifrost-centos-source/5a4e142/ : SUCCESS in 45m 40s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/71/669471/3/check/kolla-ansible-centos-source-ceph/cddb6de/ : SUCCESS in 1h 33m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/71/669471/3/check/kolla-ansible-centos-source-upgrade-ceph/031d029/ : SUCCESS in 1h 46m 35s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm http://logs.openstack.org/71/669471/3/check/kolla-ansible-centos-source-cinder-lvm/694dd68/ : SUCCESS in 1h 27m 46s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv http://logs.openstack.org/71/669471/3/check/kolla-ansible-centos-source-scenario-nfv/ee3906a/ : SUCCESS in 1h 31m 00s (non-voting)\n- kolla-ansible-centos-source-ironic http://logs.openstack.org/71/669471/3/check/kolla-ansible-centos-source-ironic/3d59a2c/ : SUCCESS in 1h 14m 21s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/669471/3/check/openstack-tox-lower-constraints/09612f2/ : SUCCESS in 3m 48s\n- openstack-tox-pep8 http://logs.openstack.org/71/669471/3/check/openstack-tox-pep8/248c08a/ : SUCCESS in 3m 19s\n- openstack-tox-py27 http://logs.openstack.org/71/669471/3/check/openstack-tox-py27/a6fb8ff/ : SUCCESS in 3m 13s\n- openstack-tox-py36 http://logs.openstack.org/71/669471/3/check/openstack-tox-py36/e050e47/ : SUCCESS in 3m 08s\n- openstack-tox-py37 http://logs.openstack.org/71/669471/3/check/openstack-tox-py37/f1ba849/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/71/669471/3/check/openstack-tox-docs/2087b1c/html/ : SUCCESS in 2m 51s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/669471/3/check/kolla-build-ubuntu-binary/883446d/ : SUCCESS in 20m 11s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/669471/3/check/kolla-build-ubuntu-source/e276148/ : SUCCESS in 30m 14s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/669471/3/check/kolla-ansible-ubuntu-source/7cd798b/ : SUCCESS in 48m 05s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/669471/3/check/kolla-ansible-ubuntu-binary/6a87d10/ : SUCCESS in 45m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/71/669471/3/check/kolla-ansible-ubuntu-source-upgrade/1fb03bd/ : SUCCESS in 59m 33s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/71/669471/3/check/kolla-ansible-ubuntu-source-zun/b7ea6a9/ : SUCCESS in 47m 53s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/71/669471/3/check/kolla-ansible-ubuntu-source-ceph/31b1f26/ : SUCCESS in 1h 04m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/71/669471/3/check/kolla-ansible-ubuntu-source-upgrade-ceph/01a6005/ : SUCCESS in 1h 24m 51s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm http://logs.openstack.org/71/669471/3/check/kolla-ansible-ubuntu-source-cinder-lvm/4f16752/ : SUCCESS in 1h 07m 50s (non-voting)\n- kolla-ansible-ubuntu-source-ironic http://logs.openstack.org/71/669471/3/check/kolla-ansible-ubuntu-source-ironic/b33411d/ : SUCCESS in 44m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e02c393b41dca9cc694666f94bea75dbd4f69fe2","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-15 09:38:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"60d196035b756a3c43ce4726a4987c1ffb628ae5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 10:17:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/71/669471/4/openstack-check/tripleo-build-containers-fedora-28-master/8a6f639/ : FAILURE in 5m 25s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/71/669471/4/openstack-check/tripleo-build-containers-fedora-28-stein/1734929/ : SUCCESS in 38m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2486b45951af2fbeab97a286588ea526e08bbd2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 11:26:04.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/71/669471/4/check/kolla-build-centos-binary/5d2f863/ : SUCCESS in 31m 23s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/669471/4/check/kolla-build-centos-source/2854183/ : SUCCESS in 49m 38s\n- kolla-ansible-centos-source http://logs.openstack.org/71/669471/4/check/kolla-ansible-centos-source/435c50c/ : SUCCESS in 1h 15m 44s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/669471/4/check/kolla-ansible-centos-binary/4e7fbe7/ : SUCCESS in 1h 09m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/71/669471/4/check/kolla-ansible-centos-source-upgrade/843abca/ : SUCCESS in 1h 24m 44s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/71/669471/4/check/tripleo-build-containers-centos-7/9fd7c61/ : SUCCESS in 48m 19s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/71/669471/4/check/kolla-ansible-bifrost-centos-source/0ab3b13/ : SUCCESS in 42m 48s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/71/669471/4/check/kolla-ansible-centos-source-ceph/5066555/ : SUCCESS in 1h 41m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/71/669471/4/check/kolla-ansible-centos-source-upgrade-ceph/eb45e3d/ : SUCCESS in 1h 43m 03s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm http://logs.openstack.org/71/669471/4/check/kolla-ansible-centos-source-cinder-lvm/70a680f/ : FAILURE in 1h 16m 21s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv http://logs.openstack.org/71/669471/4/check/kolla-ansible-centos-source-scenario-nfv/16280a2/ : SUCCESS in 1h 30m 24s (non-voting)\n- kolla-ansible-centos-source-ironic http://logs.openstack.org/71/669471/4/check/kolla-ansible-centos-source-ironic/a1fdb8f/ : SUCCESS in 1h 18m 15s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/669471/4/check/openstack-tox-lower-constraints/06bfa36/ : SUCCESS in 3m 12s\n- openstack-tox-pep8 http://logs.openstack.org/71/669471/4/check/openstack-tox-pep8/b01a415/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/71/669471/4/check/openstack-tox-py27/edae7e2/ : SUCCESS in 4m 37s\n- openstack-tox-py36 http://logs.openstack.org/71/669471/4/check/openstack-tox-py36/d452e36/ : SUCCESS in 4m 41s\n- openstack-tox-py37 http://logs.openstack.org/71/669471/4/check/openstack-tox-py37/99dc0c3/ : SUCCESS in 5m 04s\n- openstack-tox-docs http://logs.openstack.org/71/669471/4/check/openstack-tox-docs/7d50d9e/html/ : SUCCESS in 3m 46s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/669471/4/check/kolla-build-ubuntu-binary/837e57a/ : SUCCESS in 24m 30s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/669471/4/check/kolla-build-ubuntu-source/438afde/ : SUCCESS in 1h 02m 43s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/669471/4/check/kolla-ansible-ubuntu-source/733ea9a/ : SUCCESS in 1h 25m 16s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/669471/4/check/kolla-ansible-ubuntu-binary/fe3fd11/ : SUCCESS in 44m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/71/669471/4/check/kolla-ansible-ubuntu-source-upgrade/2a63312/ : SUCCESS in 56m 33s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/71/669471/4/check/kolla-ansible-ubuntu-source-zun/557f33d/ : SUCCESS in 51m 40s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/71/669471/4/check/kolla-ansible-ubuntu-source-ceph/f66adec/ : SUCCESS in 1h 07m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/71/669471/4/check/kolla-ansible-ubuntu-source-upgrade-ceph/45b26c9/ : SUCCESS in 1h 24m 17s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm http://logs.openstack.org/71/669471/4/check/kolla-ansible-ubuntu-source-cinder-lvm/6bd6a81/ : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-ansible-ubuntu-source-ironic http://logs.openstack.org/71/669471/4/check/kolla-ansible-ubuntu-source-ironic/a23da06/ : SUCCESS in 1h 05m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"96f723bb3ed534e9b05f794ec45d1001786c4309","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-07-15 20:57:41.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d83c8c541f1ba728442e3ceb5e3c7489824061b0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-15 21:45:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/71/669471/5/openstack-check/tripleo-build-containers-fedora-28-master/cb79b2a/ : FAILURE in 5m 55s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/71/669471/5/openstack-check/tripleo-build-containers-fedora-28-stein/ec48c4e/ : SUCCESS in 46m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"82a698f215d67a511d736b7d9df562f21f4083a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 21:59:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/71/669471/5/check/kolla-build-centos-binary/71dda3f/ : FAILURE in 5m 42s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/669471/5/check/kolla-build-centos-source/bb1c348/ : FAILURE in 55m 28s\n- kolla-ansible-centos-source http://logs.openstack.org/71/669471/5/check/kolla-ansible-centos-source/faa4f19/ : FAILURE in 34m 21s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/669471/5/check/kolla-ansible-centos-binary/70f18ca/ : FAILURE in 9m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/71/669471/5/check/kolla-ansible-centos-source-upgrade/6ba7112/ : FAILURE in 34m 15s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/71/669471/5/check/tripleo-build-containers-centos-7/db92d48/ : FAILURE in 12m 32s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/71/669471/5/check/kolla-ansible-bifrost-centos-source/00afaf8/ : SUCCESS in 40m 35s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/71/669471/5/check/kolla-ansible-centos-source-ceph/37c6718/ : FAILURE in 40m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/71/669471/5/check/kolla-ansible-centos-source-upgrade-ceph/8a3ad70/ : FAILURE in 40m 46s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm http://logs.openstack.org/71/669471/5/check/kolla-ansible-centos-source-cinder-lvm/4e9e883/ : FAILURE in 45m 23s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv http://logs.openstack.org/71/669471/5/check/kolla-ansible-centos-source-scenario-nfv/2b3cd7e/ : FAILURE in 30m 42s (non-voting)\n- kolla-ansible-centos-source-ironic http://logs.openstack.org/71/669471/5/check/kolla-ansible-centos-source-ironic/b3b3af1/ : FAILURE in 44m 12s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/669471/5/check/openstack-tox-lower-constraints/048c59e/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/71/669471/5/check/openstack-tox-pep8/8b27b2b/ : SUCCESS in 5m 10s\n- openstack-tox-py27 http://logs.openstack.org/71/669471/5/check/openstack-tox-py27/f441955/ : SUCCESS in 4m 18s\n- openstack-tox-py36 http://logs.openstack.org/71/669471/5/check/openstack-tox-py36/6df389b/ : SUCCESS in 4m 15s\n- openstack-tox-py37 http://logs.openstack.org/71/669471/5/check/openstack-tox-py37/0e2560e/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/71/669471/5/check/openstack-tox-docs/4dd850d/html/ : SUCCESS in 2m 30s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/669471/5/check/kolla-build-ubuntu-binary/2de26b1/ : FAILURE in 25m 56s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/669471/5/check/kolla-build-ubuntu-source/c0c3fbf/ : FAILURE in 31m 14s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/669471/5/check/kolla-ansible-ubuntu-source/820058a/ : FAILURE in 17m 03s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/669471/5/check/kolla-ansible-ubuntu-binary/51a6382/ : FAILURE in 15m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/71/669471/5/check/kolla-ansible-ubuntu-source-upgrade/e8206af/ : FAILURE in 21m 17s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/71/669471/5/check/kolla-ansible-ubuntu-source-zun/25757a2/ : FAILURE in 25m 08s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/71/669471/5/check/kolla-ansible-ubuntu-source-ceph/2fe4cbf/ : FAILURE in 21m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/71/669471/5/check/kolla-ansible-ubuntu-source-upgrade-ceph/ce29780/ : FAILURE in 20m 43s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm http://logs.openstack.org/71/669471/5/check/kolla-ansible-ubuntu-source-cinder-lvm/8bb127b/ : FAILURE in 18m 23s (non-voting)\n- kolla-ansible-ubuntu-source-ironic http://logs.openstack.org/71/669471/5/check/kolla-ansible-ubuntu-source-ironic/1a6dcbe/ : FAILURE in 32m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6eec75c162286ce87599fe9513ec8603cb4a904c","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-17 04:19:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cf76f1e87cc32d17807e9b41474e2e9fb3bf2598","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 05:18:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/71/669471/6/openstack-check/tripleo-build-containers-fedora-28-master/a9a1c6e/ : FAILURE in 6m 28s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/71/669471/6/openstack-check/tripleo-build-containers-fedora-28-stein/c49e356/ : SUCCESS in 56m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7df5f7bf5fcf30c4d661c78454afb2602812d5cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 06:08:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/71/669471/6/check/kolla-build-centos-binary/1217b62/ : RETRY_LIMIT in 2m 11s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/669471/6/check/kolla-build-centos-source/0e25fd0/ : SUCCESS in 52m 54s\n- kolla-ansible-centos-source http://logs.openstack.org/71/669471/6/check/kolla-ansible-centos-source/8e27cc1/ : SUCCESS in 1h 04m 01s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/669471/6/check/kolla-ansible-centos-binary/c1866d8/ : SUCCESS in 1h 07m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/71/669471/6/check/kolla-ansible-centos-source-upgrade/d803f96/ : SUCCESS in 1h 27m 14s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/71/669471/6/check/tripleo-build-containers-centos-7/1746d65/ : SUCCESS in 43m 19s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/71/669471/6/check/kolla-ansible-bifrost-centos-source/99c5022/ : SUCCESS in 47m 47s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/71/669471/6/check/kolla-ansible-centos-source-ceph/719f199/ : SUCCESS in 1h 37m 38s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/71/669471/6/check/kolla-ansible-centos-source-upgrade-ceph/6977be0/ : FAILURE in 1h 06m 58s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm http://logs.openstack.org/71/669471/6/check/kolla-ansible-centos-source-cinder-lvm/74974f5/ : FAILURE in 48m 33s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv http://logs.openstack.org/71/669471/6/check/kolla-ansible-centos-source-scenario-nfv/39c14c9/ : FAILURE in 50m 32s (non-voting)\n- kolla-ansible-centos-source-ironic http://logs.openstack.org/71/669471/6/check/kolla-ansible-centos-source-ironic/2c90e84/ : SUCCESS in 1h 25m 32s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/669471/6/check/openstack-tox-lower-constraints/f576a73/ : SUCCESS in 3m 24s\n- openstack-tox-pep8 http://logs.openstack.org/71/669471/6/check/openstack-tox-pep8/843da7c/ : SUCCESS in 5m 41s\n- openstack-tox-py27 http://logs.openstack.org/71/669471/6/check/openstack-tox-py27/ff1bac5/ : SUCCESS in 5m 36s\n- openstack-tox-py36 http://logs.openstack.org/71/669471/6/check/openstack-tox-py36/b0fc8cf/ : SUCCESS in 5m 40s\n- openstack-tox-py37 http://logs.openstack.org/71/669471/6/check/openstack-tox-py37/1a2a218/ : SUCCESS in 5m 32s\n- openstack-tox-docs http://logs.openstack.org/71/669471/6/check/openstack-tox-docs/2848972/html/ : SUCCESS in 3m 31s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/669471/6/check/kolla-build-ubuntu-binary/28e6d36/ : SUCCESS in 41m 32s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/669471/6/check/kolla-build-ubuntu-source/b3041e1/ : SUCCESS in 41m 14s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/669471/6/check/kolla-ansible-ubuntu-source/15ba344/ : SUCCESS in 57m 37s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/669471/6/check/kolla-ansible-ubuntu-binary/0543e06/ : SUCCESS in 58m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/71/669471/6/check/kolla-ansible-ubuntu-source-upgrade/03cb739/ : SUCCESS in 1h 21m 56s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/71/669471/6/check/kolla-ansible-ubuntu-source-zun/1340f56/ : SUCCESS in 59m 26s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/71/669471/6/check/kolla-ansible-ubuntu-source-ceph/ebf322e/ : SUCCESS in 1h 30m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/71/669471/6/check/kolla-ansible-ubuntu-source-upgrade-ceph/c5147f4/ : SUCCESS in 1h 26m 47s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm http://logs.openstack.org/71/669471/6/check/kolla-ansible-ubuntu-source-cinder-lvm/9fa9a10/ : FAILURE in 36m 08s (non-voting)\n- kolla-ansible-ubuntu-source-ironic http://logs.openstack.org/71/669471/6/check/kolla-ansible-ubuntu-source-ironic/f1af71b/ : SUCCESS in 53m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4c46b881b083fabf4966c90e5669e941f321438a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 08:26:33.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"47f48a58a3aad46e2906cb3ed155e98407715534","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 09:15:43.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/71/669471/7/openstack-check/tripleo-build-containers-fedora-28-master/5250d70/ : FAILURE in 6m 49s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/71/669471/7/openstack-check/tripleo-build-containers-fedora-28-stein/17ba2b3/ : SUCCESS in 48m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6b3644303f67c5ee501cc03640568b595545417a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 09:30:19.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"01102efa2d0ac7181e801e94b0d6ba4781c7ce3c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 10:27:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/71/669471/8/openstack-check/tripleo-build-containers-fedora-28-master/c3c327d/ : FAILURE in 5m 31s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/71/669471/8/openstack-check/tripleo-build-containers-fedora-28-stein/6800f3a/ : SUCCESS in 52m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d25d9082371402a8c50b014c34dd83dcbfb72f09","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 10:33:23.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"8037e63f153360d64f398a5a1f58e527b76e0478","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 11:35:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/71/669471/9/openstack-check/tripleo-build-containers-fedora-28-master/f3af950/ : FAILURE in 6m 26s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/71/669471/9/openstack-check/tripleo-build-containers-fedora-28-stein/7b23a19/ : FAILURE in 49m 09s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f61f96a379a932b59de439254327539c4e4611e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 12:26:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/71/669471/9/check/kolla-build-centos-binary/deddf12/ : SUCCESS in 53m 47s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/669471/9/check/kolla-build-centos-source/1bacc94/ : SUCCESS in 51m 22s\n- kolla-ansible-centos-source http://logs.openstack.org/71/669471/9/check/kolla-ansible-centos-source/4566578/ : SUCCESS in 1h 13m 13s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/669471/9/check/kolla-ansible-centos-binary/57914ce/ : SUCCESS in 1h 02m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/71/669471/9/check/kolla-ansible-centos-source-upgrade/dbd42c8/ : SUCCESS in 1h 17m 02s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/71/669471/9/check/tripleo-build-containers-centos-7/802d177/ : SUCCESS in 48m 57s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/71/669471/9/check/kolla-ansible-bifrost-centos-source/4c9f3f4/ : SUCCESS in 53m 53s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/71/669471/9/check/kolla-ansible-centos-source-ceph/48db59b/ : SUCCESS in 1h 33m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/71/669471/9/check/kolla-ansible-centos-source-upgrade-ceph/163dc2e/ : SUCCESS in 1h 44m 27s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm http://logs.openstack.org/71/669471/9/check/kolla-ansible-centos-source-cinder-lvm/bbc4979/ : SUCCESS in 1h 38m 29s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv http://logs.openstack.org/71/669471/9/check/kolla-ansible-centos-source-scenario-nfv/cd71ca3/ : SUCCESS in 1h 46m 29s (non-voting)\n- kolla-ansible-centos-source-ironic http://logs.openstack.org/71/669471/9/check/kolla-ansible-centos-source-ironic/7342d34/ : FAILURE in 1h 20m 38s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/669471/9/check/openstack-tox-lower-constraints/6fc71b9/ : SUCCESS in 3m 26s\n- openstack-tox-pep8 http://logs.openstack.org/71/669471/9/check/openstack-tox-pep8/d6708b9/ : SUCCESS in 4m 19s\n- openstack-tox-py27 http://logs.openstack.org/71/669471/9/check/openstack-tox-py27/a66098d/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/71/669471/9/check/openstack-tox-py36/fd12f5a/ : SUCCESS in 5m 31s\n- openstack-tox-py37 http://logs.openstack.org/71/669471/9/check/openstack-tox-py37/3cdbf94/ : SUCCESS in 3m 39s\n- openstack-tox-docs http://logs.openstack.org/71/669471/9/check/openstack-tox-docs/5eda978/html/ : SUCCESS in 3m 32s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/669471/9/check/kolla-build-ubuntu-binary/79ebad2/ : SUCCESS in 24m 52s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/669471/9/check/kolla-build-ubuntu-source/106223a/ : SUCCESS in 30m 27s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/669471/9/check/kolla-ansible-ubuntu-source/c99dfd0/ : SUCCESS in 54m 35s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/669471/9/check/kolla-ansible-ubuntu-binary/485e909/ : SUCCESS in 48m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/71/669471/9/check/kolla-ansible-ubuntu-source-upgrade/148b058/ : SUCCESS in 55m 15s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/71/669471/9/check/kolla-ansible-ubuntu-source-zun/040feac/ : SUCCESS in 50m 07s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/71/669471/9/check/kolla-ansible-ubuntu-source-ceph/39db6f4/ : SUCCESS in 1h 16m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/71/669471/9/check/kolla-ansible-ubuntu-source-upgrade-ceph/adc20ca/ : SUCCESS in 1h 36m 58s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm http://logs.openstack.org/71/669471/9/check/kolla-ansible-ubuntu-source-cinder-lvm/7ada64a/ : FAILURE in 29m 02s (non-voting)\n- kolla-ansible-ubuntu-source-ironic http://logs.openstack.org/71/669471/9/check/kolla-ansible-ubuntu-source-ironic/8410eb1/ : SUCCESS in 57m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"94add5df66c77a46710b950b6a6e223ddeaa9633","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 13:10:29.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"0af32001f3fc7b7c5df3fa5f83fe182a022169a4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 14:04:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/71/669471/10/openstack-check/tripleo-build-containers-fedora-28-master/d3fb666/ : FAILURE in 6m 26s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/71/669471/10/openstack-check/tripleo-build-containers-fedora-28-stein/2b0adb2/ : SUCCESS in 49m 42s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"91a42149932b4799868e06af5df6baf170ce898b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 15:13:27.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/71/669471/10/check/kolla-build-centos-binary/f21c8a4/ : SUCCESS in 42m 33s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/71/669471/10/check/kolla-build-centos-source/6c65406/ : SUCCESS in 54m 28s\n- kolla-ansible-centos-source http://logs.openstack.org/71/669471/10/check/kolla-ansible-centos-source/c861c99/ : SUCCESS in 1h 11m 36s\n- kolla-ansible-centos-binary http://logs.openstack.org/71/669471/10/check/kolla-ansible-centos-binary/61ba8e2/ : SUCCESS in 1h 00m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/71/669471/10/check/kolla-ansible-centos-source-upgrade/51e3ce1/ : SUCCESS in 1h 21m 21s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/71/669471/10/check/tripleo-build-containers-centos-7/b60e5a4/ : SUCCESS in 47m 58s\n- kolla-ansible-bifrost-centos-source http://logs.openstack.org/71/669471/10/check/kolla-ansible-bifrost-centos-source/3c51a70/ : SUCCESS in 48m 32s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/71/669471/10/check/kolla-ansible-centos-source-ceph/2432e00/ : SUCCESS in 1h 43m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/71/669471/10/check/kolla-ansible-centos-source-upgrade-ceph/7aa8e40/ : FAILURE in 1h 12m 27s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm http://logs.openstack.org/71/669471/10/check/kolla-ansible-centos-source-cinder-lvm/529aa41/ : SUCCESS in 1h 28m 29s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv http://logs.openstack.org/71/669471/10/check/kolla-ansible-centos-source-scenario-nfv/ce2f72b/ : SUCCESS in 1h 32m 46s (non-voting)\n- kolla-ansible-centos-source-ironic http://logs.openstack.org/71/669471/10/check/kolla-ansible-centos-source-ironic/74948ce/ : SUCCESS in 1h 28m 03s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/71/669471/10/check/openstack-tox-lower-constraints/5aeaafe/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/71/669471/10/check/openstack-tox-pep8/e3eef49/ : SUCCESS in 4m 02s\n- openstack-tox-py27 http://logs.openstack.org/71/669471/10/check/openstack-tox-py27/f5f6ed6/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/71/669471/10/check/openstack-tox-py36/1a89a04/ : SUCCESS in 3m 25s\n- openstack-tox-py37 http://logs.openstack.org/71/669471/10/check/openstack-tox-py37/253200d/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/71/669471/10/check/openstack-tox-docs/3deba13/html/ : SUCCESS in 3m 21s\n- kolla-build-ubuntu-binary http://logs.openstack.org/71/669471/10/check/kolla-build-ubuntu-binary/5e85ce1/ : SUCCESS in 27m 39s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/71/669471/10/check/kolla-build-ubuntu-source/a7edaf0/ : SUCCESS in 38m 51s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/71/669471/10/check/kolla-ansible-ubuntu-source/a74024a/ : SUCCESS in 49m 58s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/71/669471/10/check/kolla-ansible-ubuntu-binary/4049763/ : SUCCESS in 47m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/71/669471/10/check/kolla-ansible-ubuntu-source-upgrade/78f9f08/ : SUCCESS in 1h 03m 19s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/71/669471/10/check/kolla-ansible-ubuntu-source-zun/cf8c6cf/ : SUCCESS in 57m 29s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/71/669471/10/check/kolla-ansible-ubuntu-source-ceph/93f06d3/ : SUCCESS in 1h 13m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/71/669471/10/check/kolla-ansible-ubuntu-source-upgrade-ceph/bd0f7b9/ : SUCCESS in 1h 29m 40s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm http://logs.openstack.org/71/669471/10/check/kolla-ansible-ubuntu-source-cinder-lvm/660b460/ : FAILURE in 36m 01s (non-voting)\n- kolla-ansible-ubuntu-source-ironic http://logs.openstack.org/71/669471/10/check/kolla-ansible-ubuntu-source-ironic/dd66d60/ : SUCCESS in 59m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1f8b5ac8e4f6532b393b7c2b467575bc9fb9d197","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-09-20 08:17:34.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"753cff4c27cdb315c6f557edbf481e63df6da08f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 08:56:26.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/71/669471/11/openstack-check/tripleo-build-containers-rhel-8-master/db9f907/ : SUCCESS in 37m 18s","accounts_in_message":[],"_revision_number":11},{"id":"2e344560075a4c8a02d6ff9b06ebbb78bdc4497a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 11:00:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cd6ac82b55d847c6af3b01a3c9a267cb : SUCCESS in 53m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d6e850dc310c46b7963387d0f722a7f5 : SUCCESS in 1h 02m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e940f45c69454bdc9b9ae985fefd6caf : SUCCESS in 1h 29m 55s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6bb741d21845436193fca0b2762fb483 : SUCCESS in 1h 23m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5950339a9522425d81fb5dfe99b7beae : SUCCESS in 1h 30m 01s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1099ca8dbf59432491c2df8da08ea2ab : SUCCESS in 1h 00m 11s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/5265e37f06d8486bbacc6bedde857015 : SUCCESS in 54m 46s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/5cb4381181e14132b062482392429b46 : TIMED_OUT in 2h 33m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/19209f29fe004531b00770fc269d5970 : SUCCESS in 2h 02m 08s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/b6468936c3bc482d879bc442351cfb9e : SUCCESS in 1h 37m 03s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/80b8544fcf08474fb13ac3a1c9b9b66b : SUCCESS in 1h 46m 05s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/e53e9099062d41c7b2321439463a334c : SUCCESS in 1h 30m 51s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/fd1fb2333d7b452789a754da99dad83f : SUCCESS in 25m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e50908a916b74628974b704c504a9ade : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d00a8062ee64f82852e321f5398b770 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab108403c1e74216b7b42ec4a97adca8 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20dde3eed8ec42f0bb1e4cf04b55ee88 : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22fdc856d6fb446fb6e4a20479ab8ac5 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ca37bc10bbd43ef87f46315d23e4d21 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b36b96403dc54c62bab9abd5ea8c4f70 : SUCCESS in 22m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cbd19a5159d84472ac5bc484ab02fdbd : SUCCESS in 30m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dad06dc3285a44dc852f065295b971f1 : SUCCESS in 57m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bae660bd291a400eb57440d774a2459d : SUCCESS in 54m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/90c5d17f8c0746f49cff46a764b53ff3 : SUCCESS in 59m 58s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/bde8da93023042e9a03a146547e817b6 : SUCCESS in 1h 08m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/02ded4e9f7eb47f98a6b4da6483fe65a : SUCCESS in 1h 23m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/92ea43e9c3784f6aa6694ee40491f6b3 : SUCCESS in 1h 33m 05s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/3d36e2dd80594307a85a2330f83e0cbd : SUCCESS in 1h 13m 45s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/de04cdacb15f4e409001a9d77fe286ed : SUCCESS in 1h 05m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"13a2ddfbb3635ebaa01e81af78fc0de308bed2c2","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-10-03 16:47:16.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"1a76a046afe89e3a7e399923723ed335ab6ccb15","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-03 17:30:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/71/669471/12/openstack-check/tripleo-build-containers-rhel-8-master/9a12f06/ : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":12},{"id":"1371a3daaa9724177aed4dc4a52bb3818ab19a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:07:20.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/05992c992a3d4d51b0d29794bc0e14b8 : SUCCESS in 54m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0202c3777fb842a9a7da231e35e7a303 : SUCCESS in 1h 05m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a724a1409c3d49e081a432b7561ea318 : SUCCESS in 1h 22m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/113fe643ac434c61a63925519a6bd6da : SUCCESS in 1h 22m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a74dd9dbef87443a82fc3e574e795f70 : SUCCESS in 1h 35m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ada8cb1cddf9454b8cf55fed6ad12783 : SUCCESS in 59m 33s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/1165f157f25b4bbcb3af265f017c647b : SUCCESS in 56m 48s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/364d0a7195ec4c659493a92e668cd42a : SUCCESS in 2h 09m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/0a64a0fc80ac4c8287d1ea790796c087 : FAILURE in 1h 13m 52s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/c6b852a84ec84959a0bb1e3c69281409 : SUCCESS in 2h 00m 31s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/eb185f214a12401bbc08fd78555777b8 : FAILURE in 1h 58m 20s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/5153d4450f38413ca22421987b09d137 : SUCCESS in 1h 25m 37s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/10a15f0f8a83458fbdc8a508dd0c0a14 : SUCCESS in 23m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fc02078bfacc4cb6b136211c382b8655 : SUCCESS in 18m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7298ce4775443c7b39e7b7ff43f16de : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cc6ec89958f43ce864390e6a806523c : SUCCESS in 4m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d60812078a14a4e88d37549e7bb8809 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8961e1cb135548bbbb1cfb184e508d28 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66184b7795c8455b9a62cf30f8b6481a : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/448e2cee26a04bbf94d6874fb6f122bf : SUCCESS in 7m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a7d34ed954864c5f968485457ffacc1a : FAILURE in 45m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bfcc8cc2ec4f493d824bd14fa1e1086c : SUCCESS in 33m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/55edf0075fa94829b1454030deaa56f9 : SUCCESS in 59m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ede993af08264014b7adc460b6488d4a : SUCCESS in 1h 00m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d611ef514c7d47f1adfa539f642dd5b2 : SUCCESS in 1h 24m 37s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/300209fda97d47af90e2f17cce153e54 : FAILURE in 1h 48m 05s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e948a2c5ad3b494ca388d429d74ce8d6 : SUCCESS in 1h 36m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/50af6abdc84643b1a0fc4ec50222302d : FAILURE in 1h 35m 26s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/4d6424e800534ee999204b7dceb29cb3 : SUCCESS in 1h 16m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0a5daa010b24483c83192f4452254392 : SUCCESS in 1h 07m 17s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"0d9dd3432a4cba5ba1b5dbb6becf0d5d3957ed2d","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-06 11:27:23.000000000","message":"Patch Set 12: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit e034d32f74f69e168509456a159b4b8dd826a4c4","accounts_in_message":[],"_revision_number":12},{"id":"c33650e0b6f8847a82a7c6a62cb9803da7730130","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-11-25 12:12:07.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"cf2e991dc37fd0174e18922e5361f80264e26602","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-25 12:49:40.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/71/669471/13/openstack-check/tripleo-build-containers-rhel-8-master/7d6ef1b/ : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":13},{"id":"c5e8357d14b1499dc41b8dccd6a0e82061876243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 14:55:04.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6d5b722fcc0b4303a517b4c516da3247 : SUCCESS in 34m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ff42d874b8ea4ddfb0edf42fde96d62a : SUCCESS in 1h 01m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5e5e4098b2e1404382e0978cfd856412 : SUCCESS in 1h 29m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4d425007b3ab47979c85241da46f64ce : SUCCESS in 1h 16m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/678a69dbe83f40268bab961e7261d84a : SUCCESS in 1h 33m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/493dd3883a5c4a2f81a2a7355641bec3 : SUCCESS in 1h 01m 25s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/ff0c8b6b2e0b46da973e1026a326f53a : SUCCESS in 1h 03m 59s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/53bc4011f6934ceea242d738ebeba284 : SUCCESS in 1h 57m 34s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/fc336882abc541a7a68835d72bda1fcb : TIMED_OUT in 2h 33m 16s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/6c16561cda5148f995db413dae3bae4e : SUCCESS in 1h 37m 59s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/a8f7b179fc574fe7a934654b055d6cb1 : SUCCESS in 1h 52m 14s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/5bdbd1877a5d455b85a99c19653f4fe8 : SUCCESS in 1h 30m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/331afba7296a43ec9cc2a761b6387c9e : SUCCESS in 41m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/db2fb55b0f83429dbfe02472fd9e4b4a : SUCCESS in 36m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/006c2933239140abbab6af2fb26e51af : SUCCESS in 52m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f78d4e684524492ab31e054649fc685 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e31ea2717d8940f2874d6ceb7b430371 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dce6a1f22c814cfaa0baac6acb0c1824 : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ee59c5996564b0188cddd760f4905bf : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98f025c5c71148bb957e9fee815241f8 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e6fa47277f94c4082ac4c3b9edb6ade : SUCCESS in 37m 10s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/026e598156364a0293dce7e40c0890ee : SUCCESS in 38m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1566d9d5e0cb408189d1edf863b1ac47 : SUCCESS in 1h 02m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c416ff47d3c846008db81ade1f465d65 : SUCCESS in 57m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2364849a77654f45bc3b492c0f6411d9 : SUCCESS in 1h 04m 27s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/abc907754dd24a829ddf32176ca3b34e : SUCCESS in 1h 15m 57s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/ab46e340538446cda422050561b47dbc : SUCCESS in 1h 24m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f734f989ff5d402484ba77804c9f24ea : SUCCESS in 1h 35m 00s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/2439dae0dcf34335af365b1f38104e70 : SUCCESS in 1h 09m 04s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/28f54870090a4efbb4949a568d0244cb : SUCCESS in 1h 07m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8265ca9b39a96a2c3e7d9e0ff8324bb7421f4516","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-11-29 06:45:28.000000000","message":"Topic changed from rhel8-containers-build to rhel8-containers-buildupdate","accounts_in_message":[],"_revision_number":13},{"id":"91aa0c692786964af5bd2bc6a342305e059d911e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-11-29 06:49:22.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"064db6d229e80be7781a33db2b01bc4e5d9e9987","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-29 07:22:46.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/71/669471/14/openstack-check/tripleo-build-containers-rhel-8-master/ca5adb5/ : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":14},{"id":"b87eb2f058f44175c1f782854d195d01d1f06ae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 09:39:40.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ebb935ecd40944649a188a8ca031c82f : SUCCESS in 1h 07m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b4a2b9b08e3340e7ad5f0337e3ad2d4a : SUCCESS in 54m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a559ee7f8e5644aca8343ad4883e1b14 : SUCCESS in 1h 20m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f2150795eb3245f4bc299ddfb2048f6b : SUCCESS in 1h 43m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3b1b03ab38c14306840bdc2384808286 : SUCCESS in 1h 30m 42s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3d8436d2145b4b63b9e78753b353b779 : SUCCESS in 55m 47s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/0af237be97fa45f38225da6ec4abd49e : SUCCESS in 1h 18m 16s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/0973f7cb1a0a4b85ac689e005c9dc6e8 : SUCCESS in 2h 15m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8d9cc6d698474614a6899e3d122ee6cf : SUCCESS in 2h 30m 53s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/e5c6c333c9db4854808245b749ea8512 : SUCCESS in 1h 46m 17s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/228b41b5770346a68069f7aae6bf6881 : TIMED_OUT in 2h 02m 04s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/8edcf8f614fc49f3929b0e1be62681a3 : SUCCESS in 1h 31m 01s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d096de300d7044549ccc3f8b84b90381 : SUCCESS in 47m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a85ab72997b143d4a26ff12009fffbc5 : SUCCESS in 23m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d9faa628915f49ca9cffabeda97d4fcc : SUCCESS in 1h 30m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fdd247b277845f6b7f6c7ba07921912 : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f2c6086eed145e38ea6f09f6c1ecbea : SUCCESS in 16m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b83629c7448d488fb68bbecdc1f9be06 : SUCCESS in 16m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f4a17f09d0249f196245e1171240684 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac8ef28e2474deabb772b4f1b5297f5 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/048649f354424dbe88e955e06546aa75 : SUCCESS in 54m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bafa38cbfddd4d1995e19f7aef3f21cd : SUCCESS in 34m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb3a46c279e84cd594a1fd3ad09dc2a7 : SUCCESS in 54m 05s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/47a4bc94676344c0ba4cc3f752c084ce : SUCCESS in 50m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7bc5a94d7f3b481d8efe6145e59dc090 : FAILURE in 1h 03m 58s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/eb473a345dbc468bb943817c7a4bc467 : SUCCESS in 1h 07m 16s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/333a08dc3fbf4d878d14793ac6df79bf : SUCCESS in 1h 25m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/fbd2f6eed739433a8be8f39e33653529 : SUCCESS in 2h 16m 18s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/0babe0e816154d0fa3a91c44cc7e5ac9 : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/af7a5cec85f6499597129d681e03fae0 : SUCCESS in 1h 03m 09s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"692e17c62a918f11deeeafd8425c0056abc62492","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-01-31 17:25:11.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"e353a514ddaae39318500526c2e851ebd957a6b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:49:45.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"28b77600ed70584d6461c09eccc1849403213858","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-31 17:59:32.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/3ca446c5561e405cb40abe2f16bbcc76 : SUCCESS in 33m 22s","accounts_in_message":[],"_revision_number":15},{"id":"e7303b71d30c67dab5631cba193b1b62075b3c66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 19:54:55.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8595d55b2ae64fefb6aa6a2e045e0df0 : SUCCESS in 33m 17s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0864a0c18170493893ecd4a9c22be7f4 : SUCCESS in 46m 20s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c9055d3a521244118e4857ffe9e50919 : SUCCESS in 29m 49s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ae354ac95893497d969c101d3522e484 : SUCCESS in 32m 48s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d09a31a464f540b4870cc1b27a09e35e : SUCCESS in 1h 38m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ef42b24065264cc7a447ac2bcc0633ca : SUCCESS in 1h 20m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f7c5ed600ae645f4b650ed3604a9217e : SUCCESS in 1h 31m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8969ebb5aac6476e9e34b597c7667eec : SUCCESS in 1h 08m 07s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/c05f3a7a782d44dd97fbdefc7cf01fa3 : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/ba594d82469e41b2af1b81ca9dc83832 : SUCCESS in 2h 09m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/c9a93ee7ff654145a9bd07f4aa3225b4 : SUCCESS in 2h 24m 28s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/ec1d4d0fbdc149d4abe96d23368d8e5e : TIMED_OUT in 2h 04m 13s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/d9a5152f8374453393a2505cc224780e : TIMED_OUT in 2h 06m 01s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/9532bf4d857a4445910d7fd7758fa67d : SUCCESS in 1h 42m 52s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b837deffb2fa4463ba42d72bf58ddeb9 : SUCCESS in 30m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1d463e295723411ab4dabd43b449ab83 : SUCCESS in 32m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1115893a32e74941b0e865adc99137bc : SUCCESS in 1h 11m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4bbb95955004554b2b77a066e24e547 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7987a90851b480bb5695708bb8a37e5 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57c88a9ea4f6463b812e45179d78f5b7 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c223dec499794bfeb322aff43240cee4 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c44e0a5137c7427daef4b05743249912 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34af3071667d4ea88e239fd7ee6db719 : SUCCESS in 5m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/151c1c951b604a6ea3ed703389740ecd : SUCCESS in 33m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/619c435ab90c4ee3b8249e5377d19af5 : SUCCESS in 26m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/576b8dc31da94f76a5d68204e2116592 : SUCCESS in 1h 04m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c28c0ad66e0e416da76705a21957589e : SUCCESS in 1h 01m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5c66548648184a4c95953d3ed98bfae3 : SUCCESS in 1h 04m 39s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/386193d32e5c4cb5a20184d53596d95a : SUCCESS in 1h 18m 44s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/c95269118d124ed98ed4695def9c14d0 : FAILURE in 1h 10m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/bbd14c94db024ef78e369bb2ea1a9b2e : SUCCESS in 1h 37m 01s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/e5f29fdb2ef94a54a7dc139972d53248 : SUCCESS in 1h 12m 44s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/517aab5baa1a425aae47f4afbf6dfdf6 : SUCCESS in 1h 04m 53s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"12964dc4655f0edcfd077dc7aebff2aa366c63f0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-06 10:12:04.000000000","message":"Patch Set 15:\n\nAre there changes in here that actually need to land for RHEL8?","accounts_in_message":[],"_revision_number":15},{"id":"44c1d4173061ff2183c2d0767276fd3aaeaad962","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-02-07 16:28:53.000000000","message":"Patch Set 15:\n\n\u003e Are there changes in here that actually need to land for RHEL8?\n\nNo this was for centos8 early testing","accounts_in_message":[],"_revision_number":15},{"id":"5095239a79a86de11cd9d67eebe7c98112cc2380","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-02-07 16:28:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"44ee02d22af306fe9307c3b3476756735891170e","revisions":{"cfebfc2a286a15b659d6a460770cc329939e8098":{"kind":"REWORK","_number":1,"created":"2019-07-05 19:19:47.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/71/669471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/1"}}},"commit":{"parents":[{"commit":"c6b88c581dd7e856dcc241c4b46210e50face7dd","subject":"Merge \"CI: Clean up Zuul repo mirrors for CentOS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6b88c581dd7e856dcc241c4b46210e50face7dd"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:19:44.000000000","tz":-180},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfebfc2a286a15b659d6a460770cc329939e8098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cfebfc2a286a15b659d6a460770cc329939e8098"}]},"branch":"refs/heads/master"},"6d4913028e88d1f4ac0ed21cdf81e44519c1f730":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-05 19:55:48.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/71/669471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/2"}}},"commit":{"parents":[{"commit":"c6b88c581dd7e856dcc241c4b46210e50face7dd","subject":"Merge \"CI: Clean up Zuul repo mirrors for CentOS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6b88c581dd7e856dcc241c4b46210e50face7dd"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:55:28.000000000","tz":-180},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d4913028e88d1f4ac0ed21cdf81e44519c1f730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d4913028e88d1f4ac0ed21cdf81e44519c1f730"}]},"branch":"refs/heads/master"},"e16773888c877d5f42c58adf0a108f1ec572836e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-11 22:33:25.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/71/669471/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/3"}}},"commit":{"parents":[{"commit":"f1569bdee4eccb329aa6fba56772ac5b04031abb","subject":"Limit OVMF_CODE.secboot.fd to el7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1569bdee4eccb329aa6fba56772ac5b04031abb"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-07-11 22:33:09.000000000","tz":-360},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e16773888c877d5f42c58adf0a108f1ec572836e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e16773888c877d5f42c58adf0a108f1ec572836e"}]},"branch":"refs/heads/master"},"58eceed42725305f0660df2cf8d24662af296f72":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-07-15 09:38:17.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/71/669471/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/4"}}},"commit":{"parents":[{"commit":"ef6a62de1c91fec16d20485c0bef16cf93346227","subject":"Merge \"Limit OVMF_CODE.secboot.fd to el7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef6a62de1c91fec16d20485c0bef16cf93346227"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-15 09:38:17.000000000","tz":0},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58eceed42725305f0660df2cf8d24662af296f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58eceed42725305f0660df2cf8d24662af296f72"}]},"branch":"refs/heads/master"},"2597b6b0e9d686a4b8f3eb2c30aa482c6d4bafcb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-07-15 20:57:41.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/71/669471/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/5"}}},"commit":{"parents":[{"commit":"eedee919f965899b7007c41c9cd2dddd7842ddb9","subject":"DNM; WIP; add dnf module install and hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eedee919f965899b7007c41c9cd2dddd7842ddb9"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-07-15 20:57:34.000000000","tz":-360},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2597b6b0e9d686a4b8f3eb2c30aa482c6d4bafcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2597b6b0e9d686a4b8f3eb2c30aa482c6d4bafcb"}]},"branch":"refs/heads/master"},"9751d52c0031e0211f434a782a0a53f11b7e4c49":{"kind":"REWORK","_number":6,"created":"2019-07-17 04:19:47.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/71/669471/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/6"}}},"commit":{"parents":[{"commit":"3871ed73f29c8a4be3babf6514671b4503bbd7c3","subject":"DNM: Add dnf modules for rhel8/py3 containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3871ed73f29c8a4be3babf6514671b4503bbd7c3"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:18:49.000000000","tz":-180},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9751d52c0031e0211f434a782a0a53f11b7e4c49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9751d52c0031e0211f434a782a0a53f11b7e4c49"}]},"branch":"refs/heads/master"},"d9049291b8731a41d4b1826b53240c974e959bdb":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-07-17 08:26:33.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/71/669471/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/7"}}},"commit":{"parents":[{"commit":"41d674e86ede1b7ffb15530c877a66a0e1ae9267","subject":"DNM: Add dnf modules for rhel8/py3 containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41d674e86ede1b7ffb15530c877a66a0e1ae9267"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-17 08:26:33.000000000","tz":0},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9049291b8731a41d4b1826b53240c974e959bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9049291b8731a41d4b1826b53240c974e959bdb"}]},"branch":"refs/heads/master"},"bc394b8e71646c6e2bba642e5cd55622262d6d6d":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-07-17 09:30:19.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/71/669471/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/8"}}},"commit":{"parents":[{"commit":"bd7e2ee5bf14813e1ee4fafb180f8cef6d15f315","subject":"DNM: Add dnf modules for rhel8/py3 containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd7e2ee5bf14813e1ee4fafb180f8cef6d15f315"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-17 09:30:19.000000000","tz":0},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc394b8e71646c6e2bba642e5cd55622262d6d6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc394b8e71646c6e2bba642e5cd55622262d6d6d"}]},"branch":"refs/heads/master"},"07537a8e9b82ce7a6cbe500dbc206b4cbd4f9ab8":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-07-17 10:33:23.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/71/669471/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/9"}}},"commit":{"parents":[{"commit":"1c7d4b5781ee66a19f8b310d5c5f64e061689afa","subject":"DNM: Add dnf modules for rhel8/py3 containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c7d4b5781ee66a19f8b310d5c5f64e061689afa"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-17 10:33:23.000000000","tz":0},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07537a8e9b82ce7a6cbe500dbc206b4cbd4f9ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07537a8e9b82ce7a6cbe500dbc206b4cbd4f9ab8"}]},"branch":"refs/heads/master"},"57ad005d3560df97c6a635fa16900210ed1d1f40":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-07-17 13:10:29.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/71/669471/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/10"}}},"commit":{"parents":[{"commit":"0c9a419297c5621a0fc901051b29c4c41ffab3ee","subject":"DNM: Add dnf modules for rhel8/py3 containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c9a419297c5621a0fc901051b29c4c41ffab3ee"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-17 13:10:29.000000000","tz":0},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57ad005d3560df97c6a635fa16900210ed1d1f40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57ad005d3560df97c6a635fa16900210ed1d1f40"}]},"branch":"refs/heads/master"},"b76b79d29d55f79a94912b75cc30d6515dbbadf8":{"kind":"REWORK","_number":11,"created":"2019-09-20 08:17:34.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/71/669471/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/11"}}},"commit":{"parents":[{"commit":"9507aea9c67ac34a223b940f9e657e3e25e3ebfa","subject":"DNM: Add dnf modules for rhel8/py3 containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9507aea9c67ac34a223b940f9e657e3e25e3ebfa"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-09-20 08:17:13.000000000","tz":60},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b76b79d29d55f79a94912b75cc30d6515dbbadf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b76b79d29d55f79a94912b75cc30d6515dbbadf8"}]},"branch":"refs/heads/master"},"fbe60bb3143a0176c13f6b2fda6e9c44d2fe3116":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2019-10-03 16:47:16.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/71/669471/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/12"}}},"commit":{"parents":[{"commit":"986af8b0c64267686377ffc4c3db13e5247887d3","subject":"DNM: Add dnf modules for rhel8/py3 containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/986af8b0c64267686377ffc4c3db13e5247887d3"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-10-03 16:47:16.000000000","tz":0},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbe60bb3143a0176c13f6b2fda6e9c44d2fe3116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbe60bb3143a0176c13f6b2fda6e9c44d2fe3116"}]},"branch":"refs/heads/master"},"e052d39c7dc62f20816d555e4c6e1c460fd77317":{"kind":"REWORK","_number":13,"created":"2019-11-25 12:12:07.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/71/669471/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/13"}}},"commit":{"parents":[{"commit":"986af8b0c64267686377ffc4c3db13e5247887d3","subject":"DNM: Add dnf modules for rhel8/py3 containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/986af8b0c64267686377ffc4c3db13e5247887d3"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-11-25 12:11:25.000000000","tz":180},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e052d39c7dc62f20816d555e4c6e1c460fd77317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e052d39c7dc62f20816d555e4c6e1c460fd77317"}]},"branch":"refs/heads/master"},"c23c6aee40453db43196a16db300156d137edab0":{"kind":"REWORK","_number":14,"created":"2019-11-29 06:49:22.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/71/669471/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/14"}}},"commit":{"parents":[{"commit":"494029ea40416cad1b0893dfd6738fe321de9af7","subject":"DNM: Add dnf modules for rhel8/py3 containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/494029ea40416cad1b0893dfd6738fe321de9af7"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-11-29 06:49:02.000000000","tz":330},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c23c6aee40453db43196a16db300156d137edab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c23c6aee40453db43196a16db300156d137edab0"}]},"branch":"refs/heads/master"},"44ee02d22af306fe9307c3b3476756735891170e":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-01-31 17:25:11.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/71/669471/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/71/669471/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/71/669471/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/71/669471/15"}}},"commit":{"parents":[{"commit":"8dd2ea68f4f5c6c4d4bc1cde4df2d7b53f49f5ce","subject":"DNM: Add dnf modules for rhel8/py3 containers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dd2ea68f4f5c6c4d4bc1cde4df2d7b53f49f5ce"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-05 19:13:15.000000000","tz":-180},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2020-01-31 17:25:11.000000000","tz":0},"subject":"DNM: Add rhel8 support for early CentOS8 testing","message":"DNM: Add rhel8 support for early CentOS8 testing\n\nRDO is currently using a version of RHEL 8 to test the python3\npackaging for the next version of CentOS/RHEL. This change updates\nbasic support to RHEL8 as a build option.\n\nRHEL8 is used on purpose in order to match RHEL/CentOS 8.\n\nStory: https://tree.taiga.io/project/tripleo-ci-board/us/956\n\nChange-Id: I3bcef4fbba1f7c59b564446552532ea949bc34c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44ee02d22af306fe9307c3b3476756735891170e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44ee02d22af306fe9307c3b3476756735891170e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
