)]}'
{"id":"openstack%2Fkolla~669805","triplet_id":"openstack%2Fkolla~stable%2Frocky~I3f1682ed55e7d45c1e97a1b34c7993aa839f0bb2","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I3f1682ed55e7d45c1e97a1b34c7993aa839f0bb2","subject":"Write hash after compressing horizon static assets","status":"MERGED","created":"2019-07-09 07:19:33.000000000","updated":"2019-07-10 17:57:32.000000000","submitted":"2019-07-10 17:57:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669805-1562781452055-ef2d6f53","meta_rev_id":"2c794b22870e5eb0ec454b195dbd4e2fa99c103e","_number":669805,"virtual_id_number":669805,"owner":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-10 01:51:08.000000000","_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-10 17:57:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 02:47:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":2,"date":"2019-07-10 13:02:29.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-10 16:50:05.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-07-10 16:50:05.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 02:47:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 13:02:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-10 16:50:05.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-10 17:57:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a93b10fae5c6558a84646f33180ff03e5d7e87ad","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2019-07-09 07:19:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4281e8399bd1f91f40c194f85468806c000f6ce8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-09 08:05:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/05/669805/1/openstack-check/tripleo-build-containers-fedora-28-master/680df36/ : RETRY_LIMIT in 3m 31s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/05/669805/1/openstack-check/tripleo-build-containers-fedora-28-stein/7a836c4/ : SUCCESS in 44m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be22e0f74c18c30fa612455bca871e2ffcef14be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-09 09:33:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/05/669805/1/check/kolla-build-centos-binary/b3cd876/ : FAILURE in 26m 08s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/05/669805/1/check/kolla-build-centos-source/edd23f3/ : FAILURE in 26m 47s\n- kolla-ansible-centos-source http://logs.openstack.org/05/669805/1/check/kolla-ansible-centos-source/ad63b5b/ : FAILURE in 31m 02s\n- kolla-ansible-centos-binary http://logs.openstack.org/05/669805/1/check/kolla-ansible-centos-binary/a2ce333/ : FAILURE in 29m 26s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/05/669805/1/check/kolla-build-oraclelinux-binary/cf1d44a/ : FAILURE in 28m 09s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/05/669805/1/check/kolla-build-oraclelinux-source/9bcaab8/ : FAILURE in 27m 06s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/05/669805/1/check/kolla-ansible-oraclelinux-source/18c2760/ : FAILURE in 27m 45s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/05/669805/1/check/kolla-ansible-oraclelinux-binary/905fe00/ : FAILURE in 28m 34s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/05/669805/1/check/openstack-tox-lower-constraints/948ec20/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/05/669805/1/check/openstack-tox-pep8/59cd352/ : SUCCESS in 3m 31s\n- openstack-tox-py27 http://logs.openstack.org/05/669805/1/check/openstack-tox-py27/5b26ef3/ : SUCCESS in 3m 13s\n- openstack-tox-py35 http://logs.openstack.org/05/669805/1/check/openstack-tox-py35/7045ddc/ : SUCCESS in 3m 25s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/669805/1/check/build-openstack-sphinx-docs/d1c01e2/html/ : SUCCESS in 3m 14s\n- kolla-build-ubuntu-binary http://logs.openstack.org/05/669805/1/check/kolla-build-ubuntu-binary/80ae64e/ : SUCCESS in 1h 10m 15s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/05/669805/1/check/kolla-build-ubuntu-source/f8c9e17/ : SUCCESS in 33m 55s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/05/669805/1/check/kolla-ansible-ubuntu-source/6b74761/ : SUCCESS in 58m 43s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/05/669805/1/check/kolla-ansible-ubuntu-binary/6cded64/ : SUCCESS in 1h 59m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f60b0d2a8be67efd396ac1c9518e9485d2ece692","author":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"date":"2019-07-10 01:51:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5f754c7e4d9339c11bd2e05b4260489d2bf7867a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-10 02:47:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/05/669805/1/openstack-check/tripleo-build-containers-fedora-28-master/4c3b9d6/ : RETRY_LIMIT in 2m 47s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/05/669805/1/openstack-check/tripleo-build-containers-fedora-28-stein/6facfb4/ : SUCCESS in 40m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5667d57213f73bce23d65ac409d7a04fc6f1a53e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 03:59:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/05/669805/1/check/kolla-build-centos-binary/422cc6d/ : SUCCESS in 39m 14s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/05/669805/1/check/kolla-build-centos-source/b473685/ : SUCCESS in 36m 41s\n- kolla-ansible-centos-source http://logs.openstack.org/05/669805/1/check/kolla-ansible-centos-source/969276c/ : SUCCESS in 1h 16m 58s\n- kolla-ansible-centos-binary http://logs.openstack.org/05/669805/1/check/kolla-ansible-centos-binary/3929844/ : SUCCESS in 1h 00m 38s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/05/669805/1/check/kolla-build-oraclelinux-binary/0e168f4/ : SUCCESS in 30m 35s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/05/669805/1/check/kolla-build-oraclelinux-source/29eb69e/ : SUCCESS in 43m 56s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/05/669805/1/check/kolla-ansible-oraclelinux-source/e2cf2d6/ : SUCCESS in 1h 18m 17s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/05/669805/1/check/kolla-ansible-oraclelinux-binary/574a8b7/ : SUCCESS in 1h 04m 25s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/05/669805/1/check/openstack-tox-lower-constraints/fe7c1cd/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/05/669805/1/check/openstack-tox-pep8/9037625/ : SUCCESS in 4m 07s\n- openstack-tox-py27 http://logs.openstack.org/05/669805/1/check/openstack-tox-py27/fae3650/ : SUCCESS in 3m 16s\n- openstack-tox-py35 http://logs.openstack.org/05/669805/1/check/openstack-tox-py35/379c094/ : SUCCESS in 3m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/669805/1/check/build-openstack-sphinx-docs/10fb333/html/ : SUCCESS in 3m 08s\n- kolla-build-ubuntu-binary http://logs.openstack.org/05/669805/1/check/kolla-build-ubuntu-binary/381f58a/ : SUCCESS in 1h 23m 42s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/05/669805/1/check/kolla-build-ubuntu-source/63a5ae7/ : SUCCESS in 31m 00s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/05/669805/1/check/kolla-ansible-ubuntu-source/aef715c/ : SUCCESS in 1h 02m 23s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/05/669805/1/check/kolla-ansible-ubuntu-binary/d319556/ : SUCCESS in 1h 58m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d708975173c4f5b06b722a67e14434339384a24c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-10 13:02:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"875b232c4bbaa563e5ed496f37f64749aaa7c61b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-10 16:50:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0f58466a93eaa2057aa03b5d56ba252fc263916","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:50:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b853b31ce6f1e64c8f88fbc7dda44808e1f5e078","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:57:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/05/669805/1/gate/kolla-build-centos-source/900e18b/ : SUCCESS in 1h 01m 23s\n- kolla-build-oraclelinux-source http://logs.openstack.org/05/669805/1/gate/kolla-build-oraclelinux-source/8b1cd68/ : SUCCESS in 54m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/669805/1/gate/openstack-tox-lower-constraints/43ae5ab/ : SUCCESS in 5m 32s\n- openstack-tox-pep8 http://logs.openstack.org/05/669805/1/gate/openstack-tox-pep8/79db844/ : SUCCESS in 4m 35s\n- openstack-tox-py27 http://logs.openstack.org/05/669805/1/gate/openstack-tox-py27/f01683d/ : SUCCESS in 4m 41s\n- openstack-tox-py35 http://logs.openstack.org/05/669805/1/gate/openstack-tox-py35/4dbeb31/ : SUCCESS in 4m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/05/669805/1/gate/build-openstack-sphinx-docs/7d1acdc/html/ : SUCCESS in 4m 12s\n- kolla-build-ubuntu-source http://logs.openstack.org/05/669805/1/gate/kolla-build-ubuntu-source/87b7d72/ : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2c794b22870e5eb0ec454b195dbd4e2fa99c103e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 17:57:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0ef7abd54ffdd4fdc9a477ed79f95c20cfc423d","revisions":{"b0ef7abd54ffdd4fdc9a477ed79f95c20cfc423d":{"kind":"REWORK","_number":1,"created":"2019-07-09 07:19:33.000000000","uploader":{"_account_id":7488,"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","username":"Jeffrey4l"},"ref":"refs/changes/05/669805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/669805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/669805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/669805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/669805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/669805/1"}}},"commit":{"parents":[{"commit":"c7e458992cbdef9828779c57eb065b4a5ffa9a4b","subject":"Merge \"CI: Move to periodic-stable\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7e458992cbdef9828779c57eb065b4a5ffa9a4b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-04-10 11:44:37.000000000","tz":60},"committer":{"name":"Jeffrey Zhang","email":"zhang.lei.fly@gmail.com","date":"2019-07-09 07:16:28.000000000","tz":480},"subject":"Write hash after compressing horizon static assets","message":"Write hash after compressing horizon static assets\n\nIf you run the horizon container with a restart policy of never, and the\nstatic asset compression fails, the container fails to start, e.g. (seen\nin CI)\n\n++ /usr/bin/manage.py collectstatic --noinput --clear\nTraceback (most recent call last):\n  File \"/usr/bin/manage.py\", line 18, in \u003cmodule\u003e\n      from django.core.management import execute_from_command_line\n  ImportError: No module named django.core.management\n\nHowever, with a restart policy of unless-stopped (the default).  the\nhorizon container will fail the first time but will succeed after a\nrestart. This is because the settings bundle hash is updated before the\ncompression is attempted.\n\nTo fix this, update the hash after compression completes successfully.\n\nChange-Id: I3f1682ed55e7d45c1e97a1b34c7993aa839f0bb2\nCloses-Bug: #1824137\n(cherry picked from commit 92c2456851857c45e8de9a3ff8f167579e0588f4)\n(cherry picked from commit 6e010e25e3e4bf3a26a40a3a0b27055512d8a93e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0ef7abd54ffdd4fdc9a477ed79f95c20cfc423d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0ef7abd54ffdd4fdc9a477ed79f95c20cfc423d"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
