)]}'
{"id":"openstack%2Fkolla~670001","triplet_id":"openstack%2Fkolla~master~I0ca255c6358132d9e3acfa6b610b70a78756512c","project":"openstack/kolla","branch":"master","topic":"bug/1824787","hashtags":[],"change_id":"I0ca255c6358132d9e3acfa6b610b70a78756512c","subject":"Add partition detection to fix osd initialization failure","status":"MERGED","created":"2019-07-10 07:42:14.000000000","updated":"2019-08-15 09:17:49.000000000","submitted":"2019-07-16 18:48:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670001-1563302921387-6d7173bf","meta_rev_id":"c8fdbdaeecc7268ee12dd9f4945159b67b531d9f","_number":670001,"virtual_id_number":670001,"owner":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"date":"2019-07-10 09:26:45.000000000","_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-16 18:48:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-10 09:53:02.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"all":[{"value":1,"date":"2019-07-13 15:00:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2019-07-16 11:38:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-07-10 16:03:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},{"value":1,"date":"2019-07-13 15:54:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":1,"date":"2019-07-14 15:28:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":2,"date":"2019-07-16 17:44:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":1,"date":"2019-07-16 17:44:23.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 09:53:02.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-10 16:03:13.000000000","updated_by":{"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},"reviewer":{"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},"state":"REVIEWER"},{"updated":"2019-07-13 15:00:59.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-07-13 15:54:59.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-07-14 15:28:01.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-07-16 11:38:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-16 17:44:23.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-16 18:48:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cb510de8f0c630f98702ee59a9e4cae2144e776","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-07-10 07:42:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e981c9e615d6211d5648780877a4ea2baa5ad7ac","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-10 08:25:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/670001/1/openstack-check/tripleo-build-containers-fedora-28-master/5675c08/ : FAILURE in 5m 58s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/670001/1/openstack-check/tripleo-build-containers-fedora-28-stein/5b97fe6/ : SUCCESS in 42m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b73a4272fb151bcf2dbb3e9b083ef823dfc06e7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-10 08:31:08.000000000","message":"Patch Set 1:\n\nrelated (prev fix): https://review.opendev.org/652612 \u003c- maybe reference in the commit message","accounts_in_message":[],"_revision_number":1},{"id":"357112074921c863f1be4fc23fbedfbdc0f45735","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-07-10 09:12:47.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9cdb1f4b1ec5dab8c08f4d83da0784d962826182","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-07-10 09:14:46.000000000","message":"Patch Set 2:\n\n\u003e related (prev fix): https://review.opendev.org/652612 \u003c- maybe\n \u003e reference in the commit message\n\nThanks for your review, I updated the commit message to make it look clearer.","accounts_in_message":[],"_revision_number":2},{"id":"4819c56d9c5cefa9b19268763cdbe1df020239cb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-10 09:20:58.000000000","message":"Patch Set 2: Code-Review+1\n\nFine by me, looking forward to hearing from Michał. ;-)","accounts_in_message":[],"_revision_number":2},{"id":"92e64a84f69a36d447caf9fb3e7de4ccb2c52539","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-07-10 09:26:45.000000000","message":"Patch Set 2:\n\n\u003e Fine by me, looking forward to hearing from Michał. ;-)\n\nOK,^^","accounts_in_message":[],"_revision_number":2},{"id":"f96c4c69de83dc090d31cfa413d077d849ff2866","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-10 09:53:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/01/670001/2/openstack-check/tripleo-build-containers-fedora-28-master/0d6ad45/ : FAILURE in 5m 18s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/01/670001/2/openstack-check/tripleo-build-containers-fedora-28-stein/723cefa/ : SUCCESS in 38m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8f2a7109a9d829fa2c1300f472fe5293cec13774","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 11:17:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/01/670001/2/check/kolla-build-centos-binary/93c90d6/ : SUCCESS in 42m 36s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/01/670001/2/check/kolla-build-centos-source/b889f33/ : SUCCESS in 35m 18s\n- kolla-ansible-centos-source http://logs.openstack.org/01/670001/2/check/kolla-ansible-centos-source/3e28a73/ : SUCCESS in 1h 02m 22s\n- kolla-ansible-centos-binary http://logs.openstack.org/01/670001/2/check/kolla-ansible-centos-binary/f7d5cb8/ : SUCCESS in 1h 03m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/01/670001/2/check/kolla-ansible-centos-source-upgrade/e629289/ : SUCCESS in 1h 17m 03s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/670001/2/check/tripleo-build-containers-centos-7/1abb21d/ : SUCCESS in 45m 08s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/01/670001/2/check/kolla-ansible-centos-source-ceph/6f3e9b7/ : SUCCESS in 1h 32m 29s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/01/670001/2/check/kolla-ansible-centos-source-upgrade-ceph/85c2c6c/ : SUCCESS in 1h 59m 00s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/01/670001/2/check/openstack-tox-lower-constraints/a3448fd/ : SUCCESS in 3m 19s\n- openstack-tox-pep8 http://logs.openstack.org/01/670001/2/check/openstack-tox-pep8/af74c0a/ : SUCCESS in 3m 17s\n- openstack-tox-py27 http://logs.openstack.org/01/670001/2/check/openstack-tox-py27/a3af449/ : SUCCESS in 3m 09s\n- openstack-tox-py36 http://logs.openstack.org/01/670001/2/check/openstack-tox-py36/ef9643d/ : SUCCESS in 3m 24s\n- openstack-tox-py37 http://logs.openstack.org/01/670001/2/check/openstack-tox-py37/cef932b/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/01/670001/2/check/openstack-tox-docs/561d380/html/ : SUCCESS in 3m 06s\n- kolla-build-ubuntu-binary http://logs.openstack.org/01/670001/2/check/kolla-build-ubuntu-binary/15e0030/ : SUCCESS in 27m 35s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/01/670001/2/check/kolla-build-ubuntu-source/dd7d68a/ : SUCCESS in 25m 38s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/01/670001/2/check/kolla-ansible-ubuntu-source/9c6df5e/ : SUCCESS in 44m 42s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/01/670001/2/check/kolla-ansible-ubuntu-binary/a4532d9/ : SUCCESS in 45m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/01/670001/2/check/kolla-ansible-ubuntu-source-upgrade/fcaa872/ : SUCCESS in 59m 40s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/01/670001/2/check/kolla-ansible-ubuntu-source-ceph/6212683/ : SUCCESS in 1h 09m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/01/670001/2/check/kolla-ansible-ubuntu-source-upgrade-ceph/54dc54a/ : SUCCESS in 1h 21m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"18e707c114fdb5bab7e15e6ba5fd7a86ebda7783","author":{"_account_id":25238,"name":"Magnus Lööf","email":"magnus.loof@basalt.se","username":"magnusloof"},"date":"2019-07-10 16:03:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ec037a721bddca805b24edd8a5aba81048df8cb3","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-07-13 15:00:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a1f65c161ae90784bb4be406ce135d81871440f3","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-07-13 15:54:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"745a9b252f47703cacc00f10bc6001106d84e398","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-07-14 15:28:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"724cc6d332e4f0aced44452db457e701ea6c2346","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-16 11:38:10.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good, I like the structured approach - this way we can extend it if needed.","accounts_in_message":[],"_revision_number":2},{"id":"dfc47e3089ffab5aa366342930fbd428c32b321a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-16 17:44:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d5d53a956febd96c627be15491272f6b305de244","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 17:44:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"062d822943f95c5bdf770045148d30f4d2cce5f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 18:48:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/01/670001/2/gate/kolla-build-centos-source/d884790/ : SUCCESS in 48m 12s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/01/670001/2/gate/tripleo-build-containers-centos-7/597ba51/ : SUCCESS in 58m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/01/670001/2/gate/openstack-tox-lower-constraints/97c7a67/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/01/670001/2/gate/openstack-tox-pep8/48866dd/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/01/670001/2/gate/openstack-tox-py27/f453e26/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/01/670001/2/gate/openstack-tox-py36/26777a0/ : SUCCESS in 5m 12s\n- openstack-tox-py37 http://logs.openstack.org/01/670001/2/gate/openstack-tox-py37/cf00609/ : SUCCESS in 4m 41s\n- openstack-tox-docs http://logs.openstack.org/01/670001/2/gate/openstack-tox-docs/02a65c1/html/ : SUCCESS in 2m 26s\n- kolla-build-ubuntu-source http://logs.openstack.org/01/670001/2/gate/kolla-build-ubuntu-source/3ee2e2a/ : SUCCESS in 27m 52s","accounts_in_message":[],"_revision_number":2},{"id":"ecac8ee0ca3846647464692369016d7e026ae66f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 18:48:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f76716158f14c3bc4df534337e560c44fc2b7569","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-07-17 01:19:44.000000000","message":"Patch Set 2:\n\n\u003e Looks good, I like the structured approach - this way we can extend\n \u003e it if needed.\n\nThanks for your review, :-)","accounts_in_message":[],"_revision_number":2},{"id":"4db854d2992866c1470b92ee11ce8000cf6e4e66","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 07:43:20.000000000","message":"Patch Set 2:\n\nDoes this need to be backported?","accounts_in_message":[],"_revision_number":2},{"id":"68493b9ad683bc84a346070e699774611de6e5f1","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-08-15 09:13:32.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 64c1acd32537b0746f9f6ada2b9f63a676990ecc","accounts_in_message":[],"_revision_number":2},{"id":"0260c00a4d301670abcd6e1cc1b4706697cdc12c","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-08-15 09:14:40.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit e6eb8b030eaa366d3c9ef3ea58c23ac7128e14de","accounts_in_message":[],"_revision_number":2},{"id":"fe3b473ca08c7a4ebe2c7d75a8784ef6421c81fe","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-08-15 09:17:49.000000000","message":"Patch Set 2:\n\n\u003e Does this need to be backported?\n\nYes, I have cherry picked to the corresponding branch.^^","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"342c9f0cd0dc4b97870dc40fab6c5a7d241b40ed","revisions":{"683090f78d9a805e54a89acd5437d998d3d7635c":{"kind":"REWORK","_number":1,"created":"2019-07-10 07:42:14.000000000","uploader":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"ref":"refs/changes/01/670001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/670001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/670001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/670001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/670001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/670001/1"}}},"commit":{"parents":[{"commit":"008aa957f27b26caff3be8b18110e71cdaf70772","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/008aa957f27b26caff3be8b18110e71cdaf70772"}]}],"author":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-07-10 07:29:48.000000000","tz":540},"committer":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-07-10 07:29:48.000000000","tz":540},"subject":"Add partition detection to fix osd initialization failure","message":"Add partition detection to fix osd initialization failure\n\nWhen deploying osd, if the user does not use the extra block\npartition, the kolla will automatically partition the disk and then\nclean up the data on the disk partition.\nHowever, sometimes the partition does not appear immediately. In\norder to solve this problem, the partprobe command consistent with\nceph-disk is added, and then a loop is added to detect the partition,\nwhich is acquired every 1s. When the partition appears, the osd\ninitialization is continued.\n\nChange-Id: I0ca255c6358132d9e3acfa6b610b70a78756512c\nCloses-bug: #1824787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/683090f78d9a805e54a89acd5437d998d3d7635c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/683090f78d9a805e54a89acd5437d998d3d7635c"}]},"branch":"refs/heads/master"},"342c9f0cd0dc4b97870dc40fab6c5a7d241b40ed":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-10 09:12:47.000000000","uploader":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"ref":"refs/changes/01/670001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/670001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/670001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/670001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/670001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/670001/2"}}},"commit":{"parents":[{"commit":"008aa957f27b26caff3be8b18110e71cdaf70772","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/008aa957f27b26caff3be8b18110e71cdaf70772"}]}],"author":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-07-10 07:29:48.000000000","tz":540},"committer":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-07-10 09:10:53.000000000","tz":540},"subject":"Add partition detection to fix osd initialization failure","message":"Add partition detection to fix osd initialization failure\n\nThe changes in the following commit did not completely solve the\nproblem of osd initialization failure:\nhttps://review.opendev.org/652612\n\nIn order to solve this problem, the partprobe command consistent with\nceph-disk is added, and then a loop is added to detect the partition,\nwhich is acquired every 1s. When the partition appears, the osd\ninitialization is continued.\n\nChange-Id: I0ca255c6358132d9e3acfa6b610b70a78756512c\nCloses-bug: #1824787\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/342c9f0cd0dc4b97870dc40fab6c5a7d241b40ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/342c9f0cd0dc4b97870dc40fab6c5a7d241b40ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
