)]}'
{"id":"openstack%2Fkolla~670405","triplet_id":"openstack%2Fkolla~master~I5026a31d3a29b5a31001c6c1f9b90b2a32af6a4b","project":"openstack/kolla","branch":"master","topic":"centos-8","hashtags":[],"change_id":"I5026a31d3a29b5a31001c6c1f9b90b2a32af6a4b","subject":"Limit OVMF_CODE.secboot.fd to el7","status":"MERGED","created":"2019-07-11 22:29:48.000000000","updated":"2019-07-15 03:06:16.000000000","submitted":"2019-07-15 03:06:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670405-1563159976059-cc061412","meta_rev_id":"9c825e87f5733dfca668763619fab9f1cb57a4dc","_number":670405,"virtual_id_number":670405,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-15 03:06:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-12 14:01:41.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"all":[{"value":1,"date":"2019-07-13 15:57:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":1,"date":"2019-07-14 13:16:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-07-12 13:25:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-12 16:08:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-12 16:29:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":2,"date":"2019-07-15 02:09:07.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2019-07-14 15:27:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2019-07-14 13:38:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-07-15 02:09:07.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-12 03:56:36.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-12 07:45:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-12 13:25:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-12 14:01:41.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-12 16:08:13.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2019-07-12 16:29:50.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-07-13 15:57:48.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-07-14 13:16:30.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-07-14 13:38:59.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-07-14 15:27:27.000000000","updated_by":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"reviewer":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"state":"REVIEWER"},{"updated":"2019-07-15 02:09:07.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-07-15 03:06:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0028b3c1e22415b4fc94f928be7c1870f99ae29","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-11 22:29:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05582e52eaa34ba4f4b231bd0c73b0497a237ee5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-11 23:14:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/05/670405/1/openstack-check/tripleo-build-containers-fedora-28-master/fa2a5f1/ : FAILURE in 5m 46s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/05/670405/1/openstack-check/tripleo-build-containers-fedora-28-stein/0c578e7/ : SUCCESS in 37m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40a06e5fa6ec7bc06970ecbec6acc0a77556e57f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 23:54:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/05/670405/1/check/kolla-build-centos-binary/a27f300/ : SUCCESS in 40m 33s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/05/670405/1/check/kolla-build-centos-source/b165d0a/ : SUCCESS in 48m 04s\n- kolla-ansible-centos-source http://logs.openstack.org/05/670405/1/check/kolla-ansible-centos-source/476bbc1/ : SUCCESS in 1h 09m 31s\n- kolla-ansible-centos-binary http://logs.openstack.org/05/670405/1/check/kolla-ansible-centos-binary/b583c86/ : SUCCESS in 1h 08m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/05/670405/1/check/kolla-ansible-centos-source-upgrade/f126015/ : SUCCESS in 1h 19m 22s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/05/670405/1/check/tripleo-build-containers-centos-7/39035db/ : SUCCESS in 53m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/670405/1/check/openstack-tox-lower-constraints/bc165bd/ : SUCCESS in 3m 16s\n- openstack-tox-pep8 http://logs.openstack.org/05/670405/1/check/openstack-tox-pep8/d03c7b2/ : SUCCESS in 3m 03s\n- openstack-tox-py27 http://logs.openstack.org/05/670405/1/check/openstack-tox-py27/c83811f/ : SUCCESS in 3m 20s\n- openstack-tox-py36 http://logs.openstack.org/05/670405/1/check/openstack-tox-py36/8d40701/ : SUCCESS in 4m 20s\n- openstack-tox-py37 http://logs.openstack.org/05/670405/1/check/openstack-tox-py37/13c244b/ : SUCCESS in 3m 28s\n- openstack-tox-docs http://logs.openstack.org/05/670405/1/check/openstack-tox-docs/1dbe733/html/ : SUCCESS in 2m 27s\n- kolla-build-ubuntu-binary http://logs.openstack.org/05/670405/1/check/kolla-build-ubuntu-binary/293d8d8/ : SUCCESS in 23m 20s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/05/670405/1/check/kolla-build-ubuntu-source/d5772ab/ : SUCCESS in 32m 16s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/05/670405/1/check/kolla-ansible-ubuntu-source/7be23f6/ : SUCCESS in 51m 43s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/05/670405/1/check/kolla-ansible-ubuntu-binary/3d56325/ : SUCCESS in 55m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/05/670405/1/check/kolla-ansible-ubuntu-source-upgrade/2d64ba4/ : SUCCESS in 1h 08m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c79a73f9085d09c5d7d6eb1fc02c24b0da37c15e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-12 03:56:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2c89e322e2e3abf7d0f051b30e680a9d57ef8b2c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-12 07:45:32.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dbf8e124476d5f733d1668666eff605b466532f2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-12 08:35:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf9c07560563470817c8f1a9643d7d096276f576","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-12 11:21:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"62914438f457ff0c29784fbc6e0bd5378877dfe2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-12 13:13:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3accc20e8d5054a964fe69b437ab404f1869cd9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-12 13:14:27.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"21a4577d9d7536a89c624934442da0cbbc11caed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-12 13:25:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3fce88fe508e96aecd0558b0e9d374773c886f88","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-12 14:01:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/05/670405/2/openstack-check/tripleo-build-containers-fedora-28-master/1078017/ : FAILURE in 5m 57s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/05/670405/2/openstack-check/tripleo-build-containers-fedora-28-stein/fb81c34/ : SUCCESS in 39m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d31420cbae60dd2dda7056979b0d8282e3b50443","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 14:48:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/05/670405/2/check/kolla-build-centos-binary/90c3d0e/ : SUCCESS in 37m 38s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/05/670405/2/check/kolla-build-centos-source/e0d276c/ : SUCCESS in 39m 39s\n- kolla-ansible-centos-source http://logs.openstack.org/05/670405/2/check/kolla-ansible-centos-source/a0155f4/ : SUCCESS in 1h 05m 23s\n- kolla-ansible-centos-binary http://logs.openstack.org/05/670405/2/check/kolla-ansible-centos-binary/cecf47c/ : SUCCESS in 54m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/05/670405/2/check/kolla-ansible-centos-source-upgrade/aabfe9d/ : SUCCESS in 1h 28m 52s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/05/670405/2/check/tripleo-build-containers-centos-7/9a72716/ : SUCCESS in 53m 51s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/670405/2/check/openstack-tox-lower-constraints/8f8198d/ : SUCCESS in 3m 29s\n- openstack-tox-pep8 http://logs.openstack.org/05/670405/2/check/openstack-tox-pep8/d5ff638/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/05/670405/2/check/openstack-tox-py27/f31ae76/ : SUCCESS in 3m 36s\n- openstack-tox-py36 http://logs.openstack.org/05/670405/2/check/openstack-tox-py36/4d9ddfc/ : SUCCESS in 3m 31s\n- openstack-tox-py37 http://logs.openstack.org/05/670405/2/check/openstack-tox-py37/1753505/ : SUCCESS in 3m 52s\n- openstack-tox-docs http://logs.openstack.org/05/670405/2/check/openstack-tox-docs/907ea2c/html/ : SUCCESS in 2m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/05/670405/2/check/kolla-build-ubuntu-binary/7075c69/ : SUCCESS in 25m 06s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/05/670405/2/check/kolla-build-ubuntu-source/cfe33ac/ : SUCCESS in 37m 04s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/05/670405/2/check/kolla-ansible-ubuntu-source/b567d60/ : SUCCESS in 49m 07s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/05/670405/2/check/kolla-ansible-ubuntu-binary/5bd9cb2/ : SUCCESS in 52m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/05/670405/2/check/kolla-ansible-ubuntu-source-upgrade/0322b73/ : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7bbaaca93239d14f679aba5923954fdb63020834","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-12 16:08:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8b99474944258ad4f01d3aa4cf823037f2c17d3b","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-07-12 16:29:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8957c01f7d6884a98a29a8634d755d623b9457c3","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-07-13 15:57:48.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cd09e5c2da4775d21a229f0edc612b51fbeab2b4","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-07-14 13:16:30.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"888b17b306ac20a41fa70268041595d355445910","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-07-14 13:38:59.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8d9084f3ac833fe277d64b2834e4698e35ca961b","author":{"_account_id":28471,"name":"yanghongyu","email":"yang.hongyu@99cloud.net","username":"yanghongyu"},"date":"2019-07-14 15:27:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8f67db0fcb050eae91f8e27a61aeb9a4dd4f0278","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-07-15 02:09:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e320356be052604506775c2c9b197625fd7d4915","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 02:09:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4d307c5699ba5b8c72274c1b7e0a61eab5824ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 03:06:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/05/670405/2/gate/kolla-build-centos-source/f3f58ec/ : SUCCESS in 29m 52s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/05/670405/2/gate/tripleo-build-containers-centos-7/d55ba1a/ : SUCCESS in 42m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/670405/2/gate/openstack-tox-lower-constraints/6d94b4d/ : SUCCESS in 3m 04s\n- openstack-tox-pep8 http://logs.openstack.org/05/670405/2/gate/openstack-tox-pep8/a08f32a/ : SUCCESS in 3m 42s\n- openstack-tox-py27 http://logs.openstack.org/05/670405/2/gate/openstack-tox-py27/be985d7/ : SUCCESS in 2m 59s\n- openstack-tox-py36 http://logs.openstack.org/05/670405/2/gate/openstack-tox-py36/8b3296c/ : SUCCESS in 3m 10s\n- openstack-tox-py37 http://logs.openstack.org/05/670405/2/gate/openstack-tox-py37/838370b/ : SUCCESS in 3m 42s\n- openstack-tox-docs http://logs.openstack.org/05/670405/2/gate/openstack-tox-docs/356765b/html/ : SUCCESS in 2m 43s\n- kolla-build-ubuntu-source http://logs.openstack.org/05/670405/2/gate/kolla-build-ubuntu-source/6606004/ : SUCCESS in 27m 25s","accounts_in_message":[],"_revision_number":2},{"id":"c84f28413d2a94f6355544fc6cf8ad03b36e1708","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 03:06:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fe289fe61fbc852f7b6e24fbd09346a6f7e10650","revisions":{"f1569bdee4eccb329aa6fba56772ac5b04031abb":{"kind":"REWORK","_number":1,"created":"2019-07-11 22:29:48.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/05/670405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/670405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/670405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/670405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/670405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/670405/1"}}},"commit":{"parents":[{"commit":"008aa957f27b26caff3be8b18110e71cdaf70772","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/008aa957f27b26caff3be8b18110e71cdaf70772"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-11 22:29:21.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-11 22:29:21.000000000","tz":-360},"subject":"Limit OVMF_CODE.secboot.fd to el7","message":"Limit OVMF_CODE.secboot.fd to el7\n\nThis is no longer needed for el8\n\nChange-Id: I5026a31d3a29b5a31001c6c1f9b90b2a32af6a4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1569bdee4eccb329aa6fba56772ac5b04031abb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1569bdee4eccb329aa6fba56772ac5b04031abb"}]},"branch":"refs/heads/master"},"fe289fe61fbc852f7b6e24fbd09346a6f7e10650":{"kind":"REWORK","_number":2,"created":"2019-07-12 13:14:27.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/05/670405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/670405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/670405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/670405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/670405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/670405/2"}}},"commit":{"parents":[{"commit":"008aa957f27b26caff3be8b18110e71cdaf70772","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/008aa957f27b26caff3be8b18110e71cdaf70772"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-11 22:29:21.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-12 13:14:14.000000000","tz":0},"subject":"Limit OVMF_CODE.secboot.fd to el7","message":"Limit OVMF_CODE.secboot.fd to el7\n\nThis is no longer needed for el8\n\nChange-Id: I5026a31d3a29b5a31001c6c1f9b90b2a32af6a4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe289fe61fbc852f7b6e24fbd09346a6f7e10650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe289fe61fbc852f7b6e24fbd09346a6f7e10650"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
