)]}'
{"id":"openstack%2Fkolla~670816","triplet_id":"openstack%2Fkolla~master~I3a771793dfbbfc26f6d5716ba7a5b60c2e6154fa","project":"openstack/kolla","branch":"master","topic":"ci/add-centos-aarch64-job","hashtags":[],"change_id":"I3a771793dfbbfc26f6d5716ba7a5b60c2e6154fa","subject":"cleanup setup of Debian/Ubuntu CI nodes","status":"MERGED","created":"2019-07-15 11:35:38.000000000","updated":"2019-07-15 20:34:31.000000000","submitted":"2019-07-15 20:34:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670816-1563222871289-0a1ee91c","meta_rev_id":"035459bdcc597be3928b11d10a54faf2e6b4c66f","_number":670816,"virtual_id_number":670816,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-15 20:34:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-07-15 14:29:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-15 19:40:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-07-15 19:40:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-15 11:36:31.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-15 14:29:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-15 19:40:10.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-15 20:34:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c581234ee071c8bff46c0548a680b0cb9ff0d053","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-15 11:35:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbf72e707601e537fb3764da84c2e2562f5f3e3f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-15 11:37:16.000000000","message":"Patch Set 1:\n\nIt is used in both aarch64 CI jobs so it is easier to have as separate change.","accounts_in_message":[],"_revision_number":1},{"id":"c112e4f0266c65282d5d17a82780cbb7b7b264b5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-15 11:41:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c28ece6574afed4cbbd4933eb278b2f7474cc3ac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 13:08:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/16/670816/2/check/kolla-build-centos-binary/35e86d1/ : SUCCESS in 41m 54s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/16/670816/2/check/kolla-build-centos-source/b82d61c/ : SUCCESS in 55m 15s\n- kolla-ansible-centos-source http://logs.openstack.org/16/670816/2/check/kolla-ansible-centos-source/21ca11c/ : SUCCESS in 1h 20m 22s\n- kolla-ansible-centos-binary http://logs.openstack.org/16/670816/2/check/kolla-ansible-centos-binary/8ad0cf5/ : SUCCESS in 1h 06m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/16/670816/2/check/kolla-ansible-centos-source-upgrade/ee69dd7/ : SUCCESS in 1h 20m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/670816/2/check/openstack-tox-lower-constraints/5e152de/ : SUCCESS in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/16/670816/2/check/openstack-tox-pep8/80fb14f/ : SUCCESS in 4m 03s\n- openstack-tox-py27 http://logs.openstack.org/16/670816/2/check/openstack-tox-py27/4353b9e/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/16/670816/2/check/openstack-tox-py36/20e9b12/ : SUCCESS in 3m 14s\n- openstack-tox-py37 http://logs.openstack.org/16/670816/2/check/openstack-tox-py37/846c43f/ : SUCCESS in 3m 24s\n- openstack-tox-docs http://logs.openstack.org/16/670816/2/check/openstack-tox-docs/40b95d8/html/ : SUCCESS in 2m 54s\n- kolla-build-ubuntu-binary http://logs.openstack.org/16/670816/2/check/kolla-build-ubuntu-binary/8a78809/ : SUCCESS in 24m 10s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/16/670816/2/check/kolla-build-ubuntu-source/58d5ef1/ : SUCCESS in 32m 47s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/16/670816/2/check/kolla-ansible-ubuntu-source/75c4a01/ : SUCCESS in 47m 54s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/16/670816/2/check/kolla-ansible-ubuntu-binary/8ec1483/ : SUCCESS in 1h 00m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/16/670816/2/check/kolla-ansible-ubuntu-source-upgrade/02fe11c/ : SUCCESS in 1h 05m 01s","accounts_in_message":[],"_revision_number":2},{"id":"dd6d5aa59ed80f89733eeb4046db649403b8c8ba","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-15 14:29:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"abb94c5e34c8d7df1e0c12ebc7379cb7873bd999","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-15 19:40:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f0db8668ed57e0c56bdc78166ceed71526fd79b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 19:40:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"adfd2e5d2f467d9a3f5c59b6a6a9c04077ae178a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 20:34:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"29833c0149560f55ed3a4f80f84d6ea97d33a30e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-15 20:34:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/16/670816/2/gate/kolla-build-centos-source/b6ddeeb/ : SUCCESS in 48m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/16/670816/2/gate/openstack-tox-lower-constraints/9fe3a2f/ : SUCCESS in 3m 22s\n- openstack-tox-pep8 http://logs.openstack.org/16/670816/2/gate/openstack-tox-pep8/90aeeb9/ : SUCCESS in 4m 58s\n- openstack-tox-py27 http://logs.openstack.org/16/670816/2/gate/openstack-tox-py27/895ff78/ : SUCCESS in 3m 50s\n- openstack-tox-py36 http://logs.openstack.org/16/670816/2/gate/openstack-tox-py36/c6664f7/ : SUCCESS in 3m 16s\n- openstack-tox-py37 http://logs.openstack.org/16/670816/2/gate/openstack-tox-py37/f0f1402/ : SUCCESS in 4m 29s\n- openstack-tox-docs http://logs.openstack.org/16/670816/2/gate/openstack-tox-docs/a65170c/html/ : SUCCESS in 2m 40s\n- kolla-build-ubuntu-source http://logs.openstack.org/16/670816/2/gate/kolla-build-ubuntu-source/40ead48/ : SUCCESS in 30m 23s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b796fbf9834a43d708096eff435cde2ce8aefaa","revisions":{"26ed9b8f4ff785cb5f90530de16d2c9dd7edd574":{"kind":"REWORK","_number":1,"created":"2019-07-15 11:35:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/670816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/670816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/670816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/670816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/670816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/670816/1"}}},"commit":{"parents":[{"commit":"ef6a62de1c91fec16d20485c0bef16cf93346227","subject":"Merge \"Limit OVMF_CODE.secboot.fd to el7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef6a62de1c91fec16d20485c0bef16cf93346227"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:35:09.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:35:09.000000000","tz":120},"subject":"cleanup setup of Debian/Ubuntu CI nodes","message":"cleanup setup of Debian/Ubuntu CI nodes\n\nChange-Id: I3a771793dfbbfc26f6d5716ba7a5b60c2e6154fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26ed9b8f4ff785cb5f90530de16d2c9dd7edd574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26ed9b8f4ff785cb5f90530de16d2c9dd7edd574"}]},"branch":"refs/heads/master"},"0b796fbf9834a43d708096eff435cde2ce8aefaa":{"kind":"REWORK","_number":2,"created":"2019-07-15 11:41:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/16/670816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/670816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/670816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/670816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/670816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/670816/2"}}},"commit":{"parents":[{"commit":"ef6a62de1c91fec16d20485c0bef16cf93346227","subject":"Merge \"Limit OVMF_CODE.secboot.fd to el7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef6a62de1c91fec16d20485c0bef16cf93346227"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:35:09.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-15 11:41:30.000000000","tz":120},"subject":"cleanup setup of Debian/Ubuntu CI nodes","message":"cleanup setup of Debian/Ubuntu CI nodes\n\nChange-Id: I3a771793dfbbfc26f6d5716ba7a5b60c2e6154fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b796fbf9834a43d708096eff435cde2ce8aefaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b796fbf9834a43d708096eff435cde2ce8aefaa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
