)]}'
{"id":"openstack%2Fkolla~671183","triplet_id":"openstack%2Fkolla~master~I872061cf2522520f7a4208cb4ac60f4810863b4d","project":"openstack/kolla","branch":"master","topic":"rhel8-containers-build","hashtags":[],"change_id":"I872061cf2522520f7a4208cb4ac60f4810863b4d","subject":"DNM: Add dnf modules for rhel8/py3 containers","status":"ABANDONED","created":"2019-07-17 04:15:23.000000000","updated":"2020-02-07 16:19:42.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ba5bfd703e985c2c950935c59769107d98051059","_number":671183,"virtual_id_number":671183,"owner":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"date":"2020-02-06 10:11:38.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-31 19:57:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-01-31 17:24:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 08:22:43.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2019-09-20 08:17:34.000000000","updated_by":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"reviewer":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"state":"REVIEWER"},{"updated":"2019-11-08 16:53:09.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-01-31 17:24:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-31 19:57:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-06 10:11:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"7c2fe7690c6e7bf675e8ae1555f095daefd8a3d6","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2019-07-17 04:15:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b5b379d81fa7e46119662522e37b8010814efbf1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 05:11:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/83/671183/1/openstack-check/tripleo-build-containers-fedora-28-master/51e4f68/ : FAILURE in 5m 51s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/83/671183/1/openstack-check/tripleo-build-containers-fedora-28-stein/0e22fe2/ : SUCCESS in 50m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d6608452a6b652fca0696fd2f48a18dc118986c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 06:29:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/83/671183/1/check/kolla-build-centos-binary/5decd15/ : SUCCESS in 29m 41s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/83/671183/1/check/kolla-build-centos-source/2476e5f/ : SUCCESS in 35m 57s\n- kolla-ansible-centos-source http://logs.openstack.org/83/671183/1/check/kolla-ansible-centos-source/e324229/ : SUCCESS in 55m 34s\n- kolla-ansible-centos-binary http://logs.openstack.org/83/671183/1/check/kolla-ansible-centos-binary/58d808a/ : SUCCESS in 1h 06m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/83/671183/1/check/kolla-ansible-centos-source-upgrade/f20e937/ : SUCCESS in 1h 28m 07s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/83/671183/1/check/tripleo-build-containers-centos-7/0b7e5cd/ : SUCCESS in 39m 29s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/83/671183/1/check/kolla-ansible-centos-source-ceph/82094d6/ : SUCCESS in 1h 38m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/83/671183/1/check/kolla-ansible-centos-source-upgrade-ceph/00f4ebd/ : SUCCESS in 1h 46m 39s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv http://logs.openstack.org/83/671183/1/check/kolla-ansible-centos-source-scenario-nfv/eeb208d/ : SUCCESS in 1h 30m 27s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/83/671183/1/check/openstack-tox-lower-constraints/7221770/ : SUCCESS in 3m 42s\n- openstack-tox-pep8 http://logs.openstack.org/83/671183/1/check/openstack-tox-pep8/f7a14cd/ : SUCCESS in 4m 04s\n- openstack-tox-py27 http://logs.openstack.org/83/671183/1/check/openstack-tox-py27/1f55965/ : SUCCESS in 3m 30s\n- openstack-tox-py36 http://logs.openstack.org/83/671183/1/check/openstack-tox-py36/4539c82/ : SUCCESS in 3m 22s\n- openstack-tox-py37 http://logs.openstack.org/83/671183/1/check/openstack-tox-py37/fa5e286/ : SUCCESS in 3m 23s\n- openstack-tox-docs http://logs.openstack.org/83/671183/1/check/openstack-tox-docs/e0fbf6f/html/ : SUCCESS in 2m 54s\n- kolla-build-ubuntu-binary http://logs.openstack.org/83/671183/1/check/kolla-build-ubuntu-binary/fcb8b50/ : SUCCESS in 29m 07s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/83/671183/1/check/kolla-build-ubuntu-source/9af6ad6/ : SUCCESS in 25m 54s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/83/671183/1/check/kolla-ansible-ubuntu-source/49757a0/ : SUCCESS in 46m 32s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/83/671183/1/check/kolla-ansible-ubuntu-binary/3cb35f5/ : SUCCESS in 44m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/83/671183/1/check/kolla-ansible-ubuntu-source-upgrade/6967ff8/ : SUCCESS in 1h 00m 44s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/83/671183/1/check/kolla-ansible-ubuntu-source-ceph/4b82ba9/ : SUCCESS in 1h 06m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/83/671183/1/check/kolla-ansible-ubuntu-source-upgrade-ceph/f1d216d/ : SUCCESS in 1h 22m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be025f4df1b805bc2ccdc67f41fa6ea2b70c650a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2019-07-17 08:22:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03e2c9c09cd26da7ce74ebe02a5bef5a24932e38","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 08:23:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"28608bbb4873e204b6ad743b75e522cd6475b5fd","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 08:25:50.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"36671b74aef86085e3516c527ddfb6756f3b422c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 09:19:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/83/671183/3/openstack-check/tripleo-build-containers-fedora-28-master/a1f54d9/ : FAILURE in 7m 10s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/83/671183/3/openstack-check/tripleo-build-containers-fedora-28-stein/fdfdeef/ : SUCCESS in 49m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eaa7796765ffd1718e91da2caaaeec958d72d076","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 09:29:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"95d8823d7255b2e7d8281d304794f9af330f50e5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 10:23:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/83/671183/4/openstack-check/tripleo-build-containers-fedora-28-master/caa10cf/ : FAILURE in 5m 59s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/83/671183/4/openstack-check/tripleo-build-containers-fedora-28-stein/455ea70/ : SUCCESS in 49m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"309cea40b716223e470873bf9332985d37f705ce","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 10:28:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"35946da1720a5b8a393407ecdaa6bdadf8d167b4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 10:33:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1758238cfe20685a721710fab5e17654ff3ca707","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 11:30:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/83/671183/6/openstack-check/tripleo-build-containers-fedora-28-master/6f288ae/ : FAILURE in 7m 24s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/83/671183/6/openstack-check/tripleo-build-containers-fedora-28-stein/838d7f2/ : SUCCESS in 53m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6211e525d542449a9156050dd056a8cb51562fee","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 12:57:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"55f5228e2a23d1cc14c5957098d6df8b7e6b10f4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-07-17 12:59:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"cfcb35f93ccff68d9665aa139aee7c848114f3f4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 14:05:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/83/671183/8/openstack-check/tripleo-build-containers-fedora-28-master/672f79d/ : FAILURE in 5m 46s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/83/671183/8/openstack-check/tripleo-build-containers-fedora-28-stein/fb85033/ : SUCCESS in 50m 16s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cad4d17f430e72133ffaa5ec5ec6650fb357e7b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 14:59:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/83/671183/8/check/kolla-build-centos-binary/7466a4e/ : SUCCESS in 29m 21s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/83/671183/8/check/kolla-build-centos-source/8352f11/ : SUCCESS in 51m 43s\n- kolla-ansible-centos-source http://logs.openstack.org/83/671183/8/check/kolla-ansible-centos-source/33f5b26/ : SUCCESS in 1h 04m 12s\n- kolla-ansible-centos-binary http://logs.openstack.org/83/671183/8/check/kolla-ansible-centos-binary/8064617/ : SUCCESS in 1h 01m 42s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/83/671183/8/check/kolla-ansible-centos-source-upgrade/9590684/ : SUCCESS in 1h 24m 35s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/83/671183/8/check/tripleo-build-containers-centos-7/23e2153/ : SUCCESS in 41m 59s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/83/671183/8/check/kolla-ansible-centos-source-ceph/0537e66/ : SUCCESS in 1h 50m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph http://logs.openstack.org/83/671183/8/check/kolla-ansible-centos-source-upgrade-ceph/5ab3c7c/ : SUCCESS in 1h 34m 17s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv http://logs.openstack.org/83/671183/8/check/kolla-ansible-centos-source-scenario-nfv/7b37cf8/ : FAILURE in 52m 36s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/83/671183/8/check/openstack-tox-lower-constraints/20144db/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/83/671183/8/check/openstack-tox-pep8/b1e1c8a/ : SUCCESS in 4m 51s\n- openstack-tox-py27 http://logs.openstack.org/83/671183/8/check/openstack-tox-py27/8d3019b/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/83/671183/8/check/openstack-tox-py36/b82c10d/ : SUCCESS in 4m 36s\n- openstack-tox-py37 http://logs.openstack.org/83/671183/8/check/openstack-tox-py37/472743f/ : SUCCESS in 3m 58s\n- openstack-tox-docs http://logs.openstack.org/83/671183/8/check/openstack-tox-docs/9d682ba/html/ : SUCCESS in 3m 13s\n- kolla-build-ubuntu-binary http://logs.openstack.org/83/671183/8/check/kolla-build-ubuntu-binary/5cade3a/ : SUCCESS in 31m 05s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/83/671183/8/check/kolla-build-ubuntu-source/73a73f2/ : SUCCESS in 33m 54s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/83/671183/8/check/kolla-ansible-ubuntu-source/4944333/ : SUCCESS in 49m 48s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/83/671183/8/check/kolla-ansible-ubuntu-binary/6d920bd/ : SUCCESS in 50m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/83/671183/8/check/kolla-ansible-ubuntu-source-upgrade/580ca9a/ : SUCCESS in 1h 11m 03s\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/83/671183/8/check/kolla-ansible-ubuntu-source-ceph/8f50fcb/ : SUCCESS in 1h 11m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph http://logs.openstack.org/83/671183/8/check/kolla-ansible-ubuntu-source-upgrade-ceph/d44b809/ : SUCCESS in 1h 25m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5f805fb7b226e23710989c98c516c71cb2771eeb","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-09-20 08:17:34.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"13735590894c718220813abf61674d04dde2e236","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 08:57:59.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/83/671183/9/openstack-check/tripleo-build-containers-rhel-8-master/be1ce8b/ : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":9},{"id":"0bde570dd86edbb899d60c161d202ec223523ffc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 10:40:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/63ecd380f3654f948edae2d8876d9803 : SUCCESS in 48m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f591216461874da38e85003f8ee9926b : SUCCESS in 56m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fef535f400af4b6dbb8a62639b90ba1c : SUCCESS in 1h 17m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/074eac66be1549a4a89863775e2293b8 : SUCCESS in 1h 24m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1e2d4c83843946ab94ea47e71bb4371d : SUCCESS in 1h 24m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d3584f5ea8ab4034805ad45ddc09568c : SUCCESS in 58m 38s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/c1104caf6c9f41ae9553ddec5e79ea3d : SUCCESS in 1h 44m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/7ba8fe900d97454f880bf0106e90e8b3 : SUCCESS in 2h 16m 05s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/81f2c4ea6f574aba859713bbddea8e83 : SUCCESS in 1h 52m 28s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1a070af8f24449a2b53c31a806425e3f : SUCCESS in 27m 11s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e23be639211b4f76a3fc659ac77a55f0 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ef0ff93fcf4464f9732775aee500ef6 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/28852d79876b491685bd034686672346 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b5a6c23f5ed46c9aade706dd49f5442 : SUCCESS in 3m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/989efac6fed74e78b122a46fce62bd67 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17ec8f9ed3174cd68d41028d4635cb8a : SUCCESS in 5m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/39cd53f52aef4f88a480ced0369a49ab : SUCCESS in 29m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/355f1bc9806b4c5180845af56f85ce39 : SUCCESS in 30m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/459e2c9f887640e192d0f465e8171d9a : SUCCESS in 57m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/73c54bd9c1a74c56bd7d8212899d6d1a : SUCCESS in 55m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/58347dea9d4b4115943adbc8d3346db7 : SUCCESS in 1h 04m 09s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/0204f5687f53467c8ea164667406e815 : SUCCESS in 1h 23m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/3c540a0ded054c22b93f3a143a7b4e60 : FAILURE in 1h 06m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ba3a9e767cfc8d94142049b87090d1b419749f2f","author":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"date":"2019-10-03 16:46:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"2276cae9d9bc916ddf6b96d985d07fa37686fd9d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-10-03 17:27:07.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/83/671183/10/openstack-check/tripleo-build-containers-rhel-8-master/d34c3e6/ : SUCCESS in 40m 11s","accounts_in_message":[],"_revision_number":10},{"id":"8edf6e887db57dba0e1d46e72cfef668de7874f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 19:17:06.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8a462bbed9df45adb70b8478939302db : SUCCESS in 53m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bb9956640f774624836cf4969552dc1b : SUCCESS in 1h 17m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7c8c419e68584438830266844db1fdd0 : SUCCESS in 1h 34m 06s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9c4d510edc974164af569f586a780492 : SUCCESS in 1h 20m 19s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/50fee0943f824357a4a098ca2a28f808 : SUCCESS in 1h 30m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/aac53d92e0a24c798db9100ee2a40724 : SUCCESS in 57m 54s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/bdd4fc7e325247d6bd8eb93fee60bb10 : SUCCESS in 1h 58m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/7fced476cdae435e9437be180b9357d8 : FAILURE in 2h 28m 46s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/13bf7b7beeac4d29bf9bbb37db34ce7a : RETRY_LIMIT in 1m 00s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/63a4641f3393496ab2dbc9f2cd763fec : SUCCESS in 36m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/dfd8b8cefa6045e29d5172002e78b115 : SUCCESS in 29m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c80471635fe4fab88c249930431ce1d : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/babde77d3ac94173968b98a9ae86e688 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a5a1ba1250084c2e87f1375ac29aca4e : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc4395c5385c47a683af6656017fa01a : SUCCESS in 3m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6ca076dfa904c81b4ef036bae1aaed5 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0286154153cc40f6aba0c0a235ead543 : SUCCESS in 5m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/880233534402467988343c662bee7e67 : FAILURE in 31m 21s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/482ce342f1474126bd43442eb771fa91 : SUCCESS in 33m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0575b8780d4242838089a81694cdda9a : SUCCESS in 1h 17m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4041bb3975bd4926838cc547c8a71c70 : SUCCESS in 53m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ddb28f065aec45f3ad63ddb4471c3b70 : SUCCESS in 1h 25m 28s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/f5f19801be424f978c0a7efdcfc45476 : SUCCESS in 1h 29m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/eaa6524297d04256b3b7f653f86391de : FAILURE in 1h 55m 27s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"817a646a2110734ead0c5e8ab8c3c5b121ebc5ad","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-11-07 16:46:06.000000000","message":"Patch Set 10: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 8ea994c330a74a67100448f9edc475b9c2f1a7e1","accounts_in_message":[],"_revision_number":10},{"id":"032e1f53a27195545415260a04c38de16c8c8117","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-11-08 16:53:09.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"0ed722150d259d0fee0eb3a3ca1f5795d5c8ab00","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-11-29 06:42:15.000000000","message":"Patch Set 11: Patch Set 10 was rebased","accounts_in_message":[],"_revision_number":11},{"id":"ae90edf3b7c6ec8e6fcaa21805c5d18fd9a5e86d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-29 07:14:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/83/671183/11/openstack-check/tripleo-build-containers-rhel-8-master/bb66f79/ : SUCCESS in 31m 42s","accounts_in_message":[],"_revision_number":11},{"id":"a0f02bf0128afb93bd59a58f56c4f487e683235c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 09:22:04.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b58e6acbb7764fbabe539ca4336d8cf9 : SUCCESS in 35m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/325212bb1d9e4ba49bd13871d89136ed : SUCCESS in 41m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/eb99099d1d9047d0b7bde4ed8e166629 : SUCCESS in 1h 30m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/785b38f447e94532b1a071687663e0ef : SUCCESS in 1h 18m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d27829dcce1d4c079d808164ffd94ffc : FAILURE in 1h 29m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/05731e184ee74a5fa608bc180c463046 : SUCCESS in 1h 06m 37s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/a9d278a7983140508430f6684319ddcc : SUCCESS in 2h 30m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/c41d58b438b047e4b9ba954a7553cb9e : SUCCESS in 2h 27m 29s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ff33348afb764c8bbb5937a7ce3973c2 : TIMED_OUT in 2h 02m 16s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/502e79de689b450682c6308e981dcc62 : SUCCESS in 27m 25s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a421f67fd7a64efdbfbaccbd0de46d44 : SUCCESS in 22m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9a115c65d5af47cc8610692aaaf5ee03 : SUCCESS in 55m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e52f133b3186443d896c52ce9b95e2a5 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dc140c2a84e4fb5987b8653770ee152 : SUCCESS in 11m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f72edbfcbf84b35a6c96b55dad54d19 : SUCCESS in 11m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9442325f5d8345ae867e1c8b9987dbf3 : SUCCESS in 22m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc3e5900eef242b0b009a5d953286bf6 : TIMED_OUT in 31m 21s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0afb3c67acf64d759984ef222a0020d6 : SUCCESS in 31m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4872efcfdd9243b1a84ea2d65f96758d : SUCCESS in 30m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/91409cfed57043848c71b8c1839ed6f5 : SUCCESS in 53m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/05db9e12f3a44a4e978cd751ce77df68 : SUCCESS in 1h 02m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/09013dd1a446430db86aec7acea61391 : SUCCESS in 1h 00m 05s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/9c4010940aec400bb9f8ffb8926100f2 : SUCCESS in 2h 01m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/4d6634d870c848da9a9dc3155847ead4 : SUCCESS in 1h 40m 34s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7117a35f03103b6aff42bb59b8c0b8a959b6e1c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-29 16:19:36.000000000","message":"Patch Set 11:\n\nThis patch conflicts with https://review.opendev.org/#/c/699414. Are there changes in here that actually need to land for RHEL8?","accounts_in_message":[],"_revision_number":11},{"id":"ca4eddb00bf68f4d1107c0b144bf584fbe3b4dc2","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-01-31 17:16:38.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"71a5c7b9bbe474b01f413a98abea8f3ff600c009","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-01-31 17:24:13.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master https://review.rdoproject.org/zuul/build/b61bb2faacd0409b9552778fb3de1a1d : FAILURE in 5m 56s","accounts_in_message":[],"_revision_number":12},{"id":"9092ce10e52e0925fc5d4f06bc5e41299d2ffe3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:49:44.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"344aba125b434d63b4e678d20562aed1eaa86fea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 19:57:47.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4def883040e849b09dd5eebbd3125858 : SUCCESS in 32m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1dbd106b12df4f17906e1498635862bd : SUCCESS in 50m 09s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e883c491b9d74c60852a45f374d18a74 : SUCCESS in 29m 20s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/93a1cf0a4f3e4257a59fce0713819c3e : SUCCESS in 35m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8da479a4445f44d1b69d46064009d384 : SUCCESS in 1h 14m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/beacd7b335054c94b9c225c402dad3b2 : SUCCESS in 1h 11m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9d0f53a3504742f3acb69a03cd319771 : SUCCESS in 1h 31m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2b92720a302147d38a548320372679f8 : SUCCESS in 55m 23s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/cb830246f8a44219b05cb12cbd76dce6 : SUCCESS in 1h 55m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/dd213b4810e94394bf9cad288e16b39d : SUCCESS in 2h 07m 53s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1258fef1bb2e46e48e73abdedbb53503 : SUCCESS in 1h 47m 42s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e1f3b3b51824470181345a764ad0b7dc : SUCCESS in 29m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/64fa3be254c2485eb2c9f6b5a96169a2 : SUCCESS in 29m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0cb52c0aac9d4e829392f276e9e5a279 : SUCCESS in 1h 00m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3decda94f354c569b0db34edded7c1e : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41b8567cb2c7415fbb56a49e0ec6bbdd : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95c55acd6f4147d4acaaa439c58db989 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcf4ea9ad2bd47a682606a00b1ed1725 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c27ceffc51964228b66d9be107491ff2 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbf526688e3d41a8abadde12dc4c1432 : SUCCESS in 7m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4e705153dcd46249875f25c0bedfb3f : SUCCESS in 29m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8270d276dccd498199ffaf0ecd94bc6a : SUCCESS in 26m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9cc322bfc39547daa8696ea04a086256 : SUCCESS in 59m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a85285c8260a4b83be300fd1c8f493c4 : SUCCESS in 59m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b19d86a8ef634ba28548eba51ae3e9b2 : SUCCESS in 1h 06m 57s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/1914f61a02894a6b98206d6a873d18f6 : SUCCESS in 1h 20m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/dd47350611d14d05accc364535de4745 : SUCCESS in 1h 36m 44s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"1f8774605fc5887bee1b2c3b0209bc493c794bed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-06 10:11:38.000000000","message":"Patch Set 12:\n\n\u003e This patch conflicts with https://review.opendev.org/#/c/699414.\n \u003e Are there changes in here that actually need to land for RHEL8?\n\nAre there changes in here that actually need to land for RHEL8?","accounts_in_message":[],"_revision_number":12},{"id":"1c6568f55ac48c2efe824a7b2aa27110049f301c","author":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"date":"2020-02-07 16:19:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"8dd2ea68f4f5c6c4d4bc1cde4df2d7b53f49f5ce","revisions":{"3871ed73f29c8a4be3babf6514671b4503bbd7c3":{"kind":"REWORK","_number":1,"created":"2019-07-17 04:15:23.000000000","uploader":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"ref":"refs/changes/83/671183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/1"}}},"commit":{"parents":[{"commit":"135a08c0f5d398995d2d8ea950120b1af6301a3a","subject":"Merge \"Add partition detection to fix osd initialization failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/135a08c0f5d398995d2d8ea950120b1af6301a3a"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3871ed73f29c8a4be3babf6514671b4503bbd7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3871ed73f29c8a4be3babf6514671b4503bbd7c3"}]},"branch":"refs/heads/master"},"f0f5aa13042cdec66010876d432712925d4dceb5":{"kind":"REWORK","_number":2,"created":"2019-07-17 08:23:51.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/83/671183/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/2"}}},"commit":{"parents":[{"commit":"135a08c0f5d398995d2d8ea950120b1af6301a3a","subject":"Merge \"Add partition detection to fix osd initialization failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/135a08c0f5d398995d2d8ea950120b1af6301a3a"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-17 08:23:48.000000000","tz":330},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0f5aa13042cdec66010876d432712925d4dceb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0f5aa13042cdec66010876d432712925d4dceb5"}]},"branch":"refs/heads/master"},"41d674e86ede1b7ffb15530c877a66a0e1ae9267":{"kind":"REWORK","_number":3,"created":"2019-07-17 08:25:50.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/83/671183/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/3"}}},"commit":{"parents":[{"commit":"135a08c0f5d398995d2d8ea950120b1af6301a3a","subject":"Merge \"Add partition detection to fix osd initialization failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/135a08c0f5d398995d2d8ea950120b1af6301a3a"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-17 08:25:45.000000000","tz":0},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41d674e86ede1b7ffb15530c877a66a0e1ae9267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/41d674e86ede1b7ffb15530c877a66a0e1ae9267"}]},"branch":"refs/heads/master"},"bd7e2ee5bf14813e1ee4fafb180f8cef6d15f315":{"kind":"REWORK","_number":4,"created":"2019-07-17 09:29:06.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/83/671183/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/4"}}},"commit":{"parents":[{"commit":"135a08c0f5d398995d2d8ea950120b1af6301a3a","subject":"Merge \"Add partition detection to fix osd initialization failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/135a08c0f5d398995d2d8ea950120b1af6301a3a"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-17 09:28:28.000000000","tz":330},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd7e2ee5bf14813e1ee4fafb180f8cef6d15f315"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd7e2ee5bf14813e1ee4fafb180f8cef6d15f315"}]},"branch":"refs/heads/master"},"10c1a4c28f426b2094b64ceb42587a9d56c0b55e":{"kind":"REWORK","_number":5,"created":"2019-07-17 10:28:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/83/671183/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/5"}}},"commit":{"parents":[{"commit":"135a08c0f5d398995d2d8ea950120b1af6301a3a","subject":"Merge \"Add partition detection to fix osd initialization failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/135a08c0f5d398995d2d8ea950120b1af6301a3a"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-17 10:28:25.000000000","tz":330},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10c1a4c28f426b2094b64ceb42587a9d56c0b55e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10c1a4c28f426b2094b64ceb42587a9d56c0b55e"}]},"branch":"refs/heads/master"},"1c7d4b5781ee66a19f8b310d5c5f64e061689afa":{"kind":"REWORK","_number":6,"created":"2019-07-17 10:33:00.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/83/671183/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/6"}}},"commit":{"parents":[{"commit":"135a08c0f5d398995d2d8ea950120b1af6301a3a","subject":"Merge \"Add partition detection to fix osd initialization failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/135a08c0f5d398995d2d8ea950120b1af6301a3a"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-17 10:32:44.000000000","tz":330},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c7d4b5781ee66a19f8b310d5c5f64e061689afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c7d4b5781ee66a19f8b310d5c5f64e061689afa"}]},"branch":"refs/heads/master"},"1f34c26ee0f0b0fb58f240747aad5422f5236ceb":{"kind":"REWORK","_number":7,"created":"2019-07-17 12:57:59.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/83/671183/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/7"}}},"commit":{"parents":[{"commit":"135a08c0f5d398995d2d8ea950120b1af6301a3a","subject":"Merge \"Add partition detection to fix osd initialization failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/135a08c0f5d398995d2d8ea950120b1af6301a3a"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-17 12:57:42.000000000","tz":330},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f34c26ee0f0b0fb58f240747aad5422f5236ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f34c26ee0f0b0fb58f240747aad5422f5236ceb"}]},"branch":"refs/heads/master"},"0c9a419297c5621a0fc901051b29c4c41ffab3ee":{"kind":"REWORK","_number":8,"created":"2019-07-17 12:59:37.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/83/671183/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/8"}}},"commit":{"parents":[{"commit":"135a08c0f5d398995d2d8ea950120b1af6301a3a","subject":"Merge \"Add partition detection to fix osd initialization failure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/135a08c0f5d398995d2d8ea950120b1af6301a3a"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-07-17 12:59:20.000000000","tz":330},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c9a419297c5621a0fc901051b29c4c41ffab3ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c9a419297c5621a0fc901051b29c4c41ffab3ee"}]},"branch":"refs/heads/master"},"9507aea9c67ac34a223b940f9e657e3e25e3ebfa":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2019-09-20 08:17:34.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/83/671183/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/9"}}},"commit":{"parents":[{"commit":"46a03090540dba04e2a590b235eb33685185f455","subject":"Merge \"Add volume-related packages to zun-compute image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46a03090540dba04e2a590b235eb33685185f455"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-09-20 08:16:32.000000000","tz":60},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9507aea9c67ac34a223b940f9e657e3e25e3ebfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9507aea9c67ac34a223b940f9e657e3e25e3ebfa"}]},"branch":"refs/heads/master"},"986af8b0c64267686377ffc4c3db13e5247887d3":{"kind":"REWORK","_number":10,"created":"2019-10-03 16:46:05.000000000","uploader":{"_account_id":24162,"name":"Sorin Sbârnea","display_name":"zbr","email":"ssbarnea@redhat.com","username":"ssbarnea","status":"do not feed the troll 🥕"},"ref":"refs/changes/83/671183/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/10"}}},"commit":{"parents":[{"commit":"c2b5b1029a955fcec88964fe992beb61bd8fa7b2","subject":"Merge \"Debian: publish images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2b5b1029a955fcec88964fe992beb61bd8fa7b2"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Sorin Sbarnea","email":"ssbarnea@redhat.com","date":"2019-10-03 16:45:34.000000000","tz":60},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/986af8b0c64267686377ffc4c3db13e5247887d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/986af8b0c64267686377ffc4c3db13e5247887d3"}]},"branch":"refs/heads/master"},"494029ea40416cad1b0893dfd6738fe321de9af7":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-11-29 06:42:15.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/83/671183/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/11"}}},"commit":{"parents":[{"commit":"9938857d7d331743d71328831241edf21e42d160","subject":"Merge \"Ensure lsb-release exists on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9938857d7d331743d71328831241edf21e42d160"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-11-29 06:42:15.000000000","tz":0},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/494029ea40416cad1b0893dfd6738fe321de9af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/494029ea40416cad1b0893dfd6738fe321de9af7"}]},"branch":"refs/heads/master"},"8dd2ea68f4f5c6c4d4bc1cde4df2d7b53f49f5ce":{"kind":"REWORK","_number":12,"created":"2020-01-31 17:16:38.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/83/671183/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/671183/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/671183/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/671183/12"}}},"commit":{"parents":[{"commit":"f8c6762adeb171b6c4d8a670efc68c3575d0c2de","subject":"Merge \"CentOS 8: Remove shellinabox from ironic-conductor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8c6762adeb171b6c4d8a670efc68c3575d0c2de"}]}],"author":{"name":"Rafael Folco","email":"rfolco@redhat.com","date":"2019-07-17 04:11:13.000000000","tz":-180},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2020-01-31 17:16:30.000000000","tz":-420},"subject":"DNM: Add dnf modules for rhel8/py3 containers","message":"DNM: Add dnf modules for rhel8/py3 containers\n\nThis is an extension test patch for\nhttps://review.opendev.org/#/c/670911\n\nChange-Id: I872061cf2522520f7a4208cb4ac60f4810863b4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dd2ea68f4f5c6c4d4bc1cde4df2d7b53f49f5ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8dd2ea68f4f5c6c4d4bc1cde4df2d7b53f49f5ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
