)]}'
{"id":"openstack%2Fkolla~671196","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ibfe0c2b8be98db56c61f74fb0247488ab3749ef4","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"Ibfe0c2b8be98db56c61f74fb0247488ab3749ef4","subject":"Debian/Ubuntu: move to Python 3 for source images","status":"ABANDONED","created":"2019-07-17 06:46:45.000000000","updated":"2019-07-17 11:20:47.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a3b6644828f4acf23e90f0654cfa021ae0172dd2","_number":671196,"virtual_id_number":671196,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":-1,"date":"2019-07-17 08:46:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-07-17 07:41:26.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2019-07-17 08:38:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-07-17 06:47:18.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"date":"2019-07-17 07:26:41.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 06:47:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-17 07:26:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-17 07:41:26.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-17 08:38:30.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-17 08:46:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a512b7805b83f482fe6ab4e677faff1c26561447","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-17 06:46:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bf7147d4a23faeaf668b9ebd8752ed425a0475d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-17 07:26:41.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nalso - removing infoblox plugin without a release note seems rude :-)","accounts_in_message":[],"_revision_number":1},{"id":"983d38fb7d9cedf81a8506a01bd8591e52fed982","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 07:41:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/96/671196/1/openstack-check/tripleo-build-containers-fedora-28-master/092b8e9/ : FAILURE in 7m 12s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/96/671196/1/openstack-check/tripleo-build-containers-fedora-28-stein/afe2aad/ : SUCCESS in 51m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a221fbb5fc0a930f8dfbbc94910e6ccc9cd2e37c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-07-17 08:38:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bdacc5ed19dd9e16c9b145600294ad2a0917b1d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 08:46:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/96/671196/1/check/kolla-build-centos-binary/3414451/ : SUCCESS in 51m 24s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/96/671196/1/check/kolla-build-centos-source/ed4dcb6/ : SUCCESS in 52m 40s\n- kolla-ansible-centos-source http://logs.openstack.org/96/671196/1/check/kolla-ansible-centos-source/037c8bf/ : SUCCESS in 1h 17m 05s\n- kolla-ansible-centos-binary http://logs.openstack.org/96/671196/1/check/kolla-ansible-centos-binary/931bdc2/ : SUCCESS in 1h 00m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/96/671196/1/check/kolla-ansible-centos-source-upgrade/4e4768c/ : SUCCESS in 1h 26m 55s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/96/671196/1/check/tripleo-build-containers-centos-7/ef50d01/ : SUCCESS in 48m 18s\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/96/671196/1/check/kolla-ansible-centos-source-ceph/a7f43da/ : SUCCESS in 1h 30m 15s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/96/671196/1/check/kolla-build-oraclelinux-binary/942ec65/ : SUCCESS in 44m 51s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/96/671196/1/check/kolla-build-oraclelinux-source/2d6ddd2/ : SUCCESS in 49m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/96/671196/1/check/openstack-tox-lower-constraints/b412a5c/ : SUCCESS in 3m 45s\n- openstack-tox-pep8 http://logs.openstack.org/96/671196/1/check/openstack-tox-pep8/2115b09/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/96/671196/1/check/openstack-tox-py27/f118d7c/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/96/671196/1/check/openstack-tox-py36/1d4fe59/ : SUCCESS in 3m 49s\n- openstack-tox-py37 http://logs.openstack.org/96/671196/1/check/openstack-tox-py37/c1c467e/ : SUCCESS in 6m 30s\n- openstack-tox-docs http://logs.openstack.org/96/671196/1/check/openstack-tox-docs/b10a05e/html/ : SUCCESS in 4m 06s\n- kolla-build-ubuntu-binary http://logs.openstack.org/96/671196/1/check/kolla-build-ubuntu-binary/234f1bc/ : SUCCESS in 27m 43s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/96/671196/1/check/kolla-build-ubuntu-source/3caed8c/ : FAILURE in 34m 39s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/96/671196/1/check/kolla-ansible-ubuntu-source/aedfda2/ : FAILURE in 42m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/96/671196/1/check/kolla-ansible-ubuntu-binary/f7e9d0c/ : SUCCESS in 52m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/96/671196/1/check/kolla-ansible-ubuntu-source-upgrade/eb8c57c/ : FAILURE in 59m 44s\n- kolla-ansible-ubuntu-source-zun http://logs.openstack.org/96/671196/1/check/kolla-ansible-ubuntu-source-zun/9b7e7ef/ : FAILURE in 16m 26s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/96/671196/1/check/kolla-ansible-ubuntu-source-ceph/0e3bea4/ : FAILURE in 54m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3b6644828f4acf23e90f0654cfa021ae0172dd2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-17 11:20:47.000000000","message":"Abandoned\n\ntoo many backports needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af483a22aa1f6efd40b82dc156ad79f63f6b92d8","revisions":{"af483a22aa1f6efd40b82dc156ad79f63f6b92d8":{"kind":"REWORK","_number":1,"created":"2019-07-17 06:46:45.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/96/671196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/671196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/671196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/671196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/671196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/671196/1"}}},"commit":{"parents":[{"commit":"c14e4b3f1330be1fd861647175cce57a7f4f3eac","subject":"Merge \"CI: set the same gate queue for kolla and kolla-ansible\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c14e4b3f1330be1fd861647175cce57a7f4f3eac"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-03-08 17:50:14.000000000","tz":60},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-17 06:44:11.000000000","tz":120},"subject":"Debian/Ubuntu: move to Python 3 for source images","message":"Debian/Ubuntu: move to Python 3 for source images\n\nCommit 43b74ccc157c0b50138c3785ed91dab504895571 enabled use of Python 3\nbased packages but not switched to use Python 3.\n\nSome of images still contain Python 2. There are two reasons:\n\n- Ceph (ceph-common depends on Py2)\n- python3-ldappool on Ubuntu 18.04\n\nIn Ceph situation Py3 packages were added. For second one we can not do\nanything - Py2 dependency got dropped in Ubuntu 18.10 version.\n\nRemoved neutron-server-plugin-networking-infoblox due to being not\nmaintained. Once https://review.opendev.org/#/c/657578/ get merged\nsomeone may revert that part.\n\nImplements: blueprint debian-ubuntu-python3\n\nDepends-on: Ie2a1077f7def0743f1403341985e2109aa490026\n\nChange-Id: Ibfe0c2b8be98db56c61f74fb0247488ab3749ef4\n(cherry picked from commit d436001b43f0a7e3c161ac4a2c5d09d36c6f1015 and\nedited to apply:\n- kolla/common/config.py s/master/versionnumbers\n- docker/cinder-volume/Dockerfile.j2 change was already there\n)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af483a22aa1f6efd40b82dc156ad79f63f6b92d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af483a22aa1f6efd40b82dc156ad79f63f6b92d8"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
