)]}'
{"id":"openstack%2Fkolla~671507","triplet_id":"openstack%2Fkolla~master~I0f83c3498028135751b0b887665d009e5e19410f","project":"openstack/kolla","branch":"master","topic":"671507","hashtags":[],"change_id":"I0f83c3498028135751b0b887665d009e5e19410f","subject":"There is no Java 8 in Debian \u0027buster\u0027 (part 3)","status":"MERGED","created":"2019-07-18 15:05:58.000000000","updated":"2019-10-02 13:11:54.000000000","submitted":"2019-07-22 17:29:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671507-1563816559614-4e3e88c1","meta_rev_id":"66028b6d9bc9c6498291e926fbc01eca98df544d","_number":671507,"virtual_id_number":671507,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"date":"2019-07-19 12:10:29.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-22 17:29:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-19 13:00:56.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-07-20 16:52:20.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-07-22 16:32:54.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-18 15:06:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-18 15:06:20.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-07-19 04:42:12.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-07-19 13:00:56.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-22 16:32:54.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-22 17:29:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c43ba6a552e428e9ef606248b3a8c1d50df4947a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-18 15:05:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ab52e6f0afba7f8e75873bae9ce412a0ac32f2c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-18 15:08:39.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6be18cbb2bbaeecef6a79a050836577eec438da1","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-18 15:17:53.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"96c870cc043e9e3bba11548b1cc682c24f52af39","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-18 15:21:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"245ab94ac5946ac615deab96d720076a8c5790c0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-18 16:14:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/07/671507/2/openstack-check/tripleo-build-containers-fedora-28-master/4fbb761/ : FAILURE in 6m 29s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/07/671507/2/openstack-check/tripleo-build-containers-fedora-28-stein/48a7587/ : SUCCESS in 49m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc7df710c8d5e20a5dae6873e72eba9dd3022cbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 16:49:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/07/671507/2/check/kolla-build-centos-binary/3b6dd15/ : SUCCESS in 29m 55s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/07/671507/2/check/kolla-build-centos-source/12b341c/ : SUCCESS in 40m 42s\n- kolla-ansible-centos-source http://logs.openstack.org/07/671507/2/check/kolla-ansible-centos-source/82048ce/ : SUCCESS in 1h 07m 27s\n- kolla-ansible-centos-binary http://logs.openstack.org/07/671507/2/check/kolla-ansible-centos-binary/5bf6db1/ : SUCCESS in 1h 08m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/07/671507/2/check/kolla-ansible-centos-source-upgrade/97cd40e/ : SUCCESS in 1h 26m 40s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/07/671507/2/check/tripleo-build-containers-centos-7/45481c9/ : SUCCESS in 51m 22s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/671507/2/check/openstack-tox-lower-constraints/5894265/ : SUCCESS in 4m 07s\n- openstack-tox-pep8 http://logs.openstack.org/07/671507/2/check/openstack-tox-pep8/3dc3b45/ : SUCCESS in 5m 40s\n- openstack-tox-py27 http://logs.openstack.org/07/671507/2/check/openstack-tox-py27/a85a5b8/ : SUCCESS in 4m 57s\n- openstack-tox-py36 http://logs.openstack.org/07/671507/2/check/openstack-tox-py36/eb59e08/ : SUCCESS in 4m 28s\n- openstack-tox-py37 http://logs.openstack.org/07/671507/2/check/openstack-tox-py37/025abd2/ : SUCCESS in 4m 43s\n- openstack-tox-docs http://logs.openstack.org/07/671507/2/check/openstack-tox-docs/0855615/html/ : SUCCESS in 3m 03s\n- kolla-build-ubuntu-binary http://logs.openstack.org/07/671507/2/check/kolla-build-ubuntu-binary/aa5bba0/ : SUCCESS in 35m 39s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/07/671507/2/check/kolla-build-ubuntu-source/6881960/ : FAILURE in 42m 14s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/07/671507/2/check/kolla-ansible-ubuntu-source/cdeef78/ : SUCCESS in 1h 04m 01s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/07/671507/2/check/kolla-ansible-ubuntu-binary/8e57352/ : SUCCESS in 54m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/07/671507/2/check/kolla-ansible-ubuntu-source-upgrade/97b770b/ : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":2},{"id":"2bb518d00c127269690a2f2d0ca1293b532469bf","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-07-19 04:42:12.000000000","message":"Patch Set 2: Code-Review+2\n\ngood to have, let\u0027s pass the gate.","accounts_in_message":[],"_revision_number":2},{"id":"5c0a0d9c7201b11803c267232c86596668f4e97a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-19 06:22:21.000000000","message":"Patch Set 2:\n\nmonasca failed","accounts_in_message":[],"_revision_number":2},{"id":"2a9eda14ab806fb71deff7c2c7e173c886d02e32","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-19 07:53:40.000000000","message":"Patch Set 2:\n\nI will look at monasca-thresh failure","accounts_in_message":[],"_revision_number":2},{"id":"0d262fae691e13d5231ff28f1a96b319b6404b86","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-19 09:08:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ae66677bc00c7b734dec6230a7d6516f5a891a8f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 10:00:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/07/671507/3/openstack-check/tripleo-build-containers-fedora-28-master/59ac642/ : FAILURE in 6m 04s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/07/671507/3/openstack-check/tripleo-build-containers-fedora-28-stein/3ed3043/ : SUCCESS in 47m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ec70c2c1e9ed9a3b420b8aee0f63965ae0172de2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 10:21:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/07/671507/3/check/kolla-build-centos-binary/38117ca/ : FAILURE in 37m 20s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/07/671507/3/check/kolla-build-centos-source/35df145/ : FAILURE in 37m 08s\n- kolla-ansible-centos-source http://logs.openstack.org/07/671507/3/check/kolla-ansible-centos-source/27c436d/ : SUCCESS in 1h 00m 57s\n- kolla-ansible-centos-binary http://logs.openstack.org/07/671507/3/check/kolla-ansible-centos-binary/e31f887/ : SUCCESS in 1h 05m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/07/671507/3/check/kolla-ansible-centos-source-upgrade/8b8bd6e/ : SUCCESS in 1h 09m 06s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/07/671507/3/check/tripleo-build-containers-centos-7/a06bc80/ : SUCCESS in 38m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/671507/3/check/openstack-tox-lower-constraints/0da6296/ : SUCCESS in 3m 09s\n- openstack-tox-pep8 http://logs.openstack.org/07/671507/3/check/openstack-tox-pep8/d45b79a/ : SUCCESS in 4m 11s\n- openstack-tox-py27 http://logs.openstack.org/07/671507/3/check/openstack-tox-py27/0251bcd/ : SUCCESS in 3m 14s\n- openstack-tox-py36 http://logs.openstack.org/07/671507/3/check/openstack-tox-py36/42a662f/ : SUCCESS in 3m 12s\n- openstack-tox-py37 http://logs.openstack.org/07/671507/3/check/openstack-tox-py37/77ddb4e/ : SUCCESS in 3m 21s\n- openstack-tox-docs http://logs.openstack.org/07/671507/3/check/openstack-tox-docs/5086e6e/html/ : SUCCESS in 2m 57s\n- kolla-build-ubuntu-binary http://logs.openstack.org/07/671507/3/check/kolla-build-ubuntu-binary/33c39c4/ : FAILURE in 20m 41s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/07/671507/3/check/kolla-build-ubuntu-source/f3ac94f/ : FAILURE in 30m 51s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/07/671507/3/check/kolla-ansible-ubuntu-source/9ee09d2/ : SUCCESS in 40m 01s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/07/671507/3/check/kolla-ansible-ubuntu-binary/5d4e816/ : SUCCESS in 51m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/07/671507/3/check/kolla-ansible-ubuntu-source-upgrade/f980d0a/ : SUCCESS in 52m 44s","accounts_in_message":[],"_revision_number":3},{"id":"0483a2a7b1f4777d3e8492e50bff78ff5a0d9637","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-19 12:10:29.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3028300feba624363a88f309a50109c07e37b4c5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-19 13:00:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/07/671507/3/openstack-check/tripleo-build-containers-fedora-28-master/34cee4b/ : FAILURE in 6m 45s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/07/671507/3/openstack-check/tripleo-build-containers-fedora-28-stein/94e7004/ : SUCCESS in 45m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e8accf1e0bee48b6dac33b6f059251b3b51c9369","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 13:40:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/07/671507/3/check/kolla-build-centos-binary/55c67fd/ : SUCCESS in 32m 44s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/07/671507/3/check/kolla-build-centos-source/a96162e/ : SUCCESS in 41m 47s\n- kolla-ansible-centos-source http://logs.openstack.org/07/671507/3/check/kolla-ansible-centos-source/55ba368/ : SUCCESS in 1h 13m 28s\n- kolla-ansible-centos-binary http://logs.openstack.org/07/671507/3/check/kolla-ansible-centos-binary/b37cd7b/ : SUCCESS in 1h 06m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/07/671507/3/check/kolla-ansible-centos-source-upgrade/75cc8cf/ : SUCCESS in 1h 22m 20s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/07/671507/3/check/tripleo-build-containers-centos-7/9207be3/ : SUCCESS in 50m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/671507/3/check/openstack-tox-lower-constraints/8d6ccb1/ : SUCCESS in 4m 55s\n- openstack-tox-pep8 http://logs.openstack.org/07/671507/3/check/openstack-tox-pep8/403d4f7/ : SUCCESS in 5m 16s\n- openstack-tox-py27 http://logs.openstack.org/07/671507/3/check/openstack-tox-py27/9afbdb3/ : SUCCESS in 4m 23s\n- openstack-tox-py36 http://logs.openstack.org/07/671507/3/check/openstack-tox-py36/02c6185/ : SUCCESS in 5m 04s\n- openstack-tox-py37 http://logs.openstack.org/07/671507/3/check/openstack-tox-py37/87ebd42/ : SUCCESS in 5m 10s\n- openstack-tox-docs http://logs.openstack.org/07/671507/3/check/openstack-tox-docs/72ce2e0/html/ : SUCCESS in 3m 37s\n- kolla-build-ubuntu-binary http://logs.openstack.org/07/671507/3/check/kolla-build-ubuntu-binary/85109a4/ : SUCCESS in 30m 23s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/07/671507/3/check/kolla-build-ubuntu-source/a761ac8/ : SUCCESS in 32m 47s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/07/671507/3/check/kolla-ansible-ubuntu-source/456be39/ : SUCCESS in 55m 17s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/07/671507/3/check/kolla-ansible-ubuntu-binary/2b34d8c/ : SUCCESS in 46m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/07/671507/3/check/kolla-ansible-ubuntu-source-upgrade/52bd908/ : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":3},{"id":"9bf8ecabd2770c3ad5889de67ef87265fa0daa07","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-20 16:52:20.000000000","message":"Patch Set 3: Code-Review+2\n\ngood indeed","accounts_in_message":[],"_revision_number":3},{"id":"86334886e22c56b25c9ee55bc03dc3512e23b22a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-22 16:32:54.000000000","message":"Patch Set 3: Workflow+1\n\ntrivial and some containers broken without","accounts_in_message":[],"_revision_number":3},{"id":"0b9be8906d42858fa928dbead599d5ce48d90327","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 16:33:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"81d9eaebaba9ad7fce33e27fe1e184433041408a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 17:29:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/07/671507/3/gate/kolla-build-centos-source/e41332e/ : SUCCESS in 47m 01s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/07/671507/3/gate/tripleo-build-containers-centos-7/dfe993b/ : SUCCESS in 47m 49s\n- openstack-tox-lower-constraints http://logs.openstack.org/07/671507/3/gate/openstack-tox-lower-constraints/fa0a66c/ : SUCCESS in 4m 25s\n- openstack-tox-pep8 http://logs.openstack.org/07/671507/3/gate/openstack-tox-pep8/11df925/ : SUCCESS in 5m 46s\n- openstack-tox-py27 http://logs.openstack.org/07/671507/3/gate/openstack-tox-py27/3f3eafa/ : SUCCESS in 4m 26s\n- openstack-tox-py36 http://logs.openstack.org/07/671507/3/gate/openstack-tox-py36/c3e51a4/ : SUCCESS in 3m 13s\n- openstack-tox-py37 http://logs.openstack.org/07/671507/3/gate/openstack-tox-py37/fc2fbd5/ : SUCCESS in 3m 32s\n- openstack-tox-docs http://logs.openstack.org/07/671507/3/gate/openstack-tox-docs/1ee8141/html/ : SUCCESS in 3m 43s\n- kolla-build-ubuntu-source http://logs.openstack.org/07/671507/3/gate/kolla-build-ubuntu-source/d81186e/ : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":3},{"id":"f516f050c164aa8a6c987c569a3df25c3d0e9008","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 17:29:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"42f434ae6bd9ce6614bd6a4a25f820d8d87c2d8c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 13:11:54.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 7f8dd3063b9939a4d6346bd4a114805a2484c9c2","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0383aaca7b6282e69761e4c44a7d8e509305e107","revisions":{"f5966089cc04ef2422157c9c608b2fbaaa9ed68a":{"kind":"REWORK","_number":1,"created":"2019-07-18 15:05:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/07/671507/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/671507/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/671507/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/671507/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/671507/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/671507/1"}}},"commit":{"parents":[{"commit":"0e766d57ad2cfe3c355b7b4ff9dad18ee0354129","subject":"Merge \"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e766d57ad2cfe3c355b7b4ff9dad18ee0354129"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-18 15:04:36.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-18 15:04:36.000000000","tz":120},"subject":"There is no Java 8 in Debian \u0027buster\u0027 (part 3)","message":"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\n\nGood to have proper Java version. Better is when JAVA_HOME is also\nadjusted to point into proper directory.\n\nChange-Id: I0f83c3498028135751b0b887665d009e5e19410f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5966089cc04ef2422157c9c608b2fbaaa9ed68a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5966089cc04ef2422157c9c608b2fbaaa9ed68a"}]},"branch":"refs/heads/master"},"7875ea477ec467cd977df109e8c2a8332eef1a71":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-18 15:17:53.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/07/671507/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/671507/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/671507/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/671507/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/671507/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/671507/2"}}},"commit":{"parents":[{"commit":"0e766d57ad2cfe3c355b7b4ff9dad18ee0354129","subject":"Merge \"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e766d57ad2cfe3c355b7b4ff9dad18ee0354129"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-18 15:04:36.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-18 15:17:51.000000000","tz":0},"subject":"There is no Java 8 in Debian \u0027buster\u0027 (part 3)","message":"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\n\nGood to have proper Java version. Better is when JAVA_HOME is also\nadjusted to point to proper directory.\n\nChange-Id: I0f83c3498028135751b0b887665d009e5e19410f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7875ea477ec467cd977df109e8c2a8332eef1a71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7875ea477ec467cd977df109e8c2a8332eef1a71"}]},"branch":"refs/heads/master"},"0383aaca7b6282e69761e4c44a7d8e509305e107":{"kind":"REWORK","_number":3,"created":"2019-07-19 09:08:37.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/07/671507/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/671507/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/671507/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/671507/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/671507/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/671507/3"}}},"commit":{"parents":[{"commit":"0e766d57ad2cfe3c355b7b4ff9dad18ee0354129","subject":"Merge \"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e766d57ad2cfe3c355b7b4ff9dad18ee0354129"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-18 15:04:36.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-19 09:08:09.000000000","tz":0},"subject":"There is no Java 8 in Debian \u0027buster\u0027 (part 3)","message":"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\n\nGood to have proper Java version. Better is when JAVA_HOME is also\nadjusted to point to proper directory. And to have it set proper way.\n\nChange-Id: I0f83c3498028135751b0b887665d009e5e19410f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0383aaca7b6282e69761e4c44a7d8e509305e107"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0383aaca7b6282e69761e4c44a7d8e509305e107"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
