)]}'
{"id":"openstack%2Fkolla~671888","triplet_id":"openstack%2Fkolla~master~I2970b6c0d0684a00a59ff3b2d9eebb3d04495345","project":"openstack/kolla","branch":"master","topic":"bug/1837274","hashtags":[],"change_id":"I2970b6c0d0684a00a59ff3b2d9eebb3d04495345","subject":"Fixes Fluentd on Ubuntu deploys.","status":"ABANDONED","created":"2019-07-20 08:49:44.000000000","updated":"2019-08-02 08:40:29.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8f3cbda18baec34c025da16d423fcdbdbd769430","_number":671888,"virtual_id_number":671888,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-08-02 08:29:56.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"date":"2019-07-30 11:23:20.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2019-07-21 09:38:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-07-30 10:51:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":-2,"date":"2019-07-30 10:30:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-21 09:38:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-30 10:30:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-30 10:51:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-02 08:29:56.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"7463ce3c56d5bad9c42f1f5b115186d0cd5233a5","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-07-20 08:49:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd2d0ba18a6ee5e5e5b12e149d352f9d40349752","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-20 09:24:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce66f2384fb057ff38f491b95a4871da2cbe96bc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-20 09:39:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/671888/1/openstack-check/tripleo-build-containers-fedora-28-master/d683c3f/ : FAILURE in 5m 45s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/671888/1/openstack-check/tripleo-build-containers-fedora-28-stein/0776db3/ : SUCCESS in 44m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4a7547da13e47f73b2291f84843ec619542e345","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 10:08:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/88/671888/1/check/kolla-build-centos-binary/ff35483/ : SUCCESS in 35m 19s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/671888/1/check/kolla-build-centos-source/f7493bd/ : SUCCESS in 39m 47s\n- kolla-ansible-centos-source http://logs.openstack.org/88/671888/1/check/kolla-ansible-centos-source/077510e/ : SUCCESS in 1h 05m 09s\n- kolla-ansible-centos-binary http://logs.openstack.org/88/671888/1/check/kolla-ansible-centos-binary/f2bcfc0/ : SUCCESS in 1h 15m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/88/671888/1/check/kolla-ansible-centos-source-upgrade/a08d47c/ : SUCCESS in 1h 13m 58s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/671888/1/check/tripleo-build-containers-centos-7/3f2b24f/ : SUCCESS in 43m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/671888/1/check/openstack-tox-lower-constraints/7dd0896/ : SUCCESS in 3m 47s\n- openstack-tox-pep8 http://logs.openstack.org/88/671888/1/check/openstack-tox-pep8/fc79802/ : SUCCESS in 4m 22s\n- openstack-tox-py27 http://logs.openstack.org/88/671888/1/check/openstack-tox-py27/c004a79/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/88/671888/1/check/openstack-tox-py36/c3b2437/ : SUCCESS in 4m 08s\n- openstack-tox-py37 http://logs.openstack.org/88/671888/1/check/openstack-tox-py37/707315a/ : SUCCESS in 4m 03s\n- openstack-tox-docs http://logs.openstack.org/88/671888/1/check/openstack-tox-docs/5d8ef39/html/ : SUCCESS in 2m 55s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/671888/1/check/kolla-build-ubuntu-binary/2e23733/ : SUCCESS in 35m 15s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/671888/1/check/kolla-build-ubuntu-source/b6b198e/ : SUCCESS in 30m 38s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/671888/1/check/kolla-ansible-ubuntu-source/3719548/ : SUCCESS in 49m 04s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/671888/1/check/kolla-ansible-ubuntu-binary/156a998/ : SUCCESS in 44m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/88/671888/1/check/kolla-ansible-ubuntu-source-upgrade/ffa40c9/ : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":1},{"id":"ef318ea74d6d4fb8bfab6341301b37d54bd93864","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-20 16:36:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e58ca4e80e6a149a78e76f6823e0c470847c0ff","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-07-21 08:09:11.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a199965d4818b39713ef25031392df08b6d72313","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-21 08:59:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/671888/2/openstack-check/tripleo-build-containers-fedora-28-master/53d273b/ : FAILURE in 5m 48s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/671888/2/openstack-check/tripleo-build-containers-fedora-28-stein/6385559/ : SUCCESS in 44m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d59ce20c3c6b58dbf2349f78399db6ffb1a03a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-21 09:38:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/88/671888/2/check/kolla-build-centos-binary/0af8abd/ : SUCCESS in 42m 05s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/671888/2/check/kolla-build-centos-source/2c667c7/ : SUCCESS in 52m 29s\n- kolla-ansible-centos-source http://logs.openstack.org/88/671888/2/check/kolla-ansible-centos-source/0d0fdbd/ : SUCCESS in 1h 11m 28s\n- kolla-ansible-centos-binary http://logs.openstack.org/88/671888/2/check/kolla-ansible-centos-binary/d8ffefe/ : SUCCESS in 1h 07m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/88/671888/2/check/kolla-ansible-centos-source-upgrade/fd79885/ : SUCCESS in 1h 24m 15s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/671888/2/check/tripleo-build-containers-centos-7/52ef0fd/ : SUCCESS in 47m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/671888/2/check/openstack-tox-lower-constraints/814b4c3/ : SUCCESS in 3m 08s\n- openstack-tox-pep8 http://logs.openstack.org/88/671888/2/check/openstack-tox-pep8/e950eb8/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/88/671888/2/check/openstack-tox-py27/726d953/ : SUCCESS in 3m 03s\n- openstack-tox-py36 http://logs.openstack.org/88/671888/2/check/openstack-tox-py36/0f2b93b/ : SUCCESS in 3m 14s\n- openstack-tox-py37 http://logs.openstack.org/88/671888/2/check/openstack-tox-py37/83b22c0/ : SUCCESS in 2m 52s\n- openstack-tox-docs http://logs.openstack.org/88/671888/2/check/openstack-tox-docs/0a1d751/html/ : SUCCESS in 2m 20s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/671888/2/check/kolla-build-ubuntu-binary/e0c013e/ : SUCCESS in 28m 08s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/671888/2/check/kolla-build-ubuntu-source/41fdd8c/ : SUCCESS in 28m 04s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/671888/2/check/kolla-ansible-ubuntu-source/b0da3a4/ : SUCCESS in 47m 36s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/671888/2/check/kolla-ansible-ubuntu-binary/89a80c4/ : SUCCESS in 49m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/88/671888/2/check/kolla-ansible-ubuntu-source-upgrade/a027084/ : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d278dfd67b4256218eb33aa7d044c858b45cc2c3","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-07-30 09:52:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"24407e759f701087d8588f691cd19a4fd7160580","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-30 10:30:33.000000000","message":"Patch Set 2: Code-Review-2\n\nI was wondering why we do not catch this - the problem is elsewhere. You probably run monasca - seems its fluentd config is wrong, compare 09-monasca.conf.j2 with any other - the dir should be conditional there.\nSee: https://opendev.org/openstack/kolla-ansible/raw/commit/5a9ec1a7733b4a5480d70ee28ac71c4440e8b4fc/ansible/roles/common/templates/conf/input/09-monasca.conf.j2\nvs e.g.: https://opendev.org/openstack/kolla-ansible/raw/commit/5a9ec1a7733b4a5480d70ee28ac71c4440e8b4fc/ansible/roles/common/templates/conf/input/08-opendaylight.conf.j2","accounts_in_message":[],"_revision_number":2},{"id":"f5a9ecb2ea233394aed91eb7efacffda6a95259c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-30 10:33:22.000000000","message":"Patch Set 2:\n\nAnd it\u0027s not a container-killing thingy: http://logs.openstack.org/65/673165/11/check/kolla-ansible-ubuntu-source/b14bb63/primary/logs/docker_logs/fluentd.txt.gz - we should check the CI logs in fact","accounts_in_message":[],"_revision_number":2},{"id":"c4b6e2322ac48d569c2eb47a4a45d5b52ea808a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-30 10:51:37.000000000","message":"Patch Set 2:\n\nproposed: https://review.opendev.org/673479","accounts_in_message":[],"_revision_number":2},{"id":"aa1e115fd138a7d72fadc2f0dd8316de15a0fed4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 10:51:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/88/671888/2/openstack-check/tripleo-build-containers-fedora-28-master/d99e2f3/ : FAILURE in 6m 10s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/88/671888/2/openstack-check/tripleo-build-containers-fedora-28-stein/b3d94aa/ : SUCCESS in 54m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09485d4879d2c0e8511d751343b4030c8e3b4eb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 11:18:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/88/671888/2/check/kolla-build-centos-binary/2458219/ : SUCCESS in 42m 45s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/88/671888/2/check/kolla-build-centos-source/0040584/ : SUCCESS in 41m 35s\n- kolla-ansible-centos-source http://logs.openstack.org/88/671888/2/check/kolla-ansible-centos-source/1ee4c32/ : SUCCESS in 1h 12m 49s\n- kolla-ansible-centos-binary http://logs.openstack.org/88/671888/2/check/kolla-ansible-centos-binary/712e8b2/ : SUCCESS in 1h 07m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/88/671888/2/check/kolla-ansible-centos-source-upgrade/3d54e6b/ : SUCCESS in 1h 21m 37s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/88/671888/2/check/tripleo-build-containers-centos-7/315b9a5/ : SUCCESS in 48m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/671888/2/check/openstack-tox-lower-constraints/5407042/ : SUCCESS in 5m 18s\n- openstack-tox-pep8 http://logs.openstack.org/88/671888/2/check/openstack-tox-pep8/7cdf78f/ : SUCCESS in 5m 52s\n- openstack-tox-py27 http://logs.openstack.org/88/671888/2/check/openstack-tox-py27/ebbf8c6/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/88/671888/2/check/openstack-tox-py36/8de9577/ : SUCCESS in 2m 58s\n- openstack-tox-py37 http://logs.openstack.org/88/671888/2/check/openstack-tox-py37/5626bcd/ : SUCCESS in 4m 49s\n- openstack-tox-docs http://logs.openstack.org/88/671888/2/check/openstack-tox-docs/107f7df/html/ : SUCCESS in 3m 59s\n- kolla-build-ubuntu-binary http://logs.openstack.org/88/671888/2/check/kolla-build-ubuntu-binary/4cfc8d7/ : SUCCESS in 27m 45s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/88/671888/2/check/kolla-build-ubuntu-source/a08eeae/ : SUCCESS in 29m 46s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/88/671888/2/check/kolla-ansible-ubuntu-source/98c8983/ : SUCCESS in 42m 49s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/88/671888/2/check/kolla-ansible-ubuntu-binary/1541f8d/ : SUCCESS in 46m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/88/671888/2/check/kolla-ansible-ubuntu-source-upgrade/c6b8c8d/ : SUCCESS in 54m 06s","accounts_in_message":[],"_revision_number":2},{"id":"1659b1c0a712d1151a294e52c45d5b2168d00683","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-07-30 11:23:20.000000000","message":"Patch Set 2:\n\nI\u0027m confused because Monasca is not enabled at my deployment.","accounts_in_message":[],"_revision_number":2},{"id":"e6605659b53ad828bbee4ee83ec30226fde90892","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-30 11:24:41.000000000","message":"Patch Set 2:\n\n\u003e I\u0027m confused because Monasca is not enabled at my deployment.\n\nThe config file seems to be deployed even with monasca disabled, it\u0027s disabled in CI too.","accounts_in_message":[],"_revision_number":2},{"id":"8472811af368f9a3798184377db0e939fba97168","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-30 12:00:55.000000000","message":"Patch Set 2:\n\nAlso, you must have used k-a master with k stein because k-a stein does not seem to deploy this file.","accounts_in_message":[],"_revision_number":2},{"id":"cde469272cde5e24b41554881aa2f744744f6c53","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-02 08:29:56.000000000","message":"Patch Set 2:\n\nso is this patch still needed or can be abandoned?","accounts_in_message":[],"_revision_number":2},{"id":"51bf0069bc34a0bb5b39c5cfa2c773486b5dbd14","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2019-08-02 08:40:29.000000000","message":"Abandoned\n\nChange 673479 fixed this.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7f0d6ff7cd3e7f6ec96b8fbebedfbed5e7aa0338","revisions":{"fd4fa2f5b7ba3c9c73f7112cc53dff279468c299":{"kind":"REWORK","_number":1,"created":"2019-07-20 08:49:44.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/88/671888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/671888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/671888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/671888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/671888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/671888/1"}}},"commit":{"parents":[{"commit":"8b17129bd233c84feb1e8967e1d11dd2fac92ea6","subject":"Merge \"base: enable elasticsearch and rabbitmq repos for centos/aarch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b17129bd233c84feb1e8967e1d11dd2fac92ea6"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-07-20 08:45:45.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-07-20 08:49:42.000000000","tz":180},"subject":"Fixes Fluentd on Ubuntu binary deploys.","message":"Fixes Fluentd on Ubuntu binary deploys.\n\nFluentd needs /var/run/fluentd which is missing.\n\nChange-Id: I2970b6c0d0684a00a59ff3b2d9eebb3d04495345\nCloses-Bug: #1837274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4fa2f5b7ba3c9c73f7112cc53dff279468c299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4fa2f5b7ba3c9c73f7112cc53dff279468c299"}]},"branch":"refs/heads/master"},"7f0d6ff7cd3e7f6ec96b8fbebedfbed5e7aa0338":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-21 08:09:11.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/88/671888/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/671888/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/671888/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/671888/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/671888/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/671888/2"}}},"commit":{"parents":[{"commit":"8b17129bd233c84feb1e8967e1d11dd2fac92ea6","subject":"Merge \"base: enable elasticsearch and rabbitmq repos for centos/aarch64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b17129bd233c84feb1e8967e1d11dd2fac92ea6"}]}],"author":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-07-20 08:45:45.000000000","tz":180},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2019-07-21 08:07:22.000000000","tz":180},"subject":"Fixes Fluentd on Ubuntu deploys.","message":"Fixes Fluentd on Ubuntu deploys.\n\nFluentd needs /var/run/fluentd which is missing.\n\nChange-Id: I2970b6c0d0684a00a59ff3b2d9eebb3d04495345\nCloses-Bug: #1837274\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f0d6ff7cd3e7f6ec96b8fbebedfbed5e7aa0338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f0d6ff7cd3e7f6ec96b8fbebedfbed5e7aa0338"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
