)]}'
{"id":"openstack%2Fkolla~672023","triplet_id":"openstack%2Fkolla~master~If57c34bd5c3aee3e5b1f1c7b3703fd544ce498da","project":"openstack/kolla","branch":"master","topic":"networking-mlnx","hashtags":[],"change_id":"If57c34bd5c3aee3e5b1f1c7b3703fd544ce498da","subject":"Added Neutron Mellanox agent image","status":"MERGED","created":"2019-07-22 12:02:56.000000000","updated":"2019-12-12 07:27:31.000000000","submitted":"2019-12-02 21:17:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672023-1575321459686-8403a22a","meta_rev_id":"3fc4488663a63301d8f907a47c9086687126799f","_number":672023,"virtual_id_number":672023,"owner":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-02 11:14:30.000000000","_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"date":"2019-12-02 18:10:46.000000000","_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 21:17:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-02 18:41:05.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":2,"date":"2019-12-02 07:31:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":2,"date":"2019-12-02 07:58:05.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":1,"date":"2019-12-02 07:58:05.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-12-12 07:27:31.000000000","post_submit":true,"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"value":0,"date":"2019-12-02 07:31:01.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 12:15:34.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-07-25 07:09:04.000000000","updated_by":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"reviewer":{"_account_id":28714,"name":"Adrian Chiris","email":"adrianc@nvidia.com","username":"adrianc"},"state":"REVIEWER"},{"updated":"2019-10-10 13:28:36.000000000","updated_by":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"reviewer":{"_account_id":18575,"name":"Saravanan KR","email":"krsacme@gmail.com","username":"saravanankr"},"state":"REVIEWER"},{"updated":"2019-11-29 11:30:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-02 07:31:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-02 07:58:05.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-02 18:10:46.000000000","updated_by":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"REVIEWER"},{"updated":"2019-12-02 18:41:05.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-02 21:17:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-12 07:27:31.000000000","updated_by":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"reviewer":{"_account_id":25241,"name":"waleed mousa","email":"waleedm@nvidia.com","username":"waleedm"},"state":"REVIEWER"}],"messages":[{"id":"b95dc174e028c82068df793af53d58d7da252269","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-22 12:02:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"284019a8e18b15d91b59df6a6b1b22f7fd80a243","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-07-22 12:15:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d07a7454fc6c536a68f6b50bcb28503c678508a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 12:54:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/23/672023/1/openstack-check/tripleo-build-containers-fedora-28-master/097c060/ : FAILURE in 6m 14s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/23/672023/1/openstack-check/tripleo-build-containers-fedora-28-stein/99de588/ : SUCCESS in 47m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e2375f4ed9e99e26051abb5447dfecbec2966c2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 13:42:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/23/672023/1/check/kolla-build-centos-binary/9b8a98c/ : SUCCESS in 51m 49s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/23/672023/1/check/kolla-build-centos-source/36ad957/ : SUCCESS in 56m 30s\n- kolla-ansible-centos-source http://logs.openstack.org/23/672023/1/check/kolla-ansible-centos-source/ce5671d/ : SUCCESS in 1h 08m 33s\n- kolla-ansible-centos-binary http://logs.openstack.org/23/672023/1/check/kolla-ansible-centos-binary/257caac/ : SUCCESS in 1h 12m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/23/672023/1/check/kolla-ansible-centos-source-upgrade/daa571a/ : SUCCESS in 1h 33m 37s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/23/672023/1/check/tripleo-build-containers-centos-7/0477892/ : SUCCESS in 57m 48s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/672023/1/check/openstack-tox-lower-constraints/c6463d0/ : SUCCESS in 4m 17s\n- openstack-tox-pep8 http://logs.openstack.org/23/672023/1/check/openstack-tox-pep8/eb819bd/ : SUCCESS in 5m 27s\n- openstack-tox-py27 http://logs.openstack.org/23/672023/1/check/openstack-tox-py27/8f1349f/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/23/672023/1/check/openstack-tox-py36/7210dbe/ : SUCCESS in 4m 24s\n- openstack-tox-py37 http://logs.openstack.org/23/672023/1/check/openstack-tox-py37/1d6285b/ : SUCCESS in 3m 47s\n- openstack-tox-docs http://logs.openstack.org/23/672023/1/check/openstack-tox-docs/067e9e0/html/ : SUCCESS in 2m 53s\n- build-openstack-releasenotes http://logs.openstack.org/23/672023/1/check/build-openstack-releasenotes/b967098/html/ : SUCCESS in 5m 18s\n- kolla-build-ubuntu-binary http://logs.openstack.org/23/672023/1/check/kolla-build-ubuntu-binary/94d2d15/ : FAILURE in 25m 47s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/23/672023/1/check/kolla-build-ubuntu-source/771361f/ : SUCCESS in 31m 15s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/23/672023/1/check/kolla-ansible-ubuntu-source/8f9db45/ : SUCCESS in 55m 09s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/23/672023/1/check/kolla-ansible-ubuntu-binary/decf3ef/ : FAILURE in 18m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/23/672023/1/check/kolla-ansible-ubuntu-source-upgrade/a1eac61/ : SUCCESS in 1h 22m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ffaf4d217ca7b6bb6653bf754dc125e9a48de0be","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-23 10:57:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4547ea687e199a21d1aa70ea4b180b7cfde798ea","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-23 10:58:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"008394d1460968240339ecc56a993f81da26b8e7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 11:47:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/23/672023/2/openstack-check/tripleo-build-containers-fedora-28-master/2a46099/ : FAILURE in 5m 15s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/23/672023/2/openstack-check/tripleo-build-containers-fedora-28-stein/95b736d/ : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8374e0b9f4a206af78bfc7725276aa63d5ac4ef1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 12:24:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/23/672023/2/check/kolla-build-centos-binary/8db3994/ : FAILURE in 6m 30s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/23/672023/2/check/kolla-build-centos-source/ceb0c2e/ : FAILURE in 7m 34s\n- kolla-ansible-centos-source http://logs.openstack.org/23/672023/2/check/kolla-ansible-centos-source/c70afe3/ : SUCCESS in 1h 09m 10s\n- kolla-ansible-centos-binary http://logs.openstack.org/23/672023/2/check/kolla-ansible-centos-binary/72b9a16/ : FAILURE in 9m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/23/672023/2/check/kolla-ansible-centos-source-upgrade/592efac/ : SUCCESS in 1h 20m 01s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/23/672023/2/check/tripleo-build-containers-centos-7/5fa6492/ : SUCCESS in 47m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/672023/2/check/openstack-tox-lower-constraints/aaac5da/ : SUCCESS in 4m 49s\n- openstack-tox-pep8 http://logs.openstack.org/23/672023/2/check/openstack-tox-pep8/5e1e476/ : SUCCESS in 4m 50s\n- openstack-tox-py27 http://logs.openstack.org/23/672023/2/check/openstack-tox-py27/b024d30/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/23/672023/2/check/openstack-tox-py36/163d38c/ : SUCCESS in 3m 39s\n- openstack-tox-py37 http://logs.openstack.org/23/672023/2/check/openstack-tox-py37/794a6fa/ : SUCCESS in 5m 31s\n- openstack-tox-docs http://logs.openstack.org/23/672023/2/check/openstack-tox-docs/d0b2fe3/html/ : SUCCESS in 4m 49s\n- build-openstack-releasenotes http://logs.openstack.org/23/672023/2/check/build-openstack-releasenotes/184812f/html/ : SUCCESS in 5m 05s\n- kolla-build-ubuntu-binary http://logs.openstack.org/23/672023/2/check/kolla-build-ubuntu-binary/b0d81a3/ : FAILURE in 4m 55s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/23/672023/2/check/kolla-build-ubuntu-source/cca1c34/ : FAILURE in 33m 32s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/23/672023/2/check/kolla-ansible-ubuntu-source/9546f58/ : SUCCESS in 49m 21s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/23/672023/2/check/kolla-ansible-ubuntu-binary/4e9ec1d/ : FAILURE in 19m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/23/672023/2/check/kolla-ansible-ubuntu-source-upgrade/ec5b835/ : SUCCESS in 57m 56s","accounts_in_message":[],"_revision_number":2},{"id":"298a5f0610de86d2be031745f9c2cf2b93b5e4af","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-23 13:10:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0d602f61d30c2765531a3985dacd720b08ef9fb5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-23 14:01:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/23/672023/3/openstack-check/tripleo-build-containers-fedora-28-master/767cf54/ : FAILURE in 5m 46s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/23/672023/3/openstack-check/tripleo-build-containers-fedora-28-stein/91f9fac/ : SUCCESS in 47m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9f1ad6ded4207e3bbc62678d4006fe36b01d6c06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 14:40:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/23/672023/3/check/kolla-build-centos-binary/de7a6d4/ : FAILURE in 5m 48s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/23/672023/3/check/kolla-build-centos-source/ef98594/ : SUCCESS in 42m 51s\n- kolla-ansible-centos-source http://logs.openstack.org/23/672023/3/check/kolla-ansible-centos-source/fba19d4/ : SUCCESS in 1h 07m 38s\n- kolla-ansible-centos-binary http://logs.openstack.org/23/672023/3/check/kolla-ansible-centos-binary/505d8f2/ : SUCCESS in 1h 13m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/23/672023/3/check/kolla-ansible-centos-source-upgrade/a7620e2/ : SUCCESS in 1h 24m 26s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/23/672023/3/check/tripleo-build-containers-centos-7/9e996da/ : SUCCESS in 53m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/672023/3/check/openstack-tox-lower-constraints/2e93ac9/ : SUCCESS in 4m 41s\n- openstack-tox-pep8 http://logs.openstack.org/23/672023/3/check/openstack-tox-pep8/f287de7/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/23/672023/3/check/openstack-tox-py27/4c36c18/ : SUCCESS in 4m 12s\n- openstack-tox-py36 http://logs.openstack.org/23/672023/3/check/openstack-tox-py36/00c124e/ : SUCCESS in 4m 31s\n- openstack-tox-py37 http://logs.openstack.org/23/672023/3/check/openstack-tox-py37/d3b5b46/ : SUCCESS in 3m 57s\n- openstack-tox-docs http://logs.openstack.org/23/672023/3/check/openstack-tox-docs/9599cfa/html/ : SUCCESS in 4m 58s\n- build-openstack-releasenotes http://logs.openstack.org/23/672023/3/check/build-openstack-releasenotes/557f61a/html/ : SUCCESS in 5m 39s\n- kolla-build-ubuntu-binary http://logs.openstack.org/23/672023/3/check/kolla-build-ubuntu-binary/0c6e27c/ : FAILURE in 26m 23s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/23/672023/3/check/kolla-build-ubuntu-source/b9afabe/ : SUCCESS in 31m 55s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/23/672023/3/check/kolla-ansible-ubuntu-source/3c411a7/ : SUCCESS in 49m 35s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/23/672023/3/check/kolla-ansible-ubuntu-binary/fb1a617/ : FAILURE in 19m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/23/672023/3/check/kolla-ansible-ubuntu-source-upgrade/1c796fb/ : SUCCESS in 1h 01m 21s","accounts_in_message":[],"_revision_number":3},{"id":"8acb5a4ee91e52cb2e28e0db2c8c17e2e2448a2e","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-24 07:12:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"115eaf57f75fe4d79b562bd49a20e92345a948ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 07:57:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/23/672023/4/check/kolla-build-centos-binary/485671f/ : FAILURE in 9m 25s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/23/672023/4/check/kolla-build-centos-source/ddac28f/ : FAILURE in 4m 37s\n- kolla-ansible-centos-source http://logs.openstack.org/23/672023/4/check/kolla-ansible-centos-source/63aa79d/ : FAILURE in 12m 27s\n- kolla-ansible-centos-binary http://logs.openstack.org/23/672023/4/check/kolla-ansible-centos-binary/588545a/ : FAILURE in 15m 32s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/23/672023/4/check/kolla-ansible-centos-source-upgrade/fca69a5/ : FAILURE in 8m 02s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/23/672023/4/check/tripleo-build-containers-centos-7/3217ac5/ : SUCCESS in 42m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/672023/4/check/openstack-tox-lower-constraints/9313423/ : SUCCESS in 4m 38s\n- openstack-tox-pep8 http://logs.openstack.org/23/672023/4/check/openstack-tox-pep8/6361c5c/ : SUCCESS in 8m 59s\n- openstack-tox-py27 http://logs.openstack.org/23/672023/4/check/openstack-tox-py27/afdb134/ : SUCCESS in 4m 05s\n- openstack-tox-py36 http://logs.openstack.org/23/672023/4/check/openstack-tox-py36/fd72ca3/ : SUCCESS in 9m 39s\n- openstack-tox-py37 http://logs.openstack.org/23/672023/4/check/openstack-tox-py37/3d9e630/ : SUCCESS in 3m 43s\n- openstack-tox-docs http://logs.openstack.org/23/672023/4/check/openstack-tox-docs/bc61377/html/ : SUCCESS in 3m 10s\n- build-openstack-releasenotes http://logs.openstack.org/23/672023/4/check/build-openstack-releasenotes/cd62be8/html/ : SUCCESS in 3m 53s\n- kolla-build-ubuntu-binary http://logs.openstack.org/23/672023/4/check/kolla-build-ubuntu-binary/fb6681a/ : FAILURE in 34m 19s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/23/672023/4/check/kolla-build-ubuntu-source/32763dc/ : FAILURE in 21m 08s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/23/672023/4/check/kolla-ansible-ubuntu-source/711c495/ : FAILURE in 14m 52s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/23/672023/4/check/kolla-ansible-ubuntu-binary/9ef329e/ : FAILURE in 20m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/23/672023/4/check/kolla-ansible-ubuntu-source-upgrade/c2761a5/ : FAILURE in 14m 52s","accounts_in_message":[],"_revision_number":4},{"id":"db49355a4042aaea989c590680795fb132845d1f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-24 08:09:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/23/672023/4/openstack-check/tripleo-build-containers-fedora-28-master/8a48fc9/ : FAILURE in 5m 55s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/23/672023/4/openstack-check/tripleo-build-containers-fedora-28-stein/85b876c/ : SUCCESS in 49m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"05db27b24024d8776695b71a6293a79537f910a3","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-24 12:35:23.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fb4df8bc37e6f33b8743521454c4d86b99a45a04","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-24 13:28:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/23/672023/4/openstack-check/tripleo-build-containers-fedora-28-master/29a4646/ : FAILURE in 5m 51s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/23/672023/4/openstack-check/tripleo-build-containers-fedora-28-stein/430fd6c/ : SUCCESS in 51m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"413bcf83b87f667b6fea781b8deb8ba57117db5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 14:24:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/23/672023/4/check/kolla-build-centos-binary/f0b6933/ : SUCCESS in 52m 47s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/23/672023/4/check/kolla-build-centos-source/3e08220/ : SUCCESS in 54m 38s\n- kolla-ansible-centos-source http://logs.openstack.org/23/672023/4/check/kolla-ansible-centos-source/f428653/ : SUCCESS in 1h 44m 56s\n- kolla-ansible-centos-binary http://logs.openstack.org/23/672023/4/check/kolla-ansible-centos-binary/6f0eef0/ : SUCCESS in 1h 20m 51s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/23/672023/4/check/kolla-ansible-centos-source-upgrade/a71742f/ : SUCCESS in 1h 23m 48s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/23/672023/4/check/tripleo-build-containers-centos-7/145bbf9/ : SUCCESS in 49m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/23/672023/4/check/openstack-tox-lower-constraints/8a7eb21/ : SUCCESS in 4m 22s\n- openstack-tox-pep8 http://logs.openstack.org/23/672023/4/check/openstack-tox-pep8/5048433/ : SUCCESS in 6m 13s\n- openstack-tox-py27 http://logs.openstack.org/23/672023/4/check/openstack-tox-py27/89af15f/ : SUCCESS in 3m 51s\n- openstack-tox-py36 http://logs.openstack.org/23/672023/4/check/openstack-tox-py36/8aad238/ : SUCCESS in 4m 02s\n- openstack-tox-py37 http://logs.openstack.org/23/672023/4/check/openstack-tox-py37/5f3c16f/ : SUCCESS in 4m 21s\n- openstack-tox-docs http://logs.openstack.org/23/672023/4/check/openstack-tox-docs/4b5265e/html/ : SUCCESS in 3m 26s\n- build-openstack-releasenotes http://logs.openstack.org/23/672023/4/check/build-openstack-releasenotes/3f25362/html/ : SUCCESS in 5m 07s\n- kolla-build-ubuntu-binary http://logs.openstack.org/23/672023/4/check/kolla-build-ubuntu-binary/4a4f2f7/ : FAILURE in 24m 36s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/23/672023/4/check/kolla-build-ubuntu-source/59d954e/ : SUCCESS in 38m 02s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/23/672023/4/check/kolla-ansible-ubuntu-source/f04e56c/ : SUCCESS in 49m 29s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/23/672023/4/check/kolla-ansible-ubuntu-binary/1f9aa7d/ : FAILURE in 16m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/23/672023/4/check/kolla-ansible-ubuntu-source-upgrade/ad27df6/ : SUCCESS in 54m 22s","accounts_in_message":[],"_revision_number":4},{"id":"29860c6fd921640b5a91c4abaf0fda9248bf1e21","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-30 09:03:11.000000000","message":"Patch Set 4: Code-Review-1\n\nubuntu binary does not build ;-(","accounts_in_message":[],"_revision_number":4},{"id":"f70e719be4295e7b029c91d29446fd2295d536af","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-07-30 10:02:01.000000000","message":"Patch Set 4: Workflow-1\n\n\u003e ubuntu binary does not build ;-(\n\nWe are working on adding python-networking-mlnx package to Ubuntu repos, I\u0027ll WIP the change until the package is uploaded.","accounts_in_message":[],"_revision_number":4},{"id":"3994eba330c33cba0e9faed8b1a03b56fd31bd17","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2019-09-06 22:33:24.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a39e2e22cb160b0d7f9acc0814b00c255e4043c1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-06 22:59:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/4/openstack-check/tripleo-build-containers-rhel-8-master/5e5a7ba/ : RETRY_LIMIT in 7m 34s","accounts_in_message":[],"_revision_number":4},{"id":"2358c75f86827bd61deede81f50328cb19f3eb19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 23:59:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3aa7db1e51624c6a8e76e8eadc93ce4c : SUCCESS in 43m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/733ac747f7bb44df97e78ea4633f646e : SUCCESS in 49m 35s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8f09e813efb24a6a9ca50afbbc9362b6 : SUCCESS in 1h 14m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c4d91024d1ed4a61b7b0fc8bbf6d3f0e : SUCCESS in 1h 11m 41s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/43b518b959bd4577ba6577fc4587b09c : SUCCESS in 1h 18m 14s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/251805bdc95149d38401977924001b4a : SUCCESS in 50m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ddd17e449ffb4d3292f21d46f21570d7 : SUCCESS in 24m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63591b6e88674d6bbc1e6596200d2a43 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ad6027baec1422abb0f3c8f3cd59a58 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bca5255900994e96b4f04de870cba77f : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd8dbfffca274e1ea30b59ce44a3b99a : SUCCESS in 2m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66f01fc95ccf4ff58aa77e8a3103af8e : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bde0813e053043de9357ad8d878fb6a5 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7637bdcd61754216bc58acbf36f049b3 : SUCCESS in 4m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fac27a843167412aae6d0821d5e7cebc : FAILURE in 25m 29s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/72e59a6087584c10b249f3cdac01fb04 : SUCCESS in 35m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7532698c00364a77b2d42a2cde28013a : SUCCESS in 57m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2a277b7f4f784e1fa1568ef1249462c5 : FAILURE in 21m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/61e39422491f49f2b720a96caf622a8a : SUCCESS in 1h 05m 51s","accounts_in_message":[],"_revision_number":4},{"id":"b5797c493115628564b3d469f043a2615cd79efc","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-11-25 13:45:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5aa36ecea7b2bc909bb9b0611b46e67721e54b6b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-25 14:17:02.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/5/openstack-check/tripleo-build-containers-rhel-8-master/15c5063/ : FAILURE in 30m 23s","accounts_in_message":[],"_revision_number":5},{"id":"ef1394c7521a93ff6e732e41d1bdd4c563b651f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 15:34:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cfa264dc30324dab8262f72feac3e760 : FAILURE in 6m 39s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/28b9d28d0e734e71bb43d1b1a91d91e3 : FAILURE in 6m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/db7f2fabaf024a1691dbc4735b032567 : SUCCESS in 1h 29m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4d067c35d53745a8997a2c5f6c7cea97 : SUCCESS in 1h 43m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bc89ce77f10f4ed9a127025f000405f5 : SUCCESS in 1h 27m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/11b34d32ede6480da264ce377031373d : SUCCESS in 1h 18m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6c4125c0333e405d9451c0d352fd9caa : FAILURE in 4m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a457e6b469ec443b966dd692bd9604ef : FAILURE in 4m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d2dc7afd6eb24fcea1c9d8c41c4eca02 : FAILURE in 7m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a3dc05d1afd473991034efc48ff0b1a : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d2341edabb4bcdba3d7d40546095e8 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c44f86f51c364825b3a43f8f34d842d9 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9481d877ae2a41f4ab964760ec2ca347 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/615930b1ba65488a9b6e5f8921691f2c : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecbdcfda862c44d797dd475da8c8f740 : SUCCESS in 4m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/398d1925e9fa469ebec8ac1c93d7e3d9 : FAILURE in 2m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ada6a4c4a2034ef6bd130735513f346c : FAILURE in 5m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cea4831772cb45c39da70bdd83979756 : FAILURE in 9m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/27d7090d6e3448acb4f75ac43dbcf662 : FAILURE in 7m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ad616c2cc2d649c0be043c5c1881610a : FAILURE in 9m 38s","accounts_in_message":[],"_revision_number":5},{"id":"0420f38f7c64c23e3effaa557b657b7eec728cd3","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-11-26 11:44:17.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"b9cbfab7a9379c7d7d2847bca50853c5483e505b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-26 12:15:27.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/6/openstack-check/tripleo-build-containers-rhel-8-master/82e6621/ : FAILURE in 30m 08s","accounts_in_message":[],"_revision_number":6},{"id":"1c933347171cdb5b2ab03bced99ade9bae268e25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 13:25:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/14177bf2388c4a0690236b68b48510b5 : FAILURE in 2m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/add0c4d71db94aa5a4de8ab0f1b6be63 : SUCCESS in 1h 00m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/14168ac1fa2b4bf8a8967b34fadc5d80 : SUCCESS in 1h 30m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4dc64a0ca8e34503841a1588dd878cc3 : SUCCESS in 1h 21m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/08a99c6abccb4d01adf551461440670a : SUCCESS in 1h 35m 01s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bab3e6e2d231411d80a1e2f74c467b2e : SUCCESS in 50m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f9f4ce52131b4dbdb4ab8ab96a787d6c : SUCCESS in 28m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/392a5242776e4e4faa9f4ac5622accbe : FAILURE in 3m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/aebe7a340429446bbfbfbc06e64136c4 : SUCCESS in 58m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71bd78686fde49a7ad8d87443a3a7ef6 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff37a1af1f8f4ef198f8a738c7062860 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b10d00845c34fd3882802ca7d129054 : SUCCESS in 3m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7fca5eae9eb7439ebfacc361f1c0e96e : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d6ffcbd9c64037a0a312a5eeeaffc4 : FAILURE in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7adadfca30f24f7c98e08d93ca075ee2 : SUCCESS in 4m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e27143fc2ca1489c82e72d59a3468f4e : FAILURE in 3m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cef82223406946f08882fda0ae678d93 : FAILURE in 3m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64681b7d7c8149db94a1dfcb214f1474 : SUCCESS in 1h 09m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a507ad9f0a9942d6a37d51b9127d21c0 : FAILURE in 12m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/169ff0ea284f4a1cb716b775943ccfbf : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":6},{"id":"52482c3e6528b3ff81a8b29a4426971412fcced1","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-11-27 07:16:54.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6e56c71280ec249cc09a2fd2ab88933a68720496","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-27 07:50:14.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/6/openstack-check/tripleo-build-containers-rhel-8-master/bced461/ : FAILURE in 31m 45s","accounts_in_message":[],"_revision_number":6},{"id":"f07cb18db5a5b513c8657a739bd598a8f0670fa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 08:58:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/56b3a32d449c4f39b8177ff59c447878 : SUCCESS in 38m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/706093da69334d46819dd9fc96b62940 : SUCCESS in 54m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/73c934c1d6cc45a995dad024a7ffacad : SUCCESS in 1h 32m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2e3e16547de9415a83997aca1dc5c6ca : SUCCESS in 1h 28m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b6b422df9d0c48639eadf9604145e222 : SUCCESS in 1h 22m 01s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/236ed52749254316afd1292b88f1bfc0 : SUCCESS in 1h 07m 19s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/109a56c2b76c4ac29be0edf1bc539872 : SUCCESS in 36m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d87bdd64806a47feb871302f36b4b3a6 : FAILURE in 3m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f50f3440b48a4de287659fc504f70cc0 : SUCCESS in 59m 43s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b740df29ae2e42ff9cc23dd6fd7c315a : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da36e9598e4d414289b33d15463a4356 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aaac5d2191254f34914f221e11b14b5f : SUCCESS in 3m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da766857bf7243d5a79abf4a1adb3df2 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fdb0165c7a948c69212fc8f0ca40da8 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7ecf83ad89784fc391aef53615202eaa : SUCCESS in 6m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf08f1802eb948bdb3c174ab6bf44bd6 : FAILURE in 3m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c54240d8a5f2412584b84576e0414935 : SUCCESS in 26m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e4179e6c19543018bc43552d4b647bb : SUCCESS in 53m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2256b732463941b0a4cb580d820193c2 : FAILURE in 7m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5cc84cf337674cd89060df8448a86900 : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":6},{"id":"7da33465b23a483f850eceb5e16d0150322dde66","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-11-27 15:01:15.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nneeds entry in the support matrix","accounts_in_message":[],"_revision_number":6},{"id":"0906d03caeccc76e34a8048159990017e896052c","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-11-28 08:45:19.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8482b47aa38f6fdbf6cccc77a35776102f7c72a8","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-11-28 08:45:56.000000000","message":"Patch Set 6:\n\n(1 comment)\n\n\u003e (1 comment)\n \u003e \n \u003e needs entry in the support matrix\n\nDone","accounts_in_message":[],"_revision_number":6},{"id":"f76ddfe536bf33ccabb37456da01e910159f9371","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-11-28 08:47:28.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2d66a0dd4ffa8c40994cc1370d68cae987db32a6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 09:20:13.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/8/openstack-check/tripleo-build-containers-rhel-8-master/a5216cf/ : FAILURE in 31m 33s","accounts_in_message":[],"_revision_number":8},{"id":"e24c21fd0fdf7bbf4fba68816a51c84b0d452288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 09:58:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e3c68f51a8744e7083ab555e91962256 : FAILURE in 30m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9dbef47df49048729b44160f1a58058b : FAILURE in 34m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0f6a72c3a05949349bd541dde94142e6 : FAILURE in 30m 49s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/497ee081de6240b0b3906f1456448ffd : FAILURE in 26m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f2fe28226d544c64b4afe9cfbe827b56 : FAILURE in 26m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/75235112eb4f43778299adfc0c2a6de3 : SUCCESS in 41m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3284b5341dea4a239acf571f2c0fed39 : SUCCESS in 37m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2821dee0569a41f5a4d098ce463b028f : FAILURE in 4m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dfeb45199fd54d97bedc4ad5e61ae68b : SUCCESS in 59m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7110e24a4f7447d798b0f16c858b6e11 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c30498cfc2e4b048696894b418acade : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbad992479df4ebbbddc247c2045c523 : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f8e0a099c094aa595b0d71584055e63 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7cea45611f44e48b98e843e6d3d0aaf : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52209a1a73dd44c1b8fa13466fd1f673 : SUCCESS in 4m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9ce1ec20f6a041fa8c82c168d12433ce : SUCCESS in 42m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aeeebda3c02649ec908c07d0d13cd279 : SUCCESS in 42m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e72b392a023a4f74b625bc1fd90f06c8 : SUCCESS in 56m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e915c262c134e869dd5fafbe72e910f : FAILURE in 9m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0f8fbd5ca3e44e838d3c28df9aae180a : SUCCESS in 1h 10m 25s","accounts_in_message":[],"_revision_number":8},{"id":"654283e910b74e828f9fcca88a14f54f41e6df30","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-11-28 10:03:01.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"4862a525679b34fe5f426e79a3ae5a736cc2ab17","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 10:36:49.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/8/openstack-check/tripleo-build-containers-rhel-8-master/9a9749f/ : FAILURE in 32m 32s","accounts_in_message":[],"_revision_number":8},{"id":"2ebdd9ce3b4f511a9a33732df716534b554aacb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 11:32:03.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/384cea3318e3432b9b30a8e1ac2c52c2 : SUCCESS in 48m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f91e7b7e10d7403e9eeda1ec545b9bf2 : FAILURE in 34m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8265cc886d4446afbaed4bd17ccfa5c3 : FAILURE in 39m 29s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5deffe0f7ba24930915aa9c527060951 : SUCCESS in 1h 24m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c9e343b5d8b7489ab58793a365889710 : FAILURE in 33m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/570a60d082f046a3ab43b88204f8a16f : SUCCESS in 1h 06m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f706c5f8c23f421a88dd7721f855617e : SUCCESS in 29m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ee1b5cf3c80b4cd7bbb66db3c5068397 : SUCCESS in 22m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e7e414d9b64b41c280c8426d528c9320 : SUCCESS in 1h 09m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/605e005805d843f98ff25d1adb94b882 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f3fe3b0c4bc46968e08690c8210cd4e : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69b7924a622f4955a954da50e325b5d6 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e14f7b3654b474aba11671b2eb53d6f : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89e48743f329485197a1396c1266ef26 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1909903bb9ff46b8b0c02cf8748332cc : SUCCESS in 4m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a5a592ef96884b2ab5f3252fa55e1a8b : SUCCESS in 22m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6f9c7f605fe4e288d50f201565dbf79 : SUCCESS in 27m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/18d73596bbe249ae9698cb611c674f18 : SUCCESS in 53m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/60d92ab836ac48e8a39a7695e7c2aabe : SUCCESS in 51m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/00f94c9ae0794d429cc8f6333fe04444 : SUCCESS in 56m 57s","accounts_in_message":[],"_revision_number":8},{"id":"1522e76b2e69674e30594fa8e5d3798ce001b082","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-11-28 11:32:31.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"9609169c720221c5a2131aa6b101b9e8b5b6c061","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 12:05:23.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/8/openstack-check/tripleo-build-containers-rhel-8-master/e7366d1/ : FAILURE in 31m 57s","accounts_in_message":[],"_revision_number":8},{"id":"132178531add5ad2ad906cf712fbf4677f702f64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 12:48:31.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/031324d7548f4e4891dba3b4c951fb77 : SUCCESS in 54m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/cfb4a0838f924189ba23ddb8be54e27d : FAILURE in 56m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bb265f624a354413a9bb51db33c422f4 : FAILURE in 46m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/515787deb8044081acdd5715620336ac : SUCCESS in 1h 12m 21s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a41f04d6492245f8b342bb7c5b746583 : FAILURE in 34m 13s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/41711cee08ef457c9fcb8f6da3850412 : SUCCESS in 58m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1a51fc1f677445f498bbafd57dd0263c : SUCCESS in 28m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/eb69fb0b0aa5401a97ff12888ec66aef : SUCCESS in 30m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fa1b9f0c3d984be985c80af7c1036b01 : SUCCESS in 1h 06m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/add12032e015403bbf1304290d37c677 : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63bcfe7c4bb4490da2b2fe63a37acdb5 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a61470ae5d14770b1b5c6ec5e31950a : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/733b3094fcd147a4a1603a4c3c39cb75 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/965254b57f174d3d99e7cb56ca99e22a : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae3f6ee88cdc4e0bb5639f487bb5d13f : SUCCESS in 5m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/02dbfa9da62147cf81fae29707f89d62 : SUCCESS in 24m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c1a5b21417a24c1f82412e71a4610519 : SUCCESS in 26m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/06679ea2d1384123b740bc122e200ba6 : SUCCESS in 50m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2e202e3c3d45427b83318979cb143203 : SUCCESS in 1h 01m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1248e9e959d14213a1a42fa5a39d5e57 : SUCCESS in 1h 06m 49s","accounts_in_message":[],"_revision_number":8},{"id":"2a9c3eeed5d1585bdab270ebe820d4ccac8ec106","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-11-28 13:24:57.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"214d965cdd7e0350e1315642874bbffe4b0f698d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-28 13:55:14.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/8/openstack-check/tripleo-build-containers-rhel-8-master/c4cbb33/ : FAILURE in 29m 21s","accounts_in_message":[],"_revision_number":8},{"id":"d3f03dbbb69b6d4ab4b44dd66832e1c35944edd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 15:02:31.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b7c91fc7e84f4841910899e5e1a20b08 : SUCCESS in 31m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c54d7b7ddfba4dbfa39fb4a647b1a4a8 : FAILURE in 52m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3c3b14ff908d4df7a19b65e23a73ba53 : FAILURE in 37m 09s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d63c073a07884547a88453432dbe574b : SUCCESS in 1h 27m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/53c77cead6ce482d8a6acdf098a525f0 : FAILURE in 30m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6cf7fbc2f88245aaba9401f7da93d729 : SUCCESS in 59m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8881e232aa1f44a58a2caef5b0f57c47 : SUCCESS in 41m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/410241996795476b88d4d8e521bb6b85 : SUCCESS in 38m 56s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7c7990676d134c22be8dec425b7f7ea5 : SUCCESS in 1h 08m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18b7b68e95fa41cb807193356688b9eb : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee5ecd54b7d540f6b50d3977b043f22d : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a592d8b49264530819f7f199174a00c : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e7cc78fef3b4cd1b84807b85ac7a043 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50c750f854e4891844d6a5066f186ce : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b34330e3a7f74b6d95d0d8238b6d8b11 : SUCCESS in 4m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2661ad0a875d4b79929bd20a89a46f53 : SUCCESS in 32m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b7b94349cf3489db15ac7055d6b94cf : SUCCESS in 26m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e1afacacad7f4c048e77f4957561990b : SUCCESS in 1h 00m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5f3ca46e0ebd4af9984aaa1c25fa7cb1 : SUCCESS in 55m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/31efd76a13b641b39554db176076ba19 : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":8},{"id":"d75f1051ed077c0c12d8fde2a2233e5307465f2a","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-12-01 09:32:03.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"2ffa606645d5f74e9f73acd7447d34db48bb7ed8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-01 10:05:11.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/8/openstack-check/tripleo-build-containers-rhel-8-master/40f405c/ : FAILURE in 32m 06s","accounts_in_message":[],"_revision_number":8},{"id":"4e5aa9ccb204024e26ddc7db2012373a5c8cc6dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-01 11:17:25.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/52dfe64bdbcf42fb9b8cbb767622351d : SUCCESS in 37m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/71e47c0f7a2a4f1f91b610e31415346c : SUCCESS in 34m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/58f0a24dbed949db86973f135fcc550c : SUCCESS in 1h 13m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4942524b4a834e0488aca296e21a97c4 : SUCCESS in 1h 14m 42s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/41f17e2fa9b045fc928436f099abc054 : SUCCESS in 1h 22m 22s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e42c537ab0d74aef850110f04c8cb964 : SUCCESS in 43m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f34e28da390d435d9524316521d3f7ee : SUCCESS in 22m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1f2bf935a14346789c8610b40831192d : SUCCESS in 21m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6ae0730e50a34dc887771a048b0bce28 : SUCCESS in 58m 23s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b618b74c59f1401bae5d01ae9e5b161f : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bda99e0a7a6f487fbb6a2f6c47596e0c : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/606ba15298de4c2d86a17417b0ef5b29 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/341269a912334faa9a380b36fec1d0e8 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6e6a821b624a729e921d9d0edb0ff2 : SUCCESS in 11m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f6cbab6a9b64bff84a3cb4eac01e533 : SUCCESS in 5m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ec663bf226e94688bb5d1c6577397554 : SUCCESS in 23m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/98456d00d20c43ad86bd5ec6fd8cbee1 : SUCCESS in 30m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a4f19425e3ae416ca589c404dc9f364e : SUCCESS in 59m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8672916bb2394262a5ccd72e26798b4b : SUCCESS in 54m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fd6fe83069f440f2abfd8c9b2e44b750 : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":8},{"id":"cdabd53ca1715d649d2701d0292a44b38630ae39","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-02 07:10:15.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"02d04334af5d3b533bea469dbe14df411d5b57f8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-02 07:10:53.000000000","message":"Patch Set 8: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4b39453336af6b269fc4c89279a69b7c03b26973","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-12-02 07:31:01.000000000","message":"Uploaded patch set 9: Commit message was updated.","accounts_in_message":[],"_revision_number":9},{"id":"741cb4737f884ede9156e4bd94ad09c7ceac883b","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-12-02 07:31:28.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1f0d83af834b5db5a7eac0d64834386ee46a36c4","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-02 07:31:32.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"490789350fed8cd66af7deff8e9ed07d5cc00a19","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-02 07:58:05.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nVery nice, thank you.","accounts_in_message":[],"_revision_number":9},{"id":"2cc3f21a1a196c054ad1c80a1c14e233104ffb8c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-02 08:04:48.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/9/openstack-check/tripleo-build-containers-rhel-8-master/dc25330/ : FAILURE in 32m 45s","accounts_in_message":[],"_revision_number":9},{"id":"2054303b6d8b414b8ea68f202d7720cf692f9a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:00:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/7fcd91c7ebad45b88fbf97a181bf4731 : SUCCESS in 41m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2b8b7b9776ef4d76a7aec3d82f45b565 : SUCCESS in 42m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/79aeff3476b6489d8ce4c51a1e5db350 : SUCCESS in 1h 21m 45s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ed0bff8bc4504bac8b3cdc146bc3adc2 : SUCCESS in 1h 13m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ed0db70d3c0a498f887c8332c240617a : SUCCESS in 1h 27m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0616304c5fd8446c9c44928076afa0c8 : SUCCESS in 43m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7bc61db5328e4f70bb3e49f6dbeb5d90 : SUCCESS in 22m 50s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/033d22e7583042ac980420d34389e65d : SUCCESS in 18m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9ce308fcc3ca4c93a771913b13aa75b7 : SUCCESS in 58m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc21d9e8cfdd48f68649dafe2bd2ca07 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d875a687eabe45989853589fc6f29410 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cff979a0e064f569315de32a1be55cb : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6692a4bc895145a18cb94f4a36005c9e : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ce2cba729354b9baab163d8991404c3 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc32949cf1cf4762b89be802adeae588 : SUCCESS in 4m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c68757e5b4344a65a31ab43a22207706 : SUCCESS in 23m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cfa964a88bfa4b13a292058858d61a88 : SUCCESS in 26m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/51f8a4c0c4dd4c58907d21d8b906ce00 : SUCCESS in 52m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ae1272e2d4794884bada066fd9c8b5dd : SUCCESS in 54m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bd8643c381694738b97cd047c2fed225 : SUCCESS in 1h 00m 41s","accounts_in_message":[],"_revision_number":9},{"id":"4cae6e63282fa4621bd71ac38a6331eb3329a45d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 09:40:56.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"0797041c3c16b1668943ef78378d077c4610b6d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 11:12:15.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/95f100cabb324fed963c1d344892d7f1 : SUCCESS in 39m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/04be82abedf544e7807e906e369b69ad : SUCCESS in 1h 22m 36s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bf94dbd51619486691e2f68fa575fb9c : SUCCESS in 1h 27m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/be8e5bdaedf94c95a1e95283b40cf6b3 : SUCCESS in 53m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7f3dcc36b834f8c9e287d2590f9525a : SUCCESS in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7c50e881c38491aae2fdb50e0a84a30 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d499e6142c39403499d8fdd82e1b4b4d : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2a954da1775492d9b2e9def210ff439 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bbe070ff86c4f5388e00b5fb14a8ebf : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/396bbac362af450da96b33a25c84674c : SUCCESS in 4m 28s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0789fa0f52884482a9d4351945b44177 : FAILURE in 4m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1103faa297ce4184962a60e14f413aae : SUCCESS in 1h 01m 59s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5fbdfb6ca752484c806aaf8534988cc5 : SUCCESS in 59m 34s","accounts_in_message":[],"_revision_number":9},{"id":"d1b3fdb0ae3ceb78530079e3f592cb6d4c2ed88e","author":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"date":"2019-12-02 11:14:30.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"c5fc64356b711e21d39ecf9e486059dc6c9bb809","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-02 12:00:44.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/9/openstack-check/tripleo-build-containers-rhel-8-master/5735122/ : FAILURE in 32m 26s","accounts_in_message":[],"_revision_number":9},{"id":"039ed641049b2cd0830d2245b0722711c34b6fa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 13:07:22.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3dc5d2af6f1344b0b97ed1eecc6d3933 : SUCCESS in 43m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2d76bc27bf4846959a40f98dea1510d5 : SUCCESS in 50m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d7e6ce7a06444c6aaf4749335e231ad6 : SUCCESS in 1h 28m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/87d916fa633c48c2aaf0477be855c9dc : SUCCESS in 1h 37m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0f6e5729e6064258a4de3055019e0e53 : SUCCESS in 1h 50m 42s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/50c1f56bfe324b6bb68a2ed5f56c9a4c : SUCCESS in 47m 10s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/00407130e1b74587abae93cde66b5e5e : FAILURE in 7m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/64394a3a9a4149a081baa683016f1bc2 : FAILURE in 5m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/014d9a6d0abf49fab8e94223b8dd42e9 : SUCCESS in 1h 16m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ab3df9fc6e44af89c46c8140eeeb314 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c5ad67e5df0461fa13cb30fcda255d3 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b151ceac7a04bdb84d9fc1cdd5eaca3 : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04a0c0a6441a4212beb99d4876e992ca : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36312b4779634efd8d13b71b258504d0 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7950d93ec5fb4497a3b9bb69549ee974 : SUCCESS in 4m 41s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/665f02ffb1154671824a391710abb799 : SUCCESS in 24m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4de2a9d04e0f487bae5c0194440bf116 : FAILURE in 4m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e3268429c61842aeabc3f82d396d9690 : SUCCESS in 1h 18m 33s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2798241ee49a48a88e61cc455a21ba76 : FAILURE in 7m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/197bb4c9745747e28fed56ba56baa7e0 : SUCCESS in 1h 01m 03s","accounts_in_message":[],"_revision_number":9},{"id":"b5eb74c80e39781b60b20648bcf032128dc76e29","author":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"date":"2019-12-02 18:10:46.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"25204cde32d424a8b6fa125fcaec58393b34b82f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-02 18:41:05.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/23/672023/9/openstack-check/tripleo-build-containers-rhel-8-master/87c8d86/ : FAILURE in 29m 01s","accounts_in_message":[],"_revision_number":9},{"id":"3821653ef0a31afcfbcd45cd4ec5b155e3e4bd5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 19:45:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d0030ef4817d46c891ff681d61da21ee : SUCCESS in 31m 17s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/aa5176467c214b34b1a33757b64e5297 : SUCCESS in 38m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/07952af1180f418eb150cb109ba04c6e : SUCCESS in 1h 24m 43s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ea2fd96963c646598ba9f23593a471a7 : SUCCESS in 1h 11m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9ccfd51058d7433cb9ac1c1db069c91d : SUCCESS in 1h 32m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f56936f6131543778803376cdbbdcc09 : SUCCESS in 44m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/354358ea6cb543ee93e09d8f32bbfe95 : SUCCESS in 31m 12s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/452f217b644944bbadf209037787387e : SUCCESS in 16m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8690e8dc362144ceaa3838e7fc0f473b : SUCCESS in 1h 05m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b9869f57b7b4524a64208534749a27a : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a261ef3a5a14ab2ac2daefb11650bb9 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f22d6a9b115b493388bc0698b9b68b51 : SUCCESS in 3m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9304b2778da741c4b82072273365168f : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ec7c31c98446e29ff3f85474c4c886 : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57c8225944fd4f5f9a3ab8df78801360 : SUCCESS in 5m 22s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/56688ee438f749d5a8b3e63393c84bbb : SUCCESS in 27m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8832e20ced2a48ceb481531c8091d1f7 : SUCCESS in 31m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0d79c106faf48bab799197b1bef783a : SUCCESS in 1h 04m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1a4631052f454892be601f6dc4bfff6d : SUCCESS in 1h 01m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0a54380ac9c744feb57fe9060f1ae640 : SUCCESS in 1h 06m 10s","accounts_in_message":[],"_revision_number":9},{"id":"adb954d168a222617220ab8575678a2c44dacf3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 19:46:03.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"d5eef5998536e98d1d815dc196fc47ffb71df4b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 21:17:39.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6a24bf3041974d2483a81418e9150426 : SUCCESS in 45m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9e553d714b2b4f3ab94c913c8d4ed912 : SUCCESS in 1h 23m 24s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/485e28c986364c3b8675f8cfe8cafc41 : SUCCESS in 1h 25m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b3a43b1ad49d44f39221afe184b4499e : SUCCESS in 48m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0d0c19ba8f04c3aac35057a42e33006 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf9b1ee3c9640fe99f6c52ad785caff : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff5d75dbcab643cdb36614c4576f7986 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0bd8489c2254f029c17a7c941f7da6a : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9dab2f0dbe445448b850eeb041556c7 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f299034ab5d4f408a7dba47b6258467 : SUCCESS in 4m 19s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/df1f33a0887b4fb2a2abf2c8fe6ba8e4 : SUCCESS in 30m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa190bacac3941b1b64921a4c2cfdb18 : SUCCESS in 54m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9bd767afa7c349ac84e31122c76857cd : SUCCESS in 1h 04m 22s","accounts_in_message":[],"_revision_number":9},{"id":"5fb20d8d9cc53e384ebda66711bd331b845d3978","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 21:17:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"d270680caaaf2362c3fab90e00cef57f90628a0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 21:18:47.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1fecb815d4c46d29215cae08359a4c2 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/243f3d086d1d4a5baa6eeab303c92224 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"ac12834c6761551c2b491cfe7ffcd5a69dd80c65","revisions":{"f65ca895ee2c7c1ef8dea526447cef2ef4a3d8a1":{"kind":"REWORK","_number":1,"created":"2019-07-22 12:02:56.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/23/672023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/672023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/672023/1"}}},"commit":{"parents":[{"commit":"008aa957f27b26caff3be8b18110e71cdaf70772","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/008aa957f27b26caff3be8b18110e71cdaf70772"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-22 12:02:57.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-22 12:04:40.000000000","tz":180},"subject":"Added Neutron Mellanox agent image","message":"Added Neutron Mellanox agent image\n\nNeutron Mellanox agent image is a Neutron base image plus package:\npython-networking-mlnx.\n\nChange-Id: If57c34bd5c3aee3e5b1f1c7b3703fd544ce498da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f65ca895ee2c7c1ef8dea526447cef2ef4a3d8a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f65ca895ee2c7c1ef8dea526447cef2ef4a3d8a1"}]},"branch":"refs/heads/master"},"cd087b662e0603b58aea354d1b51e1387a6e19f6":{"kind":"REWORK","_number":2,"created":"2019-07-23 10:57:05.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/23/672023/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/672023/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/672023/2"}}},"commit":{"parents":[{"commit":"008aa957f27b26caff3be8b18110e71cdaf70772","subject":"Add Python 3 Train unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/008aa957f27b26caff3be8b18110e71cdaf70772"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-22 12:02:57.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-23 10:57:58.000000000","tz":180},"subject":"Added Neutron Mellanox agent image","message":"Added Neutron Mellanox agent image\n\nNeutron Mellanox agent image is a Neutron base image plus package:\npython-networking-mlnx.\n\nChange-Id: If57c34bd5c3aee3e5b1f1c7b3703fd544ce498da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd087b662e0603b58aea354d1b51e1387a6e19f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd087b662e0603b58aea354d1b51e1387a6e19f6"}]},"branch":"refs/heads/master"},"4ebbf50ba7416fb3568b28e406802bca93afa96d":{"kind":"REWORK","_number":3,"created":"2019-07-23 13:10:55.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/23/672023/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/672023/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/672023/3"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-22 12:02:57.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-23 13:12:31.000000000","tz":180},"subject":"Added Neutron Mellanox agent image","message":"Added Neutron Mellanox agent image\n\nNeutron Mellanox agent image is a Neutron base image plus package:\npython-networking-mlnx.\n\nChange-Id: If57c34bd5c3aee3e5b1f1c7b3703fd544ce498da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ebbf50ba7416fb3568b28e406802bca93afa96d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ebbf50ba7416fb3568b28e406802bca93afa96d"}]},"branch":"refs/heads/master"},"b7bbf17a810e32d62cfbc2c58b7841c1dedbe8d2":{"kind":"REWORK","_number":4,"created":"2019-07-24 07:12:51.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/23/672023/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/672023/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/672023/4"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-22 12:02:57.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-24 07:14:30.000000000","tz":180},"subject":"Added Neutron Mellanox agent image","message":"Added Neutron Mellanox agent image\n\nNeutron Mellanox agent image is a Neutron base image plus package:\npython-networking-mlnx.\n\nChange-Id: If57c34bd5c3aee3e5b1f1c7b3703fd544ce498da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7bbf17a810e32d62cfbc2c58b7841c1dedbe8d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7bbf17a810e32d62cfbc2c58b7841c1dedbe8d2"}]},"branch":"refs/heads/master"},"2269d74e1eb84cc66568bb18c5070e4303b6bbfc":{"kind":"REWORK","_number":5,"created":"2019-11-25 13:45:23.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/23/672023/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/672023/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/672023/5"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-22 12:02:57.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-11-25 13:48:25.000000000","tz":120},"subject":"Added Neutron Mellanox agent image","message":"Added Neutron Mellanox agent image\n\nNeutron Mellanox agent image is a Neutron base image plus package:\npython-networking-mlnx.\n\nUbuntu image base is not suppotred because no networking-mlnx pacakge for Ubuntu.\n\nChange-Id: If57c34bd5c3aee3e5b1f1c7b3703fd544ce498da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2269d74e1eb84cc66568bb18c5070e4303b6bbfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2269d74e1eb84cc66568bb18c5070e4303b6bbfc"}]},"branch":"refs/heads/master"},"ce2775b1c062e311efd0e89760c857667cb236e7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-11-26 11:44:17.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/23/672023/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/672023/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/672023/6"}}},"commit":{"parents":[{"commit":"eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4","subject":"Merge \"test-requirements: bump pyflakes to get f-strings support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eab2e2b6c13c138ce4bb8fd8425a97c5800ff7a4"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-22 12:02:57.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-11-26 11:44:17.000000000","tz":0},"subject":"Added Neutron Mellanox agent image","message":"Added Neutron Mellanox agent image\n\nNeutron Mellanox agent image is a Neutron base image plus package:\npython-networking-mlnx.\n\nUbuntu image base is not suppotred because no networking-mlnx pacakge for Ubuntu.\n\nChange-Id: If57c34bd5c3aee3e5b1f1c7b3703fd544ce498da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce2775b1c062e311efd0e89760c857667cb236e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce2775b1c062e311efd0e89760c857667cb236e7"}]},"branch":"refs/heads/master"},"4e90e505babcf0edf89f2a0386b3783c0a973775":{"kind":"REWORK","_number":7,"created":"2019-11-28 08:45:19.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/23/672023/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/672023/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/672023/7"}}},"commit":{"parents":[{"commit":"b5cfa8ef73e64ba6a74eb579310b3419b7d70c94","subject":"Merge \"Update skydive version to 0.26.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5cfa8ef73e64ba6a74eb579310b3419b7d70c94"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-22 12:02:57.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-11-28 08:46:05.000000000","tz":120},"subject":"Added Neutron Mellanox agent image","message":"Added Neutron Mellanox agent image\n\nNeutron Mellanox agent image is a Neutron base image plus package:\npython-networking-mlnx.\n\nUbuntu and Debian image bases when build type is binary\nis not suppotred because of missing python-networking-mlnx pacakage.\n\nChange-Id: If57c34bd5c3aee3e5b1f1c7b3703fd544ce498da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e90e505babcf0edf89f2a0386b3783c0a973775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e90e505babcf0edf89f2a0386b3783c0a973775"}]},"branch":"refs/heads/master"},"74f65e9d1990eb95dd24c328ec2a67e55905510b":{"kind":"REWORK","_number":8,"created":"2019-11-28 08:47:28.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/23/672023/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/672023/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/672023/8"}}},"commit":{"parents":[{"commit":"b5cfa8ef73e64ba6a74eb579310b3419b7d70c94","subject":"Merge \"Update skydive version to 0.26.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5cfa8ef73e64ba6a74eb579310b3419b7d70c94"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-22 12:02:57.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-11-28 08:50:40.000000000","tz":120},"subject":"Added Neutron Mellanox agent image","message":"Added Neutron Mellanox agent image\n\nNeutron Mellanox agent image is a Neutron base image plus package:\npython-networking-mlnx.\n\nUbuntu and Debian image bases when build type is binary\nis not suppotred because of missing python-networking-mlnx pacakage.\n\nChange-Id: If57c34bd5c3aee3e5b1f1c7b3703fd544ce498da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74f65e9d1990eb95dd24c328ec2a67e55905510b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74f65e9d1990eb95dd24c328ec2a67e55905510b"}]},"branch":"refs/heads/master"},"ac12834c6761551c2b491cfe7ffcd5a69dd80c65":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2019-12-02 07:31:01.000000000","uploader":{"_account_id":22948,"name":"Hamdy Khader","email":"hamdyk@mellanox.com","username":"hamdyk"},"ref":"refs/changes/23/672023/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/672023/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/672023/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/672023/9"}}},"commit":{"parents":[{"commit":"b5cfa8ef73e64ba6a74eb579310b3419b7d70c94","subject":"Merge \"Update skydive version to 0.26.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5cfa8ef73e64ba6a74eb579310b3419b7d70c94"}]}],"author":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-07-22 12:02:57.000000000","tz":180},"committer":{"name":"Hamdy Khader","email":"hamdyk@mellanox.com","date":"2019-12-02 07:33:12.000000000","tz":120},"subject":"Added Neutron Mellanox agent image","message":"Added Neutron Mellanox agent image\n\nNeutron Mellanox agent image is a Neutron base image plus package:\npython-networking-mlnx.\n\nUbuntu and Debian image bases when build type is binary\nis not supported because of missing python-networking-mlnx package.\n\nChange-Id: If57c34bd5c3aee3e5b1f1c7b3703fd544ce498da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac12834c6761551c2b491cfe7ffcd5a69dd80c65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac12834c6761551c2b491cfe7ffcd5a69dd80c65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
