)]}'
{"id":"openstack%2Fkolla~672027","triplet_id":"openstack%2Fkolla~stable%2Frocky~I16d5f79c00473e5113aee3739defc822b516bda4","project":"openstack/kolla","branch":"stable/rocky","hashtags":[],"change_id":"I16d5f79c00473e5113aee3739defc822b516bda4","subject":"storm: fetch storm archive from upstream","status":"MERGED","created":"2019-07-22 12:21:10.000000000","updated":"2019-07-22 15:33:03.000000000","submitted":"2019-07-22 15:33:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672027-1563809583715-31ef7618","meta_rev_id":"632891ad70bd71bd56b42f999a72c44cee15628c","_number":672027,"virtual_id_number":672027,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-22 15:33:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-22 13:11:59.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-07-22 12:42:29.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-07-22 12:48:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-07-22 12:48:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"date":"2019-07-22 12:21:22.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 12:21:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-22 12:42:29.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-07-22 12:48:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-22 13:11:59.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-22 15:33:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92ae5ecb9b8375166e3b4c177a871df1482f3bcf","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-07-22 12:21:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"5473db69dccec44ac0f63c01121fbcba4d98ee26","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-07-22 12:42:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f1b21ffdb2646b28380506d747eefb515987796","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-22 12:48:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d07e84c2ed0b2cf6b3c7b33daf75f121f554152f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-22 13:11:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/27/672027/1/openstack-check/tripleo-build-containers-fedora-28-master/f87342a/ : RETRY_LIMIT in 3m 14s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/27/672027/1/openstack-check/tripleo-build-containers-fedora-28-stein/199222d/ : SUCCESS in 48m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"de961abd6966d1518bca5aa4481e128a8f47f1b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:36:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/27/672027/1/check/kolla-build-centos-binary/d5a9156/ : SUCCESS in 38m 52s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/27/672027/1/check/kolla-build-centos-source/24fca15/ : SUCCESS in 47m 49s\n- kolla-ansible-centos-source http://logs.openstack.org/27/672027/1/check/kolla-ansible-centos-source/36fa01f/ : SUCCESS in 1h 11m 22s\n- kolla-ansible-centos-binary http://logs.openstack.org/27/672027/1/check/kolla-ansible-centos-binary/b746435/ : SUCCESS in 1h 06m 38s (non-voting)\n- kolla-build-oraclelinux-binary http://logs.openstack.org/27/672027/1/check/kolla-build-oraclelinux-binary/0b87717/ : SUCCESS in 46m 12s (non-voting)\n- kolla-build-oraclelinux-source http://logs.openstack.org/27/672027/1/check/kolla-build-oraclelinux-source/a22f2d5/ : SUCCESS in 33m 18s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/27/672027/1/check/kolla-ansible-oraclelinux-source/48df60b/ : SUCCESS in 1h 00m 51s (non-voting)\n- kolla-ansible-oraclelinux-binary http://logs.openstack.org/27/672027/1/check/kolla-ansible-oraclelinux-binary/066d927/ : SUCCESS in 1h 07m 49s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/27/672027/1/check/openstack-tox-lower-constraints/b3e5dd0/ : SUCCESS in 6m 01s\n- openstack-tox-pep8 http://logs.openstack.org/27/672027/1/check/openstack-tox-pep8/1965009/ : SUCCESS in 5m 45s\n- openstack-tox-py27 http://logs.openstack.org/27/672027/1/check/openstack-tox-py27/2295550/ : SUCCESS in 4m 11s\n- openstack-tox-py35 http://logs.openstack.org/27/672027/1/check/openstack-tox-py35/7a2a94b/ : SUCCESS in 4m 11s\n- build-openstack-sphinx-docs http://logs.openstack.org/27/672027/1/check/build-openstack-sphinx-docs/a23ddf6/html/ : SUCCESS in 4m 56s\n- kolla-build-ubuntu-binary http://logs.openstack.org/27/672027/1/check/kolla-build-ubuntu-binary/608fec2/ : SUCCESS in 55m 46s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/27/672027/1/check/kolla-build-ubuntu-source/b58a98f/ : SUCCESS in 31m 23s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/27/672027/1/check/kolla-ansible-ubuntu-source/e8d8065/ : SUCCESS in 1h 04m 28s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/27/672027/1/check/kolla-ansible-ubuntu-binary/25daedd/ : TIMED_OUT in 2h 01m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e7d164fda83c5588801ed1cd2aa5cbf47cc84e8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:36:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"329e0cda468a46ccb9aa81aef1a2ae4b244054e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 15:33:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/27/672027/1/gate/kolla-build-centos-source/1bc9096/ : SUCCESS in 38m 04s\n- kolla-build-oraclelinux-source http://logs.openstack.org/27/672027/1/gate/kolla-build-oraclelinux-source/0c258b0/ : SUCCESS in 46m 40s\n- openstack-tox-lower-constraints http://logs.openstack.org/27/672027/1/gate/openstack-tox-lower-constraints/90c172b/ : SUCCESS in 5m 52s\n- openstack-tox-pep8 http://logs.openstack.org/27/672027/1/gate/openstack-tox-pep8/4137162/ : SUCCESS in 4m 27s\n- openstack-tox-py27 http://logs.openstack.org/27/672027/1/gate/openstack-tox-py27/6e6fa4c/ : SUCCESS in 4m 30s\n- openstack-tox-py35 http://logs.openstack.org/27/672027/1/gate/openstack-tox-py35/ebdd30d/ : SUCCESS in 5m 45s\n- build-openstack-sphinx-docs http://logs.openstack.org/27/672027/1/gate/build-openstack-sphinx-docs/266145c/html/ : SUCCESS in 4m 41s\n- kolla-build-ubuntu-source http://logs.openstack.org/27/672027/1/gate/kolla-build-ubuntu-source/90ff267/ : SUCCESS in 30m 30s","accounts_in_message":[],"_revision_number":1},{"id":"632891ad70bd71bd56b42f999a72c44cee15628c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 15:33:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ac2cd44a46e8c235763a56f6ef54c87901079eec","revisions":{"ac2cd44a46e8c235763a56f6ef54c87901079eec":{"kind":"REWORK","_number":1,"created":"2019-07-22 12:21:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/27/672027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/672027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/672027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/672027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/672027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/672027/1"}}},"commit":{"parents":[{"commit":"330fc3698df0bddeea0e56f6c75623ff4dd88414","subject":"Change to use Python3 dependencies for gnocchi on Ubuntu binary.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/330fc3698df0bddeea0e56f6c75623ff4dd88414"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-19 08:40:07.000000000","tz":0},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-22 12:21:10.000000000","tz":0},"subject":"storm: fetch storm archive from upstream","message":"storm: fetch storm archive from upstream\n\nURL in Dockerfile was giving us 404 error\n\nChange-Id: I16d5f79c00473e5113aee3739defc822b516bda4\n(cherry picked from commit c73815347e2e6a456c29dc1f6d93d7ed5c8f7c59)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac2cd44a46e8c235763a56f6ef54c87901079eec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac2cd44a46e8c235763a56f6ef54c87901079eec"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
