)]}'
{"id":"openstack%2Fkolla~672547","triplet_id":"openstack%2Fkolla~master~I22a26374ee00e93b4f52ec0c4f1152ca1ad777e6","project":"openstack/kolla","branch":"master","topic":"collectd-sensubility","hashtags":[],"change_id":"I22a26374ee00e93b4f52ec0c4f1152ca1ad777e6","subject":"Add collectd-sensubility to the image","status":"MERGED","created":"2019-07-24 15:30:13.000000000","updated":"2019-08-05 10:11:28.000000000","submitted":"2019-08-05 10:11:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672547-1564999888712-c33df9d0","meta_rev_id":"490d8bb5496e4b80732d186b01d6a195b6cf33eb","_number":672547,"virtual_id_number":672547,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-05 10:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-02 09:52:37.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-08-05 07:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2019-08-02 10:42:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-08-05 07:31:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 13:21:36.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2019-08-02 09:52:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-02 10:42:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-05 07:31:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-08-05 10:11:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab59d92cf1fb2df63e7d05a576ca3864659df556","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-07-24 15:30:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfbe0a804d876ddf6c30af600af0b7b8403b4599","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-24 16:43:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/47/672547/1/openstack-check/tripleo-build-containers-fedora-28-master/3b1da77/ : FAILURE in 6m 36s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/47/672547/1/openstack-check/tripleo-build-containers-fedora-28-stein/046f0dd/ : SUCCESS in 1h 02m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02599199f2a84cef9499c469cb9e416e3e1a328e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 17:03:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/47/672547/1/check/kolla-build-centos-binary/533111a/ : FAILURE in 46m 26s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/47/672547/1/check/kolla-build-centos-source/1316b36/ : FAILURE in 48m 59s\n- kolla-ansible-centos-source http://logs.openstack.org/47/672547/1/check/kolla-ansible-centos-source/b932550/ : SUCCESS in 1h 02m 49s\n- kolla-ansible-centos-binary http://logs.openstack.org/47/672547/1/check/kolla-ansible-centos-binary/b2a861c/ : SUCCESS in 1h 14m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/47/672547/1/check/kolla-ansible-centos-source-upgrade/08f0f48/ : SUCCESS in 1h 27m 26s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/47/672547/1/check/tripleo-build-containers-centos-7/5b896dc/ : SUCCESS in 50m 00s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/672547/1/check/openstack-tox-lower-constraints/97d937c/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/47/672547/1/check/openstack-tox-pep8/27ef063/ : SUCCESS in 4m 17s\n- openstack-tox-py27 http://logs.openstack.org/47/672547/1/check/openstack-tox-py27/05b3cc6/ : SUCCESS in 6m 04s\n- openstack-tox-py36 http://logs.openstack.org/47/672547/1/check/openstack-tox-py36/6e0a4a5/ : SUCCESS in 3m 57s\n- openstack-tox-py37 http://logs.openstack.org/47/672547/1/check/openstack-tox-py37/775421d/ : SUCCESS in 3m 45s\n- openstack-tox-docs http://logs.openstack.org/47/672547/1/check/openstack-tox-docs/968f056/html/ : SUCCESS in 3m 32s\n- kolla-build-ubuntu-binary http://logs.openstack.org/47/672547/1/check/kolla-build-ubuntu-binary/3cabbf2/ : SUCCESS in 33m 51s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/47/672547/1/check/kolla-build-ubuntu-source/e45f086/ : SUCCESS in 38m 47s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/47/672547/1/check/kolla-ansible-ubuntu-source/c7b2b9c/ : SUCCESS in 46m 21s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/47/672547/1/check/kolla-ansible-ubuntu-binary/053627f/ : SUCCESS in 52m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/47/672547/1/check/kolla-ansible-ubuntu-source-upgrade/fd1728e/ : SUCCESS in 1h 05m 37s","accounts_in_message":[],"_revision_number":1},{"id":"c527eb21345aca0b676eef83e523fa8d4078c771","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2019-07-25 14:24:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c189eb9affafe045d5827b07727d8fdb8cba8eca","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 15:16:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/47/672547/2/openstack-check/tripleo-build-containers-fedora-28-master/befd7aa/ : FAILURE in 5m 53s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/47/672547/2/openstack-check/tripleo-build-containers-fedora-28-stein/499e598/ : SUCCESS in 50m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21cfb3aec079b4647be1b39dd83fe0e576599e5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 15:54:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/47/672547/2/check/kolla-build-centos-binary/61fe29f/ : SUCCESS in 46m 42s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/47/672547/2/check/kolla-build-centos-source/11269aa/ : SUCCESS in 54m 58s\n- kolla-ansible-centos-source http://logs.openstack.org/47/672547/2/check/kolla-ansible-centos-source/232f656/ : SUCCESS in 1h 21m 48s\n- kolla-ansible-centos-binary http://logs.openstack.org/47/672547/2/check/kolla-ansible-centos-binary/096fc70/ : SUCCESS in 1h 15m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/47/672547/2/check/kolla-ansible-centos-source-upgrade/e80f959/ : SUCCESS in 1h 23m 37s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/47/672547/2/check/tripleo-build-containers-centos-7/06e3e18/ : SUCCESS in 45m 14s\n- openstack-tox-lower-constraints http://logs.openstack.org/47/672547/2/check/openstack-tox-lower-constraints/51d510e/ : SUCCESS in 3m 59s\n- openstack-tox-pep8 http://logs.openstack.org/47/672547/2/check/openstack-tox-pep8/b0fa9d4/ : SUCCESS in 4m 56s\n- openstack-tox-py27 http://logs.openstack.org/47/672547/2/check/openstack-tox-py27/5d6498e/ : SUCCESS in 3m 19s\n- openstack-tox-py36 http://logs.openstack.org/47/672547/2/check/openstack-tox-py36/9633c00/ : SUCCESS in 3m 27s\n- openstack-tox-py37 http://logs.openstack.org/47/672547/2/check/openstack-tox-py37/a173387/ : SUCCESS in 3m 51s\n- openstack-tox-docs http://logs.openstack.org/47/672547/2/check/openstack-tox-docs/7375482/html/ : SUCCESS in 4m 52s\n- kolla-build-ubuntu-binary http://logs.openstack.org/47/672547/2/check/kolla-build-ubuntu-binary/cfbc1e2/ : SUCCESS in 29m 52s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/47/672547/2/check/kolla-build-ubuntu-source/22bf5ad/ : SUCCESS in 38m 51s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/47/672547/2/check/kolla-ansible-ubuntu-source/b2db2dd/ : SUCCESS in 56m 09s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/47/672547/2/check/kolla-ansible-ubuntu-binary/bd88b78/ : SUCCESS in 45m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/47/672547/2/check/kolla-ansible-ubuntu-source-upgrade/beb3939/ : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":2},{"id":"7aaf16e9dbb080871f0287b61dd325e0bfeaef50","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-30 16:02:54.000000000","message":"Patch Set 2:\n\nAnd what is the reason of adding that package?","accounts_in_message":[],"_revision_number":2},{"id":"91d587eb6e9b29cec52fb8a5fdd7d0fd7406ad04","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2019-08-01 13:21:36.000000000","message":"Patch Set 2: Code-Review+1\n\nThe package can emit events when certain conditions are not met, like sensu; with that, you can drive your availability monitoring through collectd events.","accounts_in_message":[],"_revision_number":2},{"id":"b558c737b8d891fd6e99b95c42fa36f43d2f5b9a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-02 06:40:04.000000000","message":"Patch Set 2:\n\n\u003e The package can emit events when certain conditions are not met,\n \u003e like sensu; with that, you can drive your availability monitoring\n \u003e through collectd events.\n\nLet\u0027s add this to the commit message.","accounts_in_message":[],"_revision_number":2},{"id":"a07cc25f31884ec08f86ff7646d2dc16baf5c538","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2019-08-02 09:18:25.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b85693c89216428db9f14cde4fab621c7b141014","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 09:52:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/47/672547/3/openstack-check/tripleo-build-containers-rhel-8-master/4a75d97/ : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":3},{"id":"f528a8ecb4af2a58241c66e1766d237b00660854","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 10:40:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://logs.opendev.org/47/672547/3/check/kolla-build-centos-binary/a26466b/ : SUCCESS in 33m 51s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/47/672547/3/check/kolla-build-centos-source/e597472/ : SUCCESS in 50m 23s\n- kolla-ansible-centos-source https://logs.opendev.org/47/672547/3/check/kolla-ansible-centos-source/dd3e169/ : SUCCESS in 1h 07m 50s\n- kolla-ansible-centos-binary https://logs.opendev.org/47/672547/3/check/kolla-ansible-centos-binary/d4e42b4/ : SUCCESS in 1h 05m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/47/672547/3/check/kolla-ansible-centos-source-upgrade/dfb528b/ : SUCCESS in 1h 19m 56s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/47/672547/3/check/tripleo-build-containers-centos-7/0743daa/ : SUCCESS in 49m 40s\n- openstack-tox-lower-constraints https://logs.opendev.org/47/672547/3/check/openstack-tox-lower-constraints/55c7ff2/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://logs.opendev.org/47/672547/3/check/openstack-tox-pep8/a8c0405/ : SUCCESS in 6m 27s\n- openstack-tox-py27 https://logs.opendev.org/47/672547/3/check/openstack-tox-py27/bfe65c9/ : SUCCESS in 4m 27s\n- openstack-tox-py36 https://logs.opendev.org/47/672547/3/check/openstack-tox-py36/0956598/ : SUCCESS in 3m 27s\n- openstack-tox-py37 https://logs.opendev.org/47/672547/3/check/openstack-tox-py37/5f25e73/ : SUCCESS in 5m 45s\n- openstack-tox-docs https://logs.opendev.org/47/672547/3/check/openstack-tox-docs/c1b35ff/html/ : SUCCESS in 4m 13s\n- kolla-build-ubuntu-binary https://logs.opendev.org/47/672547/3/check/kolla-build-ubuntu-binary/5165619/ : SUCCESS in 27m 41s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/47/672547/3/check/kolla-build-ubuntu-source/c581398/ : SUCCESS in 38m 08s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/47/672547/3/check/kolla-ansible-ubuntu-source/414eb60/ : SUCCESS in 51m 48s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/47/672547/3/check/kolla-ansible-ubuntu-binary/5efad5c/ : SUCCESS in 49m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/47/672547/3/check/kolla-ansible-ubuntu-source-upgrade/4ef7502/ : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":3},{"id":"3a0996cae328e230a82313faef86a70ad601240a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-02 10:42:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"390635faf3c7aeb835e834bf85eaac3eac75d800","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-05 07:31:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b4ae6598b17a9e4b59615a48a9f51d590a7ee83c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 07:31:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d3f159833f49ad8980b30c528c1e9511227e7f06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 10:11:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fdc27964f56bf78da5d8809b069d46a71441d74a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 10:11:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://logs.opendev.org/47/672547/3/gate/kolla-build-centos-source/c4078f1/ : SUCCESS in 1h 04m 49s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/47/672547/3/gate/tripleo-build-containers-centos-7/d81efc6/ : SUCCESS in 53m 35s\n- openstack-tox-lower-constraints https://logs.opendev.org/47/672547/3/gate/openstack-tox-lower-constraints/4dbff6e/ : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://logs.opendev.org/47/672547/3/gate/openstack-tox-pep8/f16c89c/ : SUCCESS in 5m 10s\n- openstack-tox-py27 https://logs.opendev.org/47/672547/3/gate/openstack-tox-py27/378c265/ : SUCCESS in 3m 40s\n- openstack-tox-py36 https://logs.opendev.org/47/672547/3/gate/openstack-tox-py36/dceea91/ : SUCCESS in 3m 17s\n- openstack-tox-py37 https://logs.opendev.org/47/672547/3/gate/openstack-tox-py37/3e3bea8/ : SUCCESS in 6m 14s\n- openstack-tox-docs https://logs.opendev.org/47/672547/3/gate/openstack-tox-docs/b094fc9/html/ : SUCCESS in 4m 43s\n- kolla-build-ubuntu-source https://logs.opendev.org/47/672547/3/gate/kolla-build-ubuntu-source/461e7ef/ : SUCCESS in 32m 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"057d605d9b9526ab2150baf1867560288b416c9e","revisions":{"ad0a0c2296003b14a25187d5d847cc49160751fa":{"kind":"REWORK","_number":1,"created":"2019-07-24 15:30:13.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/47/672547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/672547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/672547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/672547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/672547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/672547/1"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-07-24 15:28:41.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-07-24 15:28:41.000000000","tz":120},"subject":"[WIP] Add collectd-sensubility to the image","message":"[WIP] Add collectd-sensubility to the image\n\nThis patch add installation of collectd-sensubility as part of collectd image.\nThe package is available in centos-release-opstools repo.\n\nChange-Id: I22a26374ee00e93b4f52ec0c4f1152ca1ad777e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad0a0c2296003b14a25187d5d847cc49160751fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad0a0c2296003b14a25187d5d847cc49160751fa"}]},"branch":"refs/heads/master"},"a47613d93365f73ff8f0d502bc449b648be97b7a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-25 14:24:12.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/47/672547/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/672547/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/672547/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/672547/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/672547/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/672547/2"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-07-24 15:28:41.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-07-25 14:23:57.000000000","tz":120},"subject":"Add collectd-sensubility to the image","message":"Add collectd-sensubility to the image\n\nThis patch add installation of collectd-sensubility as part of collectd image.\nThe package is available in centos-release-opstools repo.\n\nChange-Id: I22a26374ee00e93b4f52ec0c4f1152ca1ad777e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a47613d93365f73ff8f0d502bc449b648be97b7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a47613d93365f73ff8f0d502bc449b648be97b7a"}]},"branch":"refs/heads/master"},"057d605d9b9526ab2150baf1867560288b416c9e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-02 09:18:25.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/47/672547/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/47/672547/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/672547/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/672547/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/47/672547/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/47/672547/3"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2019-07-24 15:28:41.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2019-08-02 09:18:17.000000000","tz":0},"subject":"Add collectd-sensubility to the image","message":"Add collectd-sensubility to the image\n\nThis patch add installation of collectd-sensubility as part of collectd image.\nThe package is available in centos-release-opstools repo.\n\nCollectd-sensubility aims provide possibility to switch from Sensu based\navailability monitoring solution to monitoring solution based on\ncollectd with AMQP-1.0 messaging bus.\n\nUpstream is here: https://github.com/paramite/collectd-sensubility\nChange-Id: I22a26374ee00e93b4f52ec0c4f1152ca1ad777e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/057d605d9b9526ab2150baf1867560288b416c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/057d605d9b9526ab2150baf1867560288b416c9e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
