)]}'
{"id":"openstack%2Fkolla~672717","triplet_id":"openstack%2Fkolla~master~I0aa375ff4b26a3216839d69dff8e843584a3ecee","project":"openstack/kolla","branch":"master","topic":"monasca-grafana-v4-2-0","hashtags":[],"change_id":"I0aa375ff4b26a3216839d69dff8e843584a3ecee","subject":"Change monasca-grafana source branch to use grafana v4.2.0 with keystone changes","status":"MERGED","created":"2019-07-25 12:55:25.000000000","updated":"2019-08-07 19:12:15.000000000","submitted":"2019-08-07 19:12:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672717-1565205135545-e657b839","meta_rev_id":"e91845f590a6d9c659790850f656127aac336d1e","_number":672717,"virtual_id_number":672717,"owner":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-07 19:12:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-25 13:46:36.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-07-25 13:02:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-08-07 17:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-07-25 14:30:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-08-07 17:58:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-25 13:02:28.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-25 13:46:36.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-25 14:30:35.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-07 17:58:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-08-07 19:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af74961e5c1f0701e7659df254822daa2d3f5533","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-07-25 12:55:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb0af4776776eacb1c4d509a5866f225ba03f8e5","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-07-25 12:58:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"26388062cdeea94635c64184d34966e9565efce4","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-07-25 12:59:22.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ced153e38c52128db520dfcd4009e2b1da8a0492","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-25 13:02:28.000000000","message":"Patch Set 3: Code-Review+1\n\nThe context here is that the Monasca fork of Grafana has been updated to Grafana 4.2. This looks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"bb0b85b4e1c49a80c7dae0077853cc23b2760c6e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-25 13:46:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/17/672717/3/openstack-check/tripleo-build-containers-fedora-28-master/d21d341/ : FAILURE in 6m 08s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/17/672717/3/openstack-check/tripleo-build-containers-fedora-28-stein/9a171c8/ : SUCCESS in 45m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f5977ecc0ab7d0861a042a79367a6f006f9d32c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-25 14:30:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"01b7029a9da097177a1695d0135a6a0bfdd15d3b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 14:32:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/17/672717/3/check/kolla-build-centos-binary/5ed3082/ : SUCCESS in 45m 25s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/17/672717/3/check/kolla-build-centos-source/40cf669/ : SUCCESS in 51m 52s\n- kolla-ansible-centos-source http://logs.openstack.org/17/672717/3/check/kolla-ansible-centos-source/f9fec03/ : SUCCESS in 1h 27m 23s\n- kolla-ansible-centos-binary http://logs.openstack.org/17/672717/3/check/kolla-ansible-centos-binary/0dcd082/ : SUCCESS in 1h 16m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/17/672717/3/check/kolla-ansible-centos-source-upgrade/5ae897e/ : SUCCESS in 1h 24m 55s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/17/672717/3/check/tripleo-build-containers-centos-7/0972671/ : SUCCESS in 56m 21s\n- openstack-tox-lower-constraints http://logs.openstack.org/17/672717/3/check/openstack-tox-lower-constraints/6e09659/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 http://logs.openstack.org/17/672717/3/check/openstack-tox-pep8/e0dab62/ : SUCCESS in 5m 24s\n- openstack-tox-py27 http://logs.openstack.org/17/672717/3/check/openstack-tox-py27/36d5b72/ : SUCCESS in 3m 41s\n- openstack-tox-py36 http://logs.openstack.org/17/672717/3/check/openstack-tox-py36/b18ed92/ : SUCCESS in 3m 51s\n- openstack-tox-py37 http://logs.openstack.org/17/672717/3/check/openstack-tox-py37/f572e62/ : SUCCESS in 4m 17s\n- openstack-tox-docs http://logs.openstack.org/17/672717/3/check/openstack-tox-docs/9e65b5b/html/ : SUCCESS in 3m 16s\n- kolla-build-ubuntu-binary http://logs.openstack.org/17/672717/3/check/kolla-build-ubuntu-binary/1075e91/ : SUCCESS in 25m 19s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/17/672717/3/check/kolla-build-ubuntu-source/ccb837a/ : SUCCESS in 31m 22s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/17/672717/3/check/kolla-ansible-ubuntu-source/ed50179/ : SUCCESS in 52m 05s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/17/672717/3/check/kolla-ansible-ubuntu-binary/29d9874/ : SUCCESS in 56m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/17/672717/3/check/kolla-ansible-ubuntu-source-upgrade/5e1c059/ : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":3},{"id":"340335c5fd1504fc1cf7778e2e381b54c9300dfb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-30 15:58:29.000000000","message":"Patch Set 3: Code-Review-1\n\nCan we finally move this to proper approach with https://docs.openstack.org/kolla/latest/admin/image-building.html#plugin-functionality?\nIt is problematic to build through http proxy...","accounts_in_message":[],"_revision_number":3},{"id":"dca4d602ff188a447fda0736ed796177ef9668e8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-07 15:11:31.000000000","message":"Patch Set 3: -Code-Review\n\nActually putting it as a downloadable source in kolla/common/config.py will only work for source - can we move this to use curl?","accounts_in_message":[],"_revision_number":3},{"id":"6bcd39ddaf9a5d81539e55ef7658b5f5f4861a56","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-07 16:15:17.000000000","message":"Patch Set 3:\n\nNot sure I follow - it is using curl at the moment.","accounts_in_message":[],"_revision_number":3},{"id":"e80d762366b47cd616a4fe18743fccc41b676cba","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-07 17:58:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nRight Doug, sorry! I still had the different approach that is in fluentd (ADD instead of curl). This is not the case here - looks good.","accounts_in_message":[],"_revision_number":3},{"id":"38a6d717132119b969b780c243f05d96c4f2346b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 17:58:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f3451884db9b3dd6c905bc1d08ac6a90c69d72e5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 19:12:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"34aef18d27fb5e6b14a9bc8ac0c26a66a6ab3228","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 19:12:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://logs.opendev.org/17/672717/3/gate/kolla-build-centos-source/6b84b42/ : SUCCESS in 1h 07m 55s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/17/672717/3/gate/tripleo-build-containers-centos-7/ca26e1e/ : SUCCESS in 44m 15s\n- openstack-tox-lower-constraints https://logs.opendev.org/17/672717/3/gate/openstack-tox-lower-constraints/d275199/ : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://logs.opendev.org/17/672717/3/gate/openstack-tox-pep8/5e228d1/ : SUCCESS in 4m 42s\n- openstack-tox-py27 https://logs.opendev.org/17/672717/3/gate/openstack-tox-py27/7432e14/ : SUCCESS in 4m 36s\n- openstack-tox-py36 https://logs.opendev.org/17/672717/3/gate/openstack-tox-py36/9aa743c/ : SUCCESS in 4m 09s\n- openstack-tox-py37 https://logs.opendev.org/17/672717/3/gate/openstack-tox-py37/9cef140/ : SUCCESS in 5m 05s\n- openstack-tox-docs https://logs.opendev.org/17/672717/3/gate/openstack-tox-docs/90f1813/html/ : SUCCESS in 3m 13s\n- kolla-build-ubuntu-source https://logs.opendev.org/17/672717/3/gate/kolla-build-ubuntu-source/5270006/ : SUCCESS in 35m 28s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"438ef928f3d5ce3ebad5c471afa4d7f5df99bf7c","revisions":{"2c7a8338a7b7fb974130d5ee07803a4455b88343":{"kind":"REWORK","_number":1,"created":"2019-07-25 12:55:25.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/17/672717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/672717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/672717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/672717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/672717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/672717/1"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-07-25 12:54:47.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-07-25 12:55:23.000000000","tz":60},"subject":"Change monasca-grafana source branch to use grafana v4.2.0 with keystone changes","message":"Change monasca-grafana source branch to use grafana v4.2.0 with keystone changes\n\nChange-Id: I0aa375ff4b26a3216839d69dff8e843584a3ecee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c7a8338a7b7fb974130d5ee07803a4455b88343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c7a8338a7b7fb974130d5ee07803a4455b88343"}]},"branch":"refs/heads/master"},"ef4f97fdaa978424c791a126a22e05e3af156da6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-25 12:58:35.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/17/672717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/672717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/672717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/672717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/672717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/672717/2"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-07-25 12:54:47.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-07-25 12:57:52.000000000","tz":0},"subject":"Change monasca-grafana source branch to use grafana v4.2.0 with  keystone changes","message":"Change monasca-grafana source branch to use grafana v4.2.0 with \nkeystone changes\n\nChange-Id: I0aa375ff4b26a3216839d69dff8e843584a3ecee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef4f97fdaa978424c791a126a22e05e3af156da6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ef4f97fdaa978424c791a126a22e05e3af156da6"}]},"branch":"refs/heads/master"},"438ef928f3d5ce3ebad5c471afa4d7f5df99bf7c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-25 12:59:22.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/17/672717/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/672717/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/672717/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/672717/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/672717/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/672717/3"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-07-25 12:54:47.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-07-25 12:59:17.000000000","tz":0},"subject":"Change monasca-grafana source branch to use grafana v4.2.0 with keystone changes","message":"Change monasca-grafana source branch to use grafana v4.2.0\nwith keystone changes\n\nChange-Id: I0aa375ff4b26a3216839d69dff8e843584a3ecee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/438ef928f3d5ce3ebad5c471afa4d7f5df99bf7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/438ef928f3d5ce3ebad5c471afa4d7f5df99bf7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
