)]}'
{"id":"openstack%2Fkolla~673311","triplet_id":"openstack%2Fkolla~master~I6e893c25a9dfbbbac6bd69f2fa26e6c72a4154d7","project":"openstack/kolla","branch":"master","topic":"oe-latest-release","hashtags":[],"change_id":"I6e893c25a9dfbbbac6bd69f2fa26e6c72a4154d7","subject":"[docker/openstack-exporter] Bump to 0.2.1 version","status":"MERGED","created":"2019-07-29 15:19:43.000000000","updated":"2019-09-20 16:07:41.000000000","submitted":"2019-09-20 16:05:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673311-1568995551985-953e65f2","meta_rev_id":"e663751c40f49c09a91360400390e7a49d79f4e0","_number":673311,"virtual_id_number":673311,"owner":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-20 16:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-20 13:13:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":2,"date":"2019-09-12 07:37:52.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":1,"date":"2019-08-14 15:40:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":1,"date":"2019-08-13 15:50:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":2,"date":"2019-08-06 08:19:57.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-09-11 16:04:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2019-08-13 14:55:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-09-11 16:04:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"value":0,"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 10:00:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-07-30 11:26:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-06 08:19:57.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-13 14:55:27.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2019-08-13 15:50:55.000000000","updated_by":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"reviewer":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"state":"REVIEWER"},{"updated":"2019-08-14 15:40:07.000000000","updated_by":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"reviewer":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"state":"REVIEWER"},{"updated":"2019-09-11 16:04:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-12 07:37:52.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-09-20 13:13:35.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-20 16:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"abcf266b1d5c62fac09be4632d7c0db83febc275","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2019-07-29 15:19:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ffee884b0cee5654305aa4fe953b39a2386d8a2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-29 16:16:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/11/673311/1/openstack-check/tripleo-build-containers-fedora-28-master/e5582e2/ : FAILURE in 7m 04s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/11/673311/1/openstack-check/tripleo-build-containers-fedora-28-stein/8fdd009/ : SUCCESS in 50m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55e08ea45f6dcc9fad06dfa6b1e188422717089a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-29 16:54:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary http://logs.openstack.org/11/673311/1/check/kolla-build-centos-binary/2504888/ : FAILURE in 36m 02s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/11/673311/1/check/kolla-build-centos-source/c1f851e/ : FAILURE in 55m 16s\n- kolla-ansible-centos-source http://logs.openstack.org/11/673311/1/check/kolla-ansible-centos-source/eccc96a/ : SUCCESS in 1h 29m 11s\n- kolla-ansible-centos-binary http://logs.openstack.org/11/673311/1/check/kolla-ansible-centos-binary/93ee21a/ : SUCCESS in 56m 56s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/11/673311/1/check/kolla-ansible-centos-source-upgrade/b05b5ed/ : SUCCESS in 1h 28m 25s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/11/673311/1/check/tripleo-build-containers-centos-7/4e13e66/ : SUCCESS in 48m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/11/673311/1/check/openstack-tox-lower-constraints/ee11eed/ : SUCCESS in 5m 13s\n- openstack-tox-pep8 http://logs.openstack.org/11/673311/1/check/openstack-tox-pep8/20acdca/ : SUCCESS in 4m 20s\n- openstack-tox-py27 http://logs.openstack.org/11/673311/1/check/openstack-tox-py27/edb4a03/ : SUCCESS in 4m 51s\n- openstack-tox-py36 http://logs.openstack.org/11/673311/1/check/openstack-tox-py36/e3b3c6a/ : SUCCESS in 5m 13s\n- openstack-tox-py37 http://logs.openstack.org/11/673311/1/check/openstack-tox-py37/493aa72/ : SUCCESS in 4m 12s\n- openstack-tox-docs http://logs.openstack.org/11/673311/1/check/openstack-tox-docs/257e4e1/html/ : SUCCESS in 3m 23s\n- kolla-build-ubuntu-binary http://logs.openstack.org/11/673311/1/check/kolla-build-ubuntu-binary/7c5c9fe/ : FAILURE in 30m 41s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/11/673311/1/check/kolla-build-ubuntu-source/d2b5383/ : FAILURE in 44m 21s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/11/673311/1/check/kolla-ansible-ubuntu-source/ca43a7a/ : SUCCESS in 54m 54s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/11/673311/1/check/kolla-ansible-ubuntu-binary/527e217/ : SUCCESS in 46m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/11/673311/1/check/kolla-ansible-ubuntu-source-upgrade/1dbc04d/ : SUCCESS in 51m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b1372f5d1a52a985a157f2a7ec0dbe2c3dff75e8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-30 10:00:57.000000000","message":"Patch Set 1:\n\nThe build is failing because https://github.com/openstack-exporter/openstack-exporter/releases/tag/v0.2.1 doesn\u0027t have the amd64 binary tarball, but only source code tarballs.","accounts_in_message":[],"_revision_number":1},{"id":"b2026900c79733f0029b908cf7d5d8ef4aa1ee54","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-30 11:26:58.000000000","message":"Patch Set 1:\n\n... and we don\u0027t build from source.","accounts_in_message":[],"_revision_number":1},{"id":"582cc667b291b6d31b62a065d57653e8e1589df5","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-02 08:30:23.000000000","message":"Patch Set 1: Code-Review-1\n\nno binary releases","accounts_in_message":[],"_revision_number":1},{"id":"c078ba7ef1e39d188b2748cd282bff0a9ba18f53","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2019-08-05 22:25:01.000000000","message":"Patch Set 1:\n\nOK, the CI release task failed and therefore the artifacts weren\u0027t generated. Now they\u0027re attached into the release as binaries.\n\nThank you and I apologize for the previous issue.","accounts_in_message":[],"_revision_number":1},{"id":"927d77132b7216cbac05a69d39bc6d76bea03a56","author":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"date":"2019-08-05 22:25:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"db5939858b90db490a6c240187f6eff287c30b4e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-05 23:10:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/11/673311/2/openstack-check/tripleo-build-containers-rhel-8-master/8fb89be/ : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":2},{"id":"8edfafb67ae4b9f12ccabb7020e01c010f413587","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 23:58:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://logs.opendev.org/11/673311/2/check/kolla-build-centos-binary/c50e791/ : SUCCESS in 30m 22s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/11/673311/2/check/kolla-build-centos-source/dad49fa/ : SUCCESS in 45m 29s\n- kolla-ansible-centos-source https://logs.opendev.org/11/673311/2/check/kolla-ansible-centos-source/28009ca/ : SUCCESS in 1h 12m 55s\n- kolla-ansible-centos-binary https://logs.opendev.org/11/673311/2/check/kolla-ansible-centos-binary/e46808e/ : SUCCESS in 1h 14m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/11/673311/2/check/kolla-ansible-centos-source-upgrade/700ce52/ : SUCCESS in 1h 31m 23s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/11/673311/2/check/tripleo-build-containers-centos-7/f3f8efc/ : SUCCESS in 52m 24s\n- openstack-tox-lower-constraints https://logs.opendev.org/11/673311/2/check/openstack-tox-lower-constraints/0fffc69/ : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://logs.opendev.org/11/673311/2/check/openstack-tox-pep8/1d5de2a/ : SUCCESS in 4m 23s\n- openstack-tox-py27 https://logs.opendev.org/11/673311/2/check/openstack-tox-py27/dccf8cb/ : SUCCESS in 3m 27s\n- openstack-tox-py36 https://logs.opendev.org/11/673311/2/check/openstack-tox-py36/f3bb347/ : SUCCESS in 3m 31s\n- openstack-tox-py37 https://logs.opendev.org/11/673311/2/check/openstack-tox-py37/45520ee/ : SUCCESS in 3m 35s\n- openstack-tox-docs https://logs.opendev.org/11/673311/2/check/openstack-tox-docs/9ad06dd/html/ : SUCCESS in 3m 00s\n- kolla-build-ubuntu-binary https://logs.opendev.org/11/673311/2/check/kolla-build-ubuntu-binary/3d1043c/ : SUCCESS in 53m 15s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/11/673311/2/check/kolla-build-ubuntu-source/2dcb260/ : SUCCESS in 28m 43s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/11/673311/2/check/kolla-ansible-ubuntu-source/c4081d3/ : SUCCESS in 55m 57s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/11/673311/2/check/kolla-ansible-ubuntu-binary/ae8e380/ : SUCCESS in 1h 18m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/11/673311/2/check/kolla-ansible-ubuntu-source-upgrade/e5d5b5b/ : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":2},{"id":"1d961fdf417ffb962f17d26c6edbe37c3759aaf8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-06 08:19:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6a86163f71f65943080f65df8038a2198cc7a96f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-09 09:31:02.000000000","message":"Patch Set 2:\n\nLooks good, just want to check that this version of the exporter is compatible with the version of prometheus deployed by kolla.","accounts_in_message":[],"_revision_number":2},{"id":"56d74731717bb7062860e3b4484bca84e51cc947","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2019-08-13 14:55:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cdddbf39d693e27885efba9a1bb7c3fa083c47d0","author":{"_account_id":28743,"name":"huang liujie","email":"huang.liujie@99cloud.net","username":"ljhuang"},"date":"2019-08-13 15:50:55.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5a01b5800ea2ce5faf9d3550fc7d31faeb936899","author":{"_account_id":27781,"name":"zhulingjie","email":"easyzlj@gmail.com","username":"zhulingjie"},"date":"2019-08-14 15:40:07.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"672aacc670364c766a83a19881a61541c0b8cf61","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-11 16:04:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4d9c9a80ad5321b7e939cf86c11290602ee41d9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 16:04:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"165867aba2c23b9adce4dfd4065a5bddf6d5b4c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 18:09:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c1aa297d12d54d71bd3a6e0f1e9d3fb9 : SUCCESS in 1h 03m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bd2fb31a1db245b9a186bcf0fc07b13d : SUCCESS in 1h 19m 39s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/05ce94b8f3e54c91b37bcaa579241993 : SUCCESS in 1h 35m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/d7d0a50e9d154d1a9b2da88e61575f4f : SUCCESS in 40m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a53f308522a7496294d6c21df90d038d : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/405ee687acb2439a970c5dcff883c099 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b253c3771b4b4387a066d920e9f8d937 : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d966c6fb57a74299b04cdca0888c6f26 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/06f14bbbd0bb4b4783c1ba6b586d7e64 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d22e4b34326747d7b34191b0e3a818cd : SUCCESS in 5m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b2e9f88ebb44d4fb541ab22d17357f8 : FAILURE in 35m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a88e3c232f744c8d97254669b913dc08 : SUCCESS in 1h 00m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8d80e5e687d245d09102d393e66144e5 : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":2},{"id":"520153c6fd789a7062335e7d24e1c8b3e61e13ec","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-09-12 07:37:52.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"58f5d9ad2fb38160a8857e597444ed920fead942","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-12 08:15:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/11/673311/2/openstack-check/tripleo-build-containers-rhel-8-master/0184bea/ : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":2},{"id":"af189a77d8cd8c7adac38d95b461ebf440e11dbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 09:15:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5a51f8a991ae4d58aec43e2de524f539 : SUCCESS in 48m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/74797f8635ba4ae1b00352cf90d906b2 : SUCCESS in 1h 08m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a6e9bb4887d043ad91e9e81def8543d7 : SUCCESS in 1h 13m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9924f53ebc2d40edb8ed2888f97c1610 : SUCCESS in 1h 08m 04s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e002e68a366d4919bdcd01ac0bcd2a05 : SUCCESS in 1h 23m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a4b49a26cbaf46fb93563025adcd7cc3 : SUCCESS in 42m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bbb187d5fbb9485e8d15eaa9d61ecd53 : FAILURE in 6m 47s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04cf0b9268fa4a43a0d48ec6551dafce : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18bc483073d449339cf9baacb6ec9041 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1210b89b00a54510be30b4dde70d3ab4 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4297cf5e357f483dba920f0c807d10ce : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89fca3c5bd7442d4b54e784174d7485b : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c17da8de99bb403589a54e463da9600a : SUCCESS in 3m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9ef435a6edb44a7907df00ec983510c : FAILURE in 24m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f551154da854cfca3e3ccb11a3614fa : FAILURE in 38m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef97e15aa31a404ca847924db197b968 : SUCCESS in 1h 06m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/51ff1be6bd8e4f369d55667b8f20b4a7 : SUCCESS in 52m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3bdc268605fe4239be451dc45b5aae4f : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":2},{"id":"2e212949db018daf4013814556cc59084fa2a7f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-16 14:29:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bc8e44f2f1a831ba3ed7764e20a6f46087922afe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-16 15:07:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/11/673311/2/openstack-check/tripleo-build-containers-rhel-8-master/7175b46/ : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4c6c17a7769a3e75ce9a454dafd115529fcb597b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 15:57:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/01ace870a49c44fa9f720e2d6c0be031 : SUCCESS in 1h 01m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6116585a9ae64443b21bfe20a203e7c3 : SUCCESS in 52m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/da43f73d21244f30a601d20bbbf82098 : SUCCESS in 1h 22m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ee99faa6a3b643e581b3acfb95c016f3 : SUCCESS in 1h 13m 43s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/79715768b5a5458ca96f48654708e1b6 : FAILURE in 31m 39s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/144d3b21ce7c426ea2106a8e5a2110e4 : SUCCESS in 46m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/957aace6c20d4d28ae6513e4ddfe9d23 : SUCCESS in 24m 48s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31805562abfb4dbca3086f744f744141 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/835ee739e39144fe86dabeb0445e6aca : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/357a01d2431449c9b020fd6b7ca169da : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f5d3264ed68455e806c1cdee3869022 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/01afef65362d495fab57e5777b2f299a : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3092097c2bf4c2abf27981e3a1d2c15 : SUCCESS in 3m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/afb9674b5c584e7f840de58c9a874dcc : SUCCESS in 29m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7822c312eaa45d68e6201444f1d1f93 : SUCCESS in 35m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/90219987e366416593268c01cd229c47 : SUCCESS in 55m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/79167e430ba242e0a194d3a18ba1d662 : SUCCESS in 1h 14m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ba465cf5ed194da6a520160fd3d55362 : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":2},{"id":"3ca046b7666e8dd92e181e3cebc359f5ff089252","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-20 12:25:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"da73052775d245ad3640f4853fceb358b40e576d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-20 13:13:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/11/673311/2/openstack-check/tripleo-build-containers-rhel-8-master/6acbbdc/ : SUCCESS in 46m 54s","accounts_in_message":[],"_revision_number":2},{"id":"0a8582f15ef878a73e5db013ec834a6326d597ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:15:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/57c17ab0df1e473c8a0d7ba7057ef43b : SUCCESS in 49m 44s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a584bd844eef4d339a3693f0d745c4ac : SUCCESS in 50m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3ab49d4d3f594f0a88fae2d97d7459d6 : SUCCESS in 1h 35m 27s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f08cb4f3c1004a6d8e85f8be6d8ba838 : SUCCESS in 1h 22m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1fa275c583984c5187ccf63e8afeb769 : SUCCESS in 1h 42m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/824644e965de4d56b4a36b90ef64d387 : SUCCESS in 1h 02m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b2b168a56c1645c3b3b5e77df70030b2 : SUCCESS in 26m 01s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88560bdc69824af397f5586253fae3ad : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e24978d33245b090b0cf514f5060c6 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba21b6c4bbbd4f72ab8f75a05a5db704 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6924e1e4a554836a6bce3a18ef6e7f8 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/271065ab0ffe40468cdc4175598ba8c2 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d7714d7c7b46658f6fcf8e53836599 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/69e1a1f30f324da0af9bebfcddde7283 : SUCCESS in 26m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0b6de3b10464623abdfaf3cc42ac13b : SUCCESS in 37m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a80d80af15db4e109b38db61ade7eac9 : SUCCESS in 51m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f8e44a010efa43fd8b3841a3522ecfcf : SUCCESS in 59m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/38809712d9d14c1aa68e4072bc025d72 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":2},{"id":"7ce73335c5713a46434c73d644cdac57daf540ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 14:15:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"533f1d5c5df542a4007a306a6a7098282efec063","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:05:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/64e11c7523cc43c691fc63bcfe2221e1 : SUCCESS in 55m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e50a5a2f442848ecbd09390d49687637 : SUCCESS in 1h 36m 10s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/86e0f6bd0d324a0fbb2ed7c5b35843ba : SUCCESS in 1h 41m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/879d2bc4645842abb542b5f3afa335e7 : SUCCESS in 1h 07m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f44872ebff34bff95bc53e4fe0f6c0c : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/089585f2cd77405c947bf77e6e5aa6b9 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4efa8b62419a49689e6887c28fadf322 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a65d8cc86b584b1b92b4315c4344ea3d : SUCCESS in 2m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a46670219094d0cb1bca5a92f46d7b6 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b2952f39e84c029aa310eab8d97f71 : SUCCESS in 11m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/398b74126fe2440ba51504374e057571 : SUCCESS in 39m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0cb2b0b0841492dbc0faab759e666fb : SUCCESS in 1h 03m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7f4a279ad6f64e948b1a44525e60a8eb : SUCCESS in 1h 10m 09s","accounts_in_message":[],"_revision_number":2},{"id":"49cfada7faccff0eb892fe1f6e4ccd1e8635b71b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:05:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"82e9b708a91b16e1fcb527fa7617edca352c6534","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-20 16:07:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69cc04e88fee4075b61d4ea083be5446 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6077dde145d23eba5aa9366f4e63dd4a44fff22b","revisions":{"33dbebba7914a02e919c7e6a04d57f077d70e633":{"kind":"REWORK","_number":1,"created":"2019-07-29 15:19:43.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/11/673311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/673311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/673311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/673311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/673311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/673311/1"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2019-07-29 15:15:38.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2019-07-29 15:17:33.000000000","tz":-240},"subject":"[docker/openstack-exporter] Bump to 0.2.1 version","message":"[docker/openstack-exporter] Bump to 0.2.1 version\n\nBump the default version to 0.2.1 of the\nopenstack-exporter.\n\nChange-Id: I6e893c25a9dfbbbac6bd69f2fa26e6c72a4154d7\nSigned-off-by: Jorge Niedbalski \u003cjnr@metaklass.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33dbebba7914a02e919c7e6a04d57f077d70e633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33dbebba7914a02e919c7e6a04d57f077d70e633"}]},"branch":"refs/heads/master"},"6077dde145d23eba5aa9366f4e63dd4a44fff22b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-05 22:25:06.000000000","uploader":{"_account_id":2276,"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","username":"niedbalski"},"ref":"refs/changes/11/673311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/673311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/673311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/673311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/673311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/673311/2"}}},"commit":{"parents":[{"commit":"f95a009d2bf11748e20ee695091ea52260428c99","subject":"Merge \"Add HAcluster containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f95a009d2bf11748e20ee695091ea52260428c99"}]}],"author":{"name":"Jorge Niedbalski","email":"jnr@metaklass.org","date":"2019-07-29 15:15:38.000000000","tz":-240},"committer":{"name":"Jorge Niedbalski","email":"jorge.niedbalski@canonical.com","date":"2019-08-05 22:25:06.000000000","tz":0},"subject":"[docker/openstack-exporter] Bump to 0.2.1 version","message":"[docker/openstack-exporter] Bump to 0.2.1 version\n\nBump the default version to 0.2.1 of the\nopenstack-exporter.\n\nChange-Id: I6e893c25a9dfbbbac6bd69f2fa26e6c72a4154d7\nSigned-off-by: Jorge Niedbalski \u003cjnr@metaklass.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6077dde145d23eba5aa9366f4e63dd4a44fff22b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6077dde145d23eba5aa9366f4e63dd4a44fff22b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
