)]}'
{"id":"openstack%2Fkolla~673489","triplet_id":"openstack%2Fkolla~master~I9b0a162513ed436930c4541b758fc9cb2ef97e96","project":"openstack/kolla","branch":"master","topic":"monasca-prometheus","hashtags":[],"change_id":"I9b0a162513ed436930c4541b758fc9cb2ef97e96","subject":"Add prometheus_client to monasca-agent image","status":"MERGED","created":"2019-07-30 12:09:08.000000000","updated":"2019-07-30 16:49:18.000000000","submitted":"2019-07-30 16:49:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673489-1564505358179-cc82055b","meta_rev_id":"d8301942f6248226a6b148077286f53085fb7a59","_number":673489,"virtual_id_number":673489,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-07-30 16:49:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-30 12:32:26.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"date":"2019-07-30 13:25:10.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-07-30 15:54:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-07-30 15:56:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-07-30 15:39:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-07-30 15:54:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 13:25:10.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-30 15:39:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-07-30 15:54:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-07-30 15:56:23.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-07-30 16:49:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cda993fc91bb72b9514ff4e234a1bb5c2be49211","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-30 12:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"273e9b9bb5dcec7d1b0fd4e1d42cb228ecd17c14","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-30 12:14:43.000000000","message":"Patch Set 1: Code-Review+1\n\nsounds reasonable\n\nhow is it done now though?","accounts_in_message":[],"_revision_number":1},{"id":"b6243619d18a0076b3f065b8bd1a183e6fce4651","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-30 12:17:21.000000000","message":"Patch Set 1:\n\nWhat do you mean by \"done now\"? At the moment prometheus_client isn\u0027t installed, which prevents the use of the Monasca agent Prometheus plugin without customising the image.","accounts_in_message":[],"_revision_number":1},{"id":"16cc9df92b645275a17c9c0f7bca61ff1337f482","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-30 12:25:57.000000000","message":"Patch Set 1:\n\n\u003e What do you mean by \"done now\"? At the moment prometheus_client\n \u003e isn\u0027t installed, which prevents the use of the Monasca agent\n \u003e Prometheus plugin without customising the image.\n\nIt was re: \"given the level of integration of Prometheus in Kolla\".","accounts_in_message":[],"_revision_number":1},{"id":"63bbff8f8fa2293f56cef9661ad572e45d7ff3aa","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-30 12:31:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6bdc03d9a3c76fe67c4f8795644729adefcc8398","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-07-30 12:32:26.000000000","message":"Patch Set 2:\n\n\u003e \u003e What do you mean by \"done now\"? At the moment prometheus_client\n \u003e \u003e isn\u0027t installed, which prevents the use of the Monasca agent\n \u003e \u003e Prometheus plugin without customising the image.\n \u003e \n \u003e It was re: \"given the level of integration of Prometheus in Kolla\".\n\nI\u0027ve reworded the commit message, I hope it provides a better explanation.","accounts_in_message":[],"_revision_number":2},{"id":"bdf6629e6c3c8b1c5168a740e039589df0b6dc9a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-30 13:25:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-fedora-28-master http://logs.rdoproject.org/89/673489/2/openstack-check/tripleo-build-containers-fedora-28-master/bb02c5b/ : FAILURE in 5m 42s (non-voting)\n- tripleo-build-containers-fedora-28-stein http://logs.rdoproject.org/89/673489/2/openstack-check/tripleo-build-containers-fedora-28-stein/bd88ca8/ : SUCCESS in 49m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f4415e135425df3cd157e2c1c082a233bd861bc8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 14:06:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary http://logs.openstack.org/89/673489/2/check/kolla-build-centos-binary/f892826/ : SUCCESS in 46m 01s (non-voting)\n- kolla-build-centos-source http://logs.openstack.org/89/673489/2/check/kolla-build-centos-source/a5f5f2d/ : SUCCESS in 56m 09s\n- kolla-ansible-centos-source http://logs.openstack.org/89/673489/2/check/kolla-ansible-centos-source/bd60471/ : SUCCESS in 1h 26m 59s\n- kolla-ansible-centos-binary http://logs.openstack.org/89/673489/2/check/kolla-ansible-centos-binary/5743c22/ : SUCCESS in 1h 16m 22s (non-voting)\n- kolla-ansible-centos-source-upgrade http://logs.openstack.org/89/673489/2/check/kolla-ansible-centos-source-upgrade/d3e4e12/ : SUCCESS in 1h 22m 27s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/89/673489/2/check/tripleo-build-containers-centos-7/c3bea5c/ : SUCCESS in 43m 31s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/673489/2/check/openstack-tox-lower-constraints/efa9ede/ : SUCCESS in 6m 36s\n- openstack-tox-pep8 http://logs.openstack.org/89/673489/2/check/openstack-tox-pep8/59a0585/ : SUCCESS in 5m 03s\n- openstack-tox-py27 http://logs.openstack.org/89/673489/2/check/openstack-tox-py27/720d79b/ : SUCCESS in 4m 39s\n- openstack-tox-py36 http://logs.openstack.org/89/673489/2/check/openstack-tox-py36/040a827/ : SUCCESS in 4m 55s\n- openstack-tox-py37 http://logs.openstack.org/89/673489/2/check/openstack-tox-py37/7a44059/ : SUCCESS in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/89/673489/2/check/openstack-tox-docs/d034e8b/html/ : SUCCESS in 4m 24s\n- kolla-build-ubuntu-binary http://logs.openstack.org/89/673489/2/check/kolla-build-ubuntu-binary/ba9f6bb/ : SUCCESS in 24m 07s (non-voting)\n- kolla-build-ubuntu-source http://logs.openstack.org/89/673489/2/check/kolla-build-ubuntu-source/0a845b2/ : SUCCESS in 31m 22s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/89/673489/2/check/kolla-ansible-ubuntu-source/df3d23d/ : SUCCESS in 48m 41s\n- kolla-ansible-ubuntu-binary http://logs.openstack.org/89/673489/2/check/kolla-ansible-ubuntu-binary/eb00e1e/ : SUCCESS in 46m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade http://logs.openstack.org/89/673489/2/check/kolla-ansible-ubuntu-source-upgrade/87af811/ : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":2},{"id":"8eda84a0be0d50bd901b04ab3de72a786a6a63c6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-07-30 15:39:26.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks, Pierre. For someone less knowledgeable in this matter (me) this reads much better.","accounts_in_message":[],"_revision_number":2},{"id":"c4eaa3f7bbbfe75fd674cddb5d609347e5be5b10","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-07-30 15:54:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b8e0125a45a78104ad6a583c3d0603673ee1d543","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 15:54:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cd64f94fe229658864f28e43a39c5fbc6653574b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-07-30 15:56:23.000000000","message":"Patch Set 2: Code-Review+1\n\nThis makes sense for the reasons in the commit message. The Monasca Agent provides an alternative via the extras section:`https://github.com/openstack/monasca-agent/blob/master/setup.cfg#L52`, however the pep for that has been withdrawn and it probably needs an overhaul.","accounts_in_message":[],"_revision_number":2},{"id":"cad003dbcbe3f3f851e8cedb87dc8a48238e27c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 16:49:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source http://logs.openstack.org/89/673489/2/gate/kolla-build-centos-source/07a0ade/ : SUCCESS in 49m 50s\n- tripleo-build-containers-centos-7 http://logs.openstack.org/89/673489/2/gate/tripleo-build-containers-centos-7/d9eea48/ : SUCCESS in 48m 20s\n- openstack-tox-lower-constraints http://logs.openstack.org/89/673489/2/gate/openstack-tox-lower-constraints/7d24096/ : SUCCESS in 7m 10s\n- openstack-tox-pep8 http://logs.openstack.org/89/673489/2/gate/openstack-tox-pep8/17b2ae4/ : SUCCESS in 5m 50s\n- openstack-tox-py27 http://logs.openstack.org/89/673489/2/gate/openstack-tox-py27/c52b157/ : SUCCESS in 4m 36s\n- openstack-tox-py36 http://logs.openstack.org/89/673489/2/gate/openstack-tox-py36/7d7286d/ : SUCCESS in 5m 00s\n- openstack-tox-py37 http://logs.openstack.org/89/673489/2/gate/openstack-tox-py37/bc3dd3a/ : SUCCESS in 4m 42s\n- openstack-tox-docs http://logs.openstack.org/89/673489/2/gate/openstack-tox-docs/b9766ef/html/ : SUCCESS in 5m 11s\n- kolla-build-ubuntu-source http://logs.openstack.org/89/673489/2/gate/kolla-build-ubuntu-source/89e0b39/ : SUCCESS in 34m 37s","accounts_in_message":[],"_revision_number":2},{"id":"34a5994279967101a3a9a06ab590249109f5b426","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 16:49:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a74aa1bd21b04eceede6cf425a5e56dcacec1412","revisions":{"dccba176e0225c234508ee3d03a1b9fd5328ba85":{"kind":"REWORK","_number":1,"created":"2019-07-30 12:09:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/673489/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/673489/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/673489/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/673489/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/673489/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/673489/1"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-22 11:59:22.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-30 12:08:32.000000000","tz":120},"subject":"Add prometheus_client to monasca-agent image","message":"Add prometheus_client to monasca-agent image\n\nThe prometheus_client package is required to use the Prometheus plugin.\nIt makes sense to include this dependency by default given the level of\nintegration of Prometheus in Kolla. The increase in size of the\ncontainer image is negligible.\n\nChange-Id: I9b0a162513ed436930c4541b758fc9cb2ef97e96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dccba176e0225c234508ee3d03a1b9fd5328ba85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dccba176e0225c234508ee3d03a1b9fd5328ba85"}]},"branch":"refs/heads/master"},"a74aa1bd21b04eceede6cf425a5e56dcacec1412":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-30 12:31:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/673489/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/89/673489/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/673489/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/673489/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/89/673489/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/89/673489/2"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-22 11:59:22.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-30 12:31:53.000000000","tz":0},"subject":"Add prometheus_client to monasca-agent image","message":"Add prometheus_client to monasca-agent image\n\nThe prometheus_client package is required to use the Prometheus plugin.\nIt makes sense to include this dependency by default since Kolla Ansible\ncan deploy many Prometheus exporters, making it easy to scrape their\nmetrics and store them in Monasca. The increase in size of the container\nimage is negligible.\n\nChange-Id: I9b0a162513ed436930c4541b758fc9cb2ef97e96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a74aa1bd21b04eceede6cf425a5e56dcacec1412"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a74aa1bd21b04eceede6cf425a5e56dcacec1412"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
