)]}'
{"id":"openstack%2Fkolla~674088","triplet_id":"openstack%2Fkolla~master~I1822a8c610db72654bbacc7148e25a837090b680","project":"openstack/kolla","branch":"master","topic":"fix-ceph-mds-cap","hashtags":[],"change_id":"I1822a8c610db72654bbacc7148e25a837090b680","subject":"ceph mds cap \u0027allow *\u0027 in extend_start","status":"MERGED","created":"2019-08-01 16:36:41.000000000","updated":"2019-08-02 08:28:43.000000000","submitted":"2019-08-02 08:12:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674088-1564733538579-d227193f","meta_rev_id":"2858498df4421dbfc54a31285ef693cf278bcf51","_number":674088,"virtual_id_number":674088,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-02 08:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-01 17:14:42.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-08-02 07:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-08-01 17:58:36.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-08-02 07:03:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-08-01 17:13:18.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-01 17:14:42.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-01 17:58:36.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-02 07:03:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-08-02 08:12:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42df3c250f192f5d639c4933476591fd48125d2d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-01 16:36:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da559a7e27a300683ba3ddda2b4115725c408c0f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-01 17:13:18.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"b24cb10a3d3c870dd5a75563df38b3d6f9f3cd90","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-01 17:14:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/88/674088/1/openstack-check/tripleo-build-containers-rhel-8-master/224f773/ : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":1},{"id":"185de7d441907795d4f202de5bff2577bbea075f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-01 17:58:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce7c80664752315541ec8f2ba431278882dd7cab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-01 18:35:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://logs.opendev.org/88/674088/1/check/kolla-build-centos-binary/266867c/ : SUCCESS in 44m 13s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/88/674088/1/check/kolla-build-centos-source/e48e4ea/ : SUCCESS in 55m 30s\n- kolla-ansible-centos-source https://logs.opendev.org/88/674088/1/check/kolla-ansible-centos-source/6b1c04a/ : SUCCESS in 1h 18m 49s\n- kolla-ansible-centos-binary https://logs.opendev.org/88/674088/1/check/kolla-ansible-centos-binary/eb39ec8/ : SUCCESS in 1h 04m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/88/674088/1/check/kolla-ansible-centos-source-upgrade/d4637e5/ : SUCCESS in 1h 14m 14s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/88/674088/1/check/tripleo-build-containers-centos-7/6bc90a0/ : SUCCESS in 50m 53s\n- kolla-ansible-centos-source-ceph https://logs.opendev.org/88/674088/1/check/kolla-ansible-centos-source-ceph/110b840/ : FAILURE in 1h 46m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://logs.opendev.org/88/674088/1/check/kolla-ansible-centos-source-upgrade-ceph/488f423/ : FAILURE in 1h 50m 06s (non-voting)\n- openstack-tox-lower-constraints https://logs.opendev.org/88/674088/1/check/openstack-tox-lower-constraints/bd76361/ : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://logs.opendev.org/88/674088/1/check/openstack-tox-pep8/dae4f62/ : SUCCESS in 5m 15s\n- openstack-tox-py27 https://logs.opendev.org/88/674088/1/check/openstack-tox-py27/bb3e196/ : SUCCESS in 3m 36s\n- openstack-tox-py36 https://logs.opendev.org/88/674088/1/check/openstack-tox-py36/02e3001/ : SUCCESS in 3m 58s\n- openstack-tox-py37 https://logs.opendev.org/88/674088/1/check/openstack-tox-py37/ccb6c4f/ : SUCCESS in 4m 05s\n- openstack-tox-docs https://logs.opendev.org/88/674088/1/check/openstack-tox-docs/bde8c33/html/ : SUCCESS in 4m 26s\n- kolla-build-ubuntu-binary https://logs.opendev.org/88/674088/1/check/kolla-build-ubuntu-binary/489bf7c/ : SUCCESS in 27m 20s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/88/674088/1/check/kolla-build-ubuntu-source/36b3462/ : SUCCESS in 30m 41s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/88/674088/1/check/kolla-ansible-ubuntu-source/20c9ef7/ : SUCCESS in 52m 22s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/88/674088/1/check/kolla-ansible-ubuntu-binary/39e283a/ : SUCCESS in 48m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/88/674088/1/check/kolla-ansible-ubuntu-source-upgrade/b19d0ab/ : SUCCESS in 1h 08m 45s\n- kolla-ansible-ubuntu-source-ceph https://logs.opendev.org/88/674088/1/check/kolla-ansible-ubuntu-source-ceph/f748508/ : SUCCESS in 1h 21m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://logs.opendev.org/88/674088/1/check/kolla-ansible-ubuntu-source-upgrade-ceph/3cac3e2/ : SUCCESS in 1h 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2978d372de5d5ba788f4ef5a656df51427e2cf82","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 07:03:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0be3f9a5bf5eccc15105d7e7e4d6a3c38e4494b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 07:04:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ebfec0ce71a50a448e703d356ad4a48b8d744f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 08:12:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://logs.opendev.org/88/674088/1/gate/kolla-build-centos-source/d0d8263/ : SUCCESS in 51m 18s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/88/674088/1/gate/tripleo-build-containers-centos-7/d336154/ : SUCCESS in 47m 35s\n- openstack-tox-lower-constraints https://logs.opendev.org/88/674088/1/gate/openstack-tox-lower-constraints/f2301f8/ : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://logs.opendev.org/88/674088/1/gate/openstack-tox-pep8/6d4b576/ : SUCCESS in 4m 52s\n- openstack-tox-py27 https://logs.opendev.org/88/674088/1/gate/openstack-tox-py27/3ec0385/ : SUCCESS in 5m 58s\n- openstack-tox-py36 https://logs.opendev.org/88/674088/1/gate/openstack-tox-py36/fd62599/ : SUCCESS in 3m 05s\n- openstack-tox-py37 https://logs.opendev.org/88/674088/1/gate/openstack-tox-py37/6937a77/ : SUCCESS in 4m 45s\n- openstack-tox-docs https://logs.opendev.org/88/674088/1/gate/openstack-tox-docs/8cb8ec8/html/ : SUCCESS in 3m 27s\n- kolla-build-ubuntu-source https://logs.opendev.org/88/674088/1/gate/kolla-build-ubuntu-source/4d0ce4c/ : SUCCESS in 26m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c7b6f1afb2f0befb3327a57960266b85d94b5d6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 08:12:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d2fd46922669ef018a4cea34f5b206f3caa58b7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-02 08:28:43.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 2f45d6810e341cd89a20d5fceff0179cde768acc","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1d4ed51f41721c86a33d766fdfd80aaccf1df60","revisions":{"b1d4ed51f41721c86a33d766fdfd80aaccf1df60":{"kind":"REWORK","_number":1,"created":"2019-08-01 16:36:41.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/88/674088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/674088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/674088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/674088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/674088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/674088/1"}}},"commit":{"parents":[{"commit":"ae12359b83c514473793a9a42281a786d553b64f","subject":"Merge \"There is no Java 8 in Debian \u0027buster\u0027 (part 3)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae12359b83c514473793a9a42281a786d553b64f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-01 16:34:24.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-01 16:34:24.000000000","tz":120},"subject":"ceph mds cap \u0027allow *\u0027 in extend_start","message":"ceph mds cap \u0027allow *\u0027 in extend_start\n\nThis applies the fix made in kolla-ansible (see [1])\nto kolla ceph mon image directly.\n\n[1] https://review.opendev.org/670035\n\nChange-Id: I1822a8c610db72654bbacc7148e25a837090b680\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1d4ed51f41721c86a33d766fdfd80aaccf1df60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b1d4ed51f41721c86a33d766fdfd80aaccf1df60"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
