)]}'
{"id":"openstack%2Fkolla~674222","triplet_id":"openstack%2Fkolla~stable%2Fstein~I1822a8c610db72654bbacc7148e25a837090b680","project":"openstack/kolla","branch":"stable/stein","topic":"fix-ceph-mds-cap-stable/stein","hashtags":[],"change_id":"I1822a8c610db72654bbacc7148e25a837090b680","subject":"ceph mds cap \u0027allow *\u0027 in extend_start","status":"MERGED","created":"2019-08-02 08:28:43.000000000","updated":"2019-08-02 11:04:38.000000000","submitted":"2019-08-02 11:04:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674222-1564743878234-c33df9d0","meta_rev_id":"5a7e0a955882d1ea4e14cf495eea24fe641bcdd5","_number":674222,"virtual_id_number":674222,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-02 11:04:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-02 08:48:18.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-08-02 08:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-08-02 08:34:17.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-08-02 08:34:17.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-08-02 08:28:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 08:30:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-08-02 08:34:17.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-02 08:48:18.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-02 11:04:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25ff6cb165fc8b18461990807a055c37403346ed","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-02 08:28:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0a1817e7a90fd8ddeac31ba907f9a55f8a9f2e35","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-02 08:28:47.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"9b7cf66ff9da7dabd5cc6941f6222ff8440b6ab5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-02 08:30:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d6d4de97eb655c4daa80daf80eb2ced2d6418ae3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-02 08:34:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9eceb4614831df9c0c7f33909f86ae13e6a904d3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 08:48:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/22/674222/1/openstack-check/tripleo-build-containers-rhel-8-master/fb6b2cb/ : SUCCESS in 18m 47s","accounts_in_message":[],"_revision_number":1},{"id":"834da8ee2821ac882aded2a86a21eced644e0507","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 10:08:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://logs.opendev.org/22/674222/1/check/kolla-build-centos-binary/aec5ec4/ : SUCCESS in 45m 51s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/22/674222/1/check/kolla-build-centos-source/6804b5c/ : SUCCESS in 39m 54s\n- kolla-ansible-centos-source https://logs.opendev.org/22/674222/1/check/kolla-ansible-centos-source/02a8ee7/ : SUCCESS in 1h 09m 12s\n- kolla-ansible-centos-binary https://logs.opendev.org/22/674222/1/check/kolla-ansible-centos-binary/23e6840/ : SUCCESS in 1h 08m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/22/674222/1/check/kolla-ansible-centos-source-upgrade/ec390e1/ : SUCCESS in 1h 36m 50s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/22/674222/1/check/tripleo-build-containers-centos-7/ee17e8f/ : SUCCESS in 43m 44s\n- kolla-ansible-centos-source-ceph https://logs.opendev.org/22/674222/1/check/kolla-ansible-centos-source-ceph/8c20002/ : FAILURE in 1h 15m 19s (non-voting)\n- kolla-build-oraclelinux-binary https://logs.opendev.org/22/674222/1/check/kolla-build-oraclelinux-binary/a0530ef/ : SUCCESS in 47m 46s (non-voting)\n- kolla-build-oraclelinux-source https://logs.opendev.org/22/674222/1/check/kolla-build-oraclelinux-source/9c5b7e9/ : SUCCESS in 42m 06s\n- openstack-tox-lower-constraints https://logs.opendev.org/22/674222/1/check/openstack-tox-lower-constraints/e43ed41/ : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://logs.opendev.org/22/674222/1/check/openstack-tox-pep8/67abf27/ : SUCCESS in 4m 09s\n- openstack-tox-py27 https://logs.opendev.org/22/674222/1/check/openstack-tox-py27/276c032/ : SUCCESS in 5m 14s\n- openstack-tox-py36 https://logs.opendev.org/22/674222/1/check/openstack-tox-py36/52f52c7/ : SUCCESS in 4m 22s\n- openstack-tox-py37 https://logs.opendev.org/22/674222/1/check/openstack-tox-py37/a8b4168/ : SUCCESS in 4m 55s\n- openstack-tox-docs https://logs.opendev.org/22/674222/1/check/openstack-tox-docs/4385e7b/html/ : SUCCESS in 3m 19s\n- kolla-build-ubuntu-binary https://logs.opendev.org/22/674222/1/check/kolla-build-ubuntu-binary/1413436/ : SUCCESS in 25m 44s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/22/674222/1/check/kolla-build-ubuntu-source/32760da/ : SUCCESS in 26m 58s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/22/674222/1/check/kolla-ansible-ubuntu-source/79ec42a/ : SUCCESS in 59m 11s\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/22/674222/1/check/kolla-ansible-ubuntu-binary/2f54342/ : SUCCESS in 45m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/22/674222/1/check/kolla-ansible-ubuntu-source-upgrade/f342eae/ : SUCCESS in 1h 05m 04s\n- kolla-ansible-ubuntu-source-ceph https://logs.opendev.org/22/674222/1/check/kolla-ansible-ubuntu-source-ceph/b521d03/ : SUCCESS in 1h 07m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25ab734ab3a6210f0f10cd1bba8854385030e6c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 10:08:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"728de09db588d63cd9b8fa08a4351eae61feec6b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 11:04:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://logs.opendev.org/22/674222/1/gate/kolla-build-centos-source/435d0c9/ : SUCCESS in 40m 41s\n- tripleo-build-containers-centos-7 https://logs.opendev.org/22/674222/1/gate/tripleo-build-containers-centos-7/9dbd87b/ : SUCCESS in 50m 17s\n- kolla-build-oraclelinux-source https://logs.opendev.org/22/674222/1/gate/kolla-build-oraclelinux-source/73cdbd3/ : SUCCESS in 45m 08s\n- openstack-tox-lower-constraints https://logs.opendev.org/22/674222/1/gate/openstack-tox-lower-constraints/ddfe823/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://logs.opendev.org/22/674222/1/gate/openstack-tox-pep8/7dbc425/ : SUCCESS in 4m 30s\n- openstack-tox-py27 https://logs.opendev.org/22/674222/1/gate/openstack-tox-py27/6c03bda/ : SUCCESS in 3m 31s\n- openstack-tox-py36 https://logs.opendev.org/22/674222/1/gate/openstack-tox-py36/eeab0d4/ : SUCCESS in 6m 16s\n- openstack-tox-py37 https://logs.opendev.org/22/674222/1/gate/openstack-tox-py37/4a6707c/ : SUCCESS in 5m 14s\n- openstack-tox-docs https://logs.opendev.org/22/674222/1/gate/openstack-tox-docs/3e77765/html/ : SUCCESS in 4m 18s\n- kolla-build-ubuntu-source https://logs.opendev.org/22/674222/1/gate/kolla-build-ubuntu-source/9492932/ : SUCCESS in 33m 51s","accounts_in_message":[],"_revision_number":1},{"id":"741302cc8bccea3d137f846418df1dc7d395f6fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 11:04:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f45d6810e341cd89a20d5fceff0179cde768acc","revisions":{"2f45d6810e341cd89a20d5fceff0179cde768acc":{"kind":"REWORK","_number":1,"created":"2019-08-02 08:28:43.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/22/674222/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/674222/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/674222/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/674222/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/674222/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/674222/1"}}},"commit":{"parents":[{"commit":"9ef040a85d0b7bf4f9f2b3145d648aca75f12912","subject":"Merge \"nova-libvirt: in Debian qemu-kvm is x86 only\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ef040a85d0b7bf4f9f2b3145d648aca75f12912"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-01 16:34:24.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-02 08:28:43.000000000","tz":0},"subject":"ceph mds cap \u0027allow *\u0027 in extend_start","message":"ceph mds cap \u0027allow *\u0027 in extend_start\n\nThis applies the fix made in kolla-ansible (see [1])\nto kolla ceph mon image directly.\n\n[1] https://review.opendev.org/670035\n\nChange-Id: I1822a8c610db72654bbacc7148e25a837090b680\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit b1d4ed51f41721c86a33d766fdfd80aaccf1df60)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f45d6810e341cd89a20d5fceff0179cde768acc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f45d6810e341cd89a20d5fceff0179cde768acc"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
