)]}'
{"id":"openstack%2Fkolla~674266","triplet_id":"openstack%2Fkolla~stable%2Fqueens~I05433f5bb2313e2adff39850a970d4430340c852","project":"openstack/kolla","branch":"stable/queens","hashtags":[],"change_id":"I05433f5bb2313e2adff39850a970d4430340c852","subject":"Update build profiles for Placement split","status":"ABANDONED","created":"2019-08-02 11:05:41.000000000","updated":"2019-08-04 07:39:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9ec7c77b7eb990bc1ec252713ff4f9015f50a759","_number":674266,"virtual_id_number":674266,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-08-02 12:26:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-08-02 11:17:39.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":-1,"date":"2019-08-03 15:51:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-08-02 11:05:41.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-02 11:05:41.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-02 11:17:39.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-02 12:26:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-03 15:51:06.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"af9aceda89a91b2f284f5fdbf108d778c0db140e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-02 11:05:41.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"de382277867a662a795c14d1545d65742b615c4c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 11:17:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/66/674266/1/openstack-check/tripleo-build-containers-rhel-8-master/83e1b2f/ : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"ee81076253dd2eae353a33d986c3f223bfd67330","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 12:26:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://logs.opendev.org/66/674266/1/check/kolla-build-centos-binary/dcfae88/ : SUCCESS in 47m 43s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/66/674266/1/check/kolla-build-centos-source/d39461f/ : SUCCESS in 49m 45s\n- kolla-ansible-centos-source https://logs.opendev.org/66/674266/1/check/kolla-ansible-centos-source/ff7029e/ : SUCCESS in 1h 13m 37s (non-voting)\n- kolla-ansible-centos-binary https://logs.opendev.org/66/674266/1/check/kolla-ansible-centos-binary/66af29b/ : SUCCESS in 1h 01m 56s (non-voting)\n- kolla-build-oraclelinux-binary https://logs.opendev.org/66/674266/1/check/kolla-build-oraclelinux-binary/927b5a9/ : SUCCESS in 36m 17s (non-voting)\n- kolla-build-oraclelinux-source https://logs.opendev.org/66/674266/1/check/kolla-build-oraclelinux-source/bbcb45a/ : SUCCESS in 46m 12s\n- kolla-ansible-oraclelinux-source https://logs.opendev.org/66/674266/1/check/kolla-ansible-oraclelinux-source/2ac9d8d/ : SUCCESS in 1h 02m 04s (non-voting)\n- kolla-ansible-oraclelinux-binary https://logs.opendev.org/66/674266/1/check/kolla-ansible-oraclelinux-binary/77118f2/ : SUCCESS in 1h 09m 09s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/66/674266/1/check/openstack-tox-pep8/9edad44/ : SUCCESS in 5m 18s\n- openstack-tox-py27 https://logs.opendev.org/66/674266/1/check/openstack-tox-py27/61f65b6/ : SUCCESS in 5m 00s\n- openstack-tox-py35 https://logs.opendev.org/66/674266/1/check/openstack-tox-py35/9073287/ : SUCCESS in 4m 57s\n- build-openstack-sphinx-docs https://logs.opendev.org/66/674266/1/check/build-openstack-sphinx-docs/c6caf7e/html/ : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary https://logs.opendev.org/66/674266/1/check/kolla-build-ubuntu-binary/dba4b38/ : SUCCESS in 22m 47s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/66/674266/1/check/kolla-build-ubuntu-source/fb9a714/ : SUCCESS in 30m 22s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/66/674266/1/check/kolla-ansible-ubuntu-source/5146f68/ : SUCCESS in 58m 14s (non-voting)\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/66/674266/1/check/kolla-ansible-ubuntu-binary/a973f84/ : SUCCESS in 1h 00m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0100e8642c702265758b7e983164c070cd10b6e1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-03 15:51:06.000000000","message":"Patch Set 1: Code-Review-1\n\nPlacement split happened in stein - not necessary in queens.","accounts_in_message":[],"_revision_number":1},{"id":"9ec7c77b7eb990bc1ec252713ff4f9015f50a759","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-04 07:39:57.000000000","message":"Abandoned\n\nyeah, sorry, thanks for noticing","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1194069ee6359e6aaea8ed1a0fb52c203188e3f5","revisions":{"1194069ee6359e6aaea8ed1a0fb52c203188e3f5":{"kind":"REWORK","_number":1,"created":"2019-08-02 11:05:41.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/66/674266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/674266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/674266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/674266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/674266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/674266/1"}}},"commit":{"parents":[{"commit":"69a7cb5e9061abb474e9ffb9202cbf67c6114f6f","subject":"Fix deployment of ceph-nfs container failing due to wrong USER","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69a7cb5e9061abb474e9ffb9202cbf67c6114f6f"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-07-30 16:15:31.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-02 11:05:41.000000000","tz":0},"subject":"Update build profiles for Placement split","message":"Update build profiles for Placement split\n\nThe Placement API is now a separate project, but anyone\nbuilding Nova is likely to want it. This change ensures\nthat it is built alongside Nova.\n\nChange-Id: I05433f5bb2313e2adff39850a970d4430340c852\n(cherry picked from commit d4e89eb932d660b4e610460b3ec9792a3e08091c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1194069ee6359e6aaea8ed1a0fb52c203188e3f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1194069ee6359e6aaea8ed1a0fb52c203188e3f5"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
