)]}'
{"id":"openstack%2Fkolla~674510","triplet_id":"openstack%2Fkolla~master~I2555c640125f3fede1dd1715d70c0c36620f3d4c","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I2555c640125f3fede1dd1715d70c0c36620f3d4c","subject":"Add kayobe to deliverables.yaml","status":"MERGED","created":"2019-08-04 18:15:21.000000000","updated":"2019-09-19 21:18:28.000000000","submitted":"2019-09-19 21:16:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674510-1568927804845-7126b0df","meta_rev_id":"b33510cc08b11e2e13500f97daa3c7041350fd7d","_number":674510,"virtual_id_number":674510,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"date":"2019-09-19 17:43:16.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-09-18 16:07:58.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-19 21:16:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"all":[{"value":2,"date":"2019-09-17 06:48:59.000000000","_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-09-17 07:03:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-09-17 07:03:28.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-17 06:48:59.000000000","updated_by":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"reviewer":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"state":"REVIEWER"},{"updated":"2019-09-17 07:03:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-19 17:43:16.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-09-19 21:16:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96c3ae1b9c6b43e295530b90a45213653830562a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-04 18:15:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de85dcaebca6e7a2221896c7f9b50578f84e52d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-04 18:15:37.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"55696e3fce3cca8e70ef66b6cd0fc82086b765fc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-16 16:36:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cc121a4dc8fb8a5bd0f23ad1a5886806d897d78d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 18:15:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e63c8553e7e54201b13cdaac7f629b84 : SUCCESS in 52m 59s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/283348d57b30403bba0ec13545bf2cc8 : SUCCESS in 55m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/77b652d9c3f84f9eb9a63b1db9115ce2 : SUCCESS in 1h 32m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f5afc4d61296426cbd655d2458da2221 : SUCCESS in 1h 21m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/05740338b16d494593ed5eabe2eced7d : SUCCESS in 1h 26m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/948e897d81bf4a128d22c0b9d50b2ec7 : SUCCESS in 28m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97df2f132cb94c14a3ddd68eb9db9523 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12d3c298924b480c86d75f6c264f8f59 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fb20e13b241249f69051b120a57fa5d0 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aba7d429c8784b3ba7a91872a6bfcc0b : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53a9fdf2a0eb45168a8aa01a6322178a : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ff7f36af244a2ea3084ec8cffb2057 : SUCCESS in 3m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f6b96ea31a8246c1b2962cf7b7ee5d3b : SUCCESS in 24m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/061d538503244754bb56d5b12d808e04 : SUCCESS in 34m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2b6dd8370be4ddf8ecd057f7e2c5c69 : SUCCESS in 54m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0e94d9f81c4041bab1d000af7237d51a : SUCCESS in 48m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4a9056173bd443fc8adc6a4ae4d48838 : SUCCESS in 1h 00m 34s","accounts_in_message":[],"_revision_number":2},{"id":"17d018d0850214c9121bac22a893cac2721e9b7c","author":{"_account_id":19316,"name":"Eduardo Gonzalez","email":"dabarren@gmail.com","username":"egonzalez90"},"date":"2019-09-17 06:48:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d0dd0185edfff9b00e9c23e12e9c03bbfe5c08c8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-17 07:03:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bba7ff6730c01108fc8e412c8b7892c828a38e93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 07:03:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c46e88f95e706f6beb9cc1008731a75ae63ba6c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 09:02:16.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9a86769d33144f7c98e2167fb32658c0 : FAILURE in 28m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/916457f5d9294af4915bbcfc43c61776 : SUCCESS in 1h 56m 55s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7fd9bcd394514f66a02a0a9265f1d4db : SUCCESS in 1h 22m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/534f3146e4704cdf81c0fad8723015a0 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/022b234606294aeaadfb6236bbf29a55 : SUCCESS in 5m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9e17c465f56d42139abd90943cf0f1ca : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f844c3fbadcd47259fc99d8284f1fa55 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed79d875aa194447a6f6bcb0652f5ba3 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7056aa983eb8417da222a8c02813c7d2 : SUCCESS in 3m 20s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/318f731b743c49cba3098b68d82983c6 : SUCCESS in 29m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/52834b26999c49819641ece45c0a1f58 : SUCCESS in 1h 33m 51s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ef1541695e3440b4bfb6343a1f3c23aa : SUCCESS in 53m 04s","accounts_in_message":[],"_revision_number":2},{"id":"e891c1c6c2239826f92c75bd3bc6a8bd05811c4d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-17 09:15:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"505c3845163a55503ad14fd9d8f1965b55473e5c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-17 10:42:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b9e7648dee484f78b1a66f02ab8dfb2f : SUCCESS in 47m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/5c3ff57863ac4957b77f42d91a72657b : FAILURE in 27m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a1a32bb44daa4a938e440c92b9e4fbaf : FAILURE in 30m 02s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ce429df858d44e4887de7513da007ede : SUCCESS in 1h 09m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c885f1af3323414a8bbe8bbfcd214066 : SUCCESS in 1h 21m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/49cd882c0ff447509891b5ed93f6462e : SUCCESS in 30m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b3c6df1d6634219909d9c7d8252df3a : RETRY_LIMIT in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a93cb62491c48d89f85a8d3de61e196 : RETRY_LIMIT in 1m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f00b91bce0674c29b38244f4e4fd19e4 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a75dc44d3b2d4eb88a91148bbc7ae769 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8d5fa8d08bd4c43922a471ba7e02acf : RETRY_LIMIT in 2m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32bcb03ec66f45399212879005c66924 : SUCCESS in 5m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b90d1bb37f74b3f93d3eb20f854550a : RETRY_LIMIT in 2m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8ce322899fb4aab87310ce8d2de6d77 : RETRY_LIMIT in 2m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/10b0e6a7c04e4983aae0dd210192f155 : RETRY_LIMIT in 1m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/045abbc04d3746adbd858b68f4607429 : SUCCESS in 54m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b6c8d3a8235d4e52b4b3ef23cff20ef5 : SUCCESS in 1h 12m 56s","accounts_in_message":[],"_revision_number":2},{"id":"09ca27a5455e1daf9364d88902bbcc41d94a683a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-18 16:07:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"799bfd1ea17ab5ea8951f4bf06f05ea82bb748e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 18:06:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/0b39a6c94b4c4aa68af08ba5301e74c2 : SUCCESS in 44m 32s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/78790d886f8a49539d4cdc746f000b9a : SUCCESS in 49m 25s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c457ea21e0364fd4898258b1edbc6073 : SUCCESS in 1h 12m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e16c1592627a4ada9405f6b62aa27b6e : SUCCESS in 1h 22m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7402be42edae4a398a195fb0ff12682f : SUCCESS in 1h 45m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/eded98c8b81c481b8539c9f8b4cb47b2 : SUCCESS in 26m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7293b1cd2c024e079a996e169c61dc3d : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/567fd8f9b2684ed8935468ada8271963 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/731712f245c5495faa0d11d0b7504e3e : SUCCESS in 3m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6248b373b2374659ac8be349f84bed85 : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be940e59fb624aba90005fbe10096b0c : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c522c4143344a298b50e503a33baf16 : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c029c04a492498384eb2cea0a19f96c : SUCCESS in 26m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ac8725c1c998442586ddca8a35dffe49 : SUCCESS in 32m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b54cc5f6f28c4597a58f7156246f6367 : FAILURE in 10m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fca7e1f7058c4e558f1cb26c34c67c07 : SUCCESS in 49m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e536c2c217e9475e82c904334e83802d : SUCCESS in 1h 05m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b0f7889c86f6ae4ff73465a9bbde2af663cb6b17","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-19 08:49:39.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"878afa8d07529fbb166565a4f34854e22430f683","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 11:59:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2ac48cb4319c44f7a0aecf5501fe100a : FAILURE in 47m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7459c52a37574075b3343b8ffa50924e : FAILURE in 56m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/816d3bbf68a64b7892b03575bfb01bb0 : SUCCESS in 1h 24m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/aeee2f4967fb4b09adfbad75366b7fd6 : SUCCESS in 1h 23m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/89ffb0bb2260468e8bd651d66ce159c7 : SUCCESS in 1h 41m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b0e1389853874debad8ea2b61241a6e6 : SUCCESS in 26m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffc04ef24822481bb4eff7ffc2383d27 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0fe122d4a684f25b956a58205e05822 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea24d7a0b1b34d57846b7ba3a89cd5b1 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7985065ab074e72ad2c0fa915ab9607 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7caade912edf4024bd09d5394279b13e : SUCCESS in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2427529cb0d14ac3a8e440118bd65f9f : SUCCESS in 7m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/211bb916d7d94893a13815957914790c : SUCCESS in 31m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8cfb66ea17344dca9c34c0e6f7c493d1 : SUCCESS in 42m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c57d6df05494791a58a2a93a507e66e : SUCCESS in 1h 05m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7b806c8660494782b388f8c36b3225ae : SUCCESS in 57m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a41dd06a8dc415485eecc4e8e04fbc8 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":2},{"id":"7e6b2a66aeec2002df31b15af26dceb281dd23b8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-19 17:43:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"991676f3faa7d9541befa729c3e75c4db24e910f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-19 17:44:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7179ed757241b083721e8a3a81fb8ff8a2beb32f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:36:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f0bb26b7e05b455b960f4f5ed045b345 : SUCCESS in 57m 15s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d247a01f409648c4a47c73d6b83713ef : SUCCESS in 1h 03m 41s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3da26c20c05e4d3180d55d41b419d00e : SUCCESS in 1h 35m 10s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2fdb399f67f34bab822d3704aa0ac009 : SUCCESS in 1h 20m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6ebcfe8fe71046a7833cf4285d375cba : SUCCESS in 1h 27m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f88a042712ae490689fbd36f55eec11f : SUCCESS in 33m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c025712891844c7adabc6262240161f : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93b8062bf40f4657a4d2e9bdc7006ab9 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5df78678ef4145cdb619645ea05744fd : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a138d4c5df9d4faeaf13377b1ce5e608 : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7168de6037a45018ea3cebdafe6178a : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac2f134731b641578819b0052001a5f3 : SUCCESS in 7m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f3cbc126ecea4a388515e009b5c40fec : SUCCESS in 30m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5f63af1674044a6cad96ba8ca8f85762 : SUCCESS in 31m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2978989461d245b49803e904f110274c : SUCCESS in 46m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d4f753eb1d4645c988047c5675d4d61f : SUCCESS in 53m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8c73f39b6e824a039980f333bf70d76c : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d4ec5be95205220effa125d8eadfeca999479168","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 19:36:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9d2eeb3518ab9295bb8fe00a8a27ebd550fa3337","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 21:16:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"78c935998d1b1971c9d0620ec14cd05af86cb21a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 21:16:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b6ef4baa45fe4a60b4c631e6d1189d1f : SUCCESS in 1h 03m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5ce9dd0eb57d477abdb2f00dae2dd65a : SUCCESS in 1h 22m 51s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/608aa1d3e68e44ceaec7bd9345b82eda : SUCCESS in 1h 34m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9da17f4028f41c69102edfe48951264 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2303a295c4341f8a53e0b3f7671a104 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6df5de3a08a04cb19b98f90084873675 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ca27902e24d481cb6a36783a94da3d4 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2d48343ede64130b1be10ae421f330f : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94b53b09b8364c8095f29d2ad5ec4ebc : SUCCESS in 4m 24s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2764bdddc224811b78bffe3e0a0172e : SUCCESS in 33m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cb15184c8d6a48e7a3630070d680f653 : SUCCESS in 57m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d58cf08806b340a893d5faf8ce65b4f9 : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b33510cc08b11e2e13500f97daa3c7041350fd7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 21:18:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/181c7e4a44f54f18bf77def0061a19d4 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"28f8d46dc938e5a9bfe3d08c75c7e35d5874c44a","revisions":{"ccef25a3d9f079977033c88d2e3e32664053069f":{"kind":"REWORK","_number":1,"created":"2019-08-04 18:15:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/674510/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/674510/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/674510/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/674510/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/674510/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/674510/1"}}},"commit":{"parents":[{"commit":"2587d86b131833317236f86d2dd49e3e4c868bdf","subject":"Merge \"TrivialFix: remove unused ganesha nfs repo file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2587d86b131833317236f86d2dd49e3e4c868bdf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-04 18:14:07.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-04 18:14:07.000000000","tz":60},"subject":"Add kayobe to deliverables.yaml","message":"Add kayobe to deliverables.yaml\n\nChange-Id: I2555c640125f3fede1dd1715d70c0c36620f3d4c\nDepends-On: https://review.opendev.org/669298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccef25a3d9f079977033c88d2e3e32664053069f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccef25a3d9f079977033c88d2e3e32664053069f"}]},"branch":"refs/heads/master"},"28f8d46dc938e5a9bfe3d08c75c7e35d5874c44a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-09-16 16:36:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/674510/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/674510/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/674510/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/674510/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/674510/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/674510/2"}}},"commit":{"parents":[{"commit":"b8950dbfae16e4959e7e502b584018fc8e58eaaf","subject":"Merge \"Drop Percona XtraBackup in favour of MariaDB Backup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8950dbfae16e4959e7e502b584018fc8e58eaaf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-04 18:14:07.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-16 16:36:55.000000000","tz":0},"subject":"Add kayobe to deliverables.yaml","message":"Add kayobe to deliverables.yaml\n\nChange-Id: I2555c640125f3fede1dd1715d70c0c36620f3d4c\nDepends-On: https://review.opendev.org/669298\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28f8d46dc938e5a9bfe3d08c75c7e35d5874c44a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28f8d46dc938e5a9bfe3d08c75c7e35d5874c44a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
