)]}'
{"id":"openstack%2Fkolla~675043","triplet_id":"openstack%2Fkolla~stable%2Fqueens~I1822a8c610db72654bbacc7148e25a837090b680","project":"openstack/kolla","branch":"stable/queens","hashtags":[],"change_id":"I1822a8c610db72654bbacc7148e25a837090b680","subject":"ceph mds cap \u0027allow *\u0027 in extend_start","status":"MERGED","created":"2019-08-07 07:22:40.000000000","updated":"2019-08-09 12:15:25.000000000","submitted":"2019-08-09 12:15:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675043-1565352925961-c9b5ff1c","meta_rev_id":"9992ccf9b5abefd35d9f71ccf9b9d0b3fbee90a8","_number":675043,"virtual_id_number":675043,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-08-07 16:42:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-09 12:15:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-07 16:54:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-08-09 11:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-08-09 11:33:28.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-08-09 11:33:28.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 16:54:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-09 11:16:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-08-09 11:33:28.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-09 12:15:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea176e401a0e7f410ba84075c5c64a9d0cb18d67","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-07 07:22:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"c6946d5cffdd62498ac3417a54448a30cb7e189a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 07:35:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/43/675043/1/openstack-check/tripleo-build-containers-rhel-8-master/f33337f/ : SUCCESS in 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"fd34f45feab3daafe936e5c67dbb2461abdab42b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 09:03:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://logs.opendev.org/43/675043/1/check/kolla-build-centos-binary/14dee44/ : SUCCESS in 34m 36s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/43/675043/1/check/kolla-build-centos-source/8672b0a/ : FAILURE in 37m 22s\n- kolla-ansible-centos-source https://logs.opendev.org/43/675043/1/check/kolla-ansible-centos-source/f1e3cb2/ : SUCCESS in 59m 49s (non-voting)\n- kolla-ansible-centos-binary https://logs.opendev.org/43/675043/1/check/kolla-ansible-centos-binary/90d6a1e/ : SUCCESS in 1h 07m 35s (non-voting)\n- kolla-build-oraclelinux-binary https://logs.opendev.org/43/675043/1/check/kolla-build-oraclelinux-binary/02b4a05/ : SUCCESS in 46m 51s (non-voting)\n- kolla-build-oraclelinux-source https://logs.opendev.org/43/675043/1/check/kolla-build-oraclelinux-source/8563e15/ : RETRY_LIMIT in 12m 36s\n- kolla-ansible-oraclelinux-source https://logs.opendev.org/43/675043/1/check/kolla-ansible-oraclelinux-source/41302a0/ : SUCCESS in 1h 04m 04s (non-voting)\n- kolla-ansible-oraclelinux-binary https://logs.opendev.org/43/675043/1/check/kolla-ansible-oraclelinux-binary/c8db5fe/ : SUCCESS in 1h 04m 11s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/43/675043/1/check/openstack-tox-pep8/38b8b5c/ : SUCCESS in 6m 20s\n- openstack-tox-py27 https://logs.opendev.org/43/675043/1/check/openstack-tox-py27/3060858/ : SUCCESS in 5m 02s\n- openstack-tox-py35 https://logs.opendev.org/43/675043/1/check/openstack-tox-py35/c1c27a3/ : SUCCESS in 3m 40s\n- build-openstack-sphinx-docs https://logs.opendev.org/43/675043/1/check/build-openstack-sphinx-docs/d457b74/html/ : SUCCESS in 3m 59s\n- kolla-build-ubuntu-binary https://logs.opendev.org/43/675043/1/check/kolla-build-ubuntu-binary/84fa43d/ : SUCCESS in 26m 48s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/43/675043/1/check/kolla-build-ubuntu-source/5119e57/ : FAILURE in 1h 39m 39s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/43/675043/1/check/kolla-ansible-ubuntu-source/0f2be28/ : SUCCESS in 1h 05m 35s (non-voting)\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/43/675043/1/check/kolla-ansible-ubuntu-binary/35c3f79/ : SUCCESS in 1h 08m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e96eb32c8b2e8cd5ec366b956d4deec52bbf1107","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-07 16:42:46.000000000","message":"Patch Set 1:\n\nrecheck\n\nrandom issues with mirrors","accounts_in_message":[],"_revision_number":1},{"id":"bb046a2f9f6997f83dea38dc17f44c1e6b42f325","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-07 16:54:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/43/675043/1/openstack-check/tripleo-build-containers-rhel-8-master/042451a/ : SUCCESS in 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"8c343d41d314418275792f82cb92160d8ac187ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 18:18:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://logs.opendev.org/43/675043/1/check/kolla-build-centos-binary/36619d3/ : SUCCESS in 36m 49s (non-voting)\n- kolla-build-centos-source https://logs.opendev.org/43/675043/1/check/kolla-build-centos-source/277a90a/ : SUCCESS in 1h 00m 21s\n- kolla-ansible-centos-source https://logs.opendev.org/43/675043/1/check/kolla-ansible-centos-source/ca38633/ : SUCCESS in 1h 13m 38s (non-voting)\n- kolla-ansible-centos-binary https://logs.opendev.org/43/675043/1/check/kolla-ansible-centos-binary/22ba784/ : SUCCESS in 1h 34m 00s (non-voting)\n- kolla-build-oraclelinux-binary https://logs.opendev.org/43/675043/1/check/kolla-build-oraclelinux-binary/58f30e1/ : SUCCESS in 32m 44s (non-voting)\n- kolla-build-oraclelinux-source https://logs.opendev.org/43/675043/1/check/kolla-build-oraclelinux-source/5652aba/ : SUCCESS in 41m 42s\n- kolla-ansible-oraclelinux-source https://logs.opendev.org/43/675043/1/check/kolla-ansible-oraclelinux-source/02e8b98/ : SUCCESS in 1h 20m 31s (non-voting)\n- kolla-ansible-oraclelinux-binary https://logs.opendev.org/43/675043/1/check/kolla-ansible-oraclelinux-binary/78dee1b/ : SUCCESS in 1h 08m 39s (non-voting)\n- openstack-tox-pep8 https://logs.opendev.org/43/675043/1/check/openstack-tox-pep8/e4def40/ : SUCCESS in 5m 07s\n- openstack-tox-py27 https://logs.opendev.org/43/675043/1/check/openstack-tox-py27/9dc1c34/ : SUCCESS in 3m 36s\n- openstack-tox-py35 https://logs.opendev.org/43/675043/1/check/openstack-tox-py35/11412a2/ : SUCCESS in 4m 21s\n- build-openstack-sphinx-docs https://logs.opendev.org/43/675043/1/check/build-openstack-sphinx-docs/24b2fc0/html/ : SUCCESS in 5m 36s\n- kolla-build-ubuntu-binary https://logs.opendev.org/43/675043/1/check/kolla-build-ubuntu-binary/26a23d1/ : SUCCESS in 27m 40s (non-voting)\n- kolla-build-ubuntu-source https://logs.opendev.org/43/675043/1/check/kolla-build-ubuntu-source/89f48a1/ : SUCCESS in 24m 24s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/43/675043/1/check/kolla-ansible-ubuntu-source/d67181b/ : SUCCESS in 50m 29s (non-voting)\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/43/675043/1/check/kolla-ansible-ubuntu-binary/283b886/ : SUCCESS in 54m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26c69b4146aa56dff15d8d36e066437c70b5d504","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-09 11:16:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"be6796c82f98e700306b7910f8cd7d2d4d08b42e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-09 11:33:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2cf41b436c48a6508b0f3470af40cb5bc955716f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 11:33:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f87fc22dd87f14024de764740aa1e1513e361397","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 12:15:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9992ccf9b5abefd35d9f71ccf9b9d0b3fbee90a8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 12:15:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://logs.opendev.org/43/675043/1/gate/kolla-build-centos-source/a5cda3e/ : SUCCESS in 36m 50s\n- kolla-build-oraclelinux-source https://logs.opendev.org/43/675043/1/gate/kolla-build-oraclelinux-source/f6ec04f/ : SUCCESS in 35m 54s\n- openstack-tox-pep8 https://logs.opendev.org/43/675043/1/gate/openstack-tox-pep8/7d65176/ : SUCCESS in 4m 44s\n- openstack-tox-py27 https://logs.opendev.org/43/675043/1/gate/openstack-tox-py27/cbd5c60/ : SUCCESS in 4m 39s\n- openstack-tox-py35 https://logs.opendev.org/43/675043/1/gate/openstack-tox-py35/5269091/ : SUCCESS in 3m 36s\n- build-openstack-sphinx-docs https://logs.opendev.org/43/675043/1/gate/build-openstack-sphinx-docs/e0c4d57/html/ : SUCCESS in 4m 56s\n- kolla-build-ubuntu-source https://logs.opendev.org/43/675043/1/gate/kolla-build-ubuntu-source/93b5354/ : SUCCESS in 27m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"689506f537eda0334b26c831b0124b1f434d7021","revisions":{"689506f537eda0334b26c831b0124b1f434d7021":{"kind":"REWORK","_number":1,"created":"2019-08-07 07:22:40.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/43/675043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/675043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/675043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/675043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/675043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/675043/1"}}},"commit":{"parents":[{"commit":"69a7cb5e9061abb474e9ffb9202cbf67c6114f6f","subject":"Fix deployment of ceph-nfs container failing due to wrong USER","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69a7cb5e9061abb474e9ffb9202cbf67c6114f6f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-01 16:34:24.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-07 07:22:40.000000000","tz":0},"subject":"ceph mds cap \u0027allow *\u0027 in extend_start","message":"ceph mds cap \u0027allow *\u0027 in extend_start\n\nThis applies the fix made in kolla-ansible (see [1])\nto kolla ceph mon image directly.\n\n[1] https://review.opendev.org/670035\n\nChange-Id: I1822a8c610db72654bbacc7148e25a837090b680\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit b1d4ed51f41721c86a33d766fdfd80aaccf1df60)\n(cherry picked from commit 2f45d6810e341cd89a20d5fceff0179cde768acc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/689506f537eda0334b26c831b0124b1f434d7021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/689506f537eda0334b26c831b0124b1f434d7021"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
