)]}'
{"id":"openstack%2Fkolla~676128","triplet_id":"openstack%2Fkolla~master~Ifc32713cd4d6b45c74855c0ebf5bb65784701d44","project":"openstack/kolla","branch":"master","topic":"bp/container-health-check","hashtags":[],"change_id":"Ifc32713cd4d6b45c74855c0ebf5bb65784701d44","subject":"Build images with healthcheck scripts for core services","status":"MERGED","created":"2019-08-13 08:00:05.000000000","updated":"2019-12-16 13:16:19.000000000","submitted":"2019-12-13 10:38:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676128-1576233532419-9875c5dd","meta_rev_id":"93fe1e3108be56bd225411a47aecabb185d7b653","_number":676128,"virtual_id_number":676128,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-13 10:38:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-09 13:31:15.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-12-11 09:50:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-12-13 09:04:24.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-12-13 09:04:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"date":"2019-12-12 08:33:14.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 13:31:15.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-11 09:50:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-12-12 08:33:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-13 09:04:35.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-12-13 10:38:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35280a5a3f6e2936b4fbd9bf11065e3cf5c95c91","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-13 08:00:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e60dc64f0614faf79999601232a2925d5515fe0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-13 08:16:32.000000000","message":"Patch Set 1: Code-Review+1\n\nfor idea","accounts_in_message":[],"_revision_number":1},{"id":"b804129b01d680220b44e0a3c19620e72eb12a3b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-13 08:32:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"986466c77578e315a36bad6e177afb63896f909b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 10:35:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bf98e1a3ac1c4e6eb2968cafeeafb9f4 : SUCCESS in 42m 57s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/2ad76356c41540e6a98200d56a7fe422 : SUCCESS in 1h 06m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/770080e025084659975ec32050c79b19 : SUCCESS in 1h 13m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3e39ea8208ab434ca31b8a6163b6cdb4 : SUCCESS in 1h 13m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/1668b1c0fd9f4886bd101cfb180c0360 : SUCCESS in 1h 24m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a096e82eb48248d098373ff583123428 : SUCCESS in 54m 09s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/bc531f3dd0cf4010a07495f4666e2c30 : SUCCESS in 50m 14s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/3e2fa1cd32a44a45b00b8b5738f255f7 : SUCCESS in 1h 47m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/d69bf28ebdb8451383b2120966a79aee : SUCCESS in 1h 57m 25s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/57f49b6e349b494fa0490b690f132e39 : SUCCESS in 1h 31m 39s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/1d955f72b9a94a5484630f286c11de67 : SUCCESS in 1h 34m 31s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/ec5b3ef89e2946c185012a1570ccae14 : SUCCESS in 1h 29m 56s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14bb852bc26c48398c3581c74b1bcda5 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87fd1ecd762455db66ed697de438bbd : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5d3006d2128c415b922d8f385cbdea08 : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f59b3b037a443d9925837f50a0c81c1 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2083f8e0c2747c1ba8bbcedc0c306b7 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3da7dbe54c504568b8cc0384d73d8ef6 : SUCCESS in 3m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1567ce58bc38473c9594d629ed7508bd : SUCCESS in 29m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/599d7ba53b0b4fdcbb3d33c0ad47f774 : SUCCESS in 31m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f38018996b48427592f1e4964a6f8248 : SUCCESS in 51m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/134f8bb5d88e4707a882b6ec745a1a1b : SUCCESS in 48m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ad94960a95c64409b926192ea4e1e0f7 : SUCCESS in 57m 47s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2729861dc1f14557a900d7e93dd316c8 : SUCCESS in 58m 52s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/0992336b818b46ec817533de8e03ee2e : SUCCESS in 1h 12m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9aca557e66d94654b828f9e7105cca55 : SUCCESS in 1h 39m 20s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/cc8860b8ee0c423cbbb3d20af60a55db : SUCCESS in 1h 00m 17s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/6e64749f1b2e455088137632af76124d : SUCCESS in 1h 03m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"269665ddb68a0c566d235f6e6221191150c0b8af","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-13 11:11:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ce4f9d31ec10040e21547d7114496ff2c41539a6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 11:59:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/3/openstack-check/tripleo-build-containers-rhel-8-master/bde8ed4/ : FAILURE in 46m 56s","accounts_in_message":[],"_revision_number":3},{"id":"b6529d5778cd83589719de365cf4f087b34fecdf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-13 13:18:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"31ac0cfb2a0ce6cedde6c9d36732bfe5bf763995","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-13 13:19:58.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"4e9c494207d5f4abab0f1c2927ed73c7294ac72f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 14:07:50.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/5/openstack-check/tripleo-build-containers-rhel-8-master/d96eae1/ : FAILURE in 46m 51s","accounts_in_message":[],"_revision_number":5},{"id":"6cb59362e5889eba1830571f4f7fcb057b3a7111","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 15:43:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/5c6b0296c52b47499ee76c7fa9301304 : SUCCESS in 43m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e0d5045b16db4533a66b47b01f601e30 : SUCCESS in 35m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/68776caafd2d4ef4952ff403b39b7760 : SUCCESS in 1h 36m 12s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3fb00c2abe1a46bdb2f79a3d3b7ac169 : SUCCESS in 1h 08m 02s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b034b00c54be47c2afa1abbad6bf9a45 : SUCCESS in 1h 21m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/50eb8c9adab942848831e60990319bee : SUCCESS in 49m 06s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/b2387d5e1b154bdbab293a30d9be89a0 : SUCCESS in 39m 32s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/c413fb0bed4a4fff9b80b6a3b299902d : SUCCESS in 1h 52m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/891a0b38e9ce499998d4ceff43be87c8 : SUCCESS in 2h 13m 56s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/2e923da66c244a5088ad9600e0776919 : SUCCESS in 1h 31m 40s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/20147d3d6f6c4dfc988c324579fb61b8 : SUCCESS in 1h 34m 17s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/73ad4cb2655b46d1b79c7903d092f41a : SUCCESS in 1h 23m 49s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35045f4c9e65402da2d94415dc3baf2c : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed33ea79b5944e9db0dde1ebda18da54 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a3a088ae65444761826232f357a64ad5 : SUCCESS in 3m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f95eeece709409a9a1b5401eeabb9ce : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cec2f2f36b14779a0275460300bea86 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48dce490ef29490eb680f049fc76c45f : SUCCESS in 3m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/db49dac000c54096a28417f457485028 : SUCCESS in 28m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7dc544f587c4b849310925aebd6499d : SUCCESS in 35m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b0074c5551094d24b89631724ea9b8e2 : SUCCESS in 49m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/911ef83aa8934b399b0bc0d793ad3441 : SUCCESS in 49m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b2e1ee2504f94307918aa4cf31463f1b : SUCCESS in 56m 58s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2757de2da37b41d6b4aba30de9e13c21 : SUCCESS in 56m 32s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/0d0992d3bea941cb89dd3e28fabb9cb8 : SUCCESS in 1h 18m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/28737f3f4e6f4602a5019bb4839e35cf : SUCCESS in 1h 18m 41s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/59741b09201242e8993a3f071061e982 : SUCCESS in 1h 00m 55s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/78fd40ed2ba04898ab514536dc6e3fce : SUCCESS in 1h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c365539c0809df30c0c0523c8ab40fc7865f37ee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-13 16:08:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ceaba0018eb31a73d35dc4c1fa60127be06f0fff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 18:18:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/647d62135024429a97059afdf2e41af2 : SUCCESS in 39m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1b23bbefdf3b41a6a751f823aad42095 : SUCCESS in 47m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3442c7dd15954567846b25a30d7100fe : SUCCESS in 1h 15m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2f3782015bd4489ba7f6d2c1d0cc4d46 : SUCCESS in 1h 01m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/84b1317cb5c1498bbd1913db890bf71a : SUCCESS in 1h 26m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e85b7cad2cfc4a2e893e57d85423950f : SUCCESS in 56m 55s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/be7cc106015449439bf0e6f5c38bb78f : SUCCESS in 52m 51s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/8c31a5a550ff4ed8acc0a2d3528ba2a3 : SUCCESS in 1h 56m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8371d3282a4d45339c2e55cb957da269 : SUCCESS in 1h 53m 00s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/b62ff0592419424a9a4faa466d089e12 : SUCCESS in 1h 27m 22s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/9873eea4558b4b8ea60546d09c08d9db : SUCCESS in 1h 45m 16s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/a32a8d1f06cc488eb863b32d463d67e3 : SUCCESS in 1h 34m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5674e4a2cd6425cb150e9ed7d7a3123 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a115215ebb0d4df2a48cbb038b483595 : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d18fec8b03e14372902bd3419187a820 : SUCCESS in 2m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a443278da2564aec9fef9f32afa42fcc : SUCCESS in 3m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/471dc89b18d44d0fb07db218a2251cd9 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec3b05992acc4b72b620aa4b9959a2f2 : SUCCESS in 2m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1433970ca8dd4a15a8811741ecdf813b : SUCCESS in 23m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/254b43a275ec4c2aa2de57829c6af317 : SUCCESS in 38m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/89bc19c0c6d64060b64fdb5346bf5e9f : SUCCESS in 59m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d13ec27580bf48a1a0ef9ca337b9c077 : SUCCESS in 1h 27m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9cf1db4a36ba476da84655c2aba7bd45 : SUCCESS in 58m 43s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/c8f2905e3320433fb57f0cf5eb08e6d9 : SUCCESS in 51m 10s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/0eb2e7d16ce449f2860b52afb75fc83a : SUCCESS in 1h 21m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/a1981bc77ab045698e9210413aaec717 : SUCCESS in 1h 31m 33s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/d692073082a84c54b19d71f7dd77d1f9 : SUCCESS in 1h 00m 17s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/31a7e9e334d1410fa4f7950c2c2145b4 : SUCCESS in 1h 04m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3cf9e831513702bfd6f284d93ddfe5f05ffcf902","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 18:30:32.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/6/openstack-check/tripleo-build-containers-rhel-8-master/938126e/ : TIMED_OUT in 2h 09m 35s","accounts_in_message":[],"_revision_number":6},{"id":"488c608062cb8fd67f3f87afa78c2a22a07ee2b0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-02 15:17:51.000000000","message":"Patch Set 6: Review-Priority-1","accounts_in_message":[],"_revision_number":6},{"id":"d9f828e3febe73b1d801053ba2dc3535a2d986c1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-22 14:56:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"267277d1018b7cd6aeaedd694dd327ecde2c6ade","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-22 15:29:36.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/7/openstack-check/tripleo-build-containers-rhel-8-master/57bd2ce/ : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":7},{"id":"142731e9b5878533b3c88843d7ad032a72995ec7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 17:40:28.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2bd6663365d74b8f87b5899406838072 : FAILURE in 6m 20s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9c5402aaaf2141729fcb04bf3f9d65a7 : FAILURE in 6m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2f0188c968b846878455cc9833901c0f : FAILURE in 11m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4a2882010c5c463a91acfe053d3ffca8 : FAILURE in 11m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/7b8ebb4a7e1c4d11ac703822e955ed2b : FAILURE in 10m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7ab1e40ec3f548998feffc76ac7c42d7 : SUCCESS in 51m 10s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/9cea95bb361a4f4bade65a255a5a4bcc : FAILURE in 10m 42s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/a6545397545d4f099ad5b6d143838aae : SUCCESS in 2h 27m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/87c17635f4bf4d4a903e2695f4271332 : FAILURE in 14m 59s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/1e7d949572bf49f386dae6ecb8d44d8c : FAILURE in 13m 10s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/667004a24d7142f081693763ebcdf46d : FAILURE in 37m 34s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/e53d4f8bab69437ab1fe95c3f6e5752b : FAILURE in 11m 18s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cb490c8346674e2aa79083764a6d12d0 : FAILURE in 22m 33s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/976754b6cb48446d9385485db23350dd : SUCCESS in 25m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6f9d11bf0f8d4a8eb910e03557138bec : SUCCESS in 1h 11m 13s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e75231e8f9df4641a2e2cc62ea2735c6 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/997cee53896c46f1ab37d1d3391ea15c : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/649f1bc14b2e4e3cb25e94c3144ff859 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0db685d009024de1864eb8a0cfdd37f5 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b39ccfeca44a11a93e294795bdfa80 : SUCCESS in 6m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3afe18a2a20d4c849cd442530fff387a : FAILURE in 28m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8d9bce01953d4cc4888c0dfcca17245d : FAILURE in 5m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4ecab1fa43be424fb20276f0017dfac8 : SUCCESS in 52m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a26409677abb4700ad3af0afea7a18ce : FAILURE in 10m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f187997b45946c18db357ce8ebdd759 : SUCCESS in 59m 07s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/4fc1916332e84c2386688f3a84cc3334 : FAILURE in 11m 28s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/2a04eed5da2340d088a20e2541e495e6 : SUCCESS in 1h 17m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/1114584a39fb495395bc2a53062163e8 : SUCCESS in 1h 38m 01s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/4213324fc0dd42e5a8643f2e4d54f97d : SUCCESS in 1h 09m 24s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/979d2f70989b4d3d8e03b94ecea90df4 : SUCCESS in 1h 07m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"22ff4d4eb7bc445e75ca9486052b5d2138bee7a8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-25 08:53:01.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"387369e02e9d3833659addf0884609a493389ce2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-25 09:26:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/7/openstack-check/tripleo-build-containers-rhel-8-master/f1be3db/ : SUCCESS in 32m 34s","accounts_in_message":[],"_revision_number":7},{"id":"943b76bc6f9d8cfb29bbd6d49cab7cffe58dbddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 11:30:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/588cb64ec55b4f47a3e9a6020caff4d8 : SUCCESS in 45m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9e4c23597c304f41a3c0a0434aec9deb : SUCCESS in 59m 17s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a863df3af6e244d8ab6181a3d359aa60 : SUCCESS in 1h 28m 08s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/47b4b4136433439e964fd8cf9aeabdb1 : SUCCESS in 1h 22m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d2c29b73680647818724f114ef16c83e : SUCCESS in 1h 38m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7c3fc82ba46e417a8f44a51a372fc25c : SUCCESS in 1h 08m 56s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/3adae096cb4141ceb3e9e0966beea62c : SUCCESS in 58m 13s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/4ac45041e25b4fcc89f476691e440499 : SUCCESS in 2h 17m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e81abc4790254d04b30647e3c1e5a84b : TIMED_OUT in 2h 32m 46s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/27aabccf4fee4554a6d437c1adb0cb8d : SUCCESS in 1h 45m 37s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4337ee843dbf41c9a0c2381739c144c1 : TIMED_OUT in 2h 03m 22s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/d9e9f7736e174931a299da49b73c7cb6 : SUCCESS in 1h 39m 24s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e0e2087b25cc40628ea79c34f5a03da9 : SUCCESS in 32m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a2ab29b5af0f42dda44128cc16b01e90 : SUCCESS in 18m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a13b0089814042208db63116f0d74d1f : SUCCESS in 1h 28m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc469001d7724153b7abdc9e18e8cb82 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35a86a0c164e4f64bd3c6f58787109f2 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78612289152d46d8a6caf02234f3211c : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/233aa065184a4a21bfe5b5d961c9afb5 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0de9234ad5ea4373bdd964039be2cb2b : SUCCESS in 5m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/21d653919041485096fd40aa06977f27 : SUCCESS in 29m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/20e37ee3f296475f88235413e0b7d1e9 : SUCCESS in 1h 00m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d573034cfb76430295459bce7cf87885 : SUCCESS in 55m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/482486e98c5a4bf29db088dae344a9ad : SUCCESS in 1h 08m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/346b5a974f62463da1d4fa16367a862e : SUCCESS in 1h 15m 14s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/9f9d3faab0824c39aae775c5b94844da : SUCCESS in 2h 01m 51s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/430ba616db7d4e6790f5c6fe1c833f6d : SUCCESS in 1h 26m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/6f4710c989e343929f7683181ff675a8 : SUCCESS in 1h 59m 59s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/7366302737c34957bcfb6afc4c3ecf37 : SUCCESS in 1h 23m 36s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/9c83841f84c84df0aac2fdc6226ddee4 : SUCCESS in 1h 42m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b74a5d776e560b6cba138115260f7cf2db637c60","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-26 10:37:39.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8ae5aa2188e8c9e977150b2787c86ca0f3d7cb2b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-26 11:12:03.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/8/openstack-check/tripleo-build-containers-rhel-8-master/3bb48af/ : SUCCESS in 32m 53s","accounts_in_message":[],"_revision_number":8},{"id":"30b84d09855457f84dd14088479acd65f0a0dec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 13:19:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ed9bf57f64f74efebfbbafb3d9e41731 : SUCCESS in 41m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/49ba88a7e28340f2a51d02fab0d24e7c : SUCCESS in 56m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/404de2de7fe04ab08c50f159d9010528 : SUCCESS in 1h 30m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/17cb1d75801542948e251a831b6f5d76 : SUCCESS in 1h 18m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c98d534bc298439caadfa6522a971405 : SUCCESS in 1h 27m 21s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e1df99f4a0ab4b4e9d4695e278422ce5 : SUCCESS in 57m 21s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/0756e9bacd344a0284e3216a919bd23c : SUCCESS in 47m 37s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/94dd419a94284b088a319c54d3c973db : SUCCESS in 1h 51m 31s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/2e50b489ed1c43f89c9480fbb8e93668 : FAILURE in 2h 26m 39s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/9b25a68f5ab9406481b089a939a4ca5f : FAILURE in 1h 15m 02s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4ee6a6780c1f4b788b2d782ef7b7fa93 : FAILURE in 1h 16m 32s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/3834b1d236fd4dc789eb7f36e439d5ff : SUCCESS in 1h 46m 02s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/91709cd2dc3b49be9cf9fe710d5439de : SUCCESS in 28m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/862d30edbd444646ab8b8480ca6c9634 : SUCCESS in 17m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d7364f9b35154eab83ead5fddf386e3e : SUCCESS in 1h 10m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95af6ad362424f1aae8bbe1a230c2591 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcd3daa83bd34f25b42f3d134683617f : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a26a6c8e3d034f7da1a5ff627b73b623 : SUCCESS in 3m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/140e14beba1f48d9969f9b56105a4a87 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74600bfa1564bbab43b68cd69216fae : SUCCESS in 5m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ab36167f470041798099d4394c3206d2 : SUCCESS in 26m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a227764323d49b1884fb0da3aed4bdb : SUCCESS in 27m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae2ab38427c649a6b2704d8598cf2dbb : SUCCESS in 53m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3ff9d6ed699647aea256ac6a981c4c30 : SUCCESS in 55m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4f29f26db7e549c9a1a9bd9d72e53ee9 : SUCCESS in 1h 09m 18s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/4e05bd2512394954b856030bf1e34bdf : FAILURE in 39m 20s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/d26340a640c34f59a0b7a577f7a9e29d : SUCCESS in 1h 21m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/b6fe10cbb8714fc78427d24503e055a1 : SUCCESS in 1h 33m 29s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/39ab0abe4dd745f9907fe121e38a658a : SUCCESS in 1h 15m 48s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/9545d09c869043fcb659aa5c5f52b890 : SUCCESS in 1h 05m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"04e6360611e314ba79665aab264ef956675c820c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-29 06:46:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"027393f10db055c7b75dea533da1f69b0632931c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-29 06:52:09.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"fe0122c9e42859553708b3a598588465d3f62ffc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-29 07:24:16.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/10/openstack-check/tripleo-build-containers-rhel-8-master/a932553/ : SUCCESS in 30m 57s","accounts_in_message":[],"_revision_number":10},{"id":"e36010d2bbb864c454ed0708c5b82836e4ae4029","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 09:31:03.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/34ff83f81f8e4603bbaba75a209599ea : SUCCESS in 32m 48s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/17fbc012658449deb66aad9932cdac10 : SUCCESS in 38m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/69ea7b96e9124590a1de4a8181912f04 : SUCCESS in 1h 17m 29s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e6774f23dbef4a76bfc14bcf5a153a9e : SUCCESS in 1h 14m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/930a5ea5751643c696f90f32f69ecba2 : SUCCESS in 1h 51m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f3358037d4424c29ba8a4adee0ccfbf1 : SUCCESS in 56m 22s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/4e72e060e2db4c8c9b135baa523faf5b : SUCCESS in 57m 54s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/b79be86c08634e0199567a23c590b967 : SUCCESS in 2h 26m 26s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/499c88cdbaaf48478592ad78d6fc58b7 : SUCCESS in 2h 18m 58s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/c46bb07eb0004c0e809d6e9d8c6be05c : TIMED_OUT in 2h 02m 29s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ea1460b47d00439c8c4887bc0c312d38 : SUCCESS in 1h 51m 29s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/9075aed773894e969c081bf398cdb109 : SUCCESS in 1h 23m 10s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d74674f818ee409698fbd2d5a32baad4 : SUCCESS in 24m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/03d7c0006a804f2e8be7c050c4b11224 : SUCCESS in 17m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/276b8554e9b3459096641adda6dc0787 : SUCCESS in 1h 27m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f9f52b7b7e34ca2825f2f7cef3418b9 : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8572746c6c3a4f7c8228eeaf5a7593d9 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dcb9e2448874eb8ac514466581aef99 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38b9c4388f9a4588a95a30aee6dfcb70 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1724bc7b3234969b0e2f18c46803516 : SUCCESS in 5m 00s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fa135795e04849168c2e1a2b38180f7c : SUCCESS in 53m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc671be378244a05bb370ac9a4fe882a : SUCCESS in 32m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d4c51d9da7a4878b121bfe44c1e7275 : SUCCESS in 1h 04m 49s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d8992372b77b4d2791c8ec7a9094d392 : SUCCESS in 55m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/77a8788c252146a7ae0a530eec48eac4 : SUCCESS in 59m 49s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/5ef07f069dd84b4da790856db32a2371 : SUCCESS in 1h 12m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/09e01856811b4199a56738bab45fa6fd : SUCCESS in 1h 18m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9adef32eb1e94459a7dac971d7ac45e8 : FAILURE in 50m 35s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/fdec8c9237c24b3ebd7110382571b467 : SUCCESS in 1h 13m 14s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5e716a734ef148ec84ef761ecb931a8d : SUCCESS in 1h 42m 12s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3bc516799b8b44fc42bc62a56758fc22b6839195","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-29 13:37:43.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e17f7650a4437d31d422e2ae936d96f62094e918","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-29 13:49:04.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"4dd7af02429b3a5cc5f37774f3f4b0c9f485341c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-29 14:21:39.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/12/openstack-check/tripleo-build-containers-rhel-8-master/329df47/ : SUCCESS in 31m 18s","accounts_in_message":[],"_revision_number":12},{"id":"3c3f4991a308a0415e26dec1caa0d9a3b8e7c684","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-29 14:25:23.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"6a6738797d10d0918942886087ddeb1db20b70d6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-29 15:16:30.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/13/openstack-check/tripleo-build-containers-rhel-8-master/48dcb2b/ : SUCCESS in 50m 04s","accounts_in_message":[],"_revision_number":13},{"id":"0e2e6ca2cc1fd4f133c6ebaea3973e09c138ab9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 17:04:03.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/92370138b49948ceb9a7d4520ca18af9 : SUCCESS in 35m 48s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/72d56f9037f24f1fb43f2860ec1676f8 : SUCCESS in 59m 39s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5ca1808ec2664fa690daa496d9fbfb8c : SUCCESS in 1h 21m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ea675ae55b3948078e5f54784fb1c210 : SUCCESS in 1h 11m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/eb048aa05cf24faf8ca25e3402727dcb : SUCCESS in 1h 32m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5f77e941a952462c83ea7c55f6fec620 : SUCCESS in 42m 39s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/a7595666909d4ffcab3da7bdf5c96e8a : SUCCESS in 46m 26s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/07556f309e1044f780362e8e8aed60b7 : SUCCESS in 1h 56m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/91fbed5bbff748968fe997991a082a60 : TIMED_OUT in 2h 32m 32s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/be49d7cacc6f41c8a19bf128e93bb9cf : SUCCESS in 1h 43m 44s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/067538a77ed14813ae31b8c9e2d24338 : TIMED_OUT in 2h 02m 38s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/92585e99fa6b48dcb0437b9c1d278c6a : SUCCESS in 1h 32m 54s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6ecc265859404e6e9176f76cff916076 : SUCCESS in 25m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a51b89778adc4fdc92c46c5f2dd55f1a : SUCCESS in 20m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4404ad95021647a6a3ba61c0c4ef039f : SUCCESS in 1h 06m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67b5538da77d4ff0a4a877773ae50c2b : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f481e02264342a98c9ee900ff735c0e : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/025a9b16a56f4522b112fed07ccf0148 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b8aa4e501b694261aa340742bb828059 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd04008bbd34423f8a55496e744a4e6d : SUCCESS in 4m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3ac7cac4ac684d729b5a4aa140ca622c : SUCCESS in 25m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c187040c07014d2e97291927504f7201 : SUCCESS in 28m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c39140915514f3d8e5c5e907e7a38f7 : SUCCESS in 59m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3f3ae4fbcc2647fdb9f787e858921e44 : SUCCESS in 55m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b70589c9c2044466b51d3fb8e34d8f80 : SUCCESS in 1h 06m 43s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/cf104e4961724979b52223a235acd678 : SUCCESS in 1h 09m 38s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/0eeea7465f7e43af93bf5cbe72ecc144 : SUCCESS in 1h 26m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/37f274213a64435db63c7aa16f577ae2 : SUCCESS in 1h 52m 14s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/cf9a9054a5dd43c18e7eaa24f1d020c8 : SUCCESS in 1h 15m 08s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/dfd3929ac874455cafd0f80dbead5f70 : SUCCESS in 1h 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"2326026f0eaba9a72299b759fe926f1aeacf4370","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-02 12:09:23.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"eb99d611365f7076177b7eb6de84ab489a4c85fc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-02 12:33:17.000000000","message":"Removed Review-Priority-1 by Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":14},{"id":"cca9c413f54dafa9f42b94a1d6424886fcb1ba6c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-02 12:34:53.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"a9a753435a04bc65556c7f9d2ce2a29ef69076a3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-02 13:22:14.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"c8adfb1d029e8efc6558bdd16b077402a594db68","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-02 14:27:18.000000000","message":"Patch Set 16:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/16/openstack-check/tripleo-build-containers-rhel-8-master/e350a2e/ : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":16},{"id":"009b90d7deec8ffcfc11965392a73e38f0f607fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 15:45:35.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f4ae9be4564641678fed6084cea8888f : SUCCESS in 52m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/6479f66e0c014b2786aeafe579f2e20c : SUCCESS in 55m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/12bd5a14a52d4f88baa318048b7bb594 : SUCCESS in 1h 49m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c642178a86954dd286fb539c7d54b34c : SUCCESS in 1h 26m 14s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0a83a0618b7e4dfeb67ea35ee473164a : SUCCESS in 1h 33m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/a1a55e5008f647c8a949e2ee0989f031 : SUCCESS in 55m 30s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/7ceea041bea64b538642d37f4ecc5ff5 : SUCCESS in 49m 05s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/b865f62339264ea6b6475ed608cbb6e4 : SUCCESS in 1h 48m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/a3e596a9f2c34ec492bba6bdc98065c7 : SUCCESS in 2h 19m 16s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/f9577d5a07034c9ab4d0a711bd52efbe : SUCCESS in 1h 41m 41s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/84827a3b40b1462cb62a9090f2bdb5ea : SUCCESS in 1h 50m 33s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/012dc4f850a94b6fa78b37d0df0e8521 : SUCCESS in 1h 24m 26s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cf3f83f828df4df4a5d88c5a3dc2362d : SUCCESS in 24m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9e088d8cb4944909b213c1ae6c27a31b : SUCCESS in 20m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/24f62a3f5e544ef58e909314a0e78438 : SUCCESS in 56m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/00fcf7b310054f6aa73e52046df87749 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecca4301b028442d8448dae8a6e33731 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6039585e06984511abae0841c2d503cc : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4356848028e34eec95ed90c8dfa96963 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5a527ee73214b44b9e0c9fbdf920264 : SUCCESS in 5m 52s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/decfb81afb8f44e7b953f119f44ddd99 : SUCCESS in 33m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/22c3afaf207c496ab2c207eef0f3b961 : FAILURE in 32m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c5e57d82d6744afb0b1f82b2dc57f54 : SUCCESS in 56m 41s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18dbf5c142774e61897601ba1a65af49 : SUCCESS in 56m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9fdae611946c4469bfd02c65e46f327c : SUCCESS in 1h 34m 42s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/50dfd4540f0d4a1494ed9e1031b18553 : SUCCESS in 1h 08m 13s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/03065b4eb0814f6fbef3d78d50b0d636 : SUCCESS in 1h 28m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/483a9df0933a42b7b312ef18ae7beb52 : SUCCESS in 1h 34m 55s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/be1edc4f2bf84e61a28f49c99f7ebddf : SUCCESS in 1h 09m 09s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/79232ad8a6b44220a3a887f430db5c5c : SUCCESS in 1h 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"43c478b560d80d947cad384aa659d45651f928eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-03 05:57:27.000000000","message":"Patch Set 17: Commit message was updated.","accounts_in_message":[],"_revision_number":17},{"id":"e412097d9c4d04bb554305a692335cbe42264a98","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-03 06:26:47.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/17/openstack-check/tripleo-build-containers-rhel-8-master/a8ab5f3/ : SUCCESS in 28m 38s","accounts_in_message":[],"_revision_number":17},{"id":"25c4457b08d7f3887b2a60937cfd4f8bef024a1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 08:38:06.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/62b6aaadc80e479b9de50cb74dba8971 : SUCCESS in 39m 38s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/90b17ba91cc348caa1f9f42f057c5363 : SUCCESS in 46m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/136e740594d74b51ae526f96eb62c90c : SUCCESS in 1h 41m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5f5a809c36614e12a6f83d26f798a2eb : SUCCESS in 1h 24m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/66f14df143804ee0a6db3bd7dfa24893 : SUCCESS in 1h 39m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0c61671940fa4f61bcdc86a4bc7e4047 : SUCCESS in 1h 02m 10s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/869fdfcb4b0746a7b64085feabee5687 : SUCCESS in 58m 22s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/c6651c1ed0204d309c5f0f1871a79b25 : SUCCESS in 2h 25m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/593b5e9f717542398b65ca8256c19c1d : TIMED_OUT in 2h 33m 09s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/47f4c3c5294b4848b5e672e81344811c : TIMED_OUT in 2h 02m 08s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/98dc11548c27403698f914da9e053c16 : SUCCESS in 1h 58m 19s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/ea73ed66acc34459b894a4951beaf718 : FAILURE in 9m 57s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ed6285ec0d3d47ebb0f94717910886ca : FAILURE in 4m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ec160474e4bc4e738ae85ddc1ffcca82 : FAILURE in 5m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5ad848b19c774ecfbbd0e64b819739ff : SUCCESS in 58m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/850f1421a155460598e21c89ea036704 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69f3b2a0bde944cf8969645e33e37238 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de0169aacd2243b7bd8f45c1e42fae5f : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/544b0d7205be450696bb9e0ef829e01f : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/040697ce6a3a432b84179433354a8db8 : SUCCESS in 4m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e69a1f5ac37c410488856802493e9f39 : SUCCESS in 24m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2855f971a36940d5bd0ba5a77d291544 : SUCCESS in 39m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/15ead071b0f6461d9031dd500fda5898 : SUCCESS in 58m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/addca15a32044c29a4f4a9723afa42e9 : FAILURE in 7m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f9aad2be40b8475799d198148d0fe36e : FAILURE in 1h 05m 59s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/03424f5386e2494680145e61939a96a5 : SUCCESS in 1h 09m 29s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/08ac91b2d7ef469e85b629df8c3c12f8 : SUCCESS in 1h 31m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f67ebd38f79e4296b1364ee995645e27 : SUCCESS in 1h 49m 32s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/fb403afd2b5c4c34b68919e707fddf18 : FAILURE in 8m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/1d55ad6e269648e4b0f3368e96f8649b : FAILURE in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ed7b1197e26dc5307f23f26826bf668af05f4ca4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-03 09:02:03.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"20f691af8a50204825d0cb8001a2a310b9af8f37","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-03 09:39:00.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/17/openstack-check/tripleo-build-containers-rhel-8-master/8275ad4/ : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":17},{"id":"ba914dbc10d4451012c221dac4df593038996322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 11:18:29.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e1a5d7524869402fb39c13817c885eec : SUCCESS in 31m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/687119f3b46b4f7cb5072314147c15c0 : SUCCESS in 44m 49s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/be775317cbe042bc9125a739d3c69347 : SUCCESS in 1h 18m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fb201aa55c45459497c15a47ec44e8b0 : SUCCESS in 1h 36m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0a97ea84bb5649348efd76f44116112b : SUCCESS in 1h 28m 29s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c01947d142bc4a7993c060c3d60039e9 : SUCCESS in 1h 00m 54s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/945c1f3e09b445daad9629d03f02ef4b : SUCCESS in 49m 46s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/0012dedefd67443ca17765637a6bafe1 : SUCCESS in 2h 12m 06s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/3ab21f4a50f9449880612483bc340f22 : SUCCESS in 2h 10m 53s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/056053a37136435cae3b659c7d8e6041 : SUCCESS in 1h 44m 44s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/bdc665a885c84a8fb4ecd258280d6375 : TIMED_OUT in 2h 02m 47s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/7bffb8829f7b44a39c4b632dc6d49fc2 : SUCCESS in 1h 45m 25s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ebc027e676b244439bd64d1e0e24b846 : SUCCESS in 26m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a5fbd1c2ac9d497c8267d5b27f1528cd : SUCCESS in 22m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c8e9aa6b1a5d43c3b569c959eb4f3c94 : SUCCESS in 54m 00s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c2cc2fd49d74cdd8b45ab5ba5852a37 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f853d434c594ebb855653dd5b0ec083 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebe061f7fe6a4c2089a2a51d26cd6fea : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd90aa31690f40d18f6e5cfec8e63b1c : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a9d815f8fce45fbad38c9ef176ddfe4 : SUCCESS in 5m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a649b9c5698446c5b3205894d4713332 : SUCCESS in 25m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5df7ac68d1c2444db9273a5e1c150019 : SUCCESS in 43m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/efc9bdea45db48e2aeaad3ec43fa54be : SUCCESS in 55m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2ef7653ddef8460aa997d40eed23e048 : SUCCESS in 55m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ccf21827314141dc86621a20716ed250 : SUCCESS in 1h 07m 52s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a19f3f737c574691a23cf613b401b8d3 : SUCCESS in 1h 09m 26s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/ac9060e4fc4b4a18b0465a1d65a80340 : SUCCESS in 1h 44m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8fd7495bcfff46ed945d56f9786c168f : SUCCESS in 2h 08m 15s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/381ccccb3f09492a9f31efba43ddc4ad : SUCCESS in 1h 13m 27s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0893a7a147514227a7226183600e1803 : SUCCESS in 57m 28s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"77d71b8ec3f7546d20bf6f694b22f037a0e4944f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-05 08:48:48.000000000","message":"Topic changed from feature/healthchecks to bp/container-health-check","accounts_in_message":[],"_revision_number":17},{"id":"4fbe43bc08af09de685fe5871f03b30e20cebd40","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-05 12:27:35.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"ff139412ea296bf159f18fac2c07f261182bd14d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-05 13:01:37.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/18/openstack-check/tripleo-build-containers-rhel-8-master/b6bc431/ : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":18},{"id":"5015982fc9cb61441e1abad5bfc70009af7e27eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 14:05:43.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/3f4b24f9c0934566a992510feb1d4065 : SUCCESS in 31m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b2859dea22a74217b7b15107de37f1a3 : FAILURE in 35m 54s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/75b5c34d1ee34b25af3853f8135b7979 : FAILURE in 41m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/16b5660c4d704f6aa7897cfe19f7af7a : SUCCESS in 1h 13m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/62f1058d542b4215abd01ecb50dff8dd : FAILURE in 34m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/33a5d4afff6040cd840b4dcded951403 : SUCCESS in 49m 53s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/4b740f2887054cb8b8ef48ddedf5fe43 : SUCCESS in 58m 02s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/e057dc157bb94bd592a677b4da5e094e : FAILURE in 5m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/1549011f07a348719d751d51929c2576 : FAILURE in 39m 48s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/3d5069a9bb1249d4bbba4db4f25a8162 : FAILURE in 36m 29s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/e7f6c468a9844383b3b5318ec441c93d : FAILURE in 36m 46s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/118380ab6bfe493a8e314d69b1844b07 : FAILURE in 38m 46s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4aeb4219ffe54d289242715cdb7c04af : SUCCESS in 24m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/abe92c6a73674cdc8868e73e106376ba : FAILURE in 18m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7fae031fe47b42e989d1af2fb9799e1d : SUCCESS in 1h 01m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1e6a4768776d4777b510a8661ff7961a : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1018c5ed7184965a049c04e03124e3b : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c654e20c99fd4d3a9eecb467f8463712 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6fa4a24ece04f5f926a6a45fe15b28b : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fad5054963b44079ccdc433d8f76eb0 : SUCCESS in 4m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/30455bfca5c647a3a33aa512c32cd2c0 : FAILURE in 23m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/066e355d43614e2c8e9acfc62116c0e3 : SUCCESS in 31m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ccf8ffdb1f0f40f6b67fafad5469a775 : SUCCESS in 55m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/816b56cc3b434588934e952609e64d69 : SUCCESS in 56m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b4d824666d124045823de819654633b3 : SUCCESS in 1h 03m 09s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/ae811278361941c5a69dd7c98d6bcb2f : SUCCESS in 1h 07m 36s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/196f7dd7923c484b962ae39a19755e55 : SUCCESS in 1h 24m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/94ccbcb33ab04bce9307e881f8938199 : SUCCESS in 1h 31m 36s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/08167c7852bb481a803c640de58961fd : SUCCESS in 1h 13m 21s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/861b4ac25fda423abc7499d254a77ec6 : SUCCESS in 57m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3a037911b348aa57a32dcac8011c6d48978a7c22","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-06 08:15:26.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"230c6f10a07e0d38dcfb3cfb88faac54cbc03c56","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-06 08:51:01.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/18/openstack-check/tripleo-build-containers-rhel-8-master/883f807/ : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":18},{"id":"d130a95cf0f63ff174968bdd94a1e55c302d40de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 10:25:58.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1e36657ca5cc4cb19c24560af658f731 : SUCCESS in 31m 33s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/745c2af54769412ebc26474a2580136d : SUCCESS in 37m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8a2a622b16ac4a0cbe99c04ad37d90d3 : FAILURE in 56m 37s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6aa79ae8727d4b8db88ff087a0116eb7 : SUCCESS in 1h 14m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a9245b9d0f20497790e9357ea8cf6712 : FAILURE in 1h 28m 01s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fb6b22d6a3ff4399bcb5c7c1f9c885ef : SUCCESS in 43m 57s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/9db1e98d949c455c90d9a7237b0d45c2 : SUCCESS in 39m 36s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/4fcb8c480e444676988aaca970542873 : FAILURE in 1h 19m 16s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/efdaa34b963049939bfa064a3b388f46 : FAILURE in 2h 02m 45s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/1f51686b53de4eee8d51bdab69acbe3e : FAILURE in 1h 17m 23s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/787fbbc0a1364d8c8954d927a991b0dd : FAILURE in 5m 00s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/d3094ab8167f4e2f8bcd207a31e31ada : FAILURE in 56m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/940acec4a33b4d8cbb148be19617707c : SUCCESS in 18m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/29e25770cea84b438dcbaa8ffb2a759a : FAILURE in 18m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0d72836617284b5d9cdab82dc2f80e2b : SUCCESS in 57m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13a8cb60f7174624af1f522e52d4b16f : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc4cc6c708e4c1b892c33a0bb059479 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/979682bd56ed4e4ab7634404be502b11 : SUCCESS in 2m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/576d8720caf3440c998485af43b89e1f : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc410f819a149c393e8d103416daeb2 : SUCCESS in 4m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f42498bf8df940348e261b030e2a3b2b : FAILURE in 25m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/67b160ebcff74f84acc6cf75a568eea0 : SUCCESS in 28m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e22b34dbde744a46aeea94af0c57e557 : SUCCESS in 49m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b1a2be44e254297a9694a0254d8a32a : SUCCESS in 54m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a590765edfd1489fa0864cba3192a428 : SUCCESS in 59m 12s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/1675eac9f7f34d22809fcb9515da277a : SUCCESS in 1h 08m 09s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/a6dcb44496c147d5b2c7036c840f8700 : SUCCESS in 1h 17m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/7dcf94fa73d54ed6a48cd08aff6d478e : SUCCESS in 1h 49m 44s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/1422ecf4cf9946fea1b27ac6968a24c8 : SUCCESS in 1h 10m 06s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/99479ef534574fc487e0b41496adce16 : SUCCESS in 57m 08s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"c1c5c4cd6100fd2769a1b69aa5dc888443823c05","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-09 12:57:12.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"9c7694aeb822f5ed56b83685c254784d35504f89","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-09 13:31:15.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/28/676128/19/openstack-check/tripleo-build-containers-rhel-8-master/1cb6a80/ : FAILURE in 32m 29s","accounts_in_message":[],"_revision_number":19},{"id":"d6a92ea7c14c0dd0b43247b905118086ee9ff32e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 15:25:00.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/87ce04c426344a5782ddd4c16d5ad142 : SUCCESS in 31m 21s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/409a9f0edfba4a5ca18b027d138f735e : SUCCESS in 40m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/41eba8c86a3949af8da7172e7d44570d : SUCCESS in 1h 37m 29s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fdeeb0ece9c74e2db7fc2ccc88776205 : SUCCESS in 1h 18m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a7ad3176e4b04040b8cae2b8b7f3502d : SUCCESS in 1h 28m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9016b31950bd41ed94aca91f0152822b : SUCCESS in 52m 15s\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/aa6d27a0be584a3fb8101345b8b5bbb8 : SUCCESS in 49m 43s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/7080141b3a8946f99b403b1fda9a7d9f : SUCCESS in 1h 53m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/356dd682991741e7bd59b63fd43faade : FAILURE in 2h 21m 23s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/ad81780b7e32478f89065693189c85d0 : SUCCESS in 1h 55m 22s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ab32b4490af548cab34129ebab30cd17 : TIMED_OUT in 2h 04m 28s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/9849abf6fc8c40c68352e3534dcb9dc3 : SUCCESS in 1h 35m 42s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6cbc2e4053b841a1ab0bb8163f173ae4 : SUCCESS in 21m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/aa27f55370c248fea1c81688ecb16d23 : SUCCESS in 21m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/da373e0e12d7445190b335c3bf626d91 : SUCCESS in 56m 06s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83a91f95f1a947dabde13af9f5a2ec7d : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a15dda7d0742249a157092529b43e7 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0eff24909471498b9ce2fa77b0de321e : SUCCESS in 3m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa1554f5f6c642ecbf1792227ce9023e : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef51ae566983455eae79026d4b2885aa : SUCCESS in 4m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e6974b48e9548b08dc42822995d66b4 : SUCCESS in 23m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/169438422a964b498d7e6cc5816333a8 : SUCCESS in 24m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a83ddc642fd24c3fa408037f4d6ed376 : SUCCESS in 1h 01m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b717ac68a88442db8b6a592fd0f0c43c : SUCCESS in 55m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a0e29b39ae1f481fa309205250f553e4 : SUCCESS in 1h 05m 03s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/f6ff2ccc2e744d5cbc52e0e528e54eb5 : SUCCESS in 1h 06m 44s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/7b4052be333f418e9f4a105a370e4428 : SUCCESS in 1h 22m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/b6ab7607f2ea41bbba1aef39f1ef59d9 : FAILURE in 1h 37m 02s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/06b57dc0ce0747ed92152ffe249496d4 : SUCCESS in 1h 24m 17s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/c1c5771adb544b76a5ca75e3e367ca2c : SUCCESS in 54m 48s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"6e64ce9cb756c6ca1999234ec7a702df96e1d085","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-12-11 09:50:12.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"73c7fe1e60a4c79328249b6edc7031c403a4be9e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-13 09:04:24.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"d93ac92beec2db73916896298f8a7ff1440d70c0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-13 09:04:35.000000000","message":"Patch Set 19: Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"caee30bf0f45a9f93862e94df7d5602473b7007e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 09:04:48.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"eb0793617c0d42bad2dc2158808bcbb26c97f426","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 10:38:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":19},{"id":"b6ca2b38a7c1700d22e4ca24b74e4145b7f97780","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 10:38:52.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ca26b81e1d66471a85b7881e12265580 : SUCCESS in 1h 03m 50s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cab5476e2d2c47048478ff03a1656400 : SUCCESS in 1h 31m 38s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/470b2f029b1849d4b332cb7d507dec93 : SUCCESS in 1h 26m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/654c334e41934012b3121a6599f45349 : SUCCESS in 55m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b90e1d82b8f465f97d3b228cabe3d8c : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9813a84ee3c94b08849acd9c1edda7a2 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c52863d67b3d49c6af0503e92be5c137 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e08ed60e64d24da0aabe56c6150588ec : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/679c0c525b214c80856b0ce4003769ec : SUCCESS in 4m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b17eece3b972409a96289495c2e18f3d : SUCCESS in 34m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/18d42ca61519419db675b63626d5447f : SUCCESS in 1h 09m 54s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3d3c89b0096d4912b9edd1ca13decc1d : SUCCESS in 1h 12m 55s","accounts_in_message":[],"_revision_number":19},{"id":"aa3d604956885e271eaf4ac1b713b5ccfbecbc26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 10:41:02.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a51ddaf754d41a0a4024741c8a031de : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":19},{"id":"3721a4df35716b416183851178987b13544df796","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-16 12:22:06.000000000","message":"Patch Set 19:\n\n(1 comment)\n\nSorry I didn\u0027t review sooner.\n\nThe shell grepping and piping can be a bit error prone, so I might have gone for python but hopefully it will fine.","accounts_in_message":[],"_revision_number":19},{"id":"ebd319a62e5a4daaf048da86354f47156796e3c4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-16 13:14:33.000000000","message":"Patch Set 19:\n\n\u003e (1 comment)\n \u003e \n \u003e Sorry I didn\u0027t review sooner.\n \u003e \n \u003e The shell grepping and piping can be a bit error prone, so I might\n \u003e have gone for python but hopefully it will fine.\n\nSo - Docker Engine today does not have the capabilities to restart images based on their health status. So the statuses are purely informational - there\u0027s a custom docker image/script to restart docker containers based on health status. Here is some long running conversation on GitHub why Docker Engine is not doing that: https://github.com/moby/moby/issues/28400. TripleO is using it this way, so I don\u0027t think we need to improve it - but happy to do it based on some bug reports ;-)","accounts_in_message":[],"_revision_number":19},{"id":"93fe1e3108be56bd225411a47aecabb185d7b653","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-12-16 13:16:19.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"79851d7929027050d0d1e76aaa28e83c141dcc55","revisions":{"bd8fd0c10539c0ee2b0774ad0f931a0f29539c3c":{"kind":"REWORK","_number":1,"created":"2019-08-13 08:00:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/1"}}},"commit":{"parents":[{"commit":"38f8e6776996111e2eb60bd366abb3741a79379a","subject":"Merge \"Upgrade Monasca to InfluxDB 1.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38f8e6776996111e2eb60bd366abb3741a79379a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 08:00:02.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nThis patch imports healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd8fd0c10539c0ee2b0774ad0f931a0f29539c3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd8fd0c10539c0ee2b0774ad0f931a0f29539c3c"}]},"branch":"refs/heads/master"},"adcc7da08cc47643cd8c461845bbbb72b69d3a87":{"kind":"REWORK","_number":2,"created":"2019-08-13 08:32:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/2"}}},"commit":{"parents":[{"commit":"38f8e6776996111e2eb60bd366abb3741a79379a","subject":"Merge \"Upgrade Monasca to InfluxDB 1.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38f8e6776996111e2eb60bd366abb3741a79379a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 08:32:45.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nThis patch imports healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adcc7da08cc47643cd8c461845bbbb72b69d3a87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adcc7da08cc47643cd8c461845bbbb72b69d3a87"}]},"branch":"refs/heads/master"},"ee2a7b798f9f965d7c1f45caf93bac24de1a1dfa":{"kind":"REWORK","_number":3,"created":"2019-08-13 11:11:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/3"}}},"commit":{"parents":[{"commit":"38f8e6776996111e2eb60bd366abb3741a79379a","subject":"Merge \"Upgrade Monasca to InfluxDB 1.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38f8e6776996111e2eb60bd366abb3741a79379a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 11:10:58.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nThis patch imports healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee2a7b798f9f965d7c1f45caf93bac24de1a1dfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee2a7b798f9f965d7c1f45caf93bac24de1a1dfa"}]},"branch":"refs/heads/master"},"ce563efad6d8ca76a3ed1c37ec1c9df81844631c":{"kind":"REWORK","_number":4,"created":"2019-08-13 13:18:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/4"}}},"commit":{"parents":[{"commit":"38f8e6776996111e2eb60bd366abb3741a79379a","subject":"Merge \"Upgrade Monasca to InfluxDB 1.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38f8e6776996111e2eb60bd366abb3741a79379a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 13:18:00.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nThis patch imports healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce563efad6d8ca76a3ed1c37ec1c9df81844631c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce563efad6d8ca76a3ed1c37ec1c9df81844631c"}]},"branch":"refs/heads/master"},"7323c09e9c785b638ad748fa4ff81b44e9a2358e":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-08-13 13:19:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/5"}}},"commit":{"parents":[{"commit":"38f8e6776996111e2eb60bd366abb3741a79379a","subject":"Merge \"Upgrade Monasca to InfluxDB 1.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38f8e6776996111e2eb60bd366abb3741a79379a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 13:19:24.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nThis patch imports healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7323c09e9c785b638ad748fa4ff81b44e9a2358e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7323c09e9c785b638ad748fa4ff81b44e9a2358e"}]},"branch":"refs/heads/master"},"4b0273892a579bf32ffab0001bd19773b3622410":{"kind":"REWORK","_number":6,"created":"2019-08-13 16:08:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/6"}}},"commit":{"parents":[{"commit":"38f8e6776996111e2eb60bd366abb3741a79379a","subject":"Merge \"Upgrade Monasca to InfluxDB 1.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38f8e6776996111e2eb60bd366abb3741a79379a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 16:08:47.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nThis patch imports healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b0273892a579bf32ffab0001bd19773b3622410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b0273892a579bf32ffab0001bd19773b3622410"}]},"branch":"refs/heads/master"},"b9f267569f2121611e104741de5e028193c67ee2":{"kind":"REWORK","_number":7,"created":"2019-11-22 14:56:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/7"}}},"commit":{"parents":[{"commit":"861db34ffddab509d9801088cb37871121adce08","subject":"Merge \"Designate: pool-manager has been removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/861db34ffddab509d9801088cb37871121adce08"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-22 14:56:55.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nThis patch imports healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9f267569f2121611e104741de5e028193c67ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9f267569f2121611e104741de5e028193c67ee2"}]},"branch":"refs/heads/master"},"2a0c8ba6cbab61052be3969a431b564c1f994af4":{"kind":"REWORK","_number":8,"created":"2019-11-26 10:37:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/8"}}},"commit":{"parents":[{"commit":"861db34ffddab509d9801088cb37871121adce08","subject":"Merge \"Designate: pool-manager has been removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/861db34ffddab509d9801088cb37871121adce08"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-26 10:36:57.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nThis patch imports healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a0c8ba6cbab61052be3969a431b564c1f994af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a0c8ba6cbab61052be3969a431b564c1f994af4"}]},"branch":"refs/heads/master"},"d462a37b0ecac8be460d5d10db396cbf55efce70":{"kind":"REWORK","_number":9,"created":"2019-11-29 06:46:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/9"}}},"commit":{"parents":[{"commit":"861db34ffddab509d9801088cb37871121adce08","subject":"Merge \"Designate: pool-manager has been removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/861db34ffddab509d9801088cb37871121adce08"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-29 06:45:48.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nThis patch imports healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d462a37b0ecac8be460d5d10db396cbf55efce70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d462a37b0ecac8be460d5d10db396cbf55efce70"}]},"branch":"refs/heads/master"},"28544edc830360b4e223a59be2cea658377f119e":{"kind":"REWORK","_number":10,"created":"2019-11-29 06:52:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/10"}}},"commit":{"parents":[{"commit":"861db34ffddab509d9801088cb37871121adce08","subject":"Merge \"Designate: pool-manager has been removed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/861db34ffddab509d9801088cb37871121adce08"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-29 06:52:05.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nBase for this patch were healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28544edc830360b4e223a59be2cea658377f119e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28544edc830360b4e223a59be2cea658377f119e"}]},"branch":"refs/heads/master"},"b35342cffdea5e21f9eeaa2343631d808ccf2282":{"kind":"REWORK","_number":11,"created":"2019-11-29 13:37:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/11"}}},"commit":{"parents":[{"commit":"9938857d7d331743d71328831241edf21e42d160","subject":"Merge \"Ensure lsb-release exists on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9938857d7d331743d71328831241edf21e42d160"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-29 13:37:41.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nBase for this patch were healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b35342cffdea5e21f9eeaa2343631d808ccf2282"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b35342cffdea5e21f9eeaa2343631d808ccf2282"}]},"branch":"refs/heads/master"},"309e4efb6555120e22bcd78997b2816d644b849d":{"kind":"REWORK","_number":12,"created":"2019-11-29 13:49:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/12"}}},"commit":{"parents":[{"commit":"9938857d7d331743d71328831241edf21e42d160","subject":"Merge \"Ensure lsb-release exists on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9938857d7d331743d71328831241edf21e42d160"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-29 13:48:55.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nBase for this patch were healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/309e4efb6555120e22bcd78997b2816d644b849d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/309e4efb6555120e22bcd78997b2816d644b849d"}]},"branch":"refs/heads/master"},"3eef266ef2813be384dc2e0d55ee23f584406019":{"kind":"REWORK","_number":13,"created":"2019-11-29 14:25:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/13"}}},"commit":{"parents":[{"commit":"9938857d7d331743d71328831241edf21e42d160","subject":"Merge \"Ensure lsb-release exists on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9938857d7d331743d71328831241edf21e42d160"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-11-29 14:25:07.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nBase for this patch were healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3eef266ef2813be384dc2e0d55ee23f584406019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3eef266ef2813be384dc2e0d55ee23f584406019"}]},"branch":"refs/heads/master"},"3e79d6df6fc0148106c84e45f17a35ce03c85e46":{"kind":"REWORK","_number":14,"created":"2019-12-02 12:09:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/14"}}},"commit":{"parents":[{"commit":"9938857d7d331743d71328831241edf21e42d160","subject":"Merge \"Ensure lsb-release exists on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9938857d7d331743d71328831241edf21e42d160"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-02 12:09:00.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nBase for this patch were healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e79d6df6fc0148106c84e45f17a35ce03c85e46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3e79d6df6fc0148106c84e45f17a35ce03c85e46"}]},"branch":"refs/heads/master"},"1c00b3413c59e9f928ce145fe3f0a5c7d900caad":{"kind":"REWORK","_number":15,"created":"2019-12-02 12:34:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/15"}}},"commit":{"parents":[{"commit":"9938857d7d331743d71328831241edf21e42d160","subject":"Merge \"Ensure lsb-release exists on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9938857d7d331743d71328831241edf21e42d160"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-02 12:34:45.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nBase for this patch were healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c00b3413c59e9f928ce145fe3f0a5c7d900caad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c00b3413c59e9f928ce145fe3f0a5c7d900caad"}]},"branch":"refs/heads/master"},"903e25290ba97241699a05b37de79904095eb08f":{"kind":"REWORK","_number":16,"created":"2019-12-02 13:22:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/16"}}},"commit":{"parents":[{"commit":"9938857d7d331743d71328831241edf21e42d160","subject":"Merge \"Ensure lsb-release exists on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9938857d7d331743d71328831241edf21e42d160"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-02 13:22:03.000000000","tz":0},"subject":"WIP: Build images with healthcheck scripts","message":"WIP: Build images with healthcheck scripts\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nBase for this patch were healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/903e25290ba97241699a05b37de79904095eb08f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/903e25290ba97241699a05b37de79904095eb08f"}]},"branch":"refs/heads/master"},"563f183f5e03ecfd11126d22db5ee214c6b4392e":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2019-12-03 05:57:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/17"}}},"commit":{"parents":[{"commit":"9938857d7d331743d71328831241edf21e42d160","subject":"Merge \"Ensure lsb-release exists on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9938857d7d331743d71328831241edf21e42d160"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-03 05:57:23.000000000","tz":0},"subject":"Build images with healthcheck scripts for core services","message":"Build images with healthcheck scripts for core services\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nBase for this patch were healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/563f183f5e03ecfd11126d22db5ee214c6b4392e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/563f183f5e03ecfd11126d22db5ee214c6b4392e"}]},"branch":"refs/heads/master"},"d89b72625684eb7a5e8329dfa9c19eca90c115f3":{"kind":"REWORK","_number":18,"created":"2019-12-05 12:27:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/18"}}},"commit":{"parents":[{"commit":"9938857d7d331743d71328831241edf21e42d160","subject":"Merge \"Ensure lsb-release exists on debian/ubuntu\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9938857d7d331743d71328831241edf21e42d160"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-05 12:27:31.000000000","tz":0},"subject":"Build images with healthcheck scripts for core services","message":"Build images with healthcheck scripts for core services\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nBase for this patch were healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d89b72625684eb7a5e8329dfa9c19eca90c115f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d89b72625684eb7a5e8329dfa9c19eca90c115f3"}]},"branch":"refs/heads/master"},"79851d7929027050d0d1e76aaa28e83c141dcc55":{"kind":"REWORK","_number":19,"created":"2019-12-09 12:57:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/676128/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/676128/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/676128/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/676128/19"}}},"commit":{"parents":[{"commit":"f1e236522525361b8db1c05bee2b53b1c2a4dc5e","subject":"Merge \"releasenotes: fix file location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f1e236522525361b8db1c05bee2b53b1c2a4dc5e"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-13 07:55:58.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-12-09 12:57:10.000000000","tz":0},"subject":"Build images with healthcheck scripts for core services","message":"Build images with healthcheck scripts for core services\n\nblueprint container-health-check\n\nNow docker supports healthcheck directive[0] in Dockerfile,\nwhich is helpful for checking container aliveness.\n\nBase for this patch were healthcheck scripts from tripleo[1].\n\n[0] https://docs.docker.com/engine/reference/builder/#healthcheck\n[1] https://github.com/openstack/tripleo-common/tree/master/healthcheck\n\nChange-Id: Ifc32713cd4d6b45c74855c0ebf5bb65784701d44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79851d7929027050d0d1e76aaa28e83c141dcc55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79851d7929027050d0d1e76aaa28e83c141dcc55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
