)]}'
{"id":"openstack%2Fkolla~676185","triplet_id":"openstack%2Fkolla~master~I472af9ec8cee3462beb5aca04ea6171b79933abe","project":"openstack/kolla","branch":"master","topic":"monasca_grafana_app","hashtags":[],"change_id":"I472af9ec8cee3462beb5aca04ea6171b79933abe","subject":"Install Monasca plugin for Grafana by default","status":"MERGED","created":"2019-08-13 12:59:39.000000000","updated":"2020-08-08 21:07:58.000000000","submitted":"2020-08-08 21:06:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676185-1596920808808-12991b16","meta_rev_id":"5a8cfed3c4a168a9e75f90cd09cacc2b7dc00fcc","_number":676185,"virtual_id_number":676185,"owner":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-08 21:06:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-08-07 07:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-08-05 18:30:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":2,"date":"2020-08-08 19:37:54.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-08-07 16:14:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":1,"date":"2020-08-08 19:37:54.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-08-05 18:30:06.000000000","_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-05 15:27:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-13 16:28:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-20 15:36:26.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-05 12:14:46.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-08-05 15:27:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2020-08-07 07:43:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-08-07 16:14:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-08-08 19:37:54.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-08-08 21:06:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4990936eaf1e505d609032b8f9b05c7757aa5c6","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-08-13 12:59:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9883023210bfb43e3c0b8f28c244e7b66190c593","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 13:55:54.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/85/676185/1/openstack-check/tripleo-build-containers-rhel-8-master/1c06b20/ : FAILURE in 55m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ae44641af887dcf00b9b3248303d2a10e07a8ebb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 14:44:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/29a2f4df8ffa4d02a3314ec8fd7099fc : FAILURE in 45m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bda7b2b1941e441a8141bb1838293b23 : FAILURE in 52m 37s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f712638d1fe1414eba00f1321c8b9fe1 : SUCCESS in 1h 27m 37s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3223e39cd0f24c87b758494f7d4d9293 : SUCCESS in 1h 13m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0d02cdb5db604a5aa9df00718c7cfb34 : SUCCESS in 1h 38m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e1e4e6f9fd4941168fa5e9b900ed0680 : SUCCESS in 48m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa6b416532b849028bb026e9378f0fe8 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a28af139cc28473486c239a9e99f8cd3 : SUCCESS in 4m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58db46257ec049d582ae03be759dcb98 : SUCCESS in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a6c5075f97e49f6a5ca6528f198cb7d : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03f445df10b14c9097711a56824777eb : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa79b76a4824efd8bdb56096f0604f2 : SUCCESS in 3m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d2cff16bc71640fc8cfb1723f317cd2a : FAILURE in 27m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3a3b4c653d6847a3a51f9ba22fac48fd : FAILURE in 35m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/720cae7b4a2845bb9b17104a9d5c17b3 : SUCCESS in 1h 00m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2836fa03113144bbbe1cb8822fd0c977 : SUCCESS in 46m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/36bc84af8f674744b7936ee1b6ec1111 : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"4b50615079038e5df3833466a42a1bd452cc9215","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-08-13 15:43:29.000000000","message":"Patch Set 1:\n\nrecheck mirror issue","accounts_in_message":[],"_revision_number":1},{"id":"337eb5f9f4f5bb41d8fa9c1f3adee72724a69e2c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-13 16:28:43.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93d805de29cd0bef25c3199b34dfdfc63db78d54","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-13 16:41:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/85/676185/1/openstack-check/tripleo-build-containers-rhel-8-master/10204d6/ : FAILURE in 43m 36s","accounts_in_message":[],"_revision_number":1},{"id":"34a8c55e6bf113c2693bb85c89c23fa4b752dbc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 17:08:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/43f41b3489f8445c88bc3e669f1bc3d6 : FAILURE in 49m 31s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/908ad6d30a2f45528bd677d3a542231f : FAILURE in 48m 12s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/a37d2695572948d5aa46e6e0e5459c03 : SUCCESS in 1h 18m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/df6d8c5c52fd4d3d909c4e3903fe46d3 : SUCCESS in 1h 14m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/66cea510ef7c40a2bc2c101500124fe2 : SUCCESS in 1h 20m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/9bbbe4d859da4b4ca3e4fd7b51a8238a : SUCCESS in 53m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b5d6cc00949461ab80922037a38ad15 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21d121615334ae0b88a4bf2a763f6ed : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92ee5adc12c243f69a7642ececa97a56 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/599c495ffa9f49ad8b86ad300bec272f : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9267bc4eebd8449fb3d9ccfc03aef0eb : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5678cae3e4be49b58c59a7192515ef98 : SUCCESS in 3m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4df1a3f630b14486ac23aa0d237baff9 : FAILURE in 33m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc434780e72c45a393d88d31fbaf26b0 : FAILURE in 31m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/00ff6058466541a48ad2bc8e9d9b0d67 : SUCCESS in 57m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83fd9d303c6a46d2bf0a5bc7d0958afe : SUCCESS in 59m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/48872143295941d49e60d7f2efda49a3 : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9fe53bb8848a18a768a1a836a63160ef75533457","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-08-14 07:58:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"279fdcecd24267ee1e6b240ca563cf97ec3b3a14","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-14 09:08:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/85/676185/2/openstack-check/tripleo-build-containers-rhel-8-master/351cdd4/ : FAILURE in 1h 08m 23s","accounts_in_message":[],"_revision_number":2},{"id":"eed2f54b321ec49a71bc2fdff4c6cf41fc2a19ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 09:40:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fae61f9740724aad93f13e27f40c092c : FAILURE in 45m 04s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/448686acc7e94d17a3fcc76d4da3d3ef : FAILURE in 53m 44s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3c81033f78914d168198cda91f09d70c : SUCCESS in 1h 23m 59s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ecc92dd9414d43289ce7fe66c060cb55 : SUCCESS in 1h 07m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/903870458b394c1ba9873930b4819624 : SUCCESS in 1h 36m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7981532df5754a78b83c85dc5f25820b : SUCCESS in 48m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e43dd73484e446838f763f734cb22279 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d932cd09cba145488a88825ad8a30da8 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82f4bf1e4b9b4bb880aabcbf4ac18564 : SUCCESS in 2m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9735d21a91c6407d9d893bed9b5af350 : SUCCESS in 3m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a9a85e581402405f91ac427592eafd91 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/303ac80be29141609974fcc5f886c84c : SUCCESS in 2m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e59a7532e92446fbf6c7b5103c0c0b7 : FAILURE in 51m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1aa06c69659d46aaad0ff6be34baed26 : FAILURE in 29m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4f5014a0c524ba6bc465c1b0f73d64a : SUCCESS in 55m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a054fd9ebd584e9895c640936dfe8dcf : SUCCESS in 1h 13m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d87b1ce8d067434ead8537c99782d642 : SUCCESS in 1h 21m 53s","accounts_in_message":[],"_revision_number":2},{"id":"a73af3fe84f61e10f67e86c6aa3d1f6a18a9da33","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-08-15 10:59:39.000000000","message":"Patch Set 2: Workflow-1\n\n\u003e - kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/448686acc7e94d17a3fcc76d4da3d3ef\n \u003e : FAILURE in 53m 44s\n \u003e - kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1aa06c69659d46aaad0ff6be34baed26\n \u003e : FAILURE in 29m 08s\n\nThese two fail due to the npm registry url defined in the base container (http://mirror.iad.rax.openstack.org:8080/registry.npmjs/) preventing an npm install command in monasca_grafana from completing. Monasca Grafana App build works when using default npm registry. \nNeeds investigation into how this registry differs in config.","accounts_in_message":[],"_revision_number":2},{"id":"fc76ebda6a23c590c5a287227e943a08445aca09","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-20 14:51:57.000000000","message":"Patch Set 2:\n\nAny update on this?","accounts_in_message":[],"_revision_number":2},{"id":"d8b0791826f766a57a989d2062aed4d9981666f5","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-20 14:59:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"28f52ee17b60bf9aaa536dec01291c83b3b47595","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-11-20 15:36:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/85/676185/2/openstack-check/tripleo-build-containers-rhel-8-master/c3950e7/ : SUCCESS in 35m 24s","accounts_in_message":[],"_revision_number":2},{"id":"cb32ebf1c70aa7f43a2d9aefe92c03d7d9c490f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-20 17:10:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cddbb175fbc843fc971845bd78b64864 : FAILURE in 49m 52s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8fe78f5a37da42768e8606375cdc5342 : FAILURE in 56m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/58da1aea0af842b397e1d0ce3dfdca38 : SUCCESS in 1h 35m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e1802ff2d08b41a6b3fd90b04ce13c37 : SUCCESS in 1h 19m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/b139c2ce4434436fab359ee840da3e18 : SUCCESS in 1h 57m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/87fdda7986374553aa92cf6c808661dc : SUCCESS in 50m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/653e06f655664771813f06180a1820ac : SUCCESS in 23m 01s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4f02fd70e047433f940aee4aa825fe60 : SUCCESS in 21m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/66888e10b86545819e6be4835b336391 : SUCCESS in 54m 14s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e18a8383f9c944cdbb41e0a16d4db0e7 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/890d0434e95b4734aa689cccaa5e837e : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/702720146ff9427eb949882f7722a6aa : SUCCESS in 2m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c5ab137aa84455dba5a7e4b773a6ce3 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e3ede44037242ddbb55a7891dc1c2dc : SUCCESS in 5m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d0c81a120773475eabeab45317ac1a46 : FAILURE in 24m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/70e1fadb927e47ada0583d0422f6d6bd : FAILURE in 21m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c116bfea2814b97ba2a367bf8647bc1 : FAILURE in 12m 59s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/56548cc3dc804301945ff5b3e364fea5 : FAILURE in 28m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3331097018bd412eae7b1a43f54eb1e7 : FAILURE in 15m 24s","accounts_in_message":[],"_revision_number":2},{"id":"82622a281b7658fb88f51d9dd8feae1fbc213a62","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-08-05 09:00:58.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"d6a8694c1e6e1786cfebbb4e7dfcfad6c2e1b377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 10:23:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/aa9682523f6449d7b5bed51a2ce94095 : SUCCESS in 23m 56s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/85e5d525cbae43fc98bd2b6b71d03a3c : SUCCESS in 42m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/80e6e22e038a4627bf0fbd1948d85a72 : SUCCESS in 1h 18m 59s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/abd2c929e28b46d191449c74fe60a738 : SUCCESS in 1h 21m 04s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fb69d3c36fef4edba56a6158f6869b6f : SUCCESS in 1h 21m 08s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/5c184f9a905446dcb560bd3df864101e : SUCCESS in 46m 41s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f1ecef1e36b7419c950e181c5d0c1da0 : SUCCESS in 20m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/966e3470a33c4aabb13997feceab0bab : SUCCESS in 19m 19s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5da227d9def243c4a1c0ac673d366c22 : SUCCESS in 1h 00m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0b573186db24f9b991ab49224f06938 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b06723555e4bd0a3ce9c2b5c744a88 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cbaafc5e9904849b201ebbd6727002c : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b631eae9592140edaa281c280cbfe7eb : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5049b5fc0d724dbcb93d88c9e3f3d8b7 : SUCCESS in 8m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0736ad250a1f449ab2c1d5bd8088dd73 : SUCCESS in 28m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b11dbf3f1e849ac91a47505344de1b1 : SUCCESS in 18m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/114db0d69e0943448a222b4c0693498e : SUCCESS in 45m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ab655ff8f6944c08bb1a20ab89b43d9 : SUCCESS in 58m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d66b710f3bd04d30b9f5c63beb5bbc69 : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":3},{"id":"32896ba70cbb9bdfe0542d08146cfebb60d8f6f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 10:26:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1e60644b443c43ef9d08328f8c14b1a5 : SUCCESS in 53m 05s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fe05f55c14594563aa19748addd8cf73 : SUCCESS in 31m 12s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8fcebf96e2441e4b835c50ef42b8ada : SUCCESS in 1h 20m 06s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/eeaf9676ae9d4d40831d0825d7eca469 : SUCCESS in 34m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5ffaae3d7456f2657787fe39c480c7c3931415a0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-08-05 10:32:37.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"17a915bd88c8f5d207a7b9132cb5af4fe4ccfe81","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-08-05 12:14:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4ed936617d94257c540a4245a0aa86cdc623bec1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-08-05 13:31:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"345d75137087c054eafb4a12928542a0e361bf33","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-08-05 13:33:06.000000000","message":"Patch Set 4: Backport-Candidate+1","accounts_in_message":[],"_revision_number":4},{"id":"b0e403de5ed9e737f264c200c87d7de77b5ae6ff","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-05 13:33:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"daa1562cbf9bf1847d94e87ceb38ea25a8c53e12","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-08-05 13:52:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e3e7019aaa3468f63d34d27393d3a03e738afad1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-08-05 13:59:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"08de4abc9108787d1834c0bb0f26a8ef260964db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 15:24:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/6ccd2a2827fb4b32ab89a2031dceb219 : SUCCESS in 58m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea9e879d016f406f93063cdaea49baf8 : SUCCESS in 36m 28s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/20e4994437824d20b4a264825ef1e11b : SUCCESS in 1h 25m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/52d139fd2e014d478454dc52adb7d33d : SUCCESS in 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7e3b79857da05d2f08543bdfbbe2b911a8f2351a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 15:25:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/01411c753ef44d27875495f6cf11b3a9 : SUCCESS in 41m 46s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/938cbd64ef5647ff8ed45489a4d7d3c8 : SUCCESS in 45m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/929ec8014bbc437099aa996014c8986e : SUCCESS in 1h 17m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6859f1d692714057a7dc139aa41b043c : SUCCESS in 1h 12m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1337f4d35ab242048808a37382ba3286 : SUCCESS in 1h 21m 59s\n- tripleo-build-containers-centos-8 https://zuul.opendev.org/t/openstack/build/f566357628c94aff99ebfc0dd6c9e6df : SUCCESS in 56m 45s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1c0a1b838b334da7b3f0832dda2f2cc1 : SUCCESS in 1h 21m 14s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2dd19343d4104e0daed80029e31d8a6e : SUCCESS in 41m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d7bd509dae074355b8b65a98d0b74f61 : SUCCESS in 58m 02s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a786b2886d174133a59a640dc7806ec2 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0890a44c506f485fa3c355f48dc44511 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc3e66d788f644e9be80e6013f52af96 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad769439c9724f0cbd148f27ae670454 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7bd366191ed45b59074797f0f00448a : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3142f0c3efc4d16ab5ac0ce000fa549 : SUCCESS in 5m 37s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8d2b40eebc83451ca248eb2af0e449d2 : SUCCESS in 33m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4feae4f822f348c385e065103289e3c6 : SUCCESS in 22m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a8c661639cd4a17a9a0c2cdaf2d84c3 : SUCCESS in 1h 06m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7047ab3efe694d739584147c0f165608 : SUCCESS in 1h 06m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6133d0828a154317a49066554658a151 : SUCCESS in 1h 10m 43s","accounts_in_message":[],"_revision_number":5},{"id":"ee6ef3d9f3ca0fd83545f0104f882912a9d7df1d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2020-08-05 15:27:02.000000000","message":"Patch Set 5: Backport-Candidate+1","accounts_in_message":[],"_revision_number":5},{"id":"70f003956b02cf3325c7bcc1a8353bb217c707be","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-08-05 18:30:06.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+1\n\nLGTM. Thanks for picking this up for me folks","accounts_in_message":[],"_revision_number":5},{"id":"511508d61698d476a99c4d4744331279b014433d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-07 07:43:27.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"41f51ea37c4b994e67885417147b0d367f10d7c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-08-07 16:14:17.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c40e73aaa382044becd4d9db93c64808ee227a3a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-08-08 19:37:54.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"e55e6d4a1b8f72d21101a68c39538bb752715be0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 19:38:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f86716fd2398a0bd4712c21ee3c3b9f4be861d9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 21:06:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9cc1d8176f39496cb3bca6aaf5cd3480 : SUCCESS in 29m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9cb9bc715332476cbd1ef5226ff5b4bc : SUCCESS in 1h 27m 51s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0056cd3c86b046139717dc4e3b8a86fc : SUCCESS in 1h 17m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6c61e90efffb441b846bc1bc4f5743db : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92472fabb0b0438491dcc6c66a2f068b : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6e21f995fe14147be5aa1c507d0c276 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e415006718a94a30a270870c57f8de3e : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf905f50e514f7db8de5f5ec4265a5d : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d979f6b40ce140d78d4747d1872544f6 : SUCCESS in 4m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/888e0e7f09d94b09914d6a56d924a1c8 : SUCCESS in 31m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64d191dad4914e4e83a94c0526a0e8ca : SUCCESS in 58m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a059f8f0646444d98d942b3e75c7eb62 : SUCCESS in 1h 12m 46s","accounts_in_message":[],"_revision_number":5},{"id":"73f1a91811da140eaebfafca3860d0cbac56b000","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 21:06:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"d7d1f5b2e717e2cab7f26ee29c1c89c35e4d5b42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-08 21:07:58.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474f505fcfd345948eec2e5d3b2b5b4c : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17b1dd1905dd4e818cb95c440ed91c1d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2f92e4341f58d20e5277ebe2b1372b3ae00dc960","revisions":{"ca3eb62d9b855071c4ced440c87c5ce6eb47e2ec":{"kind":"REWORK","_number":1,"created":"2019-08-13 12:59:39.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/85/676185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/676185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/676185/1"}}},"commit":{"parents":[{"commit":"65c0e2acb9d8e968178339bfbd517c9a849fc27a","subject":"Merge \"Support customising MariaDB log directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65c0e2acb9d8e968178339bfbd517c9a849fc27a"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-08-13 09:45:35.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-08-13 12:59:05.000000000","tz":60},"subject":"Install Monasca plugin for Grafana by default","message":"Install Monasca plugin for Grafana by default\n\nChange-Id: I472af9ec8cee3462beb5aca04ea6171b79933abe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca3eb62d9b855071c4ced440c87c5ce6eb47e2ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ca3eb62d9b855071c4ced440c87c5ce6eb47e2ec"}]},"branch":"refs/heads/master"},"2e1512a676d7c761f7ded0389ab69c230cd531e5":{"kind":"REWORK","_number":2,"created":"2019-08-14 07:58:44.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/85/676185/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/676185/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/676185/2"}}},"commit":{"parents":[{"commit":"65c0e2acb9d8e968178339bfbd517c9a849fc27a","subject":"Merge \"Support customising MariaDB log directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65c0e2acb9d8e968178339bfbd517c9a849fc27a"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-08-13 09:45:35.000000000","tz":60},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-08-14 07:58:23.000000000","tz":60},"subject":"Install Monasca plugin for Grafana by default","message":"Install Monasca plugin for Grafana by default\n\nChange-Id: I472af9ec8cee3462beb5aca04ea6171b79933abe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e1512a676d7c761f7ded0389ab69c230cd531e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2e1512a676d7c761f7ded0389ab69c230cd531e5"}]},"branch":"refs/heads/master"},"e564ea2d76df06296d18b98dbeb7b255f7c24a01":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-05 09:00:58.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/85/676185/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/676185/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/676185/3"}}},"commit":{"parents":[{"commit":"e18246eb986b38f0d955e7df25ef320cfc4e562c","subject":"Merge \"Remove TODO in libvirt script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e18246eb986b38f0d955e7df25ef320cfc4e562c"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-08-13 09:45:35.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2020-08-05 09:00:58.000000000","tz":0},"subject":"Install Monasca plugin for Grafana by default","message":"Install Monasca plugin for Grafana by default\n\nChange-Id: I472af9ec8cee3462beb5aca04ea6171b79933abe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e564ea2d76df06296d18b98dbeb7b255f7c24a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e564ea2d76df06296d18b98dbeb7b255f7c24a01"}]},"branch":"refs/heads/master"},"90fa6db702601e2ef189ea3de748cc35f770d4c8":{"kind":"REWORK","_number":4,"created":"2020-08-05 13:31:08.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/85/676185/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/676185/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/676185/4"}}},"commit":{"parents":[{"commit":"e18246eb986b38f0d955e7df25ef320cfc4e562c","subject":"Merge \"Remove TODO in libvirt script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e18246eb986b38f0d955e7df25ef320cfc4e562c"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-08-13 09:45:35.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2020-08-05 13:31:07.000000000","tz":0},"subject":"Install Monasca plugin for Grafana by default","message":"Install Monasca plugin for Grafana by default\n\nChange-Id: I472af9ec8cee3462beb5aca04ea6171b79933abe\nCo-Authored-By: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90fa6db702601e2ef189ea3de748cc35f770d4c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90fa6db702601e2ef189ea3de748cc35f770d4c8"}]},"branch":"refs/heads/master"},"2f92e4341f58d20e5277ebe2b1372b3ae00dc960":{"kind":"REWORK","_number":5,"created":"2020-08-05 13:52:38.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/85/676185/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/676185/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/676185/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/676185/5"}}},"commit":{"parents":[{"commit":"e18246eb986b38f0d955e7df25ef320cfc4e562c","subject":"Merge \"Remove TODO in libvirt script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e18246eb986b38f0d955e7df25ef320cfc4e562c"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-08-13 09:45:35.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2020-08-05 13:52:21.000000000","tz":0},"subject":"Install Monasca plugin for Grafana by default","message":"Install Monasca plugin for Grafana by default\n\nChange-Id: I472af9ec8cee3462beb5aca04ea6171b79933abe\nCo-Authored-By: Bartosz Bezak \u003cbartosz@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f92e4341f58d20e5277ebe2b1372b3ae00dc960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f92e4341f58d20e5277ebe2b1372b3ae00dc960"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
