)]}'
{"id":"openstack%2Fkolla~676377","triplet_id":"openstack%2Fkolla~master~Iee51a5c340d071e653afd4ecd4996d04bd9cfb7f","project":"openstack/kolla","branch":"master","topic":"ci-ansible-gate","hashtags":[],"change_id":"Iee51a5c340d071e653afd4ecd4996d04bd9cfb7f","subject":"CI: Gate on jobs kolla-ansible already gates on","status":"MERGED","created":"2019-08-14 09:16:58.000000000","updated":"2019-08-21 09:19:48.000000000","submitted":"2019-08-21 09:16:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676377-1566379004726-c3f9ea31","meta_rev_id":"220d86f9056bf09a96dd60e1d674692797c13955","_number":676377,"virtual_id_number":676377,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-21 09:16:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-08-21 07:42:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-08-20 13:08:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":2,"date":"2019-08-20 13:17:40.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2019-08-21 07:42:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-08-20 12:51:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 13:08:01.000000000","updated_by":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"reviewer":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"state":"REVIEWER"},{"updated":"2019-08-20 13:17:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-21 07:42:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-08-21 09:16:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acf7b327e37bea3f77d5816bae4f905a01d3ea25","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-14 09:16:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62aa1b66c86452673a9a6bdf81e71902116c4af3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-14 09:23:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"edfa95ba97204ca1fe0652c39b3a37b7b2f64ebe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-14 09:25:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"438d1c88fb689141a4ac95552f4f10afdf7b221f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-14 09:34:16.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"504877ee90cbc6fb88bbb763fd95a83f20f4afb1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 10:58:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fb41e03811294f168dd02b76d05acc25 : SUCCESS in 42m 18s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/972a3e98f0a74d6d9577d18a05df35e7 : SUCCESS in 46m 33s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/937211473e054ae9a1b13c03402aa94f : SUCCESS in 1h 18m 16s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4609a76a6dfc40158f2e79348fe94b6d : SUCCESS in 1h 10m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/317ebf2ad44d4ec2bcc7db55ec7041a7 : SUCCESS in 1h 26m 12s (non-voting)\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/78f9cb4fab9343d283a4f90af0ad8768 : SUCCESS in 42m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cb6e1d989ec41e99d322b87001ae0c8 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/155b74231a6640dbadc66887b7ba3518 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b92aa03fcc574ac3b41a33fdb014cdc9 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40eecbfa07d7477691310fcd595a4f26 : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d282de7efe5b4f0ba24e6ed36befcbba : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef81acc7398441893c8741b6ba72995 : SUCCESS in 2m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/84630f3a02a7463e9e51d36aba3394b6 : SUCCESS in 26m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bfde8759a11541c489fd8c882363f3ca : SUCCESS in 40m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/486ac1d858034c90816415394539d4ad : SUCCESS in 58m 19s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f96cde00d22c41478b0c66c3215a7f70 : SUCCESS in 56m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a0202aab23034ee5a9bcf6ba9e736073 : SUCCESS in 1h 00m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29b78c70adfa2f0c91cf9275432af9eb0989026c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-14 13:23:52.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8283b561a653127fdce723c74c1900076fdbab4d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-14 14:58:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db49bc86650e3cfc7ac7c59b022a25fe98db8693","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-14 19:13:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7065647deae7bee7e222641d18a711e0462ebdfc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-14 19:15:11.000000000","message":"Patch Set 2:\n\n@mgoddard - realized this is full value when just adding better gating, the other stuff is independent and I abandonded the idea for now - need to think of something else, thanks for enlightening this poor soul of mine","accounts_in_message":[],"_revision_number":2},{"id":"41727685041edcd15e04566f9e6cdb29aee26ca3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 20:41:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/13aa32f06bb04e96ac77ac18ff65a6f0 : SUCCESS in 40m 48s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8b3316557c784581a1c8bb623d6e9867 : SUCCESS in 47m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8bbc49fbd83c442e97a842678e4659b1 : SUCCESS in 1h 13m 20s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/782649b7fe50406198bc20c81d41c351 : SUCCESS in 1h 15m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e37a8fc68c964bc4b936845c8ec04db9 : SUCCESS in 1h 24m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b99049fe78a44f5792e8c186225ea124 : SUCCESS in 47m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52823d9957dd4dc8a107124db76981e2 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7d81cb175e4487a3e004f94eb2f7e2 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/baaced4e3ff347c994b43c7487cbc457 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bff88bbf7d44a9fba400538b8a451c7 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ce1d2e3823845a6a84cceb1e825b4b2 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29198b19a89244b381db54388af8f580 : SUCCESS in 2m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e74a0b15c8e243ecb2f8f2fdc005c798 : SUCCESS in 38m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c5e69a3103a4448cacab467b884ca686 : SUCCESS in 36m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/86e24c709959452992800e45fd13609d : SUCCESS in 1h 04m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/576652b1937d49d9aa2202acc5b04221 : SUCCESS in 50m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a05b0deaaa20445983d8274bfa5af677 : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":2},{"id":"beac1cf2eb697ace9187aa97d7aea0cbf95d1c46","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-20 12:48:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0b45ddbdff0cb612ab3616f29cc960a9d2ec8c19","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-20 12:48:48.000000000","message":"Patch Set 3:\n\n(rebased)","accounts_in_message":[],"_revision_number":3},{"id":"0b3554522b92c900687d277aa91b628a061302bb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-20 12:50:16.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"88ba629715bef4c773e1fe864399ca2bd76f51fa","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-08-20 12:50:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"41e5c9289f9ee7432b144626b0fb287ee1517259","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-20 12:51:10.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"907951e59270bb5f6022c217313203b0aa9346ae","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-08-20 13:08:01.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8cb3a5c88a0b0fa10258ea2b75b9785e1140011f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-20 13:17:40.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2c47ce66c01d5cba078b3f301c58d6fb6d92a1a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 14:22:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b4f3b16fc742466db14629c823536df6 : SUCCESS in 42m 30s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b55255209bd247e5b10692f9c967b946 : SUCCESS in 52m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/844e76d6bc364017ad7f914bcc2a01ce : SUCCESS in 1h 25m 36s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f4f0b38194d84ddc914ffd44f3cca72f : SUCCESS in 1h 09m 18s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0ff255e1765d449ba24a6dac235622a9 : SUCCESS in 1h 15m 42s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/4bc41c6b346f4deb84404ad4e51e4d83 : SUCCESS in 56m 20s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ee0ec083e3304810a2ffb3ee87efad41 : FAILURE in 50m 58s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7faca9fbac8e44159041c4eab9cc3cca : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8164d2e496e4d44872d845e93b19087 : SUCCESS in 8m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/64374def22d14ad5925048d990e10055 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc2892469ad045f0a2d06eb1d9bb6faa : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87634bc04e174fbd8aaf87c7b0dffe23 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9787e899b80f49caa3bbf447cfc9f303 : SUCCESS in 6m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/279d486c157c4ff4b4c06e0013cf371d : SUCCESS in 26m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6a1bca20e1af47ceabbe1a98970160a5 : SUCCESS in 25m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6a6d9fe4b1d6496e95b4bd02ad14653f : SUCCESS in 48m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4a6e484bad48452883429416f84c3f24 : SUCCESS in 52m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/699a8009e7214a3093000745cd9a48ae : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":4},{"id":"9e0cfeb709795a6f5a007a0e9f7e24fc57d99073","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-21 07:42:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f869f083786609eac1d8ecc964c57c1cb40c906b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 07:43:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7efb02e018534b5b69cec449e89f4ddc64128ed2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 09:16:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"54691afafcb436ea4ffadcb6a342fb9d10c1827a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 09:16:44.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9ab921ab187a42ec9b34b7fc6a23c73a : SUCCESS in 51m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d484ef8dd48a44338c3e543638d90855 : SUCCESS in 1h 08m 19s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/aa11960206dd4cfea895e053a1abf3a5 : SUCCESS in 1h 21m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d035f2d94d9349678717aa795f7061c0 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3123151e0b2a463f93370418108b0c1c : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/395c907e16494249b13f4ee3fc37fc6a : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/107bb911b4e849bdba5fa588aa335f84 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90808f1480cc401eacf3c1dbeb262272 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/150c55e32bf645dbaca8708f782a32d9 : SUCCESS in 4m 43s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c54472383b8047fb8069e2830af6755f : SUCCESS in 28m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4010e1dc0319487da66b149bb86a914f : SUCCESS in 48m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cdc2f349905d47988c9b29e3386771ac : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":4},{"id":"190309b8187849bea44dc05e1a217899cf0bf5f3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 09:19:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7ec59e3928a4cd791c982434aeb567e : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f8d81e0307ad3a87e631da7ad1b6c0e7f977289c","revisions":{"8f28396b8932d9a0a56971416163ebac6361e344":{"kind":"REWORK","_number":1,"created":"2019-08-14 09:16:58.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/77/676377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/676377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/676377/1"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-14 09:09:56.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-14 09:09:56.000000000","tz":120},"subject":"CI: Gate on kolla-ansible jobs but make them not vote in check","message":"CI: Gate on kolla-ansible jobs but make them not vote in check\n\nWe should gate on kolla-ansible jobs. Period.\nThis makes kolla gate on jobs that kolla-ansible gates on already.\n\nMaking k-a jobs non-voting in checks should let us merge changes that\nneed a patch in kolla-ansible as well but otherwise cyclic deps\nare not allowed in CI.\n\nChange-Id: Iee51a5c340d071e653afd4ecd4996d04bd9cfb7f\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f28396b8932d9a0a56971416163ebac6361e344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f28396b8932d9a0a56971416163ebac6361e344"}]},"branch":"refs/heads/master"},"69d959fe0afd0931f638c67c22d3584757c7b515":{"kind":"REWORK","_number":2,"created":"2019-08-14 19:13:44.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/77/676377/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/676377/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/676377/2"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-14 09:09:56.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-14 19:13:38.000000000","tz":0},"subject":"CI: Gate on jobs kolla-ansible already gates on","message":"CI: Gate on jobs kolla-ansible already gates on\n\nWe should gate on kolla-ansible jobs. Period.\nThis makes kolla gate on jobs that kolla-ansible gates on already.\n\nChange-Id: Iee51a5c340d071e653afd4ecd4996d04bd9cfb7f\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69d959fe0afd0931f638c67c22d3584757c7b515"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69d959fe0afd0931f638c67c22d3584757c7b515"}]},"branch":"refs/heads/master"},"94cdc82ae0d8720cdf70514f5891fa329e7d6273":{"kind":"REWORK","_number":3,"created":"2019-08-20 12:48:19.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/77/676377/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/676377/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/676377/3"}}},"commit":{"parents":[{"commit":"62493243aeb7d12a7a1f87de06f2f4a8fc64b2c3","subject":"Merge \"Add Debian/source CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62493243aeb7d12a7a1f87de06f2f4a8fc64b2c3"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-14 09:09:56.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-20 12:47:57.000000000","tz":120},"subject":"CI: Gate on jobs kolla-ansible already gates on","message":"CI: Gate on jobs kolla-ansible already gates on\n\nWe should gate on kolla-ansible jobs. Period.\nThis makes kolla gate on jobs that kolla-ansible gates on already.\n\nChange-Id: Iee51a5c340d071e653afd4ecd4996d04bd9cfb7f\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94cdc82ae0d8720cdf70514f5891fa329e7d6273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94cdc82ae0d8720cdf70514f5891fa329e7d6273"}]},"branch":"refs/heads/master"},"f8d81e0307ad3a87e631da7ad1b6c0e7f977289c":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-08-20 12:51:10.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/77/676377/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/676377/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/676377/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/676377/4"}}},"commit":{"parents":[{"commit":"62493243aeb7d12a7a1f87de06f2f4a8fc64b2c3","subject":"Merge \"Add Debian/source CI job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62493243aeb7d12a7a1f87de06f2f4a8fc64b2c3"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-14 09:09:56.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-20 12:51:08.000000000","tz":0},"subject":"CI: Gate on jobs kolla-ansible already gates on","message":"CI: Gate on jobs kolla-ansible already gates on\n\nWe should gate on kolla-ansible jobs. Period.\nThis makes kolla gate on jobs that kolla-ansible gates on already.\nThey are already voting in kolla.\n\nChange-Id: Iee51a5c340d071e653afd4ecd4996d04bd9cfb7f\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8d81e0307ad3a87e631da7ad1b6c0e7f977289c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8d81e0307ad3a87e631da7ad1b6c0e7f977289c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
