)]}'
{"id":"openstack%2Fkolla~676411","triplet_id":"openstack%2Fkolla~master~Ie2257870d87c5b3a9b0572f8fd35085a3f95fedb","project":"openstack/kolla","branch":"master","topic":"feature/fluent_label","hashtags":[],"change_id":"Ie2257870d87c5b3a9b0572f8fd35085a3f95fedb","subject":"Add fluentd_binary and fluentd_version labels","status":"MERGED","created":"2019-08-14 13:47:12.000000000","updated":"2019-08-22 09:11:55.000000000","submitted":"2019-08-22 09:09:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676411-1566464958150-e43a0d63","meta_rev_id":"efc5c5175ab8b110a0bf862f70316d4276702f0c","_number":676411,"virtual_id_number":676411,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-22 09:09:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 12:16:29.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"all":[{"value":1,"date":"2019-08-21 14:39:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-08-19 14:01:09.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-22 07:48:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-08-22 07:48:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 12:16:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 14:01:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-21 14:39:01.000000000","updated_by":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"reviewer":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"state":"REVIEWER"},{"updated":"2019-08-22 07:48:00.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-22 09:09:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2673ac9cbca50dd3706cab40cbe18621ab8eddb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-14 13:47:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb1bf21c71b82da384be1e633451573b6dadf685","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-14 14:02:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/11/676411/1/openstack-check/tripleo-build-containers-rhel-8-master/c2aaca0/ : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"60752d99736fdc9633044fa31ae53e752c4b16c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 15:27:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/47d6a47e7eab46619741ba2061874ed5 : SUCCESS in 45m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/75ea2303519a48989fead3cc8b37730f : SUCCESS in 51m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/740ffa9a13034bfd8f5e9a952a5c86a3 : SUCCESS in 1h 12m 16s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/84757e0f0c704619949dca77761e2197 : SUCCESS in 1h 05m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/953ca65f8b8b43be9eb4c89b61c3e0ee : SUCCESS in 1h 27m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e830f1ba1f79412786ef831ffe1e5174 : SUCCESS in 47m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92b9abe964bf40d7ac17d09afd155fd1 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d93a6fded6f48e49b95dd3c433a5cf3 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d91e2328c7774e30950760c3621f4c14 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eddcc863858c4905a089f23e324f528e : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a3b104d47e34c15b38107ea7ecf4088 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84da15b0b1b42dd80edde32a04bb4d8 : SUCCESS in 2m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/acf2d291a06c4db6ba771334b53a0f76 : SUCCESS in 25m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/642346e661dd475e859273580bcc3f4c : SUCCESS in 31m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d96585bdee2c4ba2b79ddbc4e99330f5 : SUCCESS in 47m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b306c6022b7947b1831cd69f45ac0a3f : SUCCESS in 1h 02m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a20db5f02cbd4ab591ea5488fcece459 : SUCCESS in 1h 17m 09s","accounts_in_message":[],"_revision_number":1},{"id":"983c0ce59536fd1d10e4351fe5caef571992300e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-14 16:49:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f4b845d67596f079a3b35cbdb129b3c35f752f3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-14 16:59:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"761c1d33eacb3fde9a4474779914d6652ae2aa65","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-14 17:34:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1c1534220a5b4e446e333b8846695e508ac7129","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-14 19:59:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0275898c145fa436712b8be4a67d6e205cb11310","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-14 20:41:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/11/676411/2/openstack-check/tripleo-build-containers-rhel-8-master/103f8a2/ : SUCCESS in 40m 12s","accounts_in_message":[],"_revision_number":2},{"id":"3ba5102cbbb86f588216cc346dc6e26f5c28ba73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 21:28:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/da3531e1e7144b53be032829c2f63045 : SUCCESS in 35m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a9910c784c314555a018d9e03006af09 : SUCCESS in 44m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7b83756e3c404952b4e2c6b5fd2869b7 : SUCCESS in 1h 25m 30s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5d8bd7c501794955abb5f0aaaf986703 : SUCCESS in 1h 12m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2a32994e7885478d8ccaf571af6de6fc : SUCCESS in 1h 20m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/24183382c4fb4924b60958a6447102b2 : SUCCESS in 57m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9df6ee558a34853a1f33fec1c82dfb9 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c6654661a834fa78aed93ff9d0f12c4 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/170034679b6f4607950186c624573ded : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d711aa10b764df68410a6c3c6dc6530 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e7589b07bbd413da35c1319a7ecc26c : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa437435c70d4217844be41affa07a3d : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e3fa7ddea6a4844b8567fa8989dd58b : SUCCESS in 42m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/01e0e9893dcb4b729164143cc973021f : SUCCESS in 45m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/64b888940cf047b9a12d7228b893d3df : SUCCESS in 57m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/afdade1a727f43abab21edd907300f48 : SUCCESS in 56m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7955b471338d4acabc6550698713b1a1 : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":2},{"id":"1637e647b9856e5ab850d14c9f0caa4a8fe1d002","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 11:24:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d434511df02d286e9252d2c7120192d621ebaf5b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-19 09:05:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2aeb0a9b8a61361b1aa50dccd8e1df29d42dba60","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 09:29:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/11/676411/3/openstack-check/tripleo-build-containers-rhel-8-master/cd99f00/ : RETRY_LIMIT in 6m 29s","accounts_in_message":[],"_revision_number":3},{"id":"9601e25c21821afde641d332306d6c487b580508","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 10:53:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/67dbed76e4df460ea65bff6b4bb7b55c : SUCCESS in 46m 28s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/91dc9a4babf54f9fb9ae1385eef183e7 : SUCCESS in 1h 07m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bda66a502ba94b7c8c9eb9bc78c608fc : SUCCESS in 1h 23m 38s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/41873feca3cd4617afb3fc5a800c5fd3 : SUCCESS in 1h 23m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/db709499a3804aa0bb45ffdfbd725fc8 : SUCCESS in 1h 45m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7bf28d2b9ff043d1a2cb5e1fa3b0a85b : RETRY_LIMIT in 9m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d0decbd866d4c3a826841e4c8d86aa3 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c25433647b4049bdfafdd6eb1ce345 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38e1cce0895544fa86fdaa85776de794 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8b8e22cb7ab4fd486f3dfadc0dc2f13 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/856885a186724425a9a1e6c87ccca28f : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/918c94e8e5a14c59b8e6627e04938752 : SUCCESS in 4m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/713bf7db897c4ba1b134a2180708e78c : FAILURE in 50m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/79d696f2a70f422b8356edbebcaa4768 : FAILURE in 52m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/59abd3fd0c13414e86348a494ddf45ae : FAILURE in 33m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fbe177e7884f42b5adb7868c34364c0a : FAILURE in 28m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3c6b2a61a6904da0bc7273e04f3ea195 : FAILURE in 31m 48s","accounts_in_message":[],"_revision_number":3},{"id":"8d05a7ba5b6288ddb842d6c0f0571d3dff63e5ae","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-19 11:56:38.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"4b2887a3409f305d3eb474517b2b54169c249651","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 12:16:29.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/11/676411/4/openstack-check/tripleo-build-containers-rhel-8-master/0666eaa/ : RETRY_LIMIT in 5m 59s","accounts_in_message":[],"_revision_number":4},{"id":"e41661b2fca30d3ed4a32243ad2d9f72cdab0a3f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 13:27:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6d91fbcbc7824aaea8121954c09e8d25 : SUCCESS in 36m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/90e1751222e745fdaf4c2771a3be3245 : SUCCESS in 43m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f7e4ef0780f643e3b3f009842cd344c0 : SUCCESS in 1h 10m 33s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6593d90038e64be8981593defbb17369 : SUCCESS in 1h 14m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/50228f2ab38243a986df88f6b40eba9f : SUCCESS in 1h 29m 19s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/62553dd1ce924a7e83803d5774d4f795 : RETRY_LIMIT in 5m 24s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c14619e60854782b1e74c9e1cc2d9f8 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b46c01d8e834db3b42d54db4e09446d : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/281ce7874e1246cf820e197b75c9e3d1 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9e4688ec8a24947b8009731fe1e13cc : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/60dc2d16a44b46f99752ecec62598031 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b59dcbf1e9478f9c4fb0e6d7cd5ea6 : SUCCESS in 3m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/034c4d26193547d9b99223cace7c99fc : SUCCESS in 23m 22s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/953380d768914a4b9d931c806bc91f8f : SUCCESS in 32m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/dcd347108e1248f994b41b3118b7c609 : SUCCESS in 47m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0da9e2fd9eb84ac1ace8232cb2e84919 : SUCCESS in 47m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ddd9d491cc7648fcb86ffa12d734e4cc : SUCCESS in 57m 50s","accounts_in_message":[],"_revision_number":4},{"id":"e99fe3fc961361b7422323ef4a917b69948348c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-19 14:01:09.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f9126f6a55b4cbe5cc61b7a405799934e55b9ffb","author":{"_account_id":19779,"name":"Chason Chan","display_name":"chenxing","email":"chason.chan@foxmail.com","username":"chenxing"},"date":"2019-08-21 14:39:01.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"31196d998a1a28679c9cfcccdd8583546679163c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-22 07:48:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nvery nice","accounts_in_message":[],"_revision_number":4},{"id":"1c12e494cb7b2c95955e4abe7c94416f2a4ee9d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 07:48:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bc21d7067b26480e83138d7cc25086b8a06f6dd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 09:09:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/aca8291a948e4fa28263a94e3bab14f9 : SUCCESS in 37m 30s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1254f5aa19d44a47919c7fccbe654793 : SUCCESS in 1h 12m 13s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/bfbf5d6548ce45dca9b5b42032979238 : SUCCESS in 1h 18m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16009f5699db4aa78f0dcde9ab076de5 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ef2b8c73a84278938c407ec131b942 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6dfd09856ee14ef1b7eee50fe19d3f69 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47e8137af1c14664893110c37901703f : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91f0c12ca88b4367be344a645b1c0507 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78089d0e4dc24a288ab85fb86e3eeb7e : SUCCESS in 4m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ddb35dd9de7b4ee49f895a8d125018cc : SUCCESS in 29m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/925b66818f784392b2986f0902a8ac2c : SUCCESS in 1h 00m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/00ffd92020c043248d5d7d65bd616fe6 : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":4},{"id":"5c3c45c17cccd12b89db99bdef5775f2486a6e6d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 09:09:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"55773aff68519ca3b3051dce97f2708d5f81519e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 09:11:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4f82a292a7444258ad263ad4e55ce50 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"23f6d2ab011a1d664976ced95310d0dbc49d0d00","revisions":{"3a52523bed6b9b204c22e64fa622cfb562169dd8":{"kind":"REWORK","_number":1,"created":"2019-08-14 13:47:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/11/676411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/676411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/676411/1"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-14 13:44:05.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-14 13:47:10.000000000","tz":0},"subject":"Add fluentd_binary and fluentd_version labels","message":"Add fluentd_binary and fluentd_version labels\n\nIn order to orchestrate a smooth migration of fluentd to newer\ntd-agent on CentOS x86_64, a label that points to fluentd binary\nand version are needed.\n\nChange-Id: Ie2257870d87c5b3a9b0572f8fd35085a3f95fedb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a52523bed6b9b204c22e64fa622cfb562169dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a52523bed6b9b204c22e64fa622cfb562169dd8"}]},"branch":"refs/heads/master"},"095691d147bf95a70f7874c877f9a7790a963907":{"kind":"REWORK","_number":2,"created":"2019-08-14 19:59:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/11/676411/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/676411/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/676411/2"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-14 13:44:05.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-14 19:58:58.000000000","tz":0},"subject":"Add fluentd_binary and fluentd_version labels","message":"Add fluentd_binary and fluentd_version labels\n\nIn order to orchestrate a smooth migration of fluentd to newer\ntd-agent on CentOS x86_64, a label that points to fluentd binary\nand version are needed.\n\nChange-Id: Ie2257870d87c5b3a9b0572f8fd35085a3f95fedb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/095691d147bf95a70f7874c877f9a7790a963907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/095691d147bf95a70f7874c877f9a7790a963907"}]},"branch":"refs/heads/master"},"8c8742b31464342ef59c47a3690e5154b1778d45":{"kind":"REWORK","_number":3,"created":"2019-08-19 09:05:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/11/676411/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/676411/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/676411/3"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-14 13:44:05.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-19 09:05:47.000000000","tz":0},"subject":"Add fluentd_binary and fluentd_version labels","message":"Add fluentd_binary and fluentd_version labels\n\nIn order to orchestrate a smooth migration of fluentd to newer\ntd-agent on CentOS x86_64, a label that points to fluentd binary\nand version are needed.\n\nChange-Id: Ie2257870d87c5b3a9b0572f8fd35085a3f95fedb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c8742b31464342ef59c47a3690e5154b1778d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c8742b31464342ef59c47a3690e5154b1778d45"}]},"branch":"refs/heads/master"},"23f6d2ab011a1d664976ced95310d0dbc49d0d00":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-08-19 11:56:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/11/676411/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/676411/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/676411/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/676411/4"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-14 13:44:05.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-19 11:56:15.000000000","tz":0},"subject":"Add fluentd_binary and fluentd_version labels","message":"Add fluentd_binary and fluentd_version labels\n\nIn order to orchestrate a smooth migration of fluentd to newer\ntd-agent on CentOS x86_64, a label that points to fluentd binary\nand version are needed.\n\nDepends-On: https://review.opendev.org/677144/\nChange-Id: Ie2257870d87c5b3a9b0572f8fd35085a3f95fedb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23f6d2ab011a1d664976ced95310d0dbc49d0d00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23f6d2ab011a1d664976ced95310d0dbc49d0d00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
