)]}'
{"id":"openstack%2Fkolla~676614","triplet_id":"openstack%2Fkolla~stable%2Fstein~I78f8be44436b8c8e5f58fea40a762c17e6b510a8","project":"openstack/kolla","branch":"stable/stein","hashtags":[],"change_id":"I78f8be44436b8c8e5f58fea40a762c17e6b510a8","subject":"There is no Java 8 in Debian \u0027buster\u0027 (part 2)","status":"MERGED","created":"2019-08-15 07:44:33.000000000","updated":"2019-08-20 11:02:39.000000000","submitted":"2019-08-20 11:02:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676614-1566298959231-de1e3c07","meta_rev_id":"757a3bac1235640c1fc96d14a12fcf20cf4e8a79","_number":676614,"virtual_id_number":676614,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-20 11:02:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-20 09:05:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-08-16 09:10:17.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-15 09:49:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-16 09:11:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-08-16 09:10:17.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-08-16 09:11:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 09:10:17.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-16 09:11:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-20 09:05:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 11:02:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b847aecd219599ba2c5e6a3841ebd5cd0feca2cd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 07:44:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6eff23d93ae7db486089ea6e282623c697e0e0c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 08:06:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/14/676614/1/openstack-check/tripleo-build-containers-rhel-8-master/8e88c7b/ : SUCCESS in 21m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0175d1c28f3428f5613d3c66e37d8e2357d181ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 09:10:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fa784e179f3b400183be095d2ea1de40 : SUCCESS in 46m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f8f1b94d89824f28ab0311aa85eb81e4 : SUCCESS in 51m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/d3a2d4d4dcc04fa7940a04f605f7ceac : SUCCESS in 1h 20m 47s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1411816f00e2403cb37f312dff700b68 : SUCCESS in 1h 07m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/615e7796039341429b3cba0e61603822 : SUCCESS in 1h 19m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/bcf4166da2c54426bb6e709529b0441a : SUCCESS in 51m 58s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/7e1f2a5dda33478b8d4982611cee822b : SUCCESS in 42m 08s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c7c59e471e7b4d1cb091b5c9e01d03fe : SUCCESS in 45m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89aa618bf2ad42b88af0acc082df91f5 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fc3ed15e22b4662ac6aa8e7b87fdce0 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dbebca50a72749fd9b59a48ae842aae4 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/539707193a8a4c42a3134f69b6fb1395 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/356c52e8f643400f822d24ba3f9584ca : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d3cf9ca23494640965c7155ab8bf1c5 : SUCCESS in 3m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4b19e74b3494086b7dc4564953bb0a9 : SUCCESS in 24m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b46256105a5d4b55b03b838a0ae58ea8 : SUCCESS in 29m 15s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ebc623374b6b4eb7bd59fd4357a5808d : SUCCESS in 59m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7bfffc53194841d793d3f2fb64cd7d43 : SUCCESS in 46m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8dbfd76ccc1749798c1ff2e9649ada04 : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5e1ad87c129f57520134bef7817a566afad9c591","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 09:49:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dda1bf5491fc3033bf848dc909694545d0bcb921","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-16 09:10:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"db70bf1715aa688ffc52372e08a8cf4218c4bb67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:10:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67ca533191e169e236ccced6ca79e39631041242","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-16 09:11:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b93dfeb4e9755325f3788028c8b962b11c22c278","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:48:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c5f7d2950c8e4020ba4992afece7a20f : SUCCESS in 56m 30s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/fbc4835704664df4ab78fbe9259e11c4 : RETRY_LIMIT in 5m 40s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c8156df8d617402da29ba0a769519799 : SUCCESS in 54m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1d7cd19db384cd387b39bd355c12d7d : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85ac70399611499780c1533664872781 : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3429461c729c4286b5db4db9a773f8e9 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/978f67aabfbe43a58041492ce85dbd22 : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/428908189fc7491aab7684349fb32e42 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8dc527f6ec4432f8ce94734c085f932 : SUCCESS in 3m 31s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1d985282f514d309ced52e35381d7e9 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bbff3ec8ec9d2f4dc486d8c4c5a9622ddf50df80","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-20 08:40:35.000000000","message":"Patch Set 1:\n\nrecheck\n\ntripleo","accounts_in_message":[],"_revision_number":1},{"id":"6edea9a132b913596cf87eaa059e2aebd5366013","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-20 09:05:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/14/676614/1/openstack-check/tripleo-build-containers-rhel-8-master/460f4fa/ : RETRY_LIMIT in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d38765dbbb162f9b00d824c9241b734a5cb3f452","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 10:07:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1eea5990a0874c2ebd3c84baeab257be : SUCCESS in 47m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1d7c5923b4f348fca6821e7c083f25b7 : SUCCESS in 37m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7a1572ba84a146ba8ac18b721b61df47 : SUCCESS in 1h 18m 05s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/432ac90fa3474ffdb3105ae1ed82d70a : SUCCESS in 1h 04m 05s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a278c3d53b5b4ec590894783c77da6f8 : SUCCESS in 1h 20m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5b5a58d208244a39ba7f124c41f60663 : RETRY_LIMIT in 8m 13s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/03b4d83ce0ea49a18e89f243b88e39ad : SUCCESS in 53m 05s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/d25a6d5aeee84cb495c1d2c7ea293fd6 : SUCCESS in 46m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f603b8dc466742fe8268731c2c5a56a7 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b00f80a657984bf5843989e47bd9ea7e : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7c1d1e280904a27b32cdc7df5166f79 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/763ea2508afa424f8168395ba0ac2762 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/245351bca5f848cda570bd956d0cf33c : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2c638e3a7e40e3a9db64b5301d9363 : SUCCESS in 4m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37e32da715114e48abd011e026b3a7ad : SUCCESS in 31m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/404f333c86ce4553a01be21dd31736e4 : SUCCESS in 35m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/19ea328641cf4687b9653bc6531deb5c : SUCCESS in 1h 04m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6657e23d90924ada81fbcc05b79c6813 : SUCCESS in 50m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/46d8c43d1d6e42509b6d2c1785692959 : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"21673d725fa7b6ac384ccf12ffc2a1e5f897b855","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 10:07:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca8a6748b28429d2bef57122544c4c516d9b69a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 11:02:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"757a3bac1235640c1fc96d14a12fcf20cf4e8a79","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 11:02:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4ee8018bbeed49faa7340ef60baab3f4 : SUCCESS in 28m 23s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/bb864fffea454efaa36b7f4b5626e3ed : SUCCESS in 48m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31724cf092e944e395b4bb46e0e851d2 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/192380d3d41242e6956ab6985518273d : SUCCESS in 3m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7857fa9442a8401ca5b06ab383739452 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00b911b6c6b040e6b69ba18b89b70e6c : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d59a9e88051d4f68a5f73b65523cd475 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b50970862e4419197d7c0621f8cf23d : SUCCESS in 3m 25s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e1a71c5441394a56b05c5c30acda50fc : SUCCESS in 27m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3272c8cf487a2574d1e296416aaa4675f0324d6e","revisions":{"3272c8cf487a2574d1e296416aaa4675f0324d6e":{"kind":"REWORK","_number":1,"created":"2019-08-15 07:44:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/14/676614/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/676614/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/676614/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/676614/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/676614/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/676614/1"}}},"commit":{"parents":[{"commit":"cd4d4b37be3ad75d23a48fd58a550ff08684e270","subject":"Stein-only: update packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd4d4b37be3ad75d23a48fd58a550ff08684e270"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-07-16 12:50:15.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-15 07:44:33.000000000","tz":0},"subject":"There is no Java 8 in Debian \u0027buster\u0027 (part 2)","message":"There is no Java 8 in Debian \u0027buster\u0027 (part 2)\n\nDebian \u0027buster\u0027 release contains Java 11 only. So opendaylight and\nmonasca-thresh images needed adjustments.\n\nChange-Id: I78f8be44436b8c8e5f58fea40a762c17e6b510a8\n(cherry picked from commit e04428f8f52e3ef14997b16bb1f3f3e3bf1bdfe5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3272c8cf487a2574d1e296416aaa4675f0324d6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3272c8cf487a2574d1e296416aaa4675f0324d6e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
