)]}'
{"id":"openstack%2Fkolla~676616","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ib31e733eba12deebdf469c306dfba4191a1b89f5","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1832306-stable/stein","hashtags":[],"change_id":"Ib31e733eba12deebdf469c306dfba4191a1b89f5","subject":"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary","status":"MERGED","created":"2019-08-15 07:47:43.000000000","updated":"2019-09-02 09:42:34.000000000","submitted":"2019-08-20 11:18:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676616-1566299898475-16970a7d","meta_rev_id":"df301c984a01c6f2a44742f1ca6dca2ad8d4f039","_number":676616,"virtual_id_number":676616,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-08-20 08:41:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-20 11:18:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-20 09:07:25.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-08-16 09:12:13.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-15 09:48:29.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-08-16 09:12:13.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-15 07:47:43.000000000","_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 07:47:43.000000000","updated_by":{"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"},"reviewer":{"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"},"state":"REVIEWER"},{"updated":"2019-08-16 09:12:13.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-20 08:41:22.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-20 09:07:25.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 11:18:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2bab1dd6b1fb236c385a5f43b9adf1e3bba18497","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 07:47:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"121632196e213f167996e7589d8a1ba5a05d63ba","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 08:10:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/676616/1/openstack-check/tripleo-build-containers-rhel-8-master/7edc7b1/ : SUCCESS in 21m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b6c141c77c5f20fa467a685a8e257cc7e613b0ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 09:21:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/4a55ce1db11e457aa3a04be75acec6b9 : SUCCESS in 41m 14s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1c87bd06e7e5466184783f51f0e2827e : SUCCESS in 50m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5d7c7a1acc2c456e88ed9018253791a4 : SUCCESS in 1h 24m 42s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0e0f022c05cc4cd48c5607a15811b896 : SUCCESS in 1h 07m 13s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ba4d7371b8974ea19d9f5c8c21605caa : SUCCESS in 1h 26m 06s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/481fb496618f46fdad4ce8b3f8a2a365 : SUCCESS in 41m 38s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bdf88bd784ed448e897c11cdfd3ac409 : SUCCESS in 55m 00s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/337390475ce6449793dc8af65fdcae5a : SUCCESS in 53m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07a9b8d087dc48b8b8cb3cbbeb48b678 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af3933a5f0e54a0ea8b3a80a4ac989a7 : SUCCESS in 3m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/158aa8f60253455a9808c3a6f08e98e0 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/149200b57fc14987b9bad8ed7c421738 : SUCCESS in 3m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42a0288a9e8048bd86dcfd3d7bb072c5 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69068b9c50c4454b8f26a89c041a2977 : SUCCESS in 2m 44s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/866f902db2bd4180bc321db745deefe3 : SUCCESS in 34m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d16e011505ce4193a01b0d9763b49e2a : SUCCESS in 34m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6e560461d70f4cbaa08e983762b453c1 : SUCCESS in 1h 01m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ecee1b7dd814b7a8c0f704a02aac91b : SUCCESS in 40m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8e1cd232a4dc4d9dacbed4dbbfae3b99 : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1881e4488bed5e6997671cf447f9418ba2a19719","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 09:48:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f411298369ee94e19a5df37e9d9ad92dd327cbb3","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-16 09:12:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"18422f250450100955fe440e687d6c3e11bcf485","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:12:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42dd0ec8b77c67b2cb56e39aeb60de271d35c7a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 13:45:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f1d7a852f8504b3498281ab0082328c0 : SUCCESS in 50m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b7ca4dcb25584d4e916c2e43f8cc73fe : RETRY_LIMIT in 10m 59s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/5c3612fcf4964bdb8e7efb50d1a98a01 : SUCCESS in 54m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/634b81bba1ab41a8894cb0e310ab1cf8 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa142f043a974be89c5ebce371b3ceb0 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0cd96d5ef2bb48ff99cdf401255835af : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2fcd01c0f7d462caccfd74acadd9e12 : SUCCESS in 6m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/753cae870d734fcf99dfd3909122da09 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b95d5cf1c8674929af4300c14e6d6ae6 : SUCCESS in 3m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d6b822de9cee4af592ef52b090cd017d : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":1},{"id":"88e508af385ff75020b1c7f585901bc88644f1b2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-20 08:41:22.000000000","message":"Patch Set 1:\n\nrecheck\n\ntripleo","accounts_in_message":[],"_revision_number":1},{"id":"974321c44f0153b1d0915ce75816e8c236e89592","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-20 09:07:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/16/676616/1/openstack-check/tripleo-build-containers-rhel-8-master/34bba3f/ : RETRY_LIMIT in 7m 24s","accounts_in_message":[],"_revision_number":1},{"id":"55ccca6b19c9548778b8d12e790d7135386d650c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 10:22:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/adec1e86d7984d0c84eaddaae099f4e5 : SUCCESS in 41m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ac111dde5a5d4dd181e65ee8ca67434f : SUCCESS in 47m 11s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/100a02c90b0d43a1afaaa4ddc01ca465 : SUCCESS in 1h 26m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b0d214210a824cdd933ef8b9c32ce4cf : SUCCESS in 1h 09m 57s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5f40dd75e3304a638d16a9e9a64627d7 : SUCCESS in 1h 15m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6e88ab7615b9427ba7c4233e8e2ffc1b : SUCCESS in 50m 06s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/1dfc51bba12646908f78787349dded6f : SUCCESS in 51m 42s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e36a03f55fb04328b70283dd700f9d72 : SUCCESS in 58m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d71dbee0e8124d5c8931394f950a210d : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e109a6c9c7144c4a99ef8d4d01ddaae0 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/837e55abcd934ca0918b6c680224f722 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1393db2b5c1441feafa4d685e6e78cb4 : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae4e708c35914cf19322b4037df1f099 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/104940de86fd4482aa11062405a380ab : SUCCESS in 4m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6c8900296c04eac9a198b9df3bb748b : SUCCESS in 24m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/660fc3a434db400faf5c5b6d7ab9a48f : SUCCESS in 41m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6a2f59d2158a4cb7924303c58c3d8cb2 : SUCCESS in 56m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b0c851a72e3b4c1b862bf3dddb179f5b : SUCCESS in 46m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/782588c71d51439b9a385a23ff1fdc55 : SUCCESS in 1h 06m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a513634746e5e0163e64e8b6127e8c514ae820ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 10:22:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b39aabd8bd7533a05674094bd9e98de98b2f185f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 11:18:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4ab2494d602ee1f6c08aaecfc0b56d89f8efe07e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 11:18:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/13549759e92c475d917ab1c1bff350ac : SUCCESS in 44m 42s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3e89eab98150463f85da8ded1e7d59dd : SUCCESS in 33m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e0ec34a09174e73a7b93740382222a2 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1328ab658dca425eb0e172802bd533cd : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e51ed1b89594ae29dda31447601e554 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f6aff7a1f58406393209c0f98685804 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf639a4f34574eb2b5036c4f1f665bf2 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/461b290ceb3140c3b217d7092a7f0e35 : SUCCESS in 3m 49s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/995c3fe5133e4300b2368fe4159c5c0d : SUCCESS in 29m 28s","accounts_in_message":[],"_revision_number":1},{"id":"cb0daacd3facea698abd55e2d7303c4dbb16ad4a","author":{"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"},"date":"2019-09-02 00:38:22.000000000","message":"Patch Set 1:\n\nHi. Since this patch is want to solve the problems on Rocky, I hope this can also merge to stable/rocky. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"a08f47e71c573229cef2f7822c8b2a29351964bd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-02 09:23:18.000000000","message":"Patch Set 1:\n\n\u003e Hi. Since this patch is want to solve the problems on Rocky, I hope\n \u003e this can also merge to stable/rocky. Thanks.\n\nHi Eddie, you can propose a backport by clicking cherry-pick button in gerrit. If there is a merge conflict, you will need to manually cherry-pick the change and push yourself.","accounts_in_message":[],"_revision_number":1},{"id":"ab83497bd383521f870afd52c40e4d686b4e2fe6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-02 09:24:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e95fbf33a5c58a57c0deb858d79c59ff3b8120ca","author":{"_account_id":30687,"name":"Eddie Yen","email":"missile0407@gmail.com","username":"eddieak"},"date":"2019-09-02 09:42:34.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit ad76782797be61e737b71618ef9aa6171851a17a","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc6694dbc8907bad5085f9129fccd5b682e72b66","revisions":{"cc6694dbc8907bad5085f9129fccd5b682e72b66":{"kind":"REWORK","_number":1,"created":"2019-08-15 07:47:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/676616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/676616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/676616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/676616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/676616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/676616/1"}}},"commit":{"parents":[{"commit":"cd4d4b37be3ad75d23a48fd58a550ff08684e270","subject":"Stein-only: update packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd4d4b37be3ad75d23a48fd58a550ff08684e270"}]}],"author":{"name":"Eddie Yen","email":"missile0407@gmail.com","date":"2019-07-15 13:20:39.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-15 07:47:43.000000000","tz":0},"subject":"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary","message":"Delete default gnocchi \u0026 aodh wsgi conf in Ubuntu binary\n\nUbuntu distro package will create default wsgi conf in apache\nduring gnocchi-api and aodh-api installation. We need to delete\nthem since we only use the conf that pushed by kolla-ansible.\n\nChange-Id: Ib31e733eba12deebdf469c306dfba4191a1b89f5\nRelated-Bug: 1832306\nRelated-Bug: 1836520\n(cherry picked from commit 5eb780d0b0d12186c57b828320c3421b98b97840)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc6694dbc8907bad5085f9129fccd5b682e72b66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc6694dbc8907bad5085f9129fccd5b682e72b66"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
