)]}'
{"id":"openstack%2Fkolla~676622","triplet_id":"openstack%2Fkolla~stable%2Frocky~Ib07e32f4e67e500f10986103d781dfd3874ffdd2","project":"openstack/kolla","branch":"stable/rocky","topic":"monasca_thresh_data_clean-stable/rocky","hashtags":[],"change_id":"Ib07e32f4e67e500f10986103d781dfd3874ffdd2","subject":"Ensures monasca-thresh data and worker dirs exist and are empty on start up","status":"MERGED","created":"2019-08-15 08:05:04.000000000","updated":"2019-08-16 10:45:50.000000000","submitted":"2019-08-16 10:45:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676622-1565952350199-ac68746e","meta_rev_id":"7100105574d0ff0ae762807cc5dbfdbed37260d3","_number":676622,"virtual_id_number":676622,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-16 10:45:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-15 14:06:37.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"all":[{"value":1,"date":"2019-08-15 13:44:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":2,"date":"2019-08-16 09:09:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-15 16:09:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":1,"date":"2019-08-16 09:09:21.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 13:44:59.000000000","updated_by":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"reviewer":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"state":"REVIEWER"},{"updated":"2019-08-15 14:06:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-16 09:09:21.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-16 10:45:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20e8d50889f6610ef8b84dda7ab29a52ceb1dae9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 08:05:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2f5991a1c19d73b47c55648716e58478410846ce","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 08:18:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/22/676622/1/openstack-check/tripleo-build-containers-rhel-8-master/55283b6/ : FAILURE in 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"698f2efa9a938ee78eab63e8780b1b131a0bffa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 10:32:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/fc1d41ad81d749fda924a7d7ba4ff18d : SUCCESS in 27m 10s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/791f24b57e0b448fb37edd7e3229cfbe : SUCCESS in 44m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c9de703c99f8488b865a71eeb5fb49e1 : SUCCESS in 1h 11m 54s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/201e740722594ecf93d784dfc0789e51 : SUCCESS in 1h 10m 54s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d883cc0985b74e58883565d85b507829 : FAILURE in 11m 29s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c436dbbf3bfd447e944bb49c23a67a1d : FAILURE in 11m 43s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e9cbca8fff694bf1b957420ce4d1b5c6 : SUCCESS in 1h 17m 01s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/745606a2974e40c4b0ad23819a6e8297 : SUCCESS in 1h 02m 57s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10aa416187624005b2569df7277b788a : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09f0a42678347a98abf43e3f3e80838 : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/caead645a02f4329964f9b4f8f496adc : SUCCESS in 4m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/daacd2f1ed2d4eb6884270c3a9193d72 : SUCCESS in 3m 45s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2f4e8e3e59994c1f8a3ba8fa3c1e0a9f : SUCCESS in 3m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5e1417f4e1d44ddab898ab256a50958e : SUCCESS in 1h 39m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa6c699a21df4d818bbefdb68f77fc5f : SUCCESS in 48m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/43c3c86b75374a3390e01687689effd3 : SUCCESS in 1h 11m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d55c72e3a55447a29db258e9c1261c07 : TIMED_OUT in 2h 01m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d68c5bb73202dd3997c4358d06f786a05e61927","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2019-08-15 13:44:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ebb10f87d3136126656d962f3ea4cd61c311c697","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 13:53:56.000000000","message":"Patch Set 1:\n\nSeems to be failing on oraclelinux.","accounts_in_message":[],"_revision_number":1},{"id":"7844e375ab28f585ae8653669bd8453e83b792ef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 13:54:50.000000000","message":"Patch Set 1:\n\nrecheck\n\nproblem looks unrelated","accounts_in_message":[],"_revision_number":1},{"id":"5046650f929ea9d785cc9dc9268634ee0892dd5f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 14:06:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/22/676622/1/openstack-check/tripleo-build-containers-rhel-8-master/4949579/ : FAILURE in 10m 58s","accounts_in_message":[],"_revision_number":1},{"id":"527cb04bf391704f84d8d9e3da06c61bcdc4bc16","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 16:02:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/b5c5d8ab33e14d91bd6f710ac1295378 : SUCCESS in 37m 13s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/110d08d79ac54a2b9d1f8e65bc80ddce : SUCCESS in 36m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/89e57f8eb58f4c2d8471003635fa7325 : SUCCESS in 1h 11m 42s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d983a849f5964b449ff455c4049d0106 : SUCCESS in 1h 01m 55s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/5796ab16a158431b8f416a2afabc33e8 : SUCCESS in 35m 02s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/bb9ac13100be4ddeba55028221ce1616 : SUCCESS in 42m 48s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/1ccbdd91432740cd8101972b6bcea2d5 : SUCCESS in 1h 14m 52s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d8920addefab4b0e81609ad50f89130f : SUCCESS in 1h 14m 31s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70852e00de5a40b08dc8cc41377b60fd : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d422809693e74207a4dcb36ef26dbc87 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a2396c4b232c4b249d9f460e414f0b85 : SUCCESS in 3m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/99f1a492364e43fbb6ea0ec798180f41 : SUCCESS in 3m 01s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/58cf8b2fac76455999e4755dbfbfc9ab : SUCCESS in 3m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b8bb24358d6144ebbc6d3aea2fba827f : SUCCESS in 1h 42m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/80a211c8a7a84f58a3d23890ea673d82 : SUCCESS in 50m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/44cdd0bdf73542ce9cdadade575ea8f0 : SUCCESS in 1h 06m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c784b78be8ba4a68bbe23d0f616288eb : TIMED_OUT in 2h 01m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a645ee6c3ba59beb177dc6cc8a044c11fd230aef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 16:09:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b0a3e8cf3b12654a3160802de441e08252998d8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-16 09:09:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f12cccf0479529f744667a0b966982e19501675a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:09:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6736e856a4ce662a83322ddbbe861d8648a7de1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 10:45:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/240c5ca550524bdc8a573a0f95c5d7a7 : SUCCESS in 50m 12s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/c6d8eebe0c864cb4b8e679401bd95bab : SUCCESS in 27m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6146093faad64caeac5d2d204fd84d62 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9f39de8dd6a4bd89931c7ae26364e22 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1aa634775034a76bd9f6b9e3616a12a : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/01e2763c49344d47afe5f154a8691920 : SUCCESS in 3m 34s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0af521c239e94e8781844337cd5bab27 : SUCCESS in 4m 10s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d3d950d1a73d4ea3972b2f4f31c2c2d4 : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"352f02a5bee9acab94bc1aa289d313ee7e23d958","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 10:45:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8475c47d184020ab86334ccf561fd56a1ef13d33","revisions":{"8475c47d184020ab86334ccf561fd56a1ef13d33":{"kind":"REWORK","_number":1,"created":"2019-08-15 08:05:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/676622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/676622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/676622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/676622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/676622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/676622/1"}}},"commit":{"parents":[{"commit":"97b004af90dcf36ba67137917290614015eba8fc","subject":"ceph mds cap \u0027allow *\u0027 in extend_start","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97b004af90dcf36ba67137917290614015eba8fc"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2019-08-06 11:09:13.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-08-15 08:05:04.000000000","tz":0},"subject":"Ensures monasca-thresh data and worker dirs exist and are empty on start up","message":"Ensures monasca-thresh data and worker dirs exist and are empty on start up\n\nThis prevents the filesystem from filling up with Apache Storm\ntemporary files which will otherwise not be cleared.\n\nChange-Id: Ib07e32f4e67e500f10986103d781dfd3874ffdd2\nPartial-Bug: #1839149\n(cherry picked from commit 16d4cd91d24d48e4cdf90d5c06991b52e00803ba)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8475c47d184020ab86334ccf561fd56a1ef13d33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8475c47d184020ab86334ccf561fd56a1ef13d33"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
