)]}'
{"id":"openstack%2Fkolla~676657","triplet_id":"openstack%2Fkolla~stable%2Fstein~I0ca255c6358132d9e3acfa6b610b70a78756512c","project":"openstack/kolla","branch":"stable/stein","topic":"bug/1824787-stable/stein","hashtags":[],"change_id":"I0ca255c6358132d9e3acfa6b610b70a78756512c","subject":"Add partition detection to fix osd initialization failure","status":"MERGED","created":"2019-08-15 09:13:32.000000000","updated":"2019-08-20 03:31:01.000000000","submitted":"2019-08-20 03:31:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676657-1566271861152-ff059232","meta_rev_id":"3c3e162d3125f6d3a9567617ab9ccec071549bc7","_number":676657,"virtual_id_number":676657,"owner":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-08-20 01:01:45.000000000","_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":2,"date":"2019-08-20 03:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-20 01:24:08.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-08-16 09:10:05.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-15 11:06:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-08-16 09:10:05.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-15 11:06:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-16 09:10:05.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-20 01:24:08.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 03:31:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1fa136cf2b90de3b081a18ba6d9486a1e87201d8","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-08-15 09:13:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5ecce6d7e27247bbd92a13c4fd8b1f4c1eecf332","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 09:49:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/57/676657/1/openstack-check/tripleo-build-containers-rhel-8-master/f7fd24a/ : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"0733de2a6207b7dae6a421dbdd236804f06c3b9a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 10:59:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9006099298ec45cea396f1ca3b657d88 : SUCCESS in 42m 46s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1e454a75cada40f3af760501aff37794 : SUCCESS in 49m 34s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5e4954058158460286baca55ab58569f : SUCCESS in 1h 09m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/ad21c45f17ff4b2ea0ac898015c8eb3e : SUCCESS in 1h 11m 49s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/89b65e0ff3af49fe91b61438e2b8e892 : SUCCESS in 1h 33m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3e65f5c23da242c1809efe2273b334a7 : SUCCESS in 43m 02s\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/24636c1a7a70472a81f9076be55e1677 : SUCCESS in 1h 39m 23s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/68333ffffc164f7bae3a7e10a4281f6d : SUCCESS in 46m 10s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3a0a184eb4e14be9a52481952f2dcd45 : SUCCESS in 40m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2556d4273e624e3eaf6bad5bfa44f875 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/663b4ca445224898a0543a1c8618b990 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/951c110c5dfd4300b8902149b206ab9b : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5eea3bf200024e77a943882a3f18f66d : SUCCESS in 3m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0289dbe3795d464f9a675fe23df63108 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/496a7f0442724b658a31b49cf551a579 : SUCCESS in 2m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3c9843fb30a34e778cbd2487c8a0e2f1 : SUCCESS in 23m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/05448342cdf0453d984b154d9c3e5a2d : SUCCESS in 29m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c9e32c11c9f4436807646a48c55e862 : SUCCESS in 52m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e1813b3ceecc4192a4c2e80a23d5f5b2 : SUCCESS in 45m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/27a58a2073374bc28af21ac48be5ab3f : SUCCESS in 1h 07m 11s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/8fabd0bd8ea14622b8553f304ae4a424 : SUCCESS in 1h 26m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"71afc1ce1486432c65cc57e004e4457d75f4b907","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-15 11:06:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3ad5f2984efc04fd8d43ae0e92f6a4a6ce7ece45","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-16 09:10:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c890c61c11bca23d0f1564b1929ebb02194c178c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:10:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e778b33d8a9d418f879dbde48ae4e5bf38fb6a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:09:59.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b10e159faa6348c0953d9cd27bc5c708 : SUCCESS in 45m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/345aa245a36f4f8e9c83bb97de691b35 : RETRY_LIMIT in 6m 33s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8fe5f9484cdd4885a36a9c649d13fbf9 : SUCCESS in 50m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5380a873527247afabea8a0b2497a199 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5caebaec68014588b7cd7f634a572287 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b234e456517a4aeaa6c403f07069f435 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcd8f86f9480481b9b7326dfef4cefc8 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/533c1250a7a74cc2a9c290be629f22fd : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/707aef3c86b94e2eac2b5b704800251b : SUCCESS in 3m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb4396daba4f4c679163a9dbfe8a3028 : SUCCESS in 48m 24s","accounts_in_message":[],"_revision_number":1},{"id":"92d7498c0f76508d83144ad5cd2b117f6eef1e24","author":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"date":"2019-08-20 01:01:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"043bc5f646ae4fff9aac8406902c038842f50323","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-20 01:24:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/57/676657/1/openstack-check/tripleo-build-containers-rhel-8-master/fae7251/ : RETRY_LIMIT in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a9449a7be85f117c3df3edef290a90808eae006f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 02:35:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c1a5291175164048a281bd4a1088ce4f : SUCCESS in 44m 48s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/075c14d1655f4b1a8ac060d9c5af9229 : SUCCESS in 53m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/97ab39f910d0491f8c82eb8a57b14358 : SUCCESS in 1h 19m 30s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/cd24193d0211431da350d40f08bafa6c : SUCCESS in 1h 12m 42s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e21bc688389d4296b024e2766b4145e5 : SUCCESS in 1h 24m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/e483241ee0ac444cb48d180cfac6db2c : RETRY_LIMIT in 9m 57s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/c664091d05694a3fb2223016b9f72c07 : SUCCESS in 1h 22m 20s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bdd8a32bbbba4f19aa2ed9f852aa1bc7 : SUCCESS in 43m 08s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e8f14eb6f43642eaa98f2ab488e98aa9 : SUCCESS in 40m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47a25f3151214cb3a2c44bf5918b6120 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb10129e812a419da991c43066b701b8 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c5c09b55b824ef8bc3647f3c5dcf3fd : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03d5fc684f774eb99f3b71da22625e5c : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c7188f6c38b41ba9ffa7c2e99c93e7c : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10887e4943634ea8aaaa387cdc45e15c : SUCCESS in 3m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/078685daaac94f41bff197f43e0f20b4 : SUCCESS in 24m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/49242a0684e34f92a4df3ea9fbffff06 : SUCCESS in 30m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9db95f25a9894bdb87a3e35f8458f8c6 : SUCCESS in 54m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bbfa84c50fa54ed8846b1249506aa010 : SUCCESS in 48m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/de5a97a7599144e382954d164ac4d1de : SUCCESS in 1h 08m 03s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/c9269481a18e4343988b83c3f555ff75 : SUCCESS in 1h 31m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8595fdd08ff824f86ac3ebdec21ee54f02947e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 02:35:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcd380f5bf319ce854b9e029e4825723ebccd788","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 03:31:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/322ecf811d0e41a797fed46282837a9b : SUCCESS in 46m 23s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/1d4e45a7310a46e985f6a52e3a3f0228 : SUCCESS in 50m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de05b450ce1f4b4ca25c8664b7013145 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9b6fd34bc94d4a80643b46c01b3407 : SUCCESS in 3m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbf91612cf5b47ad8052a39af421dced : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f202c85be4be4d87a8054277af4e134e : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab45ebef7af84d84a3b345ed991045ec : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9484ee96061e4278806f4d185f708903 : SUCCESS in 3m 02s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9accdad7e93843e79e4633069e16abac : SUCCESS in 29m 08s","accounts_in_message":[],"_revision_number":1},{"id":"272c84c34978a658000bec9805fa56ed51a6edd9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 03:31:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64c1acd32537b0746f9f6ada2b9f63a676990ecc","revisions":{"64c1acd32537b0746f9f6ada2b9f63a676990ecc":{"kind":"REWORK","_number":1,"created":"2019-08-15 09:13:32.000000000","uploader":{"_account_id":24250,"name":"wangwei","email":"wangw88916@126.com","username":"Dream_x"},"ref":"refs/changes/57/676657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/676657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/676657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/676657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/676657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/676657/1"}}},"commit":{"parents":[{"commit":"cd4d4b37be3ad75d23a48fd58a550ff08684e270","subject":"Stein-only: update packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd4d4b37be3ad75d23a48fd58a550ff08684e270"}]}],"author":{"name":"wangwei","email":"wangwei@navercorp.com","date":"2019-07-10 07:29:48.000000000","tz":540},"committer":{"name":"wangwei","email":"wang.wei@navercorp.com","date":"2019-08-15 09:13:32.000000000","tz":0},"subject":"Add partition detection to fix osd initialization failure","message":"Add partition detection to fix osd initialization failure\n\nThe changes in the following commit did not completely solve the\nproblem of osd initialization failure:\nhttps://review.opendev.org/652612\n\nIn order to solve this problem, the partprobe command consistent with\nceph-disk is added, and then a loop is added to detect the partition,\nwhich is acquired every 1s. When the partition appears, the osd\ninitialization is continued.\n\nChange-Id: I0ca255c6358132d9e3acfa6b610b70a78756512c\nCloses-bug: #1824787\n(cherry picked from commit 342c9f0cd0dc4b97870dc40fab6c5a7d241b40ed)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64c1acd32537b0746f9f6ada2b9f63a676990ecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64c1acd32537b0746f9f6ada2b9f63a676990ecc"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
