)]}'
{"id":"openstack%2Fkolla~676704","triplet_id":"openstack%2Fkolla~stable%2Frocky~I9b0a162513ed436930c4541b758fc9cb2ef97e96","project":"openstack/kolla","branch":"stable/rocky","topic":"monasca-prometheus-stable/rocky","hashtags":[],"change_id":"I9b0a162513ed436930c4541b758fc9cb2ef97e96","subject":"Add prometheus_client to monasca-agent image","status":"MERGED","created":"2019-08-15 14:08:14.000000000","updated":"2019-08-16 15:40:39.000000000","submitted":"2019-08-16 15:40:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676704-1565970039937-28760403","meta_rev_id":"2661d0ee20bb281a5ffc8e475e6058e6265b9cb1","_number":676704,"virtual_id_number":676704,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-16 15:40:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-16 10:34:55.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-08-16 09:08:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-16 09:02:07.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2019-08-16 09:08:55.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 09:02:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-16 09:08:55.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-16 10:34:55.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-16 15:40:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aa16e233abd48659a169c205edc855dcbf4d49cb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-08-15 14:08:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2822a6f9d20d25c9a04bb006f8268f264e38a28f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-15 14:19:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/04/676704/1/openstack-check/tripleo-build-containers-rhel-8-master/099fb65/ : FAILURE in 10m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6aa114bf20445ae825744d03f5c7b4cdf596c0d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-15 16:09:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d42fdbdf3a084164973352e2b35c94dc : SUCCESS in 44m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/84a5921a6c7a422e8445af273f0b8c99 : SUCCESS in 45m 10s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f3b85265f3cd4b1f888380c4a91124c7 : SUCCESS in 1h 10m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/c031fff4e0b44fbc92b56a35fa855d26 : SUCCESS in 1h 20m 06s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b23e527b77964706950fae590358ff23 : SUCCESS in 45m 06s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/59d2bd7f00024c3d97179016a0f0a8f6 : SUCCESS in 1h 05m 01s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/4ed86b37cde440b782ff4f9553bba9c1 : SUCCESS in 1h 13m 28s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/bd5122baf9b847efaf89f2b16ebf4430 : SUCCESS in 1h 06m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f6cf029d9b7453eb8b90e7e15b0c495 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10ef980e60d543bcbf3d447b42b2aefd : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4ec814edb617498e91763508137ceb65 : SUCCESS in 4m 00s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/533fbb86af5e427cab83879bd16b5be2 : SUCCESS in 4m 23s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/34601d73955740a595c542ba1d8a698c : SUCCESS in 4m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e9fb8e92bee4a74b73118da1e31d331 : SUCCESS in 1h 43m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1461a1953150477da0563995da91a7c8 : SUCCESS in 55m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c2889c8fefa40738e8206306ce8ca10 : SUCCESS in 1h 28m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f735b6ac0d249ceb633f505deab540f : TIMED_OUT in 2h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"170634c5b0114725ea0204996cd8405fea353770","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-16 09:02:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3d9144a31be17c383b2e5c3fab0cd20865ef720f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-16 09:08:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"09cebf91db220d0267ce8d5a24fd3b61d8876e5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:09:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94cd122af9bc922779c352321243a06f8cb98b7e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 09:44:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/01e861546bc440de891aeadbf945eead : SUCCESS in 33m 54s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/3d55ee3c4d314365a4db396b697d0f2e : FAILURE in 4m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63b9f11e58554a278bd72cdad1827241 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8e1166b639d451a8573354223631320 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/548c6d331a9c421aaba88214f6c41938 : SUCCESS in 4m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f21dfbef78c44135a871ab8916437def : SUCCESS in 4m 06s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/599a7c5fdf1741cdb362aa4f8f4e4bda : SUCCESS in 4m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1028672c8678429ebbbc390475aa4f89 : SUCCESS in 31m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5569da49a2d8b7e6b4bc278ad5b62e6a3e88fb8c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-16 10:13:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8c1c3c284f71c6aa0e0332194229ee0ff757e26f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-16 10:34:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/04/676704/1/openstack-check/tripleo-build-containers-rhel-8-master/90d22e1/ : RETRY_LIMIT in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"a29562f6c5463de41e0a037a8efd30b2149f7cf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:05:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9079101434ca4f41ba4e465756057348 : SUCCESS in 47m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/60e7cfb86b094e6bba7010bc4a74b622 : SUCCESS in 57m 03s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e3931dc2201b46fba7c852b8fbc98106 : SUCCESS in 1h 25m 50s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/75191061a6894f99a30e60f9c0cf855f : SUCCESS in 1h 18m 59s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/d33d4816e55a4d2fb482c19844b844d4 : SUCCESS in 42m 49s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/26fd8c00969a4d5698fc54a7f4052ed3 : SUCCESS in 1h 06m 09s\n- kolla-ansible-oraclelinux-source https://zuul.opendev.org/t/openstack/build/8aff30f59317410abb7819fedb1ff5c9 : SUCCESS in 1h 26m 06s (non-voting)\n- kolla-ansible-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/582dd6df36cd4a17aca0ef206d2993c2 : SUCCESS in 1h 24m 26s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5028400dedc418180b8876b693d8763 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc2dcacf05bb4fa688c63ab859437a0a : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6aac841fb0946ba950b620d9ffb5c2c : SUCCESS in 6m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/71b474f563a847699f6f0006ef6dac24 : SUCCESS in 4m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4f143fa479664795937a3ac7f14abdb1 : SUCCESS in 6m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/36efb5b26d414997b167395586fbc01f : SUCCESS in 1h 02m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/326eb028eb4a49e18bdd647ac65c6e90 : SUCCESS in 45m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e24c0e32be9e4f09bb0af0f678d53739 : SUCCESS in 1h 06m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/efa14eb31807411fade702a8afc94aab : SUCCESS in 1h 42m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e506657665af4645ce119ada3019f9780d56f0b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 13:35:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"41b6a3ec02e89ab5a1524884c7725606e8625915","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 15:40:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fcc783ed6fbf4563b326a7c5023446b9 : SUCCESS in 46m 22s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/7df916f18db74b18b45f6bdda411ee5b : SUCCESS in 1h 00m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2449efd9da44df3a23bae582e3fe265 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21ca6a41985244459ea0d0d6f79e767b : SUCCESS in 7m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b0c7ed4fd53483cab7de765c7a70449 : SUCCESS in 4m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/735f022715d54905a07b098d36b755db : SUCCESS in 5m 43s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f13bf9cc612349149c74c599d36bb24b : SUCCESS in 4m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eb794bff9d3242909f08fdf3744ea920 : SUCCESS in 33m 05s","accounts_in_message":[],"_revision_number":1},{"id":"59b5c2c1e9b8ff414cd0043e517c50c76d8ed6d4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 15:40:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ebfd35af8b256b49acef5421487bb99f9b69fff","revisions":{"3ebfd35af8b256b49acef5421487bb99f9b69fff":{"kind":"REWORK","_number":1,"created":"2019-08-15 14:08:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/676704/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/676704/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/676704/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/676704/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/676704/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/676704/1"}}},"commit":{"parents":[{"commit":"97b004af90dcf36ba67137917290614015eba8fc","subject":"ceph mds cap \u0027allow *\u0027 in extend_start","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97b004af90dcf36ba67137917290614015eba8fc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-07-22 11:59:22.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-08-15 14:06:28.000000000","tz":120},"subject":"Add prometheus_client to monasca-agent image","message":"Add prometheus_client to monasca-agent image\n\nThe prometheus_client package is required to use the Prometheus plugin.\nIt makes sense to include this dependency by default since Kolla Ansible\ncan deploy many Prometheus exporters, making it easy to scrape their\nmetrics and store them in Monasca. The increase in size of the container\nimage is negligible.\n\nChange-Id: I9b0a162513ed436930c4541b758fc9cb2ef97e96\n(cherry picked from commit a74aa1bd21b04eceede6cf425a5e56dcacec1412)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ebfd35af8b256b49acef5421487bb99f9b69fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ebfd35af8b256b49acef5421487bb99f9b69fff"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
