)]}'
{"id":"openstack%2Fkolla~676896","triplet_id":"openstack%2Fkolla~master~I57fa5206f91abcedcd2d526e71e74cf634b81d2c","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I57fa5206f91abcedcd2d526e71e74cf634b81d2c","subject":"CI: drop docker systemd MountFlags","status":"MERGED","created":"2019-08-16 09:58:49.000000000","updated":"2019-08-16 14:53:46.000000000","submitted":"2019-08-16 14:53:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676896-1565967226068-16970a7d","meta_rev_id":"c838fe98355c2d7ca4115810445bd0074a322c99","_number":676896,"virtual_id_number":676896,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-16 14:53:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-08-16 11:28:59.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-16 10:00:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-08-16 11:28:59.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-08-16 11:28:07.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 10:00:58.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-16 11:28:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-08-16 11:28:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-16 14:53:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42c40dda87560fb36566421c0fb7376cb732afed","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-16 09:58:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71da4ab676a83834ec9998724cbcfbf228a79ee7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-16 10:00:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a1c1484673fcd61dda47e2f20ecced8bd0015621","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-16 10:00:58.000000000","message":"Patch Set 2: Code-Review+2\n\nweird, but I\u0027m all in for dropping that","accounts_in_message":[],"_revision_number":2},{"id":"d83b83e1c6c249afbdc1d1a9b9fa7d4da05284f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 11:22:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/55102e94626345909ee8a0f687052663 : SUCCESS in 38m 40s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7ca3f75d6b504d43b1689fc486f7256c : SUCCESS in 36m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ce82c07520a9449d8a14b6228ff54c48 : SUCCESS in 1h 09m 29s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/85b7f6fccd2b4f47b2fd40b1bfbb7fd3 : SUCCESS in 1h 20m 17s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/33ebb70927e944c396c7cd593c8298a4 : SUCCESS in 1h 17m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9484177a8a7466ea43dcc59b6e5c572 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caecac0bb3004855ba4d6ba47cd0ab56 : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97f3042935234b958ee9bb5c5bf372e4 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1331b8ac2ed345749086b0a6c10bd694 : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33d8a7c6b0a241ef85a956a8051aa4f4 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6f8991a0df41159f50c597a31a069e : SUCCESS in 3m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/691ac2ad187d4728a13e1e59a5f72637 : SUCCESS in 39m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa1f3dd7940e4aa8a246a1a1762dc7d3 : SUCCESS in 25m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa57dc187e3344dcbefaffb1353b9f80 : SUCCESS in 1h 06m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/01153e2586cb483a9766fb45c4c9f394 : SUCCESS in 43m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/58b63db0db2943628c51af1f6cf92d7c : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4b1078afa65e0fab18d66b317bd0bf9c392085b0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-16 11:28:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b81376f874857a144c794dd145e34d5a219275ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 12:10:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"07246d6983f141e819e3765b792db46e376342aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 14:53:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/9daa84021c3e4f55b13279dbb817fcd7 : SUCCESS in 43m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a2e0a2f59714dd6be59fc84189ffb2f : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9686cd7626564949904d6e11d741bcee : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/051d30e7297c4beebeca617ed9191854 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f3589fbe02e4226a1d55db7e0ef2ef1 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b08bcf4d19c4ba1a19912978803beb0 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d79e46931ee4e3a89c687c3319c3b71 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/26e0bafe99a5457da3f66bc12961cb1f : SUCCESS in 26m 47s","accounts_in_message":[],"_revision_number":2},{"id":"c838fe98355c2d7ca4115810445bd0074a322c99","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 14:53:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"92e2c4bb6d35f41aa4b0d38159ee53d3e104fb49","revisions":{"18869d83af83068995b17f6626807df753e51336":{"kind":"REWORK","_number":1,"created":"2019-08-16 09:58:49.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/96/676896/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/676896/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/676896/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/676896/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/676896/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/676896/1"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-08-16 09:57:44.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-08-16 09:57:49.000000000","tz":120},"subject":"drop MountFlags on CI","message":"drop MountFlags on CI\n\nhttps://review.opendev.org/#/c/674587/ dropped using of that flag in\ndeployments. Time to drop it on CI as Debian CI fails when it is\nenabled.\n\nChange-Id: I57fa5206f91abcedcd2d526e71e74cf634b81d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18869d83af83068995b17f6626807df753e51336"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18869d83af83068995b17f6626807df753e51336"}]},"branch":"refs/heads/master"},"92e2c4bb6d35f41aa4b0d38159ee53d3e104fb49":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-16 10:00:31.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/96/676896/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/676896/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/676896/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/676896/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/676896/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/676896/2"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-08-16 09:57:44.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-16 10:00:23.000000000","tz":0},"subject":"CI: drop docker systemd MountFlags","message":"CI: drop docker systemd MountFlags\n\nhttps://review.opendev.org/#/c/674587/ dropped using of that flag in\ndeployments. Time to drop it on CI as Debian CI fails when it is\nenabled.\n\nChange-Id: I57fa5206f91abcedcd2d526e71e74cf634b81d2c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92e2c4bb6d35f41aa4b0d38159ee53d3e104fb49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/92e2c4bb6d35f41aa4b0d38159ee53d3e104fb49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
