)]}'
{"id":"openstack%2Fkolla~676905","triplet_id":"openstack%2Fkolla~master~I795771569a3dcb9730063bd951c230806a08b3c0","project":"openstack/kolla","branch":"master","topic":"676905","hashtags":[],"change_id":"I795771569a3dcb9730063bd951c230806a08b3c0","subject":"debian: add some images to unbuildable list","status":"MERGED","created":"2019-08-16 11:03:54.000000000","updated":"2019-08-20 09:23:14.000000000","submitted":"2019-08-20 09:23:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676905-1566292994990-8c476a95","meta_rev_id":"32865394cc745564de2eb22e50530661a95225a7","_number":676905,"virtual_id_number":676905,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-20 09:23:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 16:23:22.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-20 08:22:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-20 06:18:21.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-08-20 08:22:37.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-08-19 21:51:31.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 11:28:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-08-19 16:23:22.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-20 06:18:21.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-20 08:22:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-20 09:23:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2887923138d57b04cd853b15e0241f5a9551d12c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-16 11:03:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6754c6e3be860324d12e1108a5c0ced88db4761d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-16 11:20:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a20219f017ceb5548f5e4dd34df9cab463ebea67","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-16 11:42:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/05/676905/2/openstack-check/tripleo-build-containers-rhel-8-master/5dbbcb9/ : RETRY_LIMIT in 6m 30s","accounts_in_message":[],"_revision_number":2},{"id":"aa0e42277cc35f5929106dc94f84adcb8112fae9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-16 12:19:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"631acaeb69a3fbfd71528b2763fe46fca2f083dd","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-16 12:20:50.000000000","message":"Patch Set 2:\n\nthanks for generalizing bifrost","accounts_in_message":[],"_revision_number":2},{"id":"34c323e6a4bb87511302cfe6c3b37356801e343b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-16 12:38:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4aae2571cc67248760d8ed27fc8153502555cad9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-16 13:05:14.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/05/676905/3/openstack-check/tripleo-build-containers-rhel-8-master/aea9a8d/ : RETRY_LIMIT in 7m 20s","accounts_in_message":[],"_revision_number":3},{"id":"353791e7bb877aee605512eae5ad2898779dfd6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 14:11:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/a7938d429a6c42db85a586ef125e2578 : SUCCESS in 1h 12m 01s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/182a2eb6d2064bd7b297685c744b0f72 : SUCCESS in 1h 01m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/48221ac2d0fe4776aefbfdbcc04cc843 : SUCCESS in 1h 31m 57s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/94f39f06ab8c4866b86a5498183397ec : SUCCESS in 1h 25m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/229c1c18049d449786b3eb01edccdb66 : RETRY_LIMIT in 1m 59s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6401259d0a30451a887415ce3b1cc602 : RETRY_LIMIT in 8m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f25584e380094c8a92867b285c4dd76c : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bef7add5df54c6baca5a0f0e767f4da : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6aa31b10f92e420bb0cf0af9b25ce80d : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88831f7021224b44b6609c4ae6bcd148 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/43c4fa5a5be54793b2bfd74f5363e2e1 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12982e836adf4d80a5efca6a23515891 : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9670e629d74c49ecb930bc2e9af92351 : SUCCESS in 23m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/61676f252307443d8b82ff58efc7f675 : SUCCESS in 31m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d0a26a46df8462a9723916fd8834816 : SUCCESS in 49m 00s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c939b14653249c68a69ed459855fd1c : SUCCESS in 53m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/76babd890e524736a6bfb45828b66ed1 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":3},{"id":"8e3a98ace84e73951a54560c06f6df4e5219a923","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-17 10:02:03.000000000","message":"Patch Set 3:\n\nbifrost tracked on https://review.opendev.org/665902","accounts_in_message":[],"_revision_number":3},{"id":"ea358e7f09c59578134c5f907954a9bcd4e8699e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-17 10:03:24.000000000","message":"Patch Set 3:\n\nrecheck\n\nshould be passing now","accounts_in_message":[],"_revision_number":3},{"id":"235c571ee22e46ce8dd0c60d1db5ece423ab67f0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-17 10:24:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/05/676905/3/openstack-check/tripleo-build-containers-rhel-8-master/2c0a5a7/ : RETRY_LIMIT in 6m 09s","accounts_in_message":[],"_revision_number":3},{"id":"a573601f51082582abf76cc61c8cfb81bce51e7a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-17 10:59:17.000000000","message":"Patch Set 3:\n\nstill not, additionally fluentd broke for Ubuntu...","accounts_in_message":[],"_revision_number":3},{"id":"606db668ac116c224009b7215fa9a945f1d6d3bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-17 11:23:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/66a390329f294fde8bf1891bd690d48e : SUCCESS in 21m 43s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c81b3a215bc64adba2fea279128efcc8 : SUCCESS in 50m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/8c3096498aaf458193026fbbae903dc0 : SUCCESS in 1h 14m 32s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e7b54145d26148d09c3728c2ba4ba6a7 : SUCCESS in 1h 14m 33s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/171b0490990a412abaa2eb61a00b9bc4 : SUCCESS in 1h 10m 36s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/315600d1fd29450ea5d64ed929940498 : RETRY_LIMIT in 6m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf09c93c341841e6a2b4d25c875f40ff : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4e7594025724c9d914119b17f3b8206 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/351b123dc19a45849e5f70de3be2e08d : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dabd07b596724bf3b057dd2d6e1e15f0 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c8d96d5fe91044798b4a0be4de159bac : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25cb57eef08245b5b478e08c76df2483 : SUCCESS in 2m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0b4ccef23feb457a85eedaf8523aece9 : FAILURE in 24m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc2a8b3fdc1640d79312181b6a8e1dca : FAILURE in 24m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad99641f4f4d4f2fa14d974f6c06598a : FAILURE in 15m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/61fa7895914247a2a7623dc81be3d27a : FAILURE in 17m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3614b4e06ec3486d87ee4e0681495946 : FAILURE in 20m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1ce8dffc016964017177371b35bcea6bdaa7b73c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 13:19:16.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2dbdc7059a53f582f1638d65988cea002a53957a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 13:44:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/05/676905/3/openstack-check/tripleo-build-containers-rhel-8-master/961840a/ : RETRY_LIMIT in 7m 11s","accounts_in_message":[],"_revision_number":3},{"id":"236fd63948d3df1be12e5167fe9870558a53a842","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 15:01:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/2c8080d3dbad43f5af9899fb1b78ab13 : SUCCESS in 54m 50s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a3aaf87a2b674705ac3a2d16d6991705 : SUCCESS in 39m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6d3040c3216c41f092acbe41cfaca9b9 : SUCCESS in 1h 17m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d5080a64e0ad49f78d2bb3f5f2a6712f : SUCCESS in 1h 18m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f06d669b61aa401c997c1cd893bf1ae2 : SUCCESS in 1h 33m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/eddb44d9601b4cff911cd079aec254e6 : RETRY_LIMIT in 11m 12s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3f9d6121bb6400bbb5765795079c37e : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31651d4957d74c1e991ecc6a97d0531b : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f14d46f0923d4f04bd16afb4e7b856aa : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cbb5cafe9124bd899ef1306b9d6a294 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a85b0975b0cb4519a341a94be4ceee4b : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de065a70939b40cd98678bab346f10da : SUCCESS in 4m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8fbebfde62a48399e36450c42c85ced : SUCCESS in 40m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e03a2e341a084d289d49066f9338111f : SUCCESS in 57m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/96a6f512a8094825937011c6bbf96229 : SUCCESS in 1h 04m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/710ebc421548463d8e42be120995d177 : SUCCESS in 1h 02m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f014525f9ebc4351b22c6678be19eed5 : SUCCESS in 1h 13m 38s","accounts_in_message":[],"_revision_number":3},{"id":"30dc9438fd6af7a8cee755b34564ece96c4babc4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-19 15:09:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"73b6217f340caa899a84298e9af6736ac03fc939","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 16:01:15.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d88a91f87681c7fc31657dd628f020deeff9cf09","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 16:23:22.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/05/676905/4/openstack-check/tripleo-build-containers-rhel-8-master/0593c2d/ : RETRY_LIMIT in 7m 01s","accounts_in_message":[],"_revision_number":4},{"id":"12829cf46412048ce4d06103845d0768581a78c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 17:44:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d0004dfc94f7467eac60f120698ef433 : SUCCESS in 40m 11s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f686d48d32984486bb65100a2d8129d0 : SUCCESS in 36m 57s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1260ba172b8d42089d596e0c936d2043 : SUCCESS in 1h 10m 40s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0f8286997b434b75b0656559e39a2c32 : SUCCESS in 1h 27m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e22ee41d7c384681af78e9d15971ba9e : SUCCESS in 1h 31m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b195d7c6504d4102b7ddd67a1cd0cd8c : RETRY_LIMIT in 10m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b05ab2dce8214e3fb52d60f8bb03a338 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/785bc7218a4942c1bedb65de966c635f : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61995f98dc5748f397d31ab2ae3b7a1b : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c719b86630bb4399a3a1ff9b5f8d3af9 : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d17f811a35f4326a254249a18438cd3 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1ac72e13b3f475a9389256afa5c1af5 : SUCCESS in 4m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a29f1bea7d1c4ef9a1a42ede58d55dfa : SUCCESS in 27m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b123ffc151f48dc9787fbe65f68a61e : SUCCESS in 1h 16m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0860e9206bec43b78841f7625aaf8621 : SUCCESS in 54m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ff549674d88948f8b4abed7bd6bf8a38 : SUCCESS in 1h 00m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/83094b4cc6cd48f5a0002089c189be5e : SUCCESS in 1h 30m 01s","accounts_in_message":[],"_revision_number":4},{"id":"4497bb4ba107238c1bebddd1fece2ca71badb70a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 21:51:31.000000000","message":"Patch Set 4: Review-Priority+1\n\n676905, 676909, 674240 give us Debian CI job","accounts_in_message":[],"_revision_number":4},{"id":"a80f490001660ae950557c5de7d9dcbb9230353d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-20 06:18:21.000000000","message":"Patch Set 4: Code-Review+2\n\nWe should strive to enable bifrost and monasca-grafana support in debian as well as they are the more used stuff here.","accounts_in_message":[],"_revision_number":4},{"id":"7ca103682cc0ab6a0c4f6c5227e85c79c6842659","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-20 08:22:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"02512f7389240771555ca45dcf58f6f1d1441cba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 08:22:51.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3b5a4d96e6a835ea91ec3e185d34a89228bfcaeb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 09:23:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"65031e36165b01613bf27e84baa0d5f5ec3c7239","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 09:23:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/adeb423dda9c4ac581b93a1d2216221c : SUCCESS in 55m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ca9a86f66e145c4b8c92638ed2b24ad : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bcb0077aa4f4d03b85edf04652f824e : SUCCESS in 6m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a601daf19174d0b9a38276659799443 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93ba5b33f6e64073817b16b53df6095f : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38f4559d5e424f739bdcf99361c81919 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcba07b8e16a4a35b924a080a46cec82 : SUCCESS in 4m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/33ae40a8befe4d159056a3df9a888f4c : SUCCESS in 31m 01s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ad16bade339fb4634fd8dd6a502144ca0914ca7b","revisions":{"1678253ea096c462674560f51a605557f40a1bd9":{"kind":"REWORK","_number":1,"created":"2019-08-16 11:03:54.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/05/676905/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/676905/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/676905/1"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-16 11:02:54.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-16 11:02:54.000000000","tz":120},"subject":"debian: mark opendaylight, telegraf, xtrabackup as not buildable","message":"debian: mark opendaylight, telegraf, xtrabackup as not buildable\n\nWe do not have package repositories for those images.\n\nChange-Id: I795771569a3dcb9730063bd951c230806a08b3c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1678253ea096c462674560f51a605557f40a1bd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1678253ea096c462674560f51a605557f40a1bd9"}]},"branch":"refs/heads/master"},"2a550d54441a166ffe6483dab5a3b7b4756ecb9a":{"kind":"REWORK","_number":2,"created":"2019-08-16 11:20:41.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/05/676905/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/676905/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/676905/2"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-16 11:02:54.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-16 11:18:54.000000000","tz":120},"subject":"debian: bifrost, opendaylight, telegraf, xtrabackup do not build","message":"debian: bifrost, opendaylight, telegraf, xtrabackup do not build\n\nWe do not have package repositories for opendaylight, telegraf,\nxtrabackup images.\n\nbifrost tries to install \u0027mysql-server\u0027 package which is not present in\nDebian \u0027buster\u0027\n\nChange-Id: I795771569a3dcb9730063bd951c230806a08b3c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a550d54441a166ffe6483dab5a3b7b4756ecb9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2a550d54441a166ffe6483dab5a3b7b4756ecb9a"}]},"branch":"refs/heads/master"},"29b7a66074b62d6d7d82aa7e66b72a2191006d7d":{"kind":"REWORK","_number":3,"created":"2019-08-16 12:38:36.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/05/676905/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/676905/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/676905/3"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-16 11:02:54.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-16 12:37:40.000000000","tz":120},"subject":"debian: add some images to unbuildable list","message":"debian: add some images to unbuildable list\n\nWe do not have package repositories for opendaylight, telegraf,\nxtrabackup images.\n\nbifrost tries to install \u0027mysql-server\u0027 package which is not present in\nDebian \u0027buster\u0027\n\nmonasca-grafana has some SSL related issues\n\nChange-Id: I795771569a3dcb9730063bd951c230806a08b3c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29b7a66074b62d6d7d82aa7e66b72a2191006d7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29b7a66074b62d6d7d82aa7e66b72a2191006d7d"}]},"branch":"refs/heads/master"},"ad16bade339fb4634fd8dd6a502144ca0914ca7b":{"kind":"REWORK","_number":4,"created":"2019-08-19 16:01:15.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/05/676905/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/676905/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/676905/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/676905/4"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-16 11:02:54.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-08-19 16:01:12.000000000","tz":0},"subject":"debian: add some images to unbuildable list","message":"debian: add some images to unbuildable list\n\nWe do not have package repositories for opendaylight, telegraf,\nxtrabackup images.\n\nbifrost tries to install \u0027mysql-server\u0027 package which is not present in\nDebian \u0027buster\u0027\n\nmonasca-grafana has some SSL related issues\n\nChange-Id: I795771569a3dcb9730063bd951c230806a08b3c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad16bade339fb4634fd8dd6a502144ca0914ca7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ad16bade339fb4634fd8dd6a502144ca0914ca7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
