)]}'
{"id":"openstack%2Fkolla~676932","triplet_id":"openstack%2Fkolla~master~Iebc1f2c05c1f57c2b6f98ca9772f1ae9d9a420fa","project":"openstack/kolla","branch":"master","topic":"add-pciutils-package","hashtags":[],"change_id":"Iebc1f2c05c1f57c2b6f98ca9772f1ae9d9a420fa","subject":"Add pciutils package to cyborg-agent and zun-compute","status":"MERGED","created":"2019-08-16 13:16:57.000000000","updated":"2019-09-16 17:23:29.000000000","submitted":"2019-08-19 17:01:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"676932-1566234060845-ac68746e","meta_rev_id":"0f39f31eaee86bb68b990d314d9a34dce38e6f31","_number":676932,"virtual_id_number":676932,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-19 17:01:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 13:44:37.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2019-08-16 14:54:27.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-19 16:02:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-08-19 16:02:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-16 14:54:27.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-19 13:44:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 16:02:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-19 17:01:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3ec538fd48d9809b03ca45cb501897d39615d24","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-16 13:16:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccc812b39e764678467f926d737fa6efc1a8bd3e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-16 13:41:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/32/676932/1/openstack-check/tripleo-build-containers-rhel-8-master/955a748/ : RETRY_LIMIT in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"26e94b895d5f10ed4cc99c077c6e9a805c3e9c3e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 13:53:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bf40b36bfe7140299e2ad9ba8a4e59af : FAILURE in 5m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bb1a5cadaaa24f31abd77e0b51a9240f : FAILURE in 5m 24s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9dbc0ca2385e4d23b798236ae00de19c : FAILURE in 9m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/4f3bee42038147939754ccd044cb39d5 : FAILURE in 9m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0b7fceb69f4e4f4b8d662356fc9f8adc : FAILURE in 8m 43s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/28846f2c3e98457da28fe96a351822ed : RETRY_LIMIT in 7m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98bd1bfab3e84d658a076d4aacdc591b : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e55595387d24e9ba3d6a4f1728bfade : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/71e9d7d986b140c785409fdff6e42b1c : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c838a40914b7483087f8c3730c408dc2 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb6f92446710478e8ae712745e6bc130 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5edd52f4b6f4d3aab29e9ae5c3cb035 : SUCCESS in 3m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0baa9461bc2c4a5fb524b25289aed036 : FAILURE in 4m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f075f4d51a7c46498a5cc5d4cb871c4c : FAILURE in 4m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e43ebed4a46d4f2cafa396c88adf0ec4 : FAILURE in 6m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ad700a9fc8a14e88970507afa844340b : FAILURE in 5m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8d6b6f51277741bca902f3677648bdae : FAILURE in 7m 16s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a05ed98db08d43be92935a5b15d39c8b : FAILURE in 6m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecfbe3aa477ab155695ec554543cfd52b89d620e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-16 14:07:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9edc0804e79f2bfd49d3ee73153a83a0c675113a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-16 14:33:26.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/32/676932/2/openstack-check/tripleo-build-containers-rhel-8-master/3790015/ : RETRY_LIMIT in 6m 36s","accounts_in_message":[],"_revision_number":2},{"id":"ccce6d09b51b751ac07fb943fa403896faee51fc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-16 14:54:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4235a668e2bab7080566d3cd7f256847566ed83a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-16 15:45:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/30db8ce97d2b49248dcb6b770adb0479 : SUCCESS in 50m 25s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/697edf746c9b47eb9d2b71d35f6e8014 : SUCCESS in 53m 55s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c4b4d358ad244eb8b5593b6c3eb29758 : SUCCESS in 1h 17m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/52bb4d9e634c420e85206a77374fed06 : SUCCESS in 1h 08m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5ff6368b71dd4c77a6280d805f0caf0f : SUCCESS in 1h 26m 47s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0bd5f8940d9b48cca2faca01d6310fb8 : RETRY_LIMIT in 14m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f84ee7799edf45f192e63dee898ba148 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0431bde3be724da5adc2d2489ac3d1ed : SUCCESS in 7m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b14512e540ad460e94d390d02058b99b : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0311709b7331470aa7f6586ef8eb9f53 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4642074ce6fa46e8849a00861f10b243 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9109b2b289894a6f8902d6e23a9b4093 : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6f6e18b282d1459db34201a7d6e9da1d : FAILURE in 29m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1329f0cd0b0449bbb9781983c4e7c96a : SUCCESS in 35m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4af2655b4a2e48fcb9d7a0dc2006be6e : SUCCESS in 57m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0d64447b30a34ff1969056e244b0e722 : SUCCESS in 51m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ae8104a009074bb5aecc0ca070cbd4d4 : SUCCESS in 51m 08s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/d213a8e520734070938843e30fabc09d : SUCCESS in 1h 00m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c261f01b77a0097fb3e4a2bdbc91f4b32f792473","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 13:19:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a73836b3b2ecd608fbc60c5586ceee7e5c91ae59","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 13:44:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/32/676932/2/openstack-check/tripleo-build-containers-rhel-8-master/bebaf2d/ : RETRY_LIMIT in 7m 09s","accounts_in_message":[],"_revision_number":2},{"id":"cca7b983738b82f70558efb83f3e6b275e137a46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 15:25:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ffa56302af37405b9d4acfaa394885d7 : SUCCESS in 53m 22s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/41896c91435440e18069b446fbb9c6f9 : SUCCESS in 1h 03m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/265e59ca0e6b4973ba6a3337d5e3449c : SUCCESS in 1h 21m 07s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0c94eeae37d34797b8af0e2f0b858d46 : SUCCESS in 1h 13m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9f803f10329c458fa09670d9e1012971 : SUCCESS in 1h 21m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/43d4661f5b224507ba7c2f3027de6735 : RETRY_LIMIT in 10m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76ed103a5a584cfeabc6d1c0021adf4d : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc409f650aa44a3880d24873c49467b9 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f3b2bb3c2654c7691c57392d1450aad : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1334cc9227914f86b27822e65cc898cc : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9af267322584cf597fd01504662acc1 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812afae50e0148409d265469181f0b48 : SUCCESS in 3m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/199b800f29c449fdbf691cb931c93875 : SUCCESS in 33m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c5049f9f2244640974b117698df12e5 : SUCCESS in 46m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6fc4dc10ebbc4d2fb4c886bb9213c798 : SUCCESS in 1h 15m 07s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d511b81643a406ab0a55586d2c17345 : SUCCESS in 1h 56m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/47f5367b4ccd4018ad407ead13c834f3 : SUCCESS in 1h 10m 36s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/0ca32e0543014ce4b367cb5adcff48a7 : SUCCESS in 1h 04m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e1a76a5fbaba2beaf410fbdb3de59259261ead0b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-19 16:02:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99a01f5bfceec80c3291e8ec46571dc63c5dad8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:02:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4eb80c8ccb124be8436923e2743d67ac48c5c9fe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 17:01:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"11b7e8b2a328d6b7f4205be8f168e44b648a44ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 17:01:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ee2e696ae80d42bfb9e77a0e33f7980c : SUCCESS in 55m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/71a0b1291a2b437e859564c548ec16cd : RETRY_LIMIT in 7m 38s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ac128fa454f47c19a12752ca285daeb : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ddb142e76344d99ea4924cce82fe24 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c828d0ee61e64355bfb44193927d4515 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cf5b6b0c23849dcb9deb90b820558cb : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36d24755be5744b8baee8f8b1b64d31a : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949391a98df04f54b08ae340091b3f36 : SUCCESS in 5m 17s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c425516920dc42f6a6b4f455d60eecae : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":2},{"id":"007505cf8e260c62bb8e0db7951766f76d9ba4b3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-20 12:17:32.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 904f1fcbd7b1bf133547980eed7807b06a51ef3b","accounts_in_message":[],"_revision_number":2},{"id":"0974decdd8c021bd9dc4b61c96c5432fa0014147","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-16 17:23:29.000000000","message":"Patch Set 2:\n\nturned out to be one of the fixes to reported zun deployment bugs: https://bugs.launchpad.net/kolla-ansible/+bug/1797448","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c323c5be7fd4ebcc71ca4e10a5c15b9ab1f29ba6","revisions":{"d71d7a400a541e63078b5193285bad63dc2d65f7":{"kind":"REWORK","_number":1,"created":"2019-08-16 13:16:57.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/32/676932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/676932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/676932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/676932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/676932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/676932/1"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-16 13:13:03.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-16 13:13:03.000000000","tz":120},"subject":"Add pciutils package to cyborg-agent and zun-compute","message":"Add pciutils package to cyborg-agent and zun-compute\n\nThese run lspci and have this requirement in respective bindeps.\n\nCentOS images seem to have it installed but Ubuntu not.\nEnsure pciutils package is always installed.\n\nzun-compute seemed to work nonetheless but generated errors\nabout being unable to take host inventory.\n\ncyborg-agent is unknown to work at all and deb is unbuildable.\n\nChange-Id: Iebc1f2c05c1f57c2b6f98ca9772f1ae9d9a420fa\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d71d7a400a541e63078b5193285bad63dc2d65f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d71d7a400a541e63078b5193285bad63dc2d65f7"}]},"branch":"refs/heads/master"},"c323c5be7fd4ebcc71ca4e10a5c15b9ab1f29ba6":{"kind":"REWORK","_number":2,"created":"2019-08-16 14:07:36.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/32/676932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/676932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/676932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/676932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/676932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/676932/2"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-16 13:13:03.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-16 14:07:28.000000000","tz":120},"subject":"Add pciutils package to cyborg-agent and zun-compute","message":"Add pciutils package to cyborg-agent and zun-compute\n\nThese run lspci and have this requirement in respective bindeps.\n\nCentOS images seem to have it installed but Ubuntu not.\nEnsure pciutils package is always installed.\n\nzun-compute seemed to work nonetheless but generated errors\nabout being unable to take host inventory.\n\ncyborg-agent is unknown to work at all and deb is unbuildable.\n\nChange-Id: Iebc1f2c05c1f57c2b6f98ca9772f1ae9d9a420fa\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c323c5be7fd4ebcc71ca4e10a5c15b9ab1f29ba6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c323c5be7fd4ebcc71ca4e10a5c15b9ab1f29ba6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
