)]}'
{"id":"openstack%2Fkolla~677144","triplet_id":"openstack%2Fkolla~master~I203f70176af7f2b53398ddeb5c21e6234a45e268","project":"openstack/kolla","branch":"master","topic":"fluentd-tripleo-ci-unbreak","hashtags":[],"change_id":"I203f70176af7f2b53398ddeb5c21e6234a45e268","subject":"Unbreak CI: fluentd and TripleO dependency","status":"MERGED","created":"2019-08-19 08:32:28.000000000","updated":"2019-08-19 13:20:34.000000000","submitted":"2019-08-19 13:18:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677144-1566220734646-ac68746e","meta_rev_id":"612574ebf356c4b51174d55a24d8d3d5077fb4fb","_number":677144,"virtual_id_number":677144,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2019-08-19 12:30:14.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-19 13:18:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 08:54:46.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2019-08-19 11:01:16.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-08-19 09:07:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-08-19 12:38:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":2,"date":"2019-08-19 08:33:14.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-19 09:31:56.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":1,"date":"2019-08-19 12:35:54.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":2,"date":"2019-08-19 08:33:14.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 08:54:46.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 09:07:01.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-08-19 09:31:56.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-19 11:01:16.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-08-19 12:35:54.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-19 12:38:22.000000000","updated_by":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"reviewer":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"state":"REVIEWER"},{"updated":"2019-08-19 13:18:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9d670fefbec062d3b9aa436ca9329c4eccb9d19b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-19 08:32:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c1b3317169d7481d9161e30e90abf6a237922bf","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 08:33:14.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"73ba033f4fd70b4770e207f463a7074500d785fc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 08:54:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/44/677144/1/openstack-check/tripleo-build-containers-rhel-8-master/e0838dc/ : RETRY_LIMIT in 7m 06s","accounts_in_message":[],"_revision_number":1},{"id":"7f2a8371ef639c3eced5a75a0cbf4145139639b3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-19 09:07:01.000000000","message":"Patch Set 1: Code-Review+1\n\n+1 to remove the external k8s plugin - if someone is deploying the images on k8s somewhere they can add it back via an image customisation.","accounts_in_message":[],"_revision_number":1},{"id":"bf2f893d262050f31de74894ce60aa626186f694","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-19 09:15:59.000000000","message":"Patch Set 1:\n\n\u003e +1 to remove the external k8s plugin - if someone is deploying the\n \u003e images on k8s somewhere they can add it back via an image\n \u003e customisation.\n\nyup, and it is extremely unlikely that someone both wants rocky/stein/master AND deploys kubernetes since they would spend some considerable time trying to make ends meet","accounts_in_message":[],"_revision_number":1},{"id":"562bfcaf53dbe00cf4a4e168f81373249903b7f8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-19 09:31:56.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e34f82192669c5f17de111032dde305a38d13692","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-19 09:36:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4bff9a0acfd8ad86e22bf4884e738b88e0bcd453","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-08-19 11:01:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nnon-voting jobs should never be in the gate queue, they are ignored and just waste resources. Remove the job instead.","accounts_in_message":[],"_revision_number":1},{"id":"ed77d61c4f78bd75fb6fd0e8547bf6a385266e9b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-19 11:05:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"422a59041caeef64edd5503fee4619bf82e2b2d2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-19 12:30:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a8e4eec4bf07ca9efad62d2168f0f021786d5025","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 12:34:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9bf5c3a070a24011882e4d65f0565774 : SUCCESS in 46m 45s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ac15f89fd8434e66b0de75c33596cb8d : SUCCESS in 53m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2b4855b09d034045a0442cbb6bf54144 : SUCCESS in 1h 08m 25s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a3d5206bc907469698b54e3cea7142a2 : SUCCESS in 1h 18m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a6edb412d13e474ea9dddfcf573efbce : SUCCESS in 1h 53m 15s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2a729fd688204f4299c6b088c5d3f2df : RETRY_LIMIT in 11m 36s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/e5d91b3eaae84f239761a0e9b5b15cf8 : SUCCESS in 42m 47s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/52aabb72e18c409696735935933728f9 : SUCCESS in 1h 50m 37s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/872317d88958495795dabe39b52a7df8 : SUCCESS in 2h 03m 42s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/ca5af3eff64f41e3817fab258c0d4086 : SUCCESS in 1h 47m 04s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/4c349acf06a246b5aa9fe65515c324b5 : SUCCESS in 2h 04m 20s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/cb5737d00ba148a59845642a74caccb1 : SUCCESS in 1h 19m 28s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/faf29bb2ac294cf28c8cad5a0fc90875 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8132d488a4424fed9a2da819b5f8df6f : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/867c1e0961ff4072807bbe81af90db13 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be15a8881f9e42c1ba4fe089579bb52b : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8eee7251da4943ad8bf1eda3a53d9c9b : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fab8311b6bb74e23adc5df77329baa27 : SUCCESS in 4m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/545dc2facb8644b5ae689298258555d9 : SUCCESS in 24m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/16beb9fcad054027804686f8599c74ab : SUCCESS in 30m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ebd286abc7294e4cb406aaa8df90f5ad : SUCCESS in 50m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6dea320c7a804d7a83f7b1ca598783e0 : SUCCESS in 57m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/34397f64265a4bb9a1e3c54c10ba3f7f : SUCCESS in 1h 07m 12s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/e63ef67b66cc4dbda9548a0a9b21907e : SUCCESS in 1h 15m 19s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/1ea60c3e9d12434aaeaac400386c5ea3 : SUCCESS in 1h 11m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/6f8d5184f5184d0ba5c571e60a61f743 : SUCCESS in 1h 28m 21s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/074fa446c7a84be0b5cd452120a4cdfc : SUCCESS in 1h 03m 34s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/288715e182064d6ebf386c8a9c9fb85a : SUCCESS in 1h 03m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f0b1275d6a6f9efbd2c0ca1dd943a1ca869f6d2","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 12:35:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8c5ee58e74c1a8b0371938bd48f54a988e7c1ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 12:36:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee501d94b79d470192fc45f7c6840a89a0e811a6","author":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"date":"2019-08-19 12:38:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"457022fa73a1ac7fd4e3b1027e35e2b6489b503b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 13:18:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a00d5f4dabf5ce5a9a047d1febff750d15206c03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 13:18:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/045dd98839c74057a78022d12620bb17 : SUCCESS in 41m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5f1aeedee1fa4eb6b57a5c2ea5268e54 : RETRY_LIMIT in 10m 05s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e70087e67b524ec38c3a99d0dec31985 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39116b39da5845298ff399ab78c3a30a : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e9ee59aa03442ea92b1c7ee0b8bd7f0 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3537e85d1a9e4966a8324172c02bd3d4 : SUCCESS in 4m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29dd4342622e4c48ad17bc9b13660018 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0b881728204adea611a62c9d6df39d : SUCCESS in 3m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9aa9b4b9c0734b8fa04b02ae895baaa4 : SUCCESS in 39m 38s","accounts_in_message":[],"_revision_number":1},{"id":"73882518869a25231b085518cc1f4ae907e839dc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 13:20:34.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 9f93f377cf17a57b3515e7bf9e7377b38953d261","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07b55048e63f27c1045735624c845d1af95a11fd","revisions":{"07b55048e63f27c1045735624c845d1af95a11fd":{"kind":"REWORK","_number":1,"created":"2019-08-19 08:32:28.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/44/677144/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/677144/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/677144/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/677144/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/677144/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/677144/1"}}},"commit":{"parents":[{"commit":"0ffd3aefda90fd7206f4b0989ff17ace414423fe","subject":"Merge \"Ensures monasca-thresh data and worker dirs exist and are empty on start up\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ffd3aefda90fd7206f4b0989ff17ace414423fe"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-19 08:30:58.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-19 08:30:58.000000000","tz":120},"subject":"Unbreak CI: fluentd and TripleO dependency","message":"Unbreak CI: fluentd and TripleO dependency\n\nfluentd - we are not deploying kubernetes, no need for plugins\nTripleO - CI currently broken (to reverse, FIXME in place)\n\nChange-Id: I203f70176af7f2b53398ddeb5c21e6234a45e268\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07b55048e63f27c1045735624c845d1af95a11fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/07b55048e63f27c1045735624c845d1af95a11fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
