)]}'
{"id":"openstack%2Fkolla~677156","triplet_id":"openstack%2Fkolla~master~I1c3b817712ae10edd45de01382b044af2fb728bd","project":"openstack/kolla","branch":"master","topic":"bug/1840631","hashtags":[],"change_id":"I1c3b817712ae10edd45de01382b044af2fb728bd","subject":"Add effective and permitted capability to blackbox exporter","status":"MERGED","created":"2019-08-19 09:27:23.000000000","updated":"2019-11-05 12:03:44.000000000","submitted":"2019-09-18 10:33:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677156-1568802804645-d6d561dc","meta_rev_id":"b83c16e7064c9502482e8940e54c54abdbcf4363","_number":677156,"virtual_id_number":677156,"owner":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":29915,"name":"Jack Heskett","email":"Jackheskett@gmail.com","username":"jackheskett"},{"value":2,"date":"2019-09-18 10:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-04 11:17:00.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-09-05 10:47:49.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-09-18 08:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-09-05 10:36:30.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":29915,"name":"Jack Heskett","email":"Jackheskett@gmail.com","username":"jackheskett"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-09-18 08:57:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":29915,"name":"Jack Heskett","email":"Jackheskett@gmail.com","username":"jackheskett"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":29915,"name":"Jack Heskett","email":"Jackheskett@gmail.com","username":"jackheskett"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":29915,"name":"Jack Heskett","email":"Jackheskett@gmail.com","username":"jackheskett"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29915,"name":"Jack Heskett","email":"Jackheskett@gmail.com","username":"jackheskett"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 12:59:58.000000000","updated_by":{"_account_id":29915,"name":"Jack Heskett","email":"Jackheskett@gmail.com","username":"jackheskett"},"reviewer":{"_account_id":29915,"name":"Jack Heskett","email":"Jackheskett@gmail.com","username":"jackheskett"},"state":"REVIEWER"},{"updated":"2019-09-04 11:17:00.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-05 10:36:30.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-05 10:47:49.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-09-18 08:57:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-09-18 10:33:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e498e3ef82ec373bb1d66eb3d16267517934b4d7","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2019-08-19 09:27:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d57dffe3f8978fa4100d9077256a5483a84ade19","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 09:50:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/677156/1/openstack-check/tripleo-build-containers-rhel-8-master/8e20607/ : RETRY_LIMIT in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"44cb25b3c0d21cb51e583519e86ad9333695c9e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-19 11:13:29.000000000","message":"Patch Set 1:\n\nI don\u0027t think we have a role to deploy this exporter yet, do we? Does the container also need CAP_NET_RAW, or are you running it as privileged?","accounts_in_message":[],"_revision_number":1},{"id":"b8cc693bad6b3cd1273d666d0d43b37a72862727","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 11:25:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/94616a7cb36943ff89a38eb27db5e754 : SUCCESS in 36m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/aac0ee6d9fa54807a363a4d34faa4582 : SUCCESS in 1h 06m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e80a9b216f1041eeaa7c7e507b83e4a8 : SUCCESS in 1h 23m 51s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/20a0b66e641f4dc19714673c826d4c2e : SUCCESS in 1h 19m 48s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/ebd0686e74c54f2d8e1fb987a690bb1b : SUCCESS in 1h 45m 38s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/20530e140c99494e82452074211a6872 : RETRY_LIMIT in 6m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4426568b41004c22ae87ddd7eb8de705 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d2397bf81ad48269143236fc386f1cc : SUCCESS in 6m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee07de1fdfda4f7a89f921494ef5fcfe : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3179c49e8c45436a9458e16cf4418876 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b64014df107942fc9eac15ec4735a186 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e7042890d0945a48ce4843c3305f54b : SUCCESS in 6m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/88ba93e5826a4f60ad0ba06610e635c2 : FAILURE in 34m 23s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e7b67f5177e4381b67ec8760874bfd2 : FAILURE in 50m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a7d80a43ba14b9fbb92a57b9cdb5a06 : FAILURE in 28m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c913a4ed01ad4ceba5e4899f5d47d3c3 : FAILURE in 33m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9b586ac1e7c04330a7af706eaab784b0 : FAILURE in 22m 07s","accounts_in_message":[],"_revision_number":1},{"id":"881010ebdd9482e1ed3cabbd45d8187077be99ba","author":{"_account_id":29915,"name":"Jack Heskett","email":"Jackheskett@gmail.com","username":"jackheskett"},"date":"2019-08-19 12:59:58.000000000","message":"Patch Set 1:\n\n\u003e I don\u0027t think we have a role to deploy this exporter yet, do we?\n \u003e Does the container also need CAP_NET_RAW, or are you running it as\n \u003e privileged?\n\n \u003e I don\u0027t think we have a role to deploy this exporter yet, do we?\n \u003e Does the container also need CAP_NET_RAW, or are you running it as\n \u003e privileged?\n\nWe have the kolla-ansible tasks done, hopefully we will get that submitted soon. It will be added to the existing prometheus role.\n\nThe setcap command is run as root during the container build, before the USER is set to prometheus.","accounts_in_message":[],"_revision_number":1},{"id":"3a8348c7d18ac96f3490e80ec1764b7cb962ab94","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-21 11:07:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6d4cab278b4595e6c5ece4b7894cccb5c0fa3c88","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-21 11:50:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/677156/1/openstack-check/tripleo-build-containers-rhel-8-master/33153d2/ : SUCCESS in 41m 49s","accounts_in_message":[],"_revision_number":1},{"id":"e03dbc19ddffc79bfbf61619c9711ac02401ab9c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 12:45:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1641daccb4ac47b89ac034f5d1f239c8 : SUCCESS in 49m 02s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/fbd9462d00c243a586df1d802c2fcdc8 : SUCCESS in 59m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/034b3fe6658c447d83242674614d6194 : SUCCESS in 1h 20m 52s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2d8e36be09cf4e0985d366bca69bbe22 : SUCCESS in 1h 16m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/156abda819a34804abc146d68bbd9608 : SUCCESS in 1h 25m 55s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cede387be5f447a2bdf817e6e35cf382 : SUCCESS in 44m 40s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/42bff80e6379426684e0dceeda1a96eb : SUCCESS in 28m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ec9f8fb7ca3472f9f243242cd893b5f : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0db4b448a2eb4f5684474a2541540279 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6bc9dceaaafa47b4b5f963c4db86e043 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db37982f24bd42008e65a85065eeb7ab : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b45483dc6d75419b95edf5c8c3aa084b : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f61d3cf9ea54ef3907c77b3b6f99cff : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/486b69883bd54f0aa97486baeb6df148 : FAILURE in 27m 31s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/af560036a71a40e3b075cc73273c8170 : FAILURE in 36m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/54aab599312c4542a765cd684dcf1c71 : SUCCESS in 52m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/70ea7b1656e84b8a8d9481a1c62be773 : SUCCESS in 57m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8530779ee47544258ba0bf7506462e55 : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9667c82fb6d2338691f639b49464cd82bfa5338d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-21 13:13:35.000000000","message":"Patch Set 1: Code-Review-1\n\nOn ubuntu:\n\nsetcap: not found\n\nhttps://12942841482e92c0cd55-0acfe5e696fda6aec28b6c718c3c4841.ssl.cf1.rackcdn.com/677156/1/check/kolla-build-ubuntu-binary/486b698/kolla/build/000_FAILED_prometheus-blackbox-exporter.txt.gz","accounts_in_message":[],"_revision_number":1},{"id":"03592a16a2d444a2232023eb60e02f224c48cb63","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2019-08-23 12:45:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5a333db3a1b2ab20762d9736f9353c2fa1f4803","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 13:09:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/c519d27922c24701a4cad2c27a1bb82a : FAILURE in 5m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/dc29d749642b4cf2ba22317faa4489a9 : FAILURE in 5m 04s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/cef081d69a56476f84916da66204886f : FAILURE in 8m 48s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/5a2df5723870481b84fa27f8ea21294c : FAILURE in 7m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/4023a035c8284940b7cd9eb34116595a : FAILURE in 8m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/5dc9c5a2f2704ac8a6da8363d2bd93d1 : FAILURE in 19m 54s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/094cf937ae41420e88c3a8865559cb53 : FAILURE in 4m 36s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8957566554f546caab4a9e4c4a21fac4 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a15ae02386f64a8fa865da087e9b734c : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/afa5dcacd6e94389a02607245a324481 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b12bc9dcd8004a6d8d86d68ec08dc655 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a5e66576d0d4f6593427081a52528fd : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/020d22ba80844e21a1928660ef90c776 : SUCCESS in 3m 26s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/422b2b6a20b4433ba8c9586ae89b53fd : FAILURE in 3m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5fe0a12f56c64b498ea062544ca4be3d : FAILURE in 3m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d86861277364c50aec5cc99aa72eb20 : FAILURE in 6m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/043ceaa47f014d9d9ed4a064b495af0a : FAILURE in 6m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2736fb7b7a814a07986bbac93ca59582 : FAILURE in 6m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4ad00b2663fa4cb6e7b1d852001e0c456f8a05e6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 13:59:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/677156/2/openstack-check/tripleo-build-containers-rhel-8-master/5e931d6/ : SUCCESS in 1h 13m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ab07a7ac69bec9ed062ff0c0ef904c92b78b4db2","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2019-08-28 14:26:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"00d715a2d65163e3bd24318ee44d43c9412d4c87","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-28 15:03:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/677156/3/openstack-check/tripleo-build-containers-rhel-8-master/9d7d81c/ : SUCCESS in 35m 48s","accounts_in_message":[],"_revision_number":3},{"id":"3d37d80d7c52032894706788b3ef748eb872ce82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 15:42:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/631b30109a6947ec89ac7d58f67c663b : SUCCESS in 40m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/b248ffa117e3465488f83debec23b744 : SUCCESS in 38m 23s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f439bb54c592463e93975fafba3df7b5 : SUCCESS in 1h 14m 11s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1da06846da314966b8ca80aad4619f0a : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/36d2ed6d82da47bfb53379d6b411b91e : RETRY_LIMIT in 2m 52s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c0d1b30c7cb04e1fa9c2a90b46a0f852 : SUCCESS in 48m 31s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3562d2f30abc4e90ac8f53c1a22dce41 : SUCCESS in 39m 21s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7889daae4004c2a8b319ca9f0d51ef6 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc93f85cf29a4c619a078cfb99ed8fc1 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14f01bd01e444ff89cca0579c68fe30f : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43594e7e1d024ff4a0f7e4147e97ebc9 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ebb3848ac644f7492124636ffc2fc10 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b729408392b64a5cb94876176b946b6f : SUCCESS in 4m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1298c661685f41cbacfb5f8af5775ca2 : FAILURE in 26m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/22b37c0bf2dd422b924a9024a49ca1b4 : FAILURE in 28m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/51e849f2930945a0bd59f1ce4d8e5785 : FAILURE in 18m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d3f80b59429145fca7d3325b2a9ac914 : SUCCESS in 50m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7c6da89a2a964fbc8668e6c202599b1f : FAILURE in 54m 47s","accounts_in_message":[],"_revision_number":3},{"id":"5e6b35ebae0c7b2ebf03f58f365797cea9d774f8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-03 15:47:06.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"789487905ff3cc444cbf4fc8d28fba273f0dbcb5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-03 16:28:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/677156/3/openstack-check/tripleo-build-containers-rhel-8-master/93b64e0/ : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":3},{"id":"09549f761e7ce9dca8da5eaaf420f8fa828cbd71","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 17:16:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/718a0bf8840846d1969d73fa6a5324af : SUCCESS in 37m 27s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/de1eb983ea0f4cc0949045fa11a068a3 : SUCCESS in 52m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/094357e0579c4731ace3d3286f19ab12 : SUCCESS in 1h 13m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f02fec16a5a24e86b0782385b29294d7 : SUCCESS in 1h 15m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9b105231bbcc49dab473e53ffdf5453c : SUCCESS in 1h 26m 56s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6a2afe12507b404f8155d4949e559271 : SUCCESS in 59m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8c98660b9df7438e88c36721a85d4dbe : SUCCESS in 26m 39s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7728f9aeb5a74493ac8cd9fe4af217e5 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0113353894c44f8bedd9f6711b7c793 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/82fbfd3425934610bba046071d18ae6b : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7acbd7b22a454d03b55fee092614a9a2 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c8161b99ce446f78d42b317e1b19fa1 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1a197f90b8d4f3a8bcc1c7d3c056709 : SUCCESS in 6m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/802b2d65607b4b33a8b04fe87b2d3b9d : FAILURE in 30m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f626eb5768014dacadc5de7793d47795 : FAILURE in 41m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3dbea64e78534ee6a9b8f518106ecc24 : SUCCESS in 58m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d09c217a9b574cf0b395a6ac26372cf8 : SUCCESS in 54m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/af2f778623f94a56b233fd91a8978907 : SUCCESS in 1h 19m 13s","accounts_in_message":[],"_revision_number":3},{"id":"a5d9c5553233136f4b7b4b69fee82314d715e59c","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2019-09-04 10:37:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"570d0d87939a19aa6493ca20f2ed3540edcc1c94","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-04 11:17:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/56/677156/4/openstack-check/tripleo-build-containers-rhel-8-master/013a2f7/ : SUCCESS in 38m 40s","accounts_in_message":[],"_revision_number":4},{"id":"df8179e30070734289be2942b2a354755ca9899d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 12:22:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/ee38139affaf4a1c8a60ecc0a784c507 : SUCCESS in 35m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d39a5f8af23a41b3b238348b897e4da5 : SUCCESS in 52m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/e9f9f5dea0d74299a7ae4762dc6f8858 : SUCCESS in 1h 25m 56s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/24ff3a10e4e34da4bb7535ce91b46610 : SUCCESS in 1h 00m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f12a69a3c8e543459b22dfeea844053d : SUCCESS in 1h 39m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/56ed934eb74a4fd2809fe17e7136e03a : SUCCESS in 47m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/06e640b5f75d4dbfbad6e15439fa5ea3 : SUCCESS in 24m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/127856a2e527475b8197becea5c92581 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f5abff1df304211ac342414877711dc : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0967d0c52bdf4de6b195b2e122cbb8fe : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/577d46f5a4024753a364f93915a8b658 : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7badf54072c4c23a593d0c7ae5207ab : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9bc50ca984741349b1cdd6899f01cf7 : SUCCESS in 4m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/33d9c7542fe348f8818017c12bea1243 : SUCCESS in 27m 58s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc6d450c48f4404eaf591a88c6b32eb1 : SUCCESS in 49m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a89ba7554e54addb4c13510c4541cd3 : SUCCESS in 59m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5722b748ad394e3886aec0089cd079b6 : SUCCESS in 54m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cac8da003d3e4059af6a4f573de5f81a : SUCCESS in 1h 12m 17s","accounts_in_message":[],"_revision_number":4},{"id":"6b45eb01bcef0a435c7cade377a3bb74e651fef2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-05 10:36:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"409bf19d34392ce618e3e630cfdacfc52f783dba","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-09-05 10:47:49.000000000","message":"Patch Set 4: Code-Review+1\n\nI\u0027ve tested this on Centos and it allows the ICMP probes to work.","accounts_in_message":[],"_revision_number":4},{"id":"bd387bc06200ad33a9bcb76a4594745c24d07eb2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-18 08:57:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ee971f30b17edb917a297262df08a40c4d9a0b2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 08:58:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"edaa157ba88552c28ae239f22e68128177bc8337","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:33:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"7bf63c0b98c4fffb2f552986b559757fcd145c9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:33:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e3c9fae43a6c45298b693eae820413bc : SUCCESS in 55m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/18e028b7788242f59960ac4b2d260233 : SUCCESS in 1h 29m 24s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/afb2b258ab0c4edfa3ffa1959696ec19 : SUCCESS in 1h 26m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/7079515f7b85419a929998f3f4c8fdd9 : SUCCESS in 53m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecd0befbfefb47a8922b8068ff047277 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6410b8a5bd4b4017a451f0a8554a9c4d : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/80bd48ef90f24aca934e4fbf5e84ff83 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9ecf70acb944a9c85d54f7229a5dc48 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8804851568fd42b99c1db97a06dd4920 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0a3ed2c647461299cb87d14a4f508b : SUCCESS in 3m 33s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/40b1b3233c364039aa0b8d8e574071f1 : SUCCESS in 35m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0652229da294c07b26c5a6956f57897 : SUCCESS in 55m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/62aa796ca3564bffb5440efeca598fb9 : SUCCESS in 1h 10m 02s","accounts_in_message":[],"_revision_number":4},{"id":"deb0d8f01c5669a7f3236bca6068c25d862f0344","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 10:35:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31c54d0c5a69466f993c379d6ce6d301 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":4},{"id":"0dc8917f9a7331d069d062c7648a43b8b2d6bf9e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 12:03:44.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 8e712fadd3d7e12163455a74a8076791d925de11","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b5a4478b0fa9b775a9b5c45545e78e9215a50588","revisions":{"5c84bb3491d84eef15844c7ea811eec45f10ecaf":{"kind":"REWORK","_number":1,"created":"2019-08-19 09:27:23.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/56/677156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/677156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/677156/1"}}},"commit":{"parents":[{"commit":"1e95081f06897a20b1496d70282cf5c96e1b8263","subject":"Merge \"CI: drop docker systemd MountFlags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e95081f06897a20b1496d70282cf5c96e1b8263"}]}],"author":{"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","date":"2019-08-16 16:10:33.000000000","tz":0},"committer":{"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","date":"2019-08-19 09:26:35.000000000","tz":0},"subject":"Add effective and permitted capability to blackbox exporter","message":"Add effective and permitted capability to blackbox exporter\n\nThe ICMP probe in the Prometheus blackbox exporter requires\nelevated privileges to function. Linux root user or CAP_NET_RAW\ncapability is required. Can be set by executing setcap\ncap_net_raw+ep blackbox_exporter.\n\n\u0027+ep\u0027 means we\u0027re adding the capability as effective and permitted.\n\nSee: https://github.com/prometheus/blackbox_exporter\n\nChange-Id: I1c3b817712ae10edd45de01382b044af2fb728bd\nCloses-Bug: #1840631\nCo-authored-by: Jack Heskett \u003cJack.Heskett@gresearch.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c84bb3491d84eef15844c7ea811eec45f10ecaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c84bb3491d84eef15844c7ea811eec45f10ecaf"}]},"branch":"refs/heads/master"},"8caa3e80ee6b450a74f53478fa29e0687f4e5888":{"kind":"REWORK","_number":2,"created":"2019-08-23 12:45:14.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/56/677156/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/677156/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/677156/2"}}},"commit":{"parents":[{"commit":"1e95081f06897a20b1496d70282cf5c96e1b8263","subject":"Merge \"CI: drop docker systemd MountFlags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e95081f06897a20b1496d70282cf5c96e1b8263"}]}],"author":{"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","date":"2019-08-16 16:10:33.000000000","tz":0},"committer":{"name":"Scott Solkhon","email":"Scott.Solkhon@gresearch.co.uk","date":"2019-08-23 12:45:13.000000000","tz":60},"subject":"Add effective and permitted capability to blackbox exporter","message":"Add effective and permitted capability to blackbox exporter\n\nThe ICMP probe in the Prometheus blackbox exporter requires\nelevated privileges to function. Linux root user or CAP_NET_RAW\ncapability is required. Can be set by executing setcap\ncap_net_raw+ep blackbox_exporter. This change also bumps the version\nof the blackbox exporter to the next minor version to allow for this\nsupport.\n\n\u0027+ep\u0027 means we\u0027re adding the capability as effective and permitted.\n\nSee: https://github.com/prometheus/blackbox_exporter\n\nChange-Id: I1c3b817712ae10edd45de01382b044af2fb728bd\nCloses-Bug: #1840631\nCo-authored-by: Jack Heskett \u003cJack.Heskett@gresearch.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8caa3e80ee6b450a74f53478fa29e0687f4e5888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8caa3e80ee6b450a74f53478fa29e0687f4e5888"}]},"branch":"refs/heads/master"},"630ef91e48d89b3c4b5ffdaf3a2ac6baf78425ef":{"kind":"REWORK","_number":3,"created":"2019-08-28 14:26:35.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/56/677156/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/677156/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/677156/3"}}},"commit":{"parents":[{"commit":"1e95081f06897a20b1496d70282cf5c96e1b8263","subject":"Merge \"CI: drop docker systemd MountFlags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e95081f06897a20b1496d70282cf5c96e1b8263"}]}],"author":{"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","date":"2019-08-16 16:10:33.000000000","tz":0},"committer":{"name":"Scott Solkhon","email":"Scott.Solkhon@gresearch.co.uk","date":"2019-08-28 14:26:16.000000000","tz":60},"subject":"Add effective and permitted capability to blackbox exporter","message":"Add effective and permitted capability to blackbox exporter\n\nThe ICMP probe in the Prometheus blackbox exporter requires\nelevated privileges to function. Linux root user or CAP_NET_RAW\ncapability is required. Can be set by executing setcap\ncap_net_raw+ep blackbox_exporter. This change also bumps the version\nof the blackbox exporter to the next minor version to allow for this\nsupport.\n\n\u0027+ep\u0027 means we\u0027re adding the capability as effective and permitted.\n\nSee: https://github.com/prometheus/blackbox_exporter\n\nChange-Id: I1c3b817712ae10edd45de01382b044af2fb728bd\nCloses-Bug: #1840631\nCo-authored-by: Jack Heskett \u003cJack.Heskett@gresearch.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/630ef91e48d89b3c4b5ffdaf3a2ac6baf78425ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/630ef91e48d89b3c4b5ffdaf3a2ac6baf78425ef"}]},"branch":"refs/heads/master"},"b5a4478b0fa9b775a9b5c45545e78e9215a50588":{"kind":"REWORK","_number":4,"created":"2019-09-04 10:37:31.000000000","uploader":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"ref":"refs/changes/56/677156/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/677156/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/677156/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/677156/4"}}},"commit":{"parents":[{"commit":"1e95081f06897a20b1496d70282cf5c96e1b8263","subject":"Merge \"CI: drop docker systemd MountFlags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e95081f06897a20b1496d70282cf5c96e1b8263"}]}],"author":{"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","date":"2019-08-16 16:10:33.000000000","tz":0},"committer":{"name":"Scott Solkhon","email":"Scott.Solkhon@gresearch.co.uk","date":"2019-09-04 10:37:20.000000000","tz":60},"subject":"Add effective and permitted capability to blackbox exporter","message":"Add effective and permitted capability to blackbox exporter\n\nThe ICMP probe in the Prometheus blackbox exporter requires\nelevated privileges to function. Linux root user or CAP_NET_RAW\ncapability is required. Can be set by executing setcap\ncap_net_raw+ep blackbox_exporter. This change also bumps the version\nof the blackbox exporter to the next minor version to allow for this\nsupport.\n\n\u0027+ep\u0027 means we\u0027re adding the capability as effective and permitted.\n\nSee: https://github.com/prometheus/blackbox_exporter\n\nChange-Id: I1c3b817712ae10edd45de01382b044af2fb728bd\nCloses-Bug: #1840631\nCo-authored-by: Jack Heskett \u003cJack.Heskett@gresearch.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5a4478b0fa9b775a9b5c45545e78e9215a50588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5a4478b0fa9b775a9b5c45545e78e9215a50588"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
