)]}'
{"id":"openstack%2Fkolla~677194","triplet_id":"openstack%2Fkolla~master~I517c8d6fefaa8393aa0c5ac2777f263108150c9c","project":"openstack/kolla","branch":"master","topic":"677194","hashtags":[],"change_id":"I517c8d6fefaa8393aa0c5ac2777f263108150c9c","subject":"Cleanup: drop checks handled by UNBUILDABLE_IMAGES","status":"ABANDONED","created":"2019-08-19 12:50:03.000000000","updated":"2020-06-18 22:17:41.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0606abece6e68ee8696d7e3a03058a45b8fcade1","_number":677194,"virtual_id_number":677194,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"date":"2020-03-09 20:38:01.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"date":"2020-03-09 20:40:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":1,"date":"2019-08-19 23:44:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-08-19 22:09:48.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":-1,"date":"2020-03-10 11:42:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"date":"2020-03-09 20:26:04.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 12:50:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-08-19 22:09:48.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 23:44:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-09 20:26:04.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-03-09 20:40:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-03-10 11:42:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"3b6e40e302cd649d98ac19777b79622cec00d45a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 12:50:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"060dc44e2685307184bcacec0a5d9a7e06ed4400","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 13:13:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/94/677194/1/openstack-check/tripleo-build-containers-rhel-8-master/8a260c0/ : RETRY_LIMIT in 7m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a3760268e50e1442b05869316edf794acca0c9a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 14:37:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/70398125d5474515a0f458dd48efb9e3 : SUCCESS in 47m 07s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/ded65d06d68e4ea882aa407c1ffaf2d5 : SUCCESS in 1h 06m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/ecb99bd34bc74fa7a61bf8e0ed050295 : SUCCESS in 1h 13m 28s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d249158e7dd34d3c8ba95419122a2478 : SUCCESS in 1h 20m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9f950ab6a8ce4a018b3f45d56c6ebfee : SUCCESS in 1h 42m 07s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/ed57c88b59e442a6b1389a8df5f17d62 : RETRY_LIMIT in 6m 30s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/7b098a4d609b40428c52e678a1ecf00d : SUCCESS in 42m 33s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/6534d3fb8747438e875d5af0d48296f9 : SUCCESS in 1h 40m 15s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c7ebfa724aa4b2da93a194bf1683372 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61c056941c304ad393457858a277cd32 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5fa7698f2e724c5ea51564621f688592 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0817801c4c54be7af55d281a7244fff : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fa2943e0fa32401abf15bc9fbf66ad16 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d30a7dd183ca42e3ae3a1272c04a66fa : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8565723afe2245c69d6776f5000ae934 : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dbd1e4bef8514907a1b36d4a55a7b636 : FAILURE in 28m 50s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4dec904c97014f2fab4aff357438e52d : FAILURE in 34m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9bf4679172bd4513a37b81850999c8c4 : FAILURE in 28m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/94d110583f3e4fa0b35573ab8a13bfff : FAILURE in 20m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0b426b1dd3184a7c9b6493234133cd9e : FAILURE in 38m 17s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/7687be62b774489eb0e731dd11120cae : FAILURE in 37m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9e389c9a19fc1d1ae38dc5ee84f0d0b18f78a84","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 21:49:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8ba171e39131ce448c99b869b3d83984a6d6ba3e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 22:09:48.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/94/677194/2/openstack-check/tripleo-build-containers-rhel-8-master/90f0f81/ : RETRY_LIMIT in 6m 16s","accounts_in_message":[],"_revision_number":2},{"id":"3465dbbb444f0234e10f4caeebb2b4bbdbe742fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 23:44:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e95c0b7481024953ad3064379303fd25 : SUCCESS in 45m 55s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0c97071ced2f4dc1af8d2c3e119697e2 : SUCCESS in 53m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9d1f951a37534c43a4b7b91cf92772ad : SUCCESS in 1h 18m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/35590087a18a45ac9a3c76a6c523d651 : SUCCESS in 1h 08m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0038f3b34f194c09a58cd554fa388227 : SUCCESS in 1h 35m 54s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/78efd9fa8bc1472bbc2c1c492e164a7d : RETRY_LIMIT in 11m 10s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/b0ffdffb82654d63942eab7c47f1224a : SUCCESS in 52m 35s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/c6d711c75a6e4f9bafab44b09083d4ae : SUCCESS in 1h 42m 51s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/461ecdf153004e50bf0b7029a59226f1 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0678298f43074071a8d0db884ad0f8c4 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bbd3d5122c97404d9be8f818f40b7091 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f8167b090b74b52a1e86f22029b7c4d : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d1a357f07af4d9a9ed0cca660ca7534 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f45afa9ac00f47dea090c7570f0e009f : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/269de962f0f74dc39550db70bdb52349 : SUCCESS in 5m 56s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/be8cb5214c46453c9c083c668c24bf82 : SUCCESS in 21m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/609d93ed8eda4a04bb23ae232efc416c : SUCCESS in 33m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1644aeacb60b490f87d41763fa94abbe : SUCCESS in 51m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/731d98c08a2946de866caec353b47956 : SUCCESS in 53m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e68f4e2b9bf4c219f4beca040867df7 : SUCCESS in 1h 13m 19s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/c0001c31277547e7bfecc52c30a71992 : SUCCESS in 1h 00m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f0d929462270bd7a72115c30716fc4ac71b58fc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-20 06:15:57.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nI think these are sanity checks. How is it better losing them? SKIPPED_IMAGES previously skipped them anyway. And if you forgot, then they failed the CI. Magnum here did not because it was buggy (why run bash? :O ).","accounts_in_message":[],"_revision_number":2},{"id":"1fb06d63d220413a7436c3367e875b6bde762386","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-05 11:02:04.000000000","message":"Patch Set 2:\n\n\u003e I think these are sanity checks. How is it better losing them?\n\u003e SKIPPED_IMAGES previously skipped them anyway. And if you forgot,\n\u003e then they failed the CI. Magnum here did not because it was buggy\n\nAnd when one day packages get created for missing distro you have to dig through all those Docker files instead of simple edit in UNBUILDABLE_IMAGES. BTDT :(","accounts_in_message":[],"_revision_number":2},{"id":"fe3a8fd158a9a011e8be4409eebca8901e1a192d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-09 20:25:57.000000000","message":"Patch Set 2:\n\nI plan to refresh that patch.\n\nNow with UNBUILDABLE_IMAGES and \"--enable-unbuildable\" switch it is easy to keep CentOS8 CI working. We limit what is buildable and at same time can check what also can be enabled. Without editing of any file.\n\nAlso we have Jim slowly working on s390x support. How many dockerfiles are ready for handling another architecture with their \"sorry, no package for you\" check?","accounts_in_message":[],"_revision_number":2},{"id":"91c2f3da72e4d5a5c9586866f143e20489b1a7b4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-09 20:34:11.000000000","message":"Patch Set 2:\n\nthe issue with this is that removing them from unbuildable will build empty images with no warning whatsoever","accounts_in_message":[],"_revision_number":2},{"id":"208d2e3be883ff5c7e7945b8b3d5132f1f8399cc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-03-09 20:38:01.000000000","message":"Patch Set 2:\n\n\u003e the issue with this is that removing them from unbuildable will\n \u003e build empty images with no warning whatsoever\n\nIf you use --enable-unbuildable then you are on your own. Deploying resulting images would not be sane thing.\n\nThose image which end with empty images being built will be removed from that patch. I mostly want to get those which fail due to missing packages to be freed.","accounts_in_message":[],"_revision_number":2},{"id":"195bdb8876b7ca1f8caf97a274458521e675d735","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-03-09 20:40:33.000000000","message":"Patch Set 2:\n\nI see your point, makes sense this way.","accounts_in_message":[],"_revision_number":2},{"id":"1cb032c878071d3b84a1bd98047da7f61a2b0dc4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-10 11:42:20.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nI\u0027m ok with removing the stubs with echo \u0026 fail, but let\u0027s keep the \u0027if source\u0027 etc branches for clarity.","accounts_in_message":[],"_revision_number":2},{"id":"6bf0002c81b8b2e2d16edc055c064212aec8bfb8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-03-10 11:42:24.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"7bd613420c2eacf87482f34141c8af0f292875bb","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-06-18 22:17:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1929b8acff4880bbc5f89613a83a812affe80243","revisions":{"0920cdc7a76aa11452acc903e8aabf3ebc9f4b1c":{"kind":"REWORK","_number":1,"created":"2019-08-19 12:50:03.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/677194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/677194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/677194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/677194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/677194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/677194/1"}}},"commit":{"parents":[{"commit":"1e95081f06897a20b1496d70282cf5c96e1b8263","subject":"Merge \"CI: drop docker systemd MountFlags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e95081f06897a20b1496d70282cf5c96e1b8263"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-19 12:44:28.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-19 12:49:47.000000000","tz":120},"subject":"Cleanup: drop checks handled by UNBUILDABLE_IMAGES","message":"Cleanup: drop checks handled by UNBUILDABLE_IMAGES\n\nDuring Train cycle we got a way to mark not buildable images via\nUNBUILDABLE_IMAGES dict. There is no need to check for\ndistro/arch/buildtype check in Dockerfile.j2 files as they are not\nattempted to build if they are present in UNBUILDABLE_IMAGES var.\n\nChange-Id: I517c8d6fefaa8393aa0c5ac2777f263108150c9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0920cdc7a76aa11452acc903e8aabf3ebc9f4b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0920cdc7a76aa11452acc903e8aabf3ebc9f4b1c"}]},"branch":"refs/heads/master"},"1929b8acff4880bbc5f89613a83a812affe80243":{"kind":"REWORK","_number":2,"created":"2019-08-19 21:49:06.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/94/677194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/94/677194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/677194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/677194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/94/677194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/94/677194/2"}}},"commit":{"parents":[{"commit":"9d80b285594768ac151a8e04c2217a46ece63a2e","subject":"Merge \"Add pciutils package to cyborg-agent and zun-compute\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d80b285594768ac151a8e04c2217a46ece63a2e"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-19 12:44:28.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-19 21:48:48.000000000","tz":120},"subject":"Cleanup: drop checks handled by UNBUILDABLE_IMAGES","message":"Cleanup: drop checks handled by UNBUILDABLE_IMAGES\n\nDuring Train cycle we got a way to mark not buildable images via\nUNBUILDABLE_IMAGES dict. There is no need to check for\ndistro/arch/buildtype check in Dockerfile.j2 files as they are not\nattempted to build if they are present in UNBUILDABLE_IMAGES var.\n\nChange-Id: I517c8d6fefaa8393aa0c5ac2777f263108150c9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1929b8acff4880bbc5f89613a83a812affe80243"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1929b8acff4880bbc5f89613a83a812affe80243"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
