)]}'
{"id":"openstack%2Fkolla~677202","triplet_id":"openstack%2Fkolla~stable%2Fstein~I203f70176af7f2b53398ddeb5c21e6234a45e268","project":"openstack/kolla","branch":"stable/stein","topic":"fluentd-tripleo-ci-unbreak-stable/stein","hashtags":[],"change_id":"I203f70176af7f2b53398ddeb5c21e6234a45e268","subject":"Unbreak CI: fluentd and TripleO dependency","status":"MERGED","created":"2019-08-19 13:20:34.000000000","updated":"2019-08-19 16:30:16.000000000","submitted":"2019-08-19 16:30:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677202-1566232216492-5b3f03e3","meta_rev_id":"5c60e6959bd2e5c24adf2d2805fb3c633309903b","_number":677202,"virtual_id_number":677202,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-19 16:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 13:49:47.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-19 13:59:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-19 14:33:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-08-19 14:33:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 13:49:47.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 13:59:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-19 14:33:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-19 16:30:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5a000e94fd503072f3d0f65861574e59e099819","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 13:20:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"62eceb87e67dc45657e509fb934315f18d6e2102","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 13:20:53.000000000","message":"Patch Set 1:\n\nwill drop gate part","accounts_in_message":[],"_revision_number":1},{"id":"4ac07128a723c2e1172c38a697354f8541ab5021","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 13:23:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c442f55f243089ea18bfb968f0a8dee4ddf87d40","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 13:49:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/02/677202/2/openstack-check/tripleo-build-containers-rhel-8-master/65d3ce0/ : RETRY_LIMIT in 6m 59s","accounts_in_message":[],"_revision_number":2},{"id":"47fedb4b1cb27ba1856bdabd24f683f9b1176ff7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-19 13:59:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"96eff96abba9be17db069521ed1aeec4df0c19a5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-19 14:33:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nw+1 because we know it works","accounts_in_message":[],"_revision_number":2},{"id":"027787500a99257aa930ea797bdea1203ad40140","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 15:30:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/26273c1f1d3c46bf979c62965dbf7029 : SUCCESS in 45m 12s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d8daf5900a7147a19e582bb7c560f1ca : SUCCESS in 1h 01m 31s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/52fbd495ef0e44a19bde944381fea28b : SUCCESS in 1h 11m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0c9da5f2b675483e92e34cb43ab022e1 : SUCCESS in 1h 09m 01s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/009fa3c6b81c43518cf94b60a63f5841 : SUCCESS in 1h 42m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b2b941575c754038b07984216461bdbf : RETRY_LIMIT in 9m 17s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/4352ebf2eed848bbbace949e3edd5cbc : SUCCESS in 56m 53s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/979ed86287514371998365a05fb50bc6 : SUCCESS in 54m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29e791d6c33741598d7d297963e0d183 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1039bd4bca054638b64cbd1014ec9152 : SUCCESS in 5m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc69523313bd46b4b08a562f5c96ccac : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17d9bd113d954a93ab321afb521464e3 : SUCCESS in 3m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bd3eacfcce34a089efe5be3a56c3cb5 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cc528c898ec48dcace30731a5001630 : SUCCESS in 3m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/44221a4cc9364fe38c841f0b1e3c2dba : SUCCESS in 1h 05m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4786c5027ad7404d931e19f1523fbd0b : SUCCESS in 1h 09m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/08b9ab03a2134db1940c595a8fa8e822 : SUCCESS in 1h 47m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/61fa6621dd24487ca3d3fa1e4be871c8 : SUCCESS in 1h 06m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8da93bbf7e8248259b02be8cd139a3bf : SUCCESS in 2h 01m 25s","accounts_in_message":[],"_revision_number":2},{"id":"f248b7f7f14e3d42a05def0b4c8352de386486b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 15:30:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7eac78c8f2c549f77643e811a6f3c8abb306961a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:30:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"bd57d26ab46d927ce66ec47e80b60a34d07e44d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:30:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/71ea071ed160401eb31b8c6ddb3a512b : SUCCESS in 39m 02s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/e28770af2dce4fc0b3ce4d99232c272e : SUCCESS in 54m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0d455dc859640d8ba9501344576f2a6 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8dfa73056d949719c9501c047555932 : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2052e515ae634938bbd3745c3f10189b : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/906cfeb3c60648059aff1eda38b4d634 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/300fd34d4fb341a387b45db60fbe17a5 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69abf4c79c064c048620bb48c930f9a1 : SUCCESS in 3m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/afb427c19fac455f83a43a00ecf84dee : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b5a6da0dd7735e272ff2e2d1b6d931036601f308","revisions":{"9f93f377cf17a57b3515e7bf9e7377b38953d261":{"kind":"REWORK","_number":1,"created":"2019-08-19 13:20:34.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/02/677202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/677202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/677202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/677202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/677202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/677202/1"}}},"commit":{"parents":[{"commit":"cd4d4b37be3ad75d23a48fd58a550ff08684e270","subject":"Stein-only: update packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd4d4b37be3ad75d23a48fd58a550ff08684e270"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-19 08:30:58.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-08-19 13:20:34.000000000","tz":0},"subject":"Unbreak CI: fluentd and TripleO dependency","message":"Unbreak CI: fluentd and TripleO dependency\n\nfluentd - we are not deploying kubernetes, no need for plugins\nTripleO - CI currently broken (to reverse, FIXME in place)\n\nChange-Id: I203f70176af7f2b53398ddeb5c21e6234a45e268\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit 07b55048e63f27c1045735624c845d1af95a11fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f93f377cf17a57b3515e7bf9e7377b38953d261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f93f377cf17a57b3515e7bf9e7377b38953d261"}]},"branch":"refs/heads/stable/stein"},"b5a6da0dd7735e272ff2e2d1b6d931036601f308":{"kind":"REWORK","_number":2,"created":"2019-08-19 13:23:58.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/02/677202/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/677202/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/677202/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/677202/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/677202/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/677202/2"}}},"commit":{"parents":[{"commit":"cd4d4b37be3ad75d23a48fd58a550ff08684e270","subject":"Stein-only: update packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd4d4b37be3ad75d23a48fd58a550ff08684e270"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-19 08:30:58.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-08-19 13:23:13.000000000","tz":120},"subject":"Unbreak CI: fluentd and TripleO dependency","message":"Unbreak CI: fluentd and TripleO dependency\n\nfluentd - we are not deploying kubernetes, no need for plugins\nTripleO - CI currently broken (to reverse, FIXME in place)\n- dropped TripleO from gate\n\nChange-Id: I203f70176af7f2b53398ddeb5c21e6234a45e268\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n(cherry picked from commit 07b55048e63f27c1045735624c845d1af95a11fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5a6da0dd7735e272ff2e2d1b6d931036601f308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5a6da0dd7735e272ff2e2d1b6d931036601f308"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
