)]}'
{"id":"openstack%2Fkolla~677212","triplet_id":"openstack%2Fkolla~stable%2Fstein~Ifb616b082c9d82f4736e4b688cafc3a58a0c7a9f","project":"openstack/kolla","branch":"stable/stein","topic":"fix_tempest","hashtags":[],"change_id":"Ifb616b082c9d82f4736e4b688cafc3a58a0c7a9f","subject":"Use openstack-tempest-all for rpm tempest container","status":"MERGED","created":"2019-08-19 14:16:30.000000000","updated":"2019-08-20 09:17:41.000000000","submitted":"2019-08-20 09:17:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677212-1566292661758-0e67f363","meta_rev_id":"7d1c5e19305e1526c896233051f9ed8023aaf860","_number":677212,"virtual_id_number":677212,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"date":"2019-08-20 06:16:23.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-20 09:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-19 17:48:06.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-19 17:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-08-19 17:28:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2019-08-20 08:21:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-20 06:22:18.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2019-08-20 08:21:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-08-19 19:56:23.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 14:25:28.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-08-19 17:28:33.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2019-08-19 17:48:06.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-19 19:56:23.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-20 06:22:18.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-20 08:21:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-20 09:17:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"651d658fa3485465211e521b6c70fcc3bd35872e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-19 14:16:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73d6fd173f3f8d1c40f1ac455ab1975c248594dc","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-19 14:18:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7913ccbe7547cfb233f063edd3a66520cfab06af","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-08-19 14:18:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1a114fa0690426152fa00bf28d2871fe1c1082e6","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-08-19 14:19:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87b751424ba386603fdb4f86deab5d10568d9f4f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-19 14:20:58.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b66e4ed6b5ee23034b4a64827619f9186c4b195b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-08-19 14:25:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"de2d2097893a56ea58508987f8b1b5e175911a9c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 14:43:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/12/677212/2/openstack-check/tripleo-build-containers-rhel-8-master/55a9470/ : RETRY_LIMIT in 7m 31s","accounts_in_message":[],"_revision_number":2},{"id":"c4fedb8dfda3f59fbd538d8b14f0fc93b107c117","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:07:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/dd8d74e74d8b4c199bdd262dcbc7017c : SUCCESS in 46m 03s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/e6f6abe001ed43d99649b2ef1cb26626 : SUCCESS in 48m 58s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/f0a92e2855bb46f58dd7e3589287829c : SUCCESS in 1h 03m 39s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8327d0867e08480ea17c68633509c62f : FAILURE in 3m 10s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/270e9e8cf6c3488a9f17cd489b48d6c0 : SUCCESS in 1h 37m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cd2a73bbddcb42e7886d6b9c9be81b3d : RETRY_LIMIT in 9m 19s\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/b9533f99877c46a880d06f47fb08524c : SUCCESS in 47m 31s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/63ebf57f1e3f43a7bf576810bdc9677c : SUCCESS in 38m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b4570ed5e59428aa9a0405c4543bfe8 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53ec218867e24e5abc851e5308853ffb : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/795a3b2f0620453aa75d00b61d928c81 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd88a18cb2514ac1ba1db5f698687101 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81d1cbc9f1cc4a8f9653f8b023cc6850 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c4d9ae5c67e4a4b9f54f5569c6c993a : SUCCESS in 3m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1559a60ecec8432fb4fbc1ed24d67d65 : FAILURE in 38m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2eb6ac58fb38455d8e9e1e58e00646b2 : FAILURE in 39m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a270838a65654250956aee235f7e6fa7 : FAILURE in 1h 26m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e1522c1d0b6f4066a22a355c18de67cb : FAILURE in 1h 07m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dab31f1406524d6db48933b04d625680 : FAILURE in 46m 04s","accounts_in_message":[],"_revision_number":2},{"id":"beabda340e56d6c5a3d1169a197c5424f2546999","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-19 17:26:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6fa34e9f1ff69fe571788f9ceb1fe2ddb39d6ffe","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2019-08-19 17:28:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"105649cbd083daba190b5a9e53a6e2aafd12ce37","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-19 17:48:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/12/677212/3/openstack-check/tripleo-build-containers-rhel-8-master/ed7b470/ : RETRY_LIMIT in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"2d051ce750d75e31eb6e0bf8231759af810bf429","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-19 19:00:12.000000000","message":"Patch Set 1:\n\nany bug to track?","accounts_in_message":[],"_revision_number":1},{"id":"dacd7393ca5f2a86a3c2f9e037cea8e8a21bc311","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 19:02:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/9e59d2ce78a04273937de3481fa6c757 : SUCCESS in 26m 06s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/06e5f8df0e6343949244dd3b3e8c1025 : SUCCESS in 43m 06s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9f36311e30fa4e39b12fbece2bc9906a : SUCCESS in 1h 20m 16s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2aaacb87df9d434f9d3294d910a06e13 : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9ea02c306d4746b5a9b3cd4790ffc1b7 : SUCCESS in 1h 32m 18s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/09524d1d0c61496598937bf2747ed8c2 : RETRY_LIMIT in 7m 10s (non-voting)\n- kolla-build-oraclelinux-binary https://zuul.opendev.org/t/openstack/build/0d29e67422784a01a7774a46fc80504d : SUCCESS in 56m 05s (non-voting)\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/468af460048b4f7db0d466ee4c77ab0f : SUCCESS in 54m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc43ba838a7a43bdb7f80bca8ac5fe29 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72351dccad534d86a5c0f6428bf1eb79 : SUCCESS in 5m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f7973b8178104caf85c39d57b3e1966e : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8922accd03843c383064cf832b8603f : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/236ad68ca66d4d819405bd38a19e8117 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e00e415c7fb4809ba87ad50a15c3d92 : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a2f969bfdd914f199a70e8557dcda10c : SUCCESS in 44m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/96edfda9002d46e09ab5b2ec19b74597 : SUCCESS in 35m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f9fab17cce6f4895b0896b84bf385611 : SUCCESS in 1h 07m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4bac1ec1a154436b88037652a1c49e77 : SUCCESS in 48m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2ffa90907e7b45ac9a6664ce271e0f5b : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":3},{"id":"843b3ab5c8589e5229190ac84ed3add1b89dbafe","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2019-08-20 06:16:23.000000000","message":"Patch Set 3:\n\n\u003e any bug to track?\n\nWe donot have bug to track but here is the logs https://logs.rdoproject.org/openstack-periodic-latest-released/opendev.org/openstack/tripleo-ci/master/periodic-tripleo-ci-centos-7-ovb-1ctlr_2comp-featureset020-stein/2d42673/logs/undercloud/home/zuul/tempest.log.txt.gz#_2019-08-20_00_15_37\n\ntempest plugin list is showing empty.","accounts_in_message":[],"_revision_number":3},{"id":"040d4f1b683d9d0e0dd491607d0c6aa561f05091","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-20 06:22:18.000000000","message":"Patch Set 3: Code-Review+2\n\nfine by me, I just wondered whether you don\u0027t prefer to track it more formally","accounts_in_message":[],"_revision_number":3},{"id":"4af2fc8237fc1e740b4dd591a42a0b3e2b8d94c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-20 08:21:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c617f60653ea5f852dfe57c5dfcb461350fb9ad4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 08:21:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7679a5d3c4975c2c35b353a8e3619b32f2aaeaa1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 09:17:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3d3fe790e30c480bb7d6f63f79308f9f : SUCCESS in 49m 25s\n- kolla-build-oraclelinux-source https://zuul.opendev.org/t/openstack/build/433d706ce4374207ab19cd3f2af101ab : SUCCESS in 41m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92ee0de35e1b46639b5874031ac2bba5 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c5e496e796a43629933dd7ec28d3781 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1a78804999c447f185d987a7d8af9f5d : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/555509ced6004521a7b041ef123d3a79 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2a3e84562986416aa21333647248f9ff : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c49d0f07124d53a65889b295056cff : SUCCESS in 3m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/afe938dc2f194ca1aaee98db8c3a82cd : SUCCESS in 30m 39s","accounts_in_message":[],"_revision_number":3},{"id":"87454b3082ce111393ac4493bafa05481cd1175d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 09:17:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"90b16debc3050b7bc2590717c6fb9a3e7c25fa4e","revisions":{"10d4a904403fd6615fcc1910aa56b2199fef9d9e":{"kind":"REWORK","_number":1,"created":"2019-08-19 14:16:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/12/677212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/677212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/677212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/677212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/677212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/677212/1"}}},"commit":{"parents":[{"commit":"cd4d4b37be3ad75d23a48fd58a550ff08684e270","subject":"Stein-only: update packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd4d4b37be3ad75d23a48fd58a550ff08684e270"}]}],"author":{"name":"Chandan kumar","email":"chkumar@redhat.com","date":"2019-08-19 14:12:57.000000000","tz":330},"committer":{"name":"Chandan kumar","email":"chkumar@redhat.com","date":"2019-08-19 14:16:27.000000000","tz":330},"subject":"Use openstack-tempest-all for rpm tempest container","message":"Use openstack-tempest-all for rpm tempest container\n\nopenstack-tempest shipped from RDO does not have the tempest\nplugins, It has just tempest cli and library.\n\nhttps://review.opendev.org/#/c/677063/ updated the rpm package\nname while solving different issue and TripleO stein CI,\nmost of the tempest tests are getting skipped as tempest plugins\nare not there.\n\nopenstack-tempest-all is the meta package contains all the plugins,\nand installing that fixes the issue.\n\nChange-Id: Ifb616b082c9d82f4736e4b688cafc3a58a0c7a9f\nSigned-off-by: Chandan kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10d4a904403fd6615fcc1910aa56b2199fef9d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10d4a904403fd6615fcc1910aa56b2199fef9d9e"}]},"branch":"refs/heads/stable/stein"},"0718fb337ed471dd95f4fe56104492898b50a65e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-19 14:20:58.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/12/677212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/677212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/677212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/677212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/677212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/677212/2"}}},"commit":{"parents":[{"commit":"cd4d4b37be3ad75d23a48fd58a550ff08684e270","subject":"Stein-only: update packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd4d4b37be3ad75d23a48fd58a550ff08684e270"}]}],"author":{"name":"Chandan kumar","email":"chkumar@redhat.com","date":"2019-08-19 14:12:57.000000000","tz":330},"committer":{"name":"Chandan kumar","email":"chkumar@redhat.com","date":"2019-08-19 14:20:33.000000000","tz":330},"subject":"Use openstack-tempest-all for rpm tempest container","message":"Use openstack-tempest-all for rpm tempest container\n\nopenstack-tempest shipped from RDO does not have the tempest\nplugins, It has just tempest cli and library.\n\nhttps://review.opendev.org/#/c/651601/ updated the rpm package\nname while solving different issue and TripleO stein CI,\nmost of the tempest tests are getting skipped as tempest plugins\nare not there.\n\nopenstack-tempest-all is the meta package contains all the plugins,\nand installing that fixes the issue.\n\nChange-Id: Ifb616b082c9d82f4736e4b688cafc3a58a0c7a9f\nSigned-off-by: Chandan kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0718fb337ed471dd95f4fe56104492898b50a65e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0718fb337ed471dd95f4fe56104492898b50a65e"}]},"branch":"refs/heads/stable/stein"},"90b16debc3050b7bc2590717c6fb9a3e7c25fa4e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-19 17:26:01.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/12/677212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/677212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/677212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/677212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/677212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/677212/3"}}},"commit":{"parents":[{"commit":"b5a6da0dd7735e272ff2e2d1b6d931036601f308","subject":"Unbreak CI: fluentd and TripleO dependency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5a6da0dd7735e272ff2e2d1b6d931036601f308"}]}],"author":{"name":"Chandan kumar","email":"chkumar@redhat.com","date":"2019-08-19 14:12:57.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2019-08-19 17:26:01.000000000","tz":0},"subject":"Use openstack-tempest-all for rpm tempest container","message":"Use openstack-tempest-all for rpm tempest container\n\nopenstack-tempest shipped from RDO does not have the tempest\nplugins, It has just tempest cli and library.\n\nhttps://review.opendev.org/#/c/651601/ updated the rpm package\nname while solving different issue and TripleO stein CI,\nmost of the tempest tests are getting skipped as tempest plugins\nare not there.\n\nopenstack-tempest-all is the meta package contains all the plugins,\nand installing that fixes the issue.\n\nChange-Id: Ifb616b082c9d82f4736e4b688cafc3a58a0c7a9f\nSigned-off-by: Chandan kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90b16debc3050b7bc2590717c6fb9a3e7c25fa4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90b16debc3050b7bc2590717c6fb9a3e7c25fa4e"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
