)]}'
{"id":"openstack%2Fkolla~677225","triplet_id":"openstack%2Fkolla~master~Ic2948953b9e8cd5148d6bde539003ab9d081b215","project":"openstack/kolla","branch":"master","topic":"ci-tripleo-no-gate","hashtags":[],"change_id":"Ic2948953b9e8cd5148d6bde539003ab9d081b215","subject":"CI: follow-up on TripleO demotion - no gating","status":"MERGED","created":"2019-08-19 14:51:47.000000000","updated":"2019-08-20 00:20:47.000000000","submitted":"2019-08-20 00:20:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677225-1566260447319-16970a7d","meta_rev_id":"8d787f0fa1d75726643ab1f34001fcb549cf93b2","_number":677225,"virtual_id_number":677225,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-20 00:20:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-08-19 16:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-08-19 16:47:23.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-08-19 16:47:23.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-19 15:06:04.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-19 16:47:23.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-19 16:53:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-08-20 00:20:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91609d3fb8dae353ed52bf04bfd2a3596387c55c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-19 14:51:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c069c0ba5e0987120274d5ea718968b74160c75","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-19 15:06:04.000000000","message":"Patch Set 1: Review-Priority+2\n\nto make infra happy, please w+1","accounts_in_message":[],"_revision_number":1},{"id":"0bfd9c1f5b39a5d5993a9410eb67a72dd99ae8ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:25:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6f73f21e207a43c1baed7b5b6654a1cd : SUCCESS in 49m 09s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f126c9640f7143ecabd9f6a093b18765 : SUCCESS in 51m 02s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/facfd9bb715f45d282b1db01acd58113 : SUCCESS in 1h 28m 33s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/50e8f0526b7e4b76a59ca2f0a8c96f4a : SUCCESS in 1h 14m 15s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/620ef4a41a7e425090946d607d028e72 : SUCCESS in 1h 20m 31s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/099425bc0ea340ca991a97dc6354fa84 : RETRY_LIMIT in 7m 04s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/551e217ff168400eb0b5434617ce1681 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab7ad63c7da44d2eb9c2982f96e7e96d : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fc18f52dfb24c55b73a350410825623 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b12f51ee76174e70b7c3a0535a9b0923 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/704d5bf91f47499691a5c0b6778ea758 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9d5971596d44639f35fd45206a772f : SUCCESS in 5m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43ce58128dfa445399de01cc0126e7a7 : SUCCESS in 39m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6cd397ddf4654de188cd1ff2e123713e : SUCCESS in 1h 02m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/23a7e14966db47749f7c4f79f8373f2c : SUCCESS in 1h 21m 20s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b681212e4b514f2897b28be97fdba120 : SUCCESS in 1h 05m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/381361737c754def91b2855996c89cd6 : SUCCESS in 1h 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ff70b8ccaf5453a6e1fd90645531bb1199ceb930","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 16:47:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"548a6422d2ac70a9896469b5d059a15c5b8c1d45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 16:47:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d7b202485aebd50db48a027b8cc3d7b2e949bb13","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-19 16:53:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2cd7c9f498f1d2136b7ae0b7206e014eff535145","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 17:46:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8e19585c06ab44659b57c4e4de30bd28 : FAILURE in 4m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83158ec975aa466892fc7d051eb99cff : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd7c2cec18b84efc8007d213e83b1ccf : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b96d008d988540a499f5b7522567e288 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e6c4ecca02b4d97b913244c791302bb : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15b7799a7acf412da5903020f3be1bce : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/935b425da2754f97937b87cb73cb6e60 : SUCCESS in 5m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eb12ba2cb6004f57b1522aa672307c37 : SUCCESS in 37m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1c5831c42959265c0cb02c5149cef52fd9414413","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-19 21:45:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f1a39311a395bd136f843111e22840c4b7909cdd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 23:12:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/28c34dc3904848e18a10c43c4dfadc3b : SUCCESS in 26m 17s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/17b676edf74641aabb9e7aa8001286bc : SUCCESS in 43m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7abe6d2d09974e1c989c2e47c5450d62 : SUCCESS in 1h 12m 49s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8b3fba1c8bbf444c9e64a85787691ac9 : SUCCESS in 1h 18m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/6f84a71e1c68455b9af0a433efa0dd27 : SUCCESS in 1h 22m 03s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0eda7d5b86af4863b1a32647c4a14a5e : RETRY_LIMIT in 11m 35s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9bd56c9a58f748d085cbd335ac28448a : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2f06b834c054915af829a60a5b95ae5 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92c4f2912e8442e2b4b3f7af2b95510d : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88ab37d79e24443aa57d35b7551534b2 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e68dd003d2e4316a927009eb9f15c10 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df3d1c9afb284584a614a613db81259e : SUCCESS in 5m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c482746faa124003830543600a6ecedf : SUCCESS in 24m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3343cc326eca46788824d065f9a215ec : SUCCESS in 32m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c77038ac2be4134aa40bd4c81686dd3 : SUCCESS in 50m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/78a57a8be2754c799436e948f038c999 : SUCCESS in 51m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/972e0ad4e7f340b395a507531f2ccd73 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":1},{"id":"2446ab2dce02d26c8ec04b85278656cbf6b1d17a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-19 23:13:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9628ed1c81ccc0d992fd5a41a3c5a79b8cc60e42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 00:20:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"550829ebdc1c0f2a51f152a14dba79aa543b9aec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 00:20:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7d3c8e40a5cf4b8e80fff23c31505629 : SUCCESS in 1h 01m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1659902f0e54ed693b0f2ce5567dda4 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4abc4015497849e790230d38c198d024 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e420f3769ec48de811d58af3285f65d : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad0409bcf74b48b49c8efb0d39a5f98d : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e17bc2d1ab844ef48b28448bf42a9442 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62998e1e9462438cba4bf89b2b98ac09 : SUCCESS in 3m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2c0a619b0cc461d95f3a22e87205bc7 : SUCCESS in 32m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b16bfdd4bb03ee05ca5518f70ff4a4147ab31286","revisions":{"b16bfdd4bb03ee05ca5518f70ff4a4147ab31286":{"kind":"REWORK","_number":1,"created":"2019-08-19 14:51:47.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/25/677225/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/25/677225/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/677225/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/677225/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/25/677225/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/25/677225/1"}}},"commit":{"parents":[{"commit":"6320b05f71fdb2af52e697b9938a29385f65a443","subject":"Merge \"Unbreak CI: fluentd and TripleO dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6320b05f71fdb2af52e697b9938a29385f65a443"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-19 14:49:04.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-19 14:49:04.000000000","tz":120},"subject":"CI: follow-up on TripleO demotion - no gating","message":"CI: follow-up on TripleO demotion - no gating\n\nBackported to Stein already.\nSee also: https://review.opendev.org/677144\n\nChange-Id: Ic2948953b9e8cd5148d6bde539003ab9d081b215\nSigned-off-by: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b16bfdd4bb03ee05ca5518f70ff4a4147ab31286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b16bfdd4bb03ee05ca5518f70ff4a4147ab31286"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
