)]}'
{"id":"openstack%2Fkolla~677500","triplet_id":"openstack%2Fkolla~master~I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1","project":"openstack/kolla","branch":"master","topic":"feature/support_matrix","hashtags":[],"change_id":"I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1","subject":"Support matrix for Kolla","status":"MERGED","created":"2019-08-20 16:05:26.000000000","updated":"2019-10-10 13:25:21.000000000","submitted":"2019-10-10 13:23:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":51,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677500-1570713827330-e8c316f0","meta_rev_id":"77667ea986e522004338b397df76b8dccd394d83","_number":677500,"virtual_id_number":677500,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 13:23:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":2,"date":"2019-10-10 13:13:18.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-10-10 12:34:48.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-10-10 13:13:18.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-10-10 13:25:21.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-10-10 13:25:21.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-20 16:27:22.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-09-18 08:38:18.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-09 14:12:52.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-10-10 09:25:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-10 12:34:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-10 13:13:18.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-10-10 13:23:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f61066d6308199a198760687c96cb0d6dc0e58c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-20 16:05:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8578936640beef856d2e8c366eb98a02ecdf9ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-20 16:16:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2c88d32040844b39d89eef22494c699 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/286fb239f0c640b6b26fe7af4038ad83 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9c6e00e5b1015b3d6b43086150515735749a6bd3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-20 16:20:00.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nwe should have also some level for: only deployment tested vs more sophisticated tests\n\nthen we could cover k-a at once\n\nand what about marking images for tripleo?","accounts_in_message":[],"_revision_number":1},{"id":"d125ea1ea86adbcc6599ef65d2f4a60cf9a522f7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-20 17:14:36.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n \u003e \n \u003e we should have also some level for: only deployment tested vs more\n \u003e sophisticated tests\n \u003e \n \u003e then we could cover k-a at once\n \u003e \n \u003e and what about marking images for tripleo?\n\nWell, I guess TripleO would need to mark images as they use it - I don\u0027t like to do reverse engineering.\nAround deployment vs sophisticated tests - I feel we should cover it in detail in k-a docs. Trying to create a holistic view will not help I guess.\nMy idea would be to describe tests service by service in k-a, or role by role - while in Kolla we should cover images (distros and build types) - if they are buildable or not, and if they are linked to a k-a CI job that checks if k-a code + k image \u003d working service.\n\nAround Community Supported - this was the wording from PTG/whiteboard, but I also feel Community Maintained makes more sense :-)","accounts_in_message":[],"_revision_number":1},{"id":"bc25378c6808b254c9b923edf9ed023b79324500","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-21 07:14:40.000000000","message":"Patch Set 1:\n\nAll right. For details see k-a. Sweet.","accounts_in_message":[],"_revision_number":1},{"id":"e16e2b3c813fdc75898723d4dc1412e12923d8e0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-21 11:33:41.000000000","message":"Patch Set 1:\n\n(5 comments)\n\nLike the approach overall.","accounts_in_message":[],"_revision_number":1},{"id":"4e18e577d14ef39d3141141ad88748fa151e937c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-21 11:36:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f51cc6e686db5c0dc6675723c67ba6b963f70f17","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-21 11:40:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"131ad2927c532abc5afe0b6e857b9fe7d5f9cce4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-21 11:41:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd5f75ea093d78532f8a6f1745b0679e08932b6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 11:48:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0dd7bb9fe6c4e46b522af5160ed5c68 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab34edbaeda04a33b3428e45cdfded01 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"d5aed026f943584549737476090a0fbdc654fad2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-21 12:19:24.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f1f26774741c0779333bc1ae85554654eae45f45","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-21 12:30:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b7c79e5ee1d1b1e1baa58071468eef2b551a86f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-21 12:31:14.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"df486d659b1863bcb37d9d38078d9aae63531d28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 12:40:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e88c5bf1d3ef44faabf0fb937587ba22 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24e31f59e93f4ec9a10c89e46b489a02 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"37a4f4fd117e5ffab1e2f397b4cc3ede8f1c5c64","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-23 07:04:53.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ddb7e4db0452d8b984cc2ad594ea0a8868ed1e64","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-23 10:58:43.000000000","message":"Patch Set 3:\n\nooo has some matrix too: https://opendev.org/openstack/tripleo-heat-templates/src/branch/master","accounts_in_message":[],"_revision_number":3},{"id":"32870fb6648c3abf56bf4448b48fac4cc7724550","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-23 13:39:53.000000000","message":"Patch Set 3:\n\n\u003e ooo has some matrix too: https://opendev.org/openstack/tripleo-heat-templates/src/branch/master\n\nA link to a file would be better? ;-)","accounts_in_message":[],"_revision_number":3},{"id":"d442f71dcdc8b4d81444af73b5fd6b6ae7b89fab","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-23 13:47:27.000000000","message":"Patch Set 3:\n\n\u003e \u003e ooo has some matrix too: https://opendev.org/openstack/tripleo-heat-templates/src/branch/master\n \u003e \n \u003e A link to a file would be better? ;-)\nAh - that one: https://opendev.org/openstack/tripleo-heat-templates/src/branch/master/README.rst\n\nService testing matrix - but this would be more similar to our k-a matrix I guess.","accounts_in_message":[],"_revision_number":3},{"id":"530880d6f237dead80824268f5f47a080dcd2b73","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-23 17:35:06.000000000","message":"Patch Set 3:\n\n\u003e \u003e \u003e ooo has some matrix too: https://opendev.org/openstack/tripleo-heat-templates/src/branch/master\n \u003e \u003e\n \u003e \u003e A link to a file would be better? ;-)\n \u003e Ah - that one: https://opendev.org/openstack/tripleo-heat-templates/src/branch/master/README.rst\n \u003e \n \u003e Service testing matrix - but this would be more similar to our k-a\n \u003e matrix I guess.\n\nYeah, it rendered for me on the path I linked.\n\nPlease see my inline comments on rev 1.","accounts_in_message":[],"_revision_number":3},{"id":"9681b05eae20cd73fee8ca4391cb6910097baaf0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-26 07:50:19.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"b665ee322714c9b526fe609e8d64eff5f8a4b951","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 08:00:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f676459306f9432f89991daa19d82dca : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cbecbf816da4a568828d0b654b10ba3 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":4},{"id":"c8b0aecc4481f9c63d7a96151835d9a52c466209","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-26 08:09:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f9772f5ede15c125531e693b78acb888391bbffd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-26 08:20:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"31bf2d525b4b0ff5582b3818c244a0bc26a32677","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 08:36:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7836b94ff7d47f1ba9222a0b80bc7dc : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f115cb6a7de49ff81dc51be11c26d57 : SUCCESS in 10m 41s","accounts_in_message":[],"_revision_number":5},{"id":"a0a4c727d73ef41500317bd5452f3697d3139ccf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-26 10:15:12.000000000","message":"Patch Set 5:\n\nwow, you are making a serious promise in here ;-)","accounts_in_message":[],"_revision_number":5},{"id":"fa84cbe794cd4041eb54b037dd4eb60fdf1970ac","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-26 12:15:47.000000000","message":"Patch Set 5: Workflow-1\n\nOnly finished the structure... now working on the content (where I might know what I\u0027m doing for x86_64 - as for the rest - not at all) :-)","accounts_in_message":[],"_revision_number":5},{"id":"9c458e13404b2df520e495dc6e71465f43e268e6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-26 15:10:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"16932ab8c85fe60dc0d3214987dbcdc1e3333988","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 15:23:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/318692d0d3714fdbabf7028c47d769fa : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e75d9466c34499eb08c1a32fa765b08 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":6},{"id":"a604925fdf8660a49803de706352f350193eaf27","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-26 22:43:24.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nI will run builds on aarch64 to get a current list of status. And similar for ppc64le (here we need to populate UNBUILDABLE_IMAGES as well).","accounts_in_message":[],"_revision_number":6},{"id":"0f6d9277de60e14d448be9827a5e4fd1ba73820a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-27 06:41:04.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"497e66badb6a0aab150708939e9ef4014a8e1fc5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-27 10:44:54.000000000","message":"Patch Set 6:\n\n\u003e (1 comment)\n\nSo maybe let\u0027s rephrase Not Available to Unbuildable?","accounts_in_message":[],"_revision_number":6},{"id":"112dc421a70ba45618e2d47f434cb477eef7dcc8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-30 13:37:52.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"59b3ea7246e517d8a3e41d0ab06e355a421f627b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-04 12:54:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"81d5e4669bdd8a670c2dc1d42739e8caf39a02e8","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-04 12:56:04.000000000","message":"Patch Set 7:\n\nx86-64 and aarch64 entries added in \"if any component of it built for target distro then mark as \u0027C\u0027\"\n\nDebian/binary is not supportable due to lack of OpenStack packages.","accounts_in_message":[],"_revision_number":7},{"id":"2f9302c5f0a97c70eaa102c90b1e2f193830b39f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 13:25:52.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fbf72524954e2890739b6fce8697f7 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac4fc4c1af1404d876dd7e7ac1840b2 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":7},{"id":"99d5b5f88aef86a7535268416e1c8a8539c087b1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-10 12:33:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fcc920f5098b7656a8251840c20c6e23fa59e466","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 12:41:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6eff7995b194c229caf56c9410050aa : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d62f2f6baec74006a85677b7440fbc13 : FAILURE in 2m 58s","accounts_in_message":[],"_revision_number":8},{"id":"e07751734623c6f38f7b76f25f500ed235910390","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-10 12:44:28.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"20d3a4f434a2e51de2fd1499ba26602cd48e8999","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-10 12:50:20.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"e92ce8b38d2398c11c7d35354ca5fc951459f7a8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-10 13:02:50.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6b135ba8297a07c5187970893bb60c4bf64eabb5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-10 13:10:12.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master tripleo-build-containers-rhel-8-master : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"81311b100df5b9d0f5e18ffd4dfb82500c71480e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-10 13:19:18.000000000","message":"Patch Set 10:\n\nMaybe we should move UNBUILDABLE_IMAGES content to external json and parse it on runtime?\n\nThis way it can be reused.","accounts_in_message":[],"_revision_number":10},{"id":"20e87938d8fe86662e06e26e11245212b141a020","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-10 13:26:51.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"31745b27aaf366b08f8939a4cd1e328853a4ddec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-10 13:28:57.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master tripleo-build-containers-rhel-8-master : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":11},{"id":"3680bd73a2425995722ab333ee5118490481f8f1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-10 13:33:48.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"137a7c25ce317895f89db5fd13e56e89b18c34d6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-10 13:34:32.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master tripleo-build-containers-rhel-8-master : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":12},{"id":"0f3446768871deb4d63850bb47bc5146f493e7ef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-10 13:53:27.000000000","message":"Patch Set 12:\n\n\u003e Maybe we should move UNBUILDABLE_IMAGES content to external json\n \u003e and parse it on runtime?\n \u003e \n \u003e This way it can be reused.\n\nI had a quick go at adding a command to output a support matrix from the CLI.\n\nhttp://paste.openstack.org/show/774836/\n\nIt works, but does not collapse down multiple images into one service. It could be done, but would get more complicated. We\u0027d also need a way to define which images to promote from C to M or S.","accounts_in_message":[],"_revision_number":12},{"id":"ad2ce4c56bfc3c5127546338ba7376649b95dc0d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-10 14:33:45.000000000","message":"Patch Set 12:\n\n\u003e \u003e Maybe we should move UNBUILDABLE_IMAGES content to external json\n \u003e \u003e and parse it on runtime?\n \u003e \u003e\n \u003e \u003e This way it can be reused.\n \u003e \n \u003e I had a quick go at adding a command to output a support matrix\n \u003e from the CLI.\n \u003e \n \u003e http://paste.openstack.org/show/774836/\n \u003e \n \u003e It works, but does not collapse down multiple images into one\n \u003e service. It could be done, but would get more complicated. We\u0027d\n \u003e also need a way to define which images to promote from C to M or S.\n\nWell, with some work added we could dump a csv and render it in docs using csv-table or a list-table. I think we could survive having a non-collapsed list, if we had a subpage per arch. Still promoting images to M or T require some additional data, which we would need to maintain I think - similar to UNBUILDABLE.","accounts_in_message":[],"_revision_number":12},{"id":"423d15025d0644438c553af135af535693512e97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 15:16:05.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/bf93b0674b6a4e7dbe9aa31967e670e9 : SUCCESS in 47m 16s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/8bee65615220433883bd1d4fb42302f6 : SUCCESS in 53m 40s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/971688ceec44445e9ac54168204237a9 : SUCCESS in 1h 33m 22s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/67c8a80cf5ea4c599be7a34802d4ca93 : SUCCESS in 1h 06m 35s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/25eb9f9d0f554f9d947dada4e73b9938 : SUCCESS in 1h 36m 49s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8172365c2c6c4bffa8531f266a437c2d : SUCCESS in 59m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f77554f89fd64ad1a269c6b31bf949a6 : SUCCESS in 26m 27s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c96cce25c34a48cfbc399c80911a9d54 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9759681a49749aaa1ab572f8d8008c6 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c99e82db77a4b68a5cffba31d2f7d6a : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e4d9d8bde6d4b7d934a499680458a69 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0ad84ce6bbbd431fa30c10f09c7ec61d : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5855dc818054fd9849642f48df93357 : SUCCESS in 3m 34s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/71aba35f7bad49cbb2792f9e2e728f40 : FAILURE in 33m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/462cf262a78a443eb8c16453a63b6cef : FAILURE in 27m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ee8b6ce5e3744cf48f3080c96ad98fc6 : FAILURE in 16m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/59e31b7ff51745b29c6211e0e5ac6db1 : FAILURE in 16m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6e80ce3372ca401d9a6fd8b54df4df09 : FAILURE in 15m 23s","accounts_in_message":[],"_revision_number":12},{"id":"1c4b236204cb890046c113f2e98bc61d69770d38","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-11 08:46:56.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"9bd2ebc41fe3bd119bc454fe577aaa15aa54fc3e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-11 09:24:04.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/00/677500/12/openstack-check/tripleo-build-containers-rhel-8-master/1128c19/ : SUCCESS in 36m 07s","accounts_in_message":[],"_revision_number":12},{"id":"93ab336dfe279eea3d4236e3c859c5613e68d584","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 10:19:06.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/98d1731e40a8464290b18d79514d3cc8 : SUCCESS in 49m 35s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f0bc5974a1d6487cbcb1a45039072b16 : SUCCESS in 47m 20s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/dbf5e6b2d9c4404eb1c30b5e6ffb4918 : SUCCESS in 1h 18m 37s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f0d6bfad9cb04297af37a7bf27998e32 : SUCCESS in 1h 08m 24s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/daec861fc0484fc69b4a05e75ea7fc81 : SUCCESS in 1h 30m 17s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/301f622245e04f9a9b62206227a36ac2 : SUCCESS in 45m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9a657c2326f64d6da39aba43c920ff19 : SUCCESS in 31m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d023248c70424061af397616f9fc4451 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69f341807276452abee52b33d7b46e91 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/931b9d4e2ca6499f953d0d0113a42cde : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d28ebb7fa2984b928e81811af98028a7 : SUCCESS in 3m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d9fe6b428b84374b3311ae26af7621f : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3690031e03416399d61d58f0c17a8b : SUCCESS in 3m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/af4bdf00e96d45429947142c44d18ed3 : FAILURE in 27m 37s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c0885400ff2c4429844924ba7d1a4635 : FAILURE in 37m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76bc838e256e43dd9e2b51d0562cf8aa : SUCCESS in 50m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f16e9fa2c51c4e4ab758b6e790acda46 : SUCCESS in 51m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/007606df643d4cb199421ce6fcd8b5fc : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":12},{"id":"0359438b42a0ff5af473b88f5d881834671e8169","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-18 07:59:15.000000000","message":"Patch Set 12:\n\nrecheck","accounts_in_message":[],"_revision_number":12},{"id":"83c5b00e33380d442b523ea9a4457c149ef29540","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-18 08:38:18.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/00/677500/12/openstack-check/tripleo-build-containers-rhel-8-master/8c66285/ : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":12},{"id":"cbaae0948083200f48e3593a92c96bdb3dd483d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 09:45:02.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/cdb333c66a634b2bbdd6d114691de0f3 : SUCCESS in 37m 53s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/46a68db879014a0a8122748fba8cb82b : SUCCESS in 1h 00m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/210e06cc3e274ababbe5fd0931c99cd6 : SUCCESS in 1h 17m 06s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2422c12f54154c98bb5fc3918bdc5ab9 : SUCCESS in 1h 24m 59s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a5c76c4c390843459eb4374b094c6fb7 : SUCCESS in 1h 37m 28s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/115b4754514045fe99e96df9b1b4dbe4 : SUCCESS in 1h 02m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9817fa2375f845e5911afad36490b926 : SUCCESS in 30m 42s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3eef64316b8a4af8a6b3d53251343c6e : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df7a9c31d29b4b8b8612e7f0c278274c : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1aeb5d14afe544b6be25331146550c08 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/549698b5bb66474ea04e70805af7153f : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6dd86402eda34c34b5859c34219ba9d6 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0db3df0cf648d6b7f0f48d9a5b8179 : SUCCESS in 5m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/152a5e25c57041778350902b237c5f03 : SUCCESS in 34m 34s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/085aa90faab0459ba8cec055015c0dc9 : SUCCESS in 39m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/55376ab276574db4a8abaf68252f6da4 : SUCCESS in 53m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/03b092f210b146b98fd9f846ea57ad18 : SUCCESS in 52m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ee57c0d158a74327aff9ba5ed42fcd41 : SUCCESS in 59m 47s","accounts_in_message":[],"_revision_number":12},{"id":"9ceb120b06f04b5ef563ecf98dcf091787e034c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-19 10:55:50.000000000","message":"Patch Set 12:\n\nStruck me we might want a \u0027deprecated\u0027 state for some images.","accounts_in_message":[],"_revision_number":12},{"id":"56600e2f1888dac5d08d294acdf58491211e9667","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-19 11:29:17.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"c701ec3cf4069b8c54e029fac492c56b562f8559","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-19 11:58:06.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"991323e46edf6a2550f7e633f5251e6b0a788610","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-30 09:17:04.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8ecb914f1283bda34c579ebacd523e4539ee6bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 09:26:33.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/041205cfdc054d33aa7f257a472f10a4 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb33c5050c6e4e81a6f21f6161f24588 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":13},{"id":"950cd1dfb0f3e26ce0a23da9af61d5df9829d248","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-09-30 09:37:54.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"3c49ac8c0d867b92cfe0f4e09b7b5cc3681a250b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 09:49:26.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd85b4d3ae54d22b610a1d4e61eaa86 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79613ca5c3a4a06909e79efe2103173 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":14},{"id":"c5cab9bc90927332087413729fcf63d91ae75bf0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-01 06:23:52.000000000","message":"Patch Set 14:\n\nas the next step - what about categorizing stuff like the supporting services, optional supporting services, core openstack, extra openstack, external extras (or whatever naming of that) - now the table is quite lengthy and hard to decipher\n\nactually the support status depends strongly on the categorization so it\u0027s a win-win if we want to fill that in anyway","accounts_in_message":[],"_revision_number":14},{"id":"8bbdbda625184d8f031103652522c6dc51f80cbe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 16:24:34.000000000","message":"Patch Set 14:\n\n\u003e as the next step - what about categorizing stuff like the\n \u003e supporting services, optional supporting services, core openstack,\n \u003e extra openstack, external extras (or whatever naming of that) - now\n \u003e the table is quite lengthy and hard to decipher\n \u003e \n \u003e actually the support status depends strongly on the categorization\n \u003e so it\u0027s a win-win if we want to fill that in anyway\n\nLet\u0027s get a first pass in and iterate.","accounts_in_message":[],"_revision_number":14},{"id":"bb1e6627309e5a4074d7a305c6cc816330c0c389","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-03 12:45:02.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"aa95ecd85b7bed1ce9d5e73d32f7c7c11abe0cab","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-03 12:50:13.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)\n\nplease rename matrix_x86.csv to matrix_x86-64.csv as we do not support 32bit x86.","accounts_in_message":[],"_revision_number":15},{"id":"be4bdad2e3e7c41ce8b5deefed2f5b3a510e6cf6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-03 12:56:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"315296802d81083ff0dca7ccc9f691776a290f7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 12:56:41.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c071e0364e4145e48019342dcfbfbced : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3b61a4259744ef88db48b410248ac8d : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":15},{"id":"677e08788b73ad3d33d24747be1fb1f4b96d27df","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-03 13:09:12.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"303099bc6f8f0653fd35474c16c005ef6850e0a9","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-03 15:45:07.000000000","message":"Patch Set 15: -Code-Review\n\n+-----------------------+-----------------+-----------------+-----------------+\n| Image                 | CentOS          | Ubuntu          | Debian          |\n|                       +--------+--------+--------+--------+--------+--------+\n|                       | binary | source | binary | source | binary | source |\n+\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d+\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d+\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d+\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d+\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d+\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d+\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d+\n\nThe order of columns was different...","accounts_in_message":[],"_revision_number":15},{"id":"5533b5e09de65f4cfe0be6f2f08b7be57b178514","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-03 16:39:36.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"dc74c40b984b1af18650a94c95ba967dd496dbdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-03 16:49:44.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcea49ba58f343058739887a7a09ccc3 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342000aeeb094038a08d589345951b5e : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":16},{"id":"3d4a69f1f25acef8351829cb2ac8ed283cd71759","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-04 07:46:39.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"ff9cafac32fdb10e05a72e1822a3234209084253","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-04 07:47:57.000000000","message":"Patch Set 17:\n\nAdded T for services tested I the ,,base\u0027\u0027 Zuul jobs, need to add scenario based still.","accounts_in_message":[],"_revision_number":17},{"id":"432a5127d1208711457083e31eb0fbbc00f2f90d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-04 07:49:21.000000000","message":"Patch Set 17:\n\n\u003e Added T for services tested I the ,,base\u0027\u0027 Zuul jobs, need to add\n \u003e scenario based still.\n\nyou can already do the same for debian, it\u0027s going to be tested since today","accounts_in_message":[],"_revision_number":17},{"id":"39f14e1b4ca36091cdf69786884c9af5f2ee2022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 07:57:09.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5edb31a8833d4499a2af8ee002270e2f : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd9ddd5c574f4c70ab18e6c4763d58b7 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":17},{"id":"ebb2763eb0a6da29f5ddc400f40c2e5c77895618","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-04 08:44:26.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"fbdfb4e4dd8abba4309c2c060ccb9a3b3d1b7466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 08:51:33.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22456045c53944c38b244d3d8cae42f0 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/606bd6a6e7eb4088a4d64bd79e338aef : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":18},{"id":"0d24c7b30666eeb0b1fb98d591ad09928c3c0443","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-04 09:21:58.000000000","message":"Patch Set 18:\n\njust noticed, we removed xtrabackup.","accounts_in_message":[],"_revision_number":18},{"id":"bb2bc694e1f1577167efc15603e196eba101f03b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-04 10:05:29.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"12f00e58d57b077e79edbdcad38beda7dc72b36c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-04 10:13:55.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/383f3c68ed2f4cf59f60a9016355c5c3 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3db8f0dc0f24d08b4614fc608510ace : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":19},{"id":"592d5e826f2c4b05ee2c18f5da7d4c322dbb2936","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-09 09:17:36.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"633e372852478c5f4bddb365041ca290b91138e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 09:23:56.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2a5c9dc914f41f1a3102194d05f324f : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5010325bc0c74e35a818e0a6a111542e : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":20},{"id":"8d00e794b699eb29f31b2d2efc8048949fc71628","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-09 09:35:57.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"93d7f5ca951fdac10bf65223c4807d47af0ca409","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-09 09:37:44.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"e92fa75e89b6bebf3ca5f58adeed3f39cc9eb555","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-09 09:48:50.000000000","message":"Patch Set 20:\n\n(10 comments)","accounts_in_message":[],"_revision_number":20},{"id":"5a4b3469746bb3c50308800b828deefa11da670b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-09 09:49:25.000000000","message":"Patch Set 20:\n\nare we going to slap any M? :-)","accounts_in_message":[],"_revision_number":20},{"id":"3dbf1819e06e35c38cc918b51f046c222c7a549c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-09 09:51:07.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"76044c54fdb72847cb36f4b634b96c838ef14dbb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-09 09:53:44.000000000","message":"Patch Set 20:\n\n(8 comments)","accounts_in_message":[],"_revision_number":20},{"id":"c73112d4f8d1228fa17af3a7580d0e15ca082eef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-09 09:55:19.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"68b22f04994a6ef4d3bd03766147649bb0dc4934","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-09 09:56:07.000000000","message":"Patch Set 20:\n\n\u003e are we going to slap any M? :-)\n\nGood question. This is the grey area that we need to discuss as a group. I\u0027d be happy to merge this without and upgrade support later.","accounts_in_message":[],"_revision_number":20},{"id":"b153f2ac37f54b8906852b3492ca670da43679af","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-09 09:59:39.000000000","message":"Patch Set 20:\n\n\u003e \u003e are we going to slap any M? :-)\n \u003e \n \u003e Good question. This is the grey area that we need to discuss as a\n \u003e group. I\u0027d be happy to merge this without and upgrade support\n \u003e later.\n\n+1","accounts_in_message":[],"_revision_number":20},{"id":"2c0780688be5ea7e80f00c52f1df5b5286604893","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-09 10:45:56.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"bd5dc590c620bdaf3e0bb118a2fca7e2e93d4672","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-10-09 10:53:56.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"8205aaef36324ef5f8fd34215105dd71a8d23ff0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 11:04:32.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/537f6673f37748b0a0e5313ae054076a : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4677402679fd41f1ac6027040f1ae5a0 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":21},{"id":"af4254d06e1973e171751741ee03a036a1f58384","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-10-09 14:12:52.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"da45dcaee1d449ce31303e3398741e5e52ff8e78","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-09 18:32:47.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"75a652d58fa67bb1181812dd76b0a7268b33dd60","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-09 18:36:06.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"386d8e2578820af9eceba2553539092b39429a51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 18:44:20.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4aa4560cc3e4988ae2649f50a927e85 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c5ecbe2d7c84e1f9fd5991a1bb64764 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":22},{"id":"e9cecd1a02139b1884535f89489544f992abb2a3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-10 09:25:26.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"91e93642c620b559b314749ec4a7e5204b852c4c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-10 09:25:30.000000000","message":"Patch Set 22: Review-Priority+1","accounts_in_message":[],"_revision_number":22},{"id":"45e8c1b94b79bffd9097624eb0a7e61fa88f0417","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-10 12:34:33.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"f31200416ae72e151b17df43fb9f86e6b34153a8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-10 12:34:48.000000000","message":"Patch Set 23: Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"f60aa20f28a5ef4a298d8735f0806ee509317b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 12:44:04.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7356ae63468a40f0858c99f260836bd6 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d763549a31d342ee94717b0ebd398150 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":23},{"id":"dbdbbfd614319ab3a3af927a8a448ad9de5a54a7","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-10 13:13:18.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":23},{"id":"a5d5ab7c2d31a81aa9ecbddbf48106c4dd211398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:13:30.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"e366ef9d65356a4a2d72664201a4d29c9b974f51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:23:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":23},{"id":"50f5dc0c680d4aba55cfb49432712a7f1cea0dbd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:23:47.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54e6e74f60304716bffe4b505472c1f9 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83e388ebffdf4b7b80463f20e261d969 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":23},{"id":"23383aaacf0c4440188c617ca6db322a724e5e55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:25:21.000000000","message":"Patch Set 23:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c24c930cbf9a45b68b162807b9529ec4 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"3f8bf6ca100f3a11b4f18104dc3ca98f28bf2431","revisions":{"5219d5e59f1049cc3cbbe991d5f64f7c30c0bc75":{"kind":"REWORK","_number":1,"created":"2019-08-20 16:05:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/1"}}},"commit":{"parents":[{"commit":"6320b05f71fdb2af52e697b9938a29385f65a443","subject":"Merge \"Unbreak CI: fluentd and TripleO dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6320b05f71fdb2af52e697b9938a29385f65a443"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:05:21.000000000","tz":120},"subject":"WIP: Support matrix for Kolla","message":"WIP: Support matrix for Kolla\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5219d5e59f1049cc3cbbe991d5f64f7c30c0bc75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5219d5e59f1049cc3cbbe991d5f64f7c30c0bc75"}]},"branch":"refs/heads/master"},"36e1b094a78ab05eb5b093983ada4cd77710a8f0":{"kind":"REWORK","_number":2,"created":"2019-08-21 11:40:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/2"}}},"commit":{"parents":[{"commit":"6320b05f71fdb2af52e697b9938a29385f65a443","subject":"Merge \"Unbreak CI: fluentd and TripleO dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6320b05f71fdb2af52e697b9938a29385f65a443"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-21 11:40:46.000000000","tz":0},"subject":"WIP: Support matrix for Kolla","message":"WIP: Support matrix for Kolla\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36e1b094a78ab05eb5b093983ada4cd77710a8f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36e1b094a78ab05eb5b093983ada4cd77710a8f0"}]},"branch":"refs/heads/master"},"d1a32c7e6b99c96428a56b375bb3b4caae703b26":{"kind":"REWORK","_number":3,"created":"2019-08-21 12:31:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/3"}}},"commit":{"parents":[{"commit":"6320b05f71fdb2af52e697b9938a29385f65a443","subject":"Merge \"Unbreak CI: fluentd and TripleO dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6320b05f71fdb2af52e697b9938a29385f65a443"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-21 12:31:11.000000000","tz":0},"subject":"WIP: Support matrix for Kolla","message":"WIP: Support matrix for Kolla\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1a32c7e6b99c96428a56b375bb3b4caae703b26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1a32c7e6b99c96428a56b375bb3b4caae703b26"}]},"branch":"refs/heads/master"},"65cb864bddff14a3522fd4ab064b459962ec934c":{"kind":"REWORK","_number":4,"created":"2019-08-26 07:50:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/4"}}},"commit":{"parents":[{"commit":"6320b05f71fdb2af52e697b9938a29385f65a443","subject":"Merge \"Unbreak CI: fluentd and TripleO dependency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6320b05f71fdb2af52e697b9938a29385f65a443"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-26 07:49:19.000000000","tz":0},"subject":"WIP: Support matrix for Kolla","message":"WIP: Support matrix for Kolla\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65cb864bddff14a3522fd4ab064b459962ec934c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65cb864bddff14a3522fd4ab064b459962ec934c"}]},"branch":"refs/heads/master"},"e12bde3af671cd014ab2917ccd44a8a0037b7bb0":{"kind":"REWORK","_number":5,"created":"2019-08-26 08:20:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/5"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-26 08:20:27.000000000","tz":120},"subject":"WIP: Support matrix for Kolla","message":"WIP: Support matrix for Kolla\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e12bde3af671cd014ab2917ccd44a8a0037b7bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e12bde3af671cd014ab2917ccd44a8a0037b7bb0"}]},"branch":"refs/heads/master"},"1c1b6efe78853254ef7a71f18bc90a58b33a3aa8":{"kind":"REWORK","_number":6,"created":"2019-08-26 15:10:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/6"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-26 15:10:35.000000000","tz":120},"subject":"WIP: Support matrix for Kolla","message":"WIP: Support matrix for Kolla\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c1b6efe78853254ef7a71f18bc90a58b33a3aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c1b6efe78853254ef7a71f18bc90a58b33a3aa8"}]},"branch":"refs/heads/master"},"338f2ef8f4a7ff57822d6e123fcda236fdd91ea4":{"kind":"REWORK","_number":7,"created":"2019-09-04 12:54:39.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/00/677500/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/7"}}},"commit":{"parents":[{"commit":"90892a25bd9aa96d7c070e49b0afbcee9bf5fca7","subject":"Merge \"ironic-inspector: install the dnsmasq subpackage explicitly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/90892a25bd9aa96d7c070e49b0afbcee9bf5fca7"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2019-09-04 12:54:11.000000000","tz":120},"subject":"WIP: Support matrix for Kolla","message":"WIP: Support matrix for Kolla\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/338f2ef8f4a7ff57822d6e123fcda236fdd91ea4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/338f2ef8f4a7ff57822d6e123fcda236fdd91ea4"}]},"branch":"refs/heads/master"},"9cafa5b9b2d18e89e03edc3d397d10727ce859a3":{"kind":"REWORK","_number":8,"created":"2019-09-10 12:33:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/8"}}},"commit":{"parents":[{"commit":"7cda7ac96834c72ca818a3af449b2af6132bad9b","subject":"Merge \"base: drop not needed Linaro OBS key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cda7ac96834c72ca818a3af449b2af6132bad9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-09-10 12:33:17.000000000","tz":60},"subject":"WIP: Support matrix for Kolla","message":"WIP: Support matrix for Kolla\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cafa5b9b2d18e89e03edc3d397d10727ce859a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9cafa5b9b2d18e89e03edc3d397d10727ce859a3"}]},"branch":"refs/heads/master"},"2c0ebd95c74ebdc0519bd7a94281d22f28eb2745":{"kind":"REWORK","_number":9,"created":"2019-09-10 12:50:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/9"}}},"commit":{"parents":[{"commit":"7cda7ac96834c72ca818a3af449b2af6132bad9b","subject":"Merge \"base: drop not needed Linaro OBS key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cda7ac96834c72ca818a3af449b2af6132bad9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-09-10 12:50:13.000000000","tz":60},"subject":"WIP: Support matrix for Kolla","message":"WIP: Support matrix for Kolla\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c0ebd95c74ebdc0519bd7a94281d22f28eb2745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c0ebd95c74ebdc0519bd7a94281d22f28eb2745"}]},"branch":"refs/heads/master"},"34f200910212bde1f66d2bab434c316a376be952":{"kind":"REWORK","_number":10,"created":"2019-09-10 13:02:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/10"}}},"commit":{"parents":[{"commit":"7cda7ac96834c72ca818a3af449b2af6132bad9b","subject":"Merge \"base: drop not needed Linaro OBS key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cda7ac96834c72ca818a3af449b2af6132bad9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-09-10 13:01:44.000000000","tz":60},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n* Added markers in kolla/image/build.py to include unbuildable\n  images list\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34f200910212bde1f66d2bab434c316a376be952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34f200910212bde1f66d2bab434c316a376be952"}]},"branch":"refs/heads/master"},"26a179498306bce51bdb5a9457f257e89f4aa33b":{"kind":"REWORK","_number":11,"created":"2019-09-10 13:26:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/11"}}},"commit":{"parents":[{"commit":"7cda7ac96834c72ca818a3af449b2af6132bad9b","subject":"Merge \"base: drop not needed Linaro OBS key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cda7ac96834c72ca818a3af449b2af6132bad9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-09-10 13:26:43.000000000","tz":60},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n* Added markers in kolla/image/build.py to include unbuildable\n  images list\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26a179498306bce51bdb5a9457f257e89f4aa33b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26a179498306bce51bdb5a9457f257e89f4aa33b"}]},"branch":"refs/heads/master"},"08b51d3851bc5d3e025e780d638af7483b3bd133":{"kind":"REWORK","_number":12,"created":"2019-09-10 13:33:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/12"}}},"commit":{"parents":[{"commit":"7cda7ac96834c72ca818a3af449b2af6132bad9b","subject":"Merge \"base: drop not needed Linaro OBS key\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cda7ac96834c72ca818a3af449b2af6132bad9b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-09-10 13:33:32.000000000","tz":60},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n* Added markers in kolla/image/build.py to include unbuildable\n  images list\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08b51d3851bc5d3e025e780d638af7483b3bd133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08b51d3851bc5d3e025e780d638af7483b3bd133"}]},"branch":"refs/heads/master"},"65d5969ecc9bf69b781d00878556ab27c7b1ab68":{"kind":"REWORK","_number":13,"created":"2019-09-30 09:17:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/13"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-09-30 09:17:01.000000000","tz":120},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65d5969ecc9bf69b781d00878556ab27c7b1ab68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65d5969ecc9bf69b781d00878556ab27c7b1ab68"}]},"branch":"refs/heads/master"},"4ac1e8d039febf76a78d24c339842b742f6c6797":{"kind":"REWORK","_number":14,"created":"2019-09-30 09:37:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/14"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-09-30 09:37:44.000000000","tz":120},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ac1e8d039febf76a78d24c339842b742f6c6797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4ac1e8d039febf76a78d24c339842b742f6c6797"}]},"branch":"refs/heads/master"},"1faa324cc9524bc655faf043290441f14b528aac":{"kind":"REWORK","_number":15,"created":"2019-10-03 12:45:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/15"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-03 12:44:48.000000000","tz":120},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1faa324cc9524bc655faf043290441f14b528aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1faa324cc9524bc655faf043290441f14b528aac"}]},"branch":"refs/heads/master"},"7c73b770f6e258cc67515fa0d36f6249072b1af7":{"kind":"REWORK","_number":16,"created":"2019-10-03 16:39:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/16"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-03 16:39:25.000000000","tz":120},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c73b770f6e258cc67515fa0d36f6249072b1af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7c73b770f6e258cc67515fa0d36f6249072b1af7"}]},"branch":"refs/heads/master"},"dffc07a90517f13a4a16ec9aaac6f9aad3b7728e":{"kind":"REWORK","_number":17,"created":"2019-10-04 07:46:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/17"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-04 07:46:26.000000000","tz":120},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dffc07a90517f13a4a16ec9aaac6f9aad3b7728e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dffc07a90517f13a4a16ec9aaac6f9aad3b7728e"}]},"branch":"refs/heads/master"},"383468a8f4ab6ccd6a2434c4c94b60268919de34":{"kind":"REWORK","_number":18,"created":"2019-10-04 08:44:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/18"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-04 08:44:16.000000000","tz":120},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/383468a8f4ab6ccd6a2434c4c94b60268919de34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/383468a8f4ab6ccd6a2434c4c94b60268919de34"}]},"branch":"refs/heads/master"},"b0dd822a6562be9cc79cf524b7b9ce54694d28f7":{"kind":"REWORK","_number":19,"created":"2019-10-04 10:05:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/19"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-04 10:05:03.000000000","tz":0},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0dd822a6562be9cc79cf524b7b9ce54694d28f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0dd822a6562be9cc79cf524b7b9ce54694d28f7"}]},"branch":"refs/heads/master"},"b551c850713a8ccd129feadcd9bcbc50426c4d93":{"kind":"REWORK","_number":20,"created":"2019-10-09 09:17:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/20"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-09 09:15:03.000000000","tz":120},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n* Populated with buildable vs unbuildable images\n* Populated with images tested by k-a marked by T\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b551c850713a8ccd129feadcd9bcbc50426c4d93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b551c850713a8ccd129feadcd9bcbc50426c4d93"}]},"branch":"refs/heads/master"},"ff796bfe8e90a89d9f88b2dde15db22de38d49a7":{"kind":"REWORK","_number":21,"created":"2019-10-09 10:53:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/677500/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/21"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-10-09 10:53:44.000000000","tz":120},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n* Populated with buildable vs unbuildable images\n* Populated with images tested by k-a marked by T\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff796bfe8e90a89d9f88b2dde15db22de38d49a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff796bfe8e90a89d9f88b2dde15db22de38d49a7"}]},"branch":"refs/heads/master"},"d89c8504f5b954c07de945731d5bbc9ecf1a4b54":{"kind":"REWORK","_number":22,"created":"2019-10-09 18:36:06.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/00/677500/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/22"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-10-09 18:35:45.000000000","tz":0},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n* Populated with buildable vs unbuildable images\n* Populated with images tested by k-a marked by T\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d89c8504f5b954c07de945731d5bbc9ecf1a4b54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d89c8504f5b954c07de945731d5bbc9ecf1a4b54"}]},"branch":"refs/heads/master"},"3f8bf6ca100f3a11b4f18104dc3ca98f28bf2431":{"kind":"REWORK","_number":23,"created":"2019-10-10 12:34:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/677500/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/677500/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/677500/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/677500/23"}}},"commit":{"parents":[{"commit":"f9b229841db5d27fe11b95d055d74af289893ef2","subject":"Merge \"ppc64le: Ensure that rabbitmq-server.repo exists before trying sed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b229841db5d27fe11b95d055d74af289893ef2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-20 16:04:04.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-10 12:33:44.000000000","tz":0},"subject":"Support matrix for Kolla","message":"Support matrix for Kolla\n\n* Created support matrix for x86_64 and aarch64\n* Populated with buildable vs unbuildable images\n* Populated with images tested by k-a marked by T\n\nChange-Id: I3f71dafc5f50dfdd2c44a70ba3c294c9308ce3b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f8bf6ca100f3a11b4f18104dc3ca98f28bf2431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f8bf6ca100f3a11b4f18104dc3ca98f28bf2431"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
