)]}'
{"id":"openstack%2Fkolla~678196","triplet_id":"openstack%2Fkolla~master~Iac142c056938f1ecd45807163460b040250882fd","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Iac142c056938f1ecd45807163460b040250882fd","subject":"Remove Nova legacy upgrade","status":"ABANDONED","created":"2019-08-23 10:36:30.000000000","updated":"2019-10-31 12:04:52.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"77a31bbdc1cd3195f04d560f8d2fac0115540073","_number":678196,"virtual_id_number":678196,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-10-31 11:30:00.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"date":"2019-10-26 13:29:10.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2019-08-23 16:46:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-08-23 16:01:28.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 16:01:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-23 16:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-26 13:29:10.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-31 11:30:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"d254c55d14bf29ee8130eef9c89a8d787b7cf250","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-23 10:36:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"455ca1888c5e78789c6d6092b77239b1cd06de51","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-23 11:01:35.000000000","message":"Patch Set 1:\n\nI think we should use BOOTSTRAP or UPGRADE really. Otherwise it looks as if nova does not support/have/need bootstrapping.","accounts_in_message":[],"_revision_number":1},{"id":"c4798b120b23b8c183b9cfbc5737331d45f724e4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 11:16:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/96/678196/1/openstack-check/tripleo-build-containers-rhel-8-master/033321a/ : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9f74c3144aa7d1fda539c7af282bb167cbb14400","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 12:16:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/973afdc98fd14d239c08a1b92d612778 : SUCCESS in 37m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0018d404186b4768a1159685ef47dfe2 : SUCCESS in 38m 42s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/0be08dbb3b964222b303fa87a90bea6a : SUCCESS in 1h 34m 13s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/f2764890bc7d48f695e4d94be89dee7e : SUCCESS in 1h 12m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/efe6011d81c14a6b8b26b9248aa162aa : SUCCESS in 1h 21m 10s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/2128796bdf0d46efa5d9419853bb3cbc : SUCCESS in 52m 32s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b41b1d8c6ecd468fba95e4471dd544b4 : SUCCESS in 26m 33s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52fa1ce2be2f4d71b790f354a97c0288 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/029209d931eb44f7887588ef61ce706e : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f8771652af2f4c94bf7cf70f75aff013 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07cb6ea979ff441cb7ec6a3a7d9fb6ea : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40fbc5690f8042d7bde027b5093cfeb1 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b74b89d094564cbe8a51201f496cc66f : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cff6f50120c741a9bdd39d846c6ee137 : SUCCESS in 6m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a8fe064c0ce4e6aa55b536229002038 : SUCCESS in 25m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8d9aee91cc64c1f886b4cba4f8540af : SUCCESS in 56m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e8bdc09656a347fab225ab789815ead2 : SUCCESS in 48m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/87568ad985c94eb2b66bb97e0403e10c : SUCCESS in 48m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/32eeaeeffd724016bcabddb47b0cc322 : SUCCESS in 1h 01m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1f793e5ee830a28459630b1e06eaef659b32c80f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-23 13:02:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"209ec8d696488c0fa08910477f757fcd273aa945","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 14:08:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/96/678196/2/openstack-check/tripleo-build-containers-rhel-8-master/01eed7a/ : SUCCESS in 1h 05m 02s","accounts_in_message":[],"_revision_number":2},{"id":"e0086e396147f0590e247f29322e4c3999e16692","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:06:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/12a55618075a48bdb48cc9aac2d4bba8 : SUCCESS in 42m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d3e566ebc0db4aa5926ebc7c27e36c94 : SUCCESS in 52m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/06290676076b4f9aa56f375fa7ae6d5d : TIMED_OUT in 2h 01m 58s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/553d7b63412a42a1a283f88e503b9c98 : TIMED_OUT in 2h 02m 25s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/3659cb574ecc4ea5bd186df8456de640 : TIMED_OUT in 2h 02m 58s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1b7f3ea96ff641498363dd6dab432503 : SUCCESS in 45m 34s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/33df04bcc56e4e67bfefee5da59bcf00 : SUCCESS in 29m 55s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c1180a12ffc4e61a1eaf4d47ac13288 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d616669d2b34276bbcd8c1dbe6cfdf9 : SUCCESS in 4m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf04d64d197244faaa4ca48c7aa43e79 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78ec865fa9274abe86e8cee850c408a8 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c567224bc0b344d1a645e1cee0d09e5f : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69941292ef4149cbadce50d494f2a429 : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/642e495f4d604ecfb882a9e39cbde81c : SUCCESS in 4m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85da3bd1b0d74d4e99d58478f2cc95e5 : SUCCESS in 1h 02m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8def0c1c99d746caae01ef213d1800ca : SUCCESS in 30m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0779603a71a4000a18c9aa929dc5a4a : TIMED_OUT in 2h 02m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cc178eb3bdaf479582cdfeb3c30ccc8f : FAILURE in 33m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/847b377a3f094541b46db955eed7adba : TIMED_OUT in 2h 02m 09s","accounts_in_message":[],"_revision_number":2},{"id":"86352621e2b64a6d37af6ddd6a697b02fa35dca8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-23 15:10:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"307f6e0ac796b22cf9514760618a6e8f35187ac8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-08-23 15:15:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"07bd6c90099b558a0016b77b266c72df73a8616a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-23 16:01:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/96/678196/4/openstack-check/tripleo-build-containers-rhel-8-master/4684757/ : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":4},{"id":"a31f56cc9f1fe45a8f73f5fedea59e8ab2420e81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 16:46:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f7b11ea9e5d04847ba20d83a84249919 : SUCCESS in 23m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/284facba599f4cf7b0a83916444b69b1 : SUCCESS in 48m 38s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fa45f38fa68244cc8c6267944b0891a5 : SUCCESS in 1h 19m 44s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/d92ffcfe765346a39e0af729fe9e4bae : SUCCESS in 1h 01m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/19029558b1db4d37b504c34c9203ab14 : SUCCESS in 1h 24m 00s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/6d80fe54570e4844b2949c9a1ef9e998 : SUCCESS in 1h 02m 38s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ab92390a1bf4451f854e76eb6d1223e5 : SUCCESS in 35m 41s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31859ff9c72a46438bc561a9cf14e19a : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2375b4a23c244f7d8d907d08a577e5a2 : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d12621d7c7594db6a80099414b224dd1 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/480d1109e6804029a58997f98fcd1dce : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af8af86e0e934bf784e0e02d62cc66a0 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/598795dfa07d45e5902a5b01165e10b6 : SUCCESS in 2m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5c13a3d97e74ade9baeac151bdb57e7 : SUCCESS in 5m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/16b23561e20e4d63b657163a2c44c710 : SUCCESS in 24m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3280810dabbb4fa0be5efa14741b8bb5 : SUCCESS in 34m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e20af6843ea54a128a8be5c6f47c89a8 : SUCCESS in 52m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/963c02de88504ca09eeca40145588738 : SUCCESS in 52m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/985f650507014770a50270e298af347c : SUCCESS in 1h 26m 40s","accounts_in_message":[],"_revision_number":4},{"id":"4a9936571fceae1b287b09c80f7f068b25d936bf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-23 17:12:21.000000000","message":"Patch Set 4:\n\nI read our docs again and I think we need to take care of this situation more thoroughly. Nova should make it clear that it is using the rolling upgrade procedure now and only rolling one. We could amend this by modifying docs that rolling upgrade consists of repeated boostrap (however dumb that sound) and OSM (should not be ODM?).","accounts_in_message":[],"_revision_number":4},{"id":"525a4e8ab8e0ab198bd42b4ffb77fd8ea07ba546","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-27 11:11:32.000000000","message":"Patch Set 4:\n\nI thought that a potential use for the legacy upgrade might be fast-forward upgrades. I think you stop all containers, then it just looks like a normal deploy, but you still need to do OSM. We don\u0027t support FFU currently, but there has been some work done towards it.","accounts_in_message":[],"_revision_number":4},{"id":"fe6330fb98b2c53c1b100debd656c16ec9baacd6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-26 13:29:10.000000000","message":"Patch Set 4:\n\nwhere are we with this currently?","accounts_in_message":[],"_revision_number":4},{"id":"4967d0d0d9bd4b34327f2a07c1b1ba0d446a2692","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-31 11:30:00.000000000","message":"Patch Set 4:\n\n(2 comments)\n\n\u003e where are we with this currently?\n\nIt\u0027s not necessary, but we could clean up some of these. I suggest we wait for Train, to improve compatibility between images and ansible across versions.","accounts_in_message":[],"_revision_number":4},{"id":"77a31bbdc1cd3195f04d560f8d2fac0115540073","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-10-31 12:04:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4107a55c6c50219494e279e7cd9ca815527dd529","revisions":{"f0e0bc096668aa587a62b6e5e5e70dfffb82f2d2":{"kind":"REWORK","_number":1,"created":"2019-08-23 10:36:30.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/96/678196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/678196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/678196/1"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-08-23 10:32:10.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-08-23 10:35:13.000000000","tz":60},"subject":"Remove Nova legacy upgrade","message":"Remove Nova legacy upgrade\n\nThe legacy upgrade procedure has been removed from Kolla Ansible\nand is no longer required here.\n\nChange-Id: Iac142c056938f1ecd45807163460b040250882fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0e0bc096668aa587a62b6e5e5e70dfffb82f2d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0e0bc096668aa587a62b6e5e5e70dfffb82f2d2"}]},"branch":"refs/heads/master"},"ba4eea45e7b5e20f2c5068081fcec62265f2621c":{"kind":"REWORK","_number":2,"created":"2019-08-23 13:02:16.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/96/678196/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/678196/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/678196/2"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-08-23 10:32:10.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-08-23 13:01:35.000000000","tz":60},"subject":"Remove Nova legacy upgrade","message":"Remove Nova legacy upgrade\n\nThe legacy upgrade procedure has been removed from Kolla Ansible\nand is no longer required here. This also removes the online\nservice migrations from the conductor container which are not\nrequired.\n\nChange-Id: Iac142c056938f1ecd45807163460b040250882fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba4eea45e7b5e20f2c5068081fcec62265f2621c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba4eea45e7b5e20f2c5068081fcec62265f2621c"}]},"branch":"refs/heads/master"},"084d94cabcebfe8f50c686d8b1c76441f07eaac7":{"kind":"REWORK","_number":3,"created":"2019-08-23 15:10:26.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/96/678196/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/678196/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/678196/3"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-08-23 10:32:10.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-08-23 15:10:08.000000000","tz":60},"subject":"Remove Nova legacy upgrade","message":"Remove Nova legacy upgrade\n\nThe legacy upgrade procedure has been removed from Kolla Ansible\nand is no longer required here.\n\nChange-Id: Iac142c056938f1ecd45807163460b040250882fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/084d94cabcebfe8f50c686d8b1c76441f07eaac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/084d94cabcebfe8f50c686d8b1c76441f07eaac7"}]},"branch":"refs/heads/master"},"4107a55c6c50219494e279e7cd9ca815527dd529":{"kind":"REWORK","_number":4,"created":"2019-08-23 15:15:19.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/96/678196/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/678196/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/678196/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/678196/4"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-08-23 10:32:10.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-08-23 15:15:09.000000000","tz":60},"subject":"Remove Nova legacy upgrade","message":"Remove Nova legacy upgrade\n\nThe legacy upgrade procedure has been removed from Kolla Ansible\nand is no longer required here.\n\nChange-Id: Iac142c056938f1ecd45807163460b040250882fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4107a55c6c50219494e279e7cd9ca815527dd529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4107a55c6c50219494e279e7cd9ca815527dd529"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
