)]}'
{"id":"openstack%2Fkolla~678910","triplet_id":"openstack%2Fkolla~master~I3000136fa883f510b24133907a66efd194b40266","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I3000136fa883f510b24133907a66efd194b40266","subject":"disable \u0027hacluster-pcs\u0027 on aarch64 due to lack of package","status":"MERGED","created":"2019-08-27 17:03:19.000000000","updated":"2019-09-03 11:14:28.000000000","submitted":"2019-09-03 11:12:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678910-1567509139023-727564a4","meta_rev_id":"a05a608f4d1375e1f710eb5b32c4ad72b8e099b5","_number":678910,"virtual_id_number":678910,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"date":"2019-09-03 07:49:51.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-09-03 11:12:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-03 08:03:15.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-31 06:50:35.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-31 06:50:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-09-02 12:45:40.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-08-31 06:50:35.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-27 17:03:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-08-27 17:04:12.000000000","updated_by":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"reviewer":{"_account_id":16006,"name":"goldyfruit","email":"gaetan.trellu@incloudus.com","username":"goldyfruit"},"state":"REVIEWER"},{"updated":"2019-08-28 07:27:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-09-02 12:45:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-09-03 08:03:15.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-03 11:12:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a740b8818c809411c58431af82d4cb410afddb5c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-27 17:03:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64205e25ec3a9221bc3e074994653f48993334ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 18:34:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/34df06d98473438889d6cf1ef4947343 : SUCCESS in 45m 56s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7bf9a6b2924c45558b8ffd5fa4b7bed9 : SUCCESS in 45m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fab86afc2a6c4e46a52370b9a673367d : SUCCESS in 1h 13m 31s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/3169cdf6e6644674acd2e8455b6f8cd6 : SUCCESS in 1h 15m 11s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/905ecb1a66594c0d9c8ab4fc0f6197ce : SUCCESS in 1h 20m 46s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/13a55250493241ab8728e23965fee213 : SUCCESS in 56m 23s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a1013ca934f2488dbe16c2dc07bbc512 : SUCCESS in 33m 10s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39100b8b74c1440c98cab71d0ecd55df : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4183ab4b13d4f6bb20415e0ca63932f : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/94431962a77843c9805ed06fc1319b3b : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fed4d7dc869f40159914853bfae0bacf : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c185ec511ab44723842f3c6ef239bc13 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a0522a6f31b41c4a536fb324e6c2432 : SUCCESS in 4m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/13529334957d4632b5b3af364da41928 : SUCCESS in 26m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8dd559635f21430fb4deb473cc877837 : SUCCESS in 32m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d93c275dc1da4e3caa1495b765ca95a8 : SUCCESS in 58m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5be8b62f8faa482f81c01690b8307c8a : SUCCESS in 52m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/941a2acd0cea474a9f4556930ebff153 : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2441e831a42079f2daf7af1447ba9233661fab26","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-27 19:11:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/10/678910/1/openstack-check/tripleo-build-containers-rhel-8-master/dcc1adc/ : SUCCESS in 40m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8d8969fba502a03ea19b417a44b8633bdc7610d1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-28 07:27:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3fff03554b601ce118c9fbf30129623c0fb85a26","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-28 08:49:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c97fd257954d0dd9537490e99d8dd517d6a6d88","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-28 08:49:47.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99690cbefd11dc7f3cec10530d6e24d1be29e434","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 08:50:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7171f500f64387d53cd8751a34c0e3186a9fb5cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 10:19:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3101abe9e8814c8d99cf60d96fd7abc0 : SUCCESS in 33m 37s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/bd6ad8010d7741c98ee9f7d856e48015 : SUCCESS in 1h 24m 33s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/5c341468de254d4faf04e0f9ad29474d : SUCCESS in 1h 24m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/76ae2124cf3f4a4d82b663ac5a75cff1 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a850484a30994b368c4944e8b9d64064 : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a021c59ed8c6478f927441b5e42dd973 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/716955d919b04f7fa289f5161162442f : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ab12e358afa4f2794fb166c2dbf1fb9 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/390fc726aaa943c6bed3439461c8ca68 : SUCCESS in 3m 52s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0113efbf369e49988741c353de4b0a98 : SUCCESS in 31m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ea97871356b94a66ad60b9c0d65b69b8 : FAILURE in 32m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1936fd1e16e04d27ae4795e4da4ee1aa : FAILURE in 52m 34s","accounts_in_message":[],"_revision_number":1},{"id":"940f1a12b92c48343ece0923a1042a6276500235","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-31 06:50:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"36b11d1ba649fe42c0d43e7fbb321cf609b77220","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-31 07:20:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/10/678910/2/openstack-check/tripleo-build-containers-rhel-8-master/bc1c56a/ : FAILURE in 28m 48s","accounts_in_message":[],"_revision_number":2},{"id":"bbf91462582ee4feccb5d626c10fb303040c8a97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 08:37:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/32ef85d7c3234dfcb0ec81a8bbcbda89 : SUCCESS in 53m 42s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f3f3d52fef094ea8ac95a59722ddb643 : SUCCESS in 1h 00m 09s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4fdb9d6c7791415ba6f332984ec5ae61 : SUCCESS in 1h 36m 17s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0931dfa80f354323959d78b134a38bfa : SUCCESS in 1h 25m 44s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e8ac00289b764c6a958645cbf2bba7ea : SUCCESS in 1h 41m 27s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/0c85686e73c44320bab75a7721d48464 : SUCCESS in 45m 33s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4d4e03c2104b46a8b425d2ac8f7369d7 : SUCCESS in 27m 32s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1559de1be184e96a93c37473d940c34 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bed400bf6a9048e3ad01ec8d53ec7fd8 : SUCCESS in 5m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9af024f6be54419a9f1cdc1503bd272d : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/250e634c284d48a4b4692dc607a30958 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/36adb842936f4fa38655705a17d19566 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8e840268c043c5b5247e5a6a6c762f : SUCCESS in 5m 23s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ee865d05140146f2a39d170e3f1c087f : SUCCESS in 21m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3916346f300b4a07b769cb1b5c2e44be : SUCCESS in 46m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f02c503d98814ceab20acced20606974 : SUCCESS in 59m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/accf364d425b4763918ad60042736f51 : SUCCESS in 1h 20m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bf91b939064a40ba9d8cd8120861e21c : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":2},{"id":"344d76f8d0a1f395ccb5e337e258bdc0c56c8bc2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-02 12:45:40.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c27aec53aaf56c548a8095d37d7b061e21e5550b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 12:45:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f3fc19d0df666e098c6e73d13a77ca52115d9e89","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 14:27:47.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/4d200b68cbb64ecd8142ec17a9cf4e1b : SUCCESS in 1h 01m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/c1206e6e09d14a88818cc16af5d02e1c : SUCCESS in 1h 28m 02s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/81a3600b3afc43c9942ccd39a8d3b734 : SUCCESS in 1h 36m 42s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/455bc86a3bc5413c9ade25a8be115200 : RETRY_LIMIT in 9m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe2b18167a3f495984c98c4b42c50aa3 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1eeb65470814307ad83c6c078f12460 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/43d1690f1f3a4b6abfe382ea7207426d : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5623121a5db4718937ed72bd295675a : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84970fcead514fd0aa9534970b4576a4 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e634c60f328f4404bfdd3133884f7e52 : SUCCESS in 4m 09s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/286fa306c73f4e17af6943ca6da35f16 : SUCCESS in 31m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc0be85da1f74b31a53d5ac37c2004a0 : SUCCESS in 54m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/30bdcb82d4734e4386e40640b30230d5 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":2},{"id":"df605a273adae19951bcd9d189ed17b3ecfa0092","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-09-02 15:16:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6e3c42896669c82d59fcd7f0fc61e133eae84334","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-02 15:54:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/10/678910/2/openstack-check/tripleo-build-containers-rhel-8-master/154e7ee/ : FAILURE in 36m 39s","accounts_in_message":[],"_revision_number":2},{"id":"effd2de9fdef3a0821ff142ec1afcc879144e206","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 18:11:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6d292d92c35e4243b4b3567361dcbb53 : FAILURE in 15m 19s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/c3970f63e63d47e092aff9665d3e3b9c : TIMED_OUT in 2h 52m 18s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/70d8bda2ad934eeba2536a8af93b0808 : FAILURE in 19m 24s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/938c746f7359451eb644102a784d6a5b : FAILURE in 21m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/9dbeac6bd3a54604b6d66550091fa228 : FAILURE in 19m 31s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3dca1e59f0c04277bf3b6b278964e14d : RETRY_LIMIT in 11m 20s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/787ad7c3fd6d48069f66c0f37ed6582e : RETRY_LIMIT in 2m 34s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2613800b568e419a823782a6ffd984b5 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12d71a147a4141169349a79d53b14634 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23275c8ab17c4586bd41adc824dd709b : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4cc9de36e124a2ab934355b87656875 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14aede797a284d678bf548eea034eab3 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1abde24406fb4f9eb0b3b4297c68367e : SUCCESS in 4m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d237c43ea69640eea3f0a08b9d915fd8 : SUCCESS in 28m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/bb8b094a8788474b992efb9519ba6260 : SUCCESS in 37m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f51b35e2ca04c299408f2f389942260 : SUCCESS in 57m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc7dadf1c1b54908b89a18c8400c1b41 : SUCCESS in 52m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/60520b05741a43698bec5cf673e5dc96 : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":2},{"id":"bc52cf425793123b8d07bf4246253886c129864d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-09-03 07:49:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d65ac3da220636b20a82e6d4b4e91f3914ac8b85","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-03 08:03:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/10/678910/2/openstack-check/tripleo-build-containers-rhel-8-master/2ec8cb0/ : RETRY_LIMIT in 2m 04s","accounts_in_message":[],"_revision_number":2},{"id":"70b03c231cd15bd5442d38b62f8f4fa57231c995","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 09:27:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6d4eec10f4cc4dccb7443bacbd5786fc : SUCCESS in 47m 41s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/a85e105556724d25952740ef26126cc2 : SUCCESS in 49m 43s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3247e138b49b4407bed8939b72bf2585 : SUCCESS in 1h 18m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/7d640b27a249411f9981f0e4ef18dfb2 : SUCCESS in 1h 14m 27s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a10b8a2152ba4816af35410abc67a008 : SUCCESS in 1h 35m 40s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f08e99509ebd4643ad8e8d2719a356a5 : SUCCESS in 52m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a76f5be514f24adcbf889bfdf14e610c : SUCCESS in 26m 54s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24486f66a7144ebbbef1c0c310f6c496 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c3294ffd3bd453b9fc5828c3137e6de : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16859d5bb3ce4cb0aac707ed6f42584f : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7e5badb5a484f3db838303741946e40 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/154e1115ac0541578758a54eab1905ba : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e636a277c434b3bbf311f36ef459052 : SUCCESS in 4m 16s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/420f8c67d57147f1929ca72162ed2f00 : SUCCESS in 28m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4edb42a7f824a678db18b9cee94bd22 : SUCCESS in 33m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0210f76ae0f944dea32c87390fe7bfea : SUCCESS in 1h 01m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9c28e02c5ec44aa6be1a1fde57ae5316 : SUCCESS in 53m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/da162ffc353a4ecea265172b861f1a59 : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a2266821c64275be6b2bb6d3a99c047214428b03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 09:27:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5858dbc98655fd04bb9180bee6f82b58e0fcb1ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 11:12:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/d6f09ff67f4d4373beba3cd657962fa4 : SUCCESS in 57m 47s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/1fa2c7384ecc4ea88a077ab9655df284 : SUCCESS in 1h 20m 01s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/2dc73a5644a54b2690fb05619068d77c : SUCCESS in 1h 34m 26s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/90a3b0ed6872422cbb0f1fce522e02c5 : SUCCESS in 48m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46bd0bcdb1ed4df48187fbeba4c080a3 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d6135fb8ab14d5198b90a318bee3059 : SUCCESS in 5m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4392fa47f4f6449d819c623568c14287 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b3b085c1d1f496b9b214b7d73339bc6 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17d62394484c488f82b99e2687a7536f : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c142f687e849118d431e70c67cb98b : SUCCESS in 3m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/83b5a545e55e49b6abc91fefaf3f4aab : SUCCESS in 33m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa73f56f1306404e8c792f3b537e8ec1 : SUCCESS in 53m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/14c84f426f134c7aadc4e382d2de293c : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":2},{"id":"4dfa132bc43c15e41cae66cb7f1cb2cfce4c9e48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 11:12:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a05a608f4d1375e1f710eb5b32c4ad72b8e099b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 11:14:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce87689a03344b4fb3058a42fbcab039 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8d74a5ea095df32f2725d771bd05059b96efb3cc","revisions":{"f0feccdb714fc4cd75db97f6b0132f93bc770cd3":{"kind":"REWORK","_number":1,"created":"2019-08-27 17:03:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/10/678910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/678910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/678910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/678910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/678910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/678910/1"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-08-27 17:03:08.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-08-27 17:03:08.000000000","tz":120},"subject":"disable \u0027hacluster-pcs\u0027 on aarch64 due to lack of package","message":"disable \u0027hacluster-pcs\u0027 on aarch64 due to lack of package\n\nChange-Id: I3000136fa883f510b24133907a66efd194b40266\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0feccdb714fc4cd75db97f6b0132f93bc770cd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0feccdb714fc4cd75db97f6b0132f93bc770cd3"}]},"branch":"refs/heads/master"},"8d74a5ea095df32f2725d771bd05059b96efb3cc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-31 06:50:35.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/10/678910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/678910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/678910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/678910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/678910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/678910/2"}}},"commit":{"parents":[{"commit":"761d58717731dd72ce50f6dbb6cd8e2b5e608060","subject":"Move to MariaDB upstream repo and new version (10.3) on Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/761d58717731dd72ce50f6dbb6cd8e2b5e608060"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2019-08-27 17:03:08.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-31 06:50:35.000000000","tz":0},"subject":"disable \u0027hacluster-pcs\u0027 on aarch64 due to lack of package","message":"disable \u0027hacluster-pcs\u0027 on aarch64 due to lack of package\n\nChange-Id: I3000136fa883f510b24133907a66efd194b40266\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d74a5ea095df32f2725d771bd05059b96efb3cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d74a5ea095df32f2725d771bd05059b96efb3cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
