)]}'
{"id":"openstack%2Fkolla~679209","triplet_id":"openstack%2Fkolla~master~I16e4c40208912691e3015bfe2065ea0ce2358edb","project":"openstack/kolla","branch":"master","topic":"feature/ubuntu_mariadb_repo","hashtags":[],"change_id":"I16e4c40208912691e3015bfe2065ea0ce2358edb","subject":"Move to MariaDB upstream repo and new version (10.3) on Ubuntu","status":"MERGED","created":"2019-08-29 09:07:27.000000000","updated":"2019-09-10 19:36:08.000000000","submitted":"2019-08-31 10:03:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"679209-1567245835152-26e757ed","meta_rev_id":"708db2cbf6c4c0cde5cb585bc105381417da303c","_number":679209,"virtual_id_number":679209,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-31 10:03:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-31 06:54:09.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-08-31 06:40:57.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-08-31 06:41:09.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2019-08-31 06:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-30 13:28:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-08-31 06:41:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-31 06:54:09.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-31 10:03:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"85af90efa519efc4a05b61e487efa8249d797517","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-29 09:07:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0c8a4f37557753139fcee53142b502dc78ba041","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 09:49:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/09/679209/1/openstack-check/tripleo-build-containers-rhel-8-master/9799dda/ : SUCCESS in 39m 59s","accounts_in_message":[],"_revision_number":1},{"id":"eef956d61d4178f0d92750949cedaa489d0c668e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-29 10:25:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd317dc0e08d43cdcf559b7e42785ad74935bd31","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 11:08:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/09/679209/2/openstack-check/tripleo-build-containers-rhel-8-master/36fc646/ : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":2},{"id":"21e2f312da86b28bc53dcc52adfe639acec5fbcc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 12:27:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/e58833532d6c4dad83c851fce5b872d0 : SUCCESS in 47m 08s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/0c957d7215b64fc393efccd3fd85920e : SUCCESS in 59m 07s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fa78c076c2a64d3491cc0c34de426695 : SUCCESS in 1h 26m 49s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/e436db4edf924abda9022ae356cd35b7 : SUCCESS in 1h 17m 50s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/83eae708e0a448a9ad054b12af178b8c : SUCCESS in 1h 19m 02s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/558e4eedda934f1aaa152bd20312e416 : SUCCESS in 53m 26s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/9304865b66744177b4014b3f5f5c819f : SUCCESS in 46m 04s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/72361127818a47b79bd79f9fbeaa5d15 : SUCCESS in 1h 44m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/8e91758277e14af1b8c93692359a9eb7 : SUCCESS in 1h 54m 44s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/4368b6e326734ff0a28ed64ed711bd4b : SUCCESS in 1h 50m 17s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/e71cb4bdfdb142d5843a2290f9be720c : SUCCESS in 1h 35m 10s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/f08167f5f65d4b5d9465cde5a3589fa5 : SUCCESS in 1h 10m 44s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e8db82e4da9a4b5cb73a9bcf70d5b426 : SUCCESS in 27m 20s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60ff806ec7d14a49b9d40e0a46042975 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4fa0a274b314b96bdd57b2982847d8c : SUCCESS in 5m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c8ea4c6283746a1a00ea0b234765b9b : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3fc02bf6f374f05ab9e50e35c437fa8 : SUCCESS in 4m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d65c30b2d5f4daf8db4777714087454 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c355d4b17734a88b4bc3311dba1198c : SUCCESS in 4m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aaf95b0e9b37474cbef1cacc1138b2b5 : SUCCESS in 28m 08s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/53c12d79ed95474db12dce53befe75f3 : SUCCESS in 42m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5d83eee4df242aa98b35186dff4f608 : SUCCESS in 52m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b149fbc5294405cab12e3e8ae92fb42 : SUCCESS in 54m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0d7de92522d3481eaa640266eee373d0 : SUCCESS in 58m 57s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/43965c7cf09c42a1a98f19b080415a40 : SUCCESS in 1h 00m 45s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/44b0e44f23364df395634e29ef8edef9 : SUCCESS in 1h 21m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/7058cbf0dcc44f14bda872586109118a : SUCCESS in 1h 40m 41s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/2c695e928df74cadacdfc16057970024 : SUCCESS in 1h 08m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/fc14bab8df1149dfb673179bd0bf1d01 : SUCCESS in 1h 02m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f1db6154aa9121ba6cf6297d657ce03215099d7e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-29 14:35:46.000000000","message":"Patch Set 2:\n\nrecheck\n\nmake it run a few more upgrades now","accounts_in_message":[],"_revision_number":2},{"id":"1fb6b05ef18fde060b0aed5e4eca3bd6221182aa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-29 15:21:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/09/679209/2/openstack-check/tripleo-build-containers-rhel-8-master/794c999/ : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":2},{"id":"fdaf56fbc801771dc312b9eb60b6472c21a71fd1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-29 15:34:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"08d2ad09b1d2c890a3fb538e4b62a2d94e76426f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 16:50:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/1ee473b9ce0543e29111a5c9922d8760 : SUCCESS in 42m 05s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/96d39465eda141598ae7755396b88bb6 : SUCCESS in 43m 16s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/9af19923187c444786f4b527325666a9 : SUCCESS in 1h 18m 37s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8305becf49ab46f08167cf85b99c5995 : SUCCESS in 1h 00m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/aa07b5553dc943d5afba1973ff48eb30 : SUCCESS in 1h 23m 04s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/cc279e38a15e42e1b0d8f72e4e849425 : SUCCESS in 41m 24s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/ed1155cc218d4ec48e05685622b9eb22 : SUCCESS in 46m 31s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/d09d87c3dd8640bc8a4dc3704357d6d6 : SUCCESS in 1h 56m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/05b7865630e24e33b721325425b45560 : SUCCESS in 2h 05m 40s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/bc60905708e04cc4a9298084f1a6e352 : SUCCESS in 1h 29m 56s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/da063444cb50402ca47b58ef77ad9ec8 : SUCCESS in 1h 52m 00s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/b8b29ac3cdf044ebbf3e326461622c45 : SUCCESS in 1h 38m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0f2e56319d4e4b3e80e9cb7e7858127b : SUCCESS in 23m 09s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/471abe9fc94044eba1ea4fae9cf16fbd : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/528f16ba8c30438b9ef6b09b6bf717dd : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/588b6b8f05ac4121a86e10879368c47e : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7153617a3cd44f9597a0b893fe5da901 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/701e9dc1cafc4e439c0a9c6800148b43 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/513dff44e420471283d3a634d9bf37c7 : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a5b24f498ef34142b8e6bcc41ffe5563 : SUCCESS in 25m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/385d83fd341447f7935984148b6fa00a : SUCCESS in 34m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/83c9fa9cf99c4b36bc713ed0a2673ef4 : SUCCESS in 50m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2727694edda248ac9cf7c4c740f0db6d : SUCCESS in 54m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/11534bb4a0a34021aad70604e2e637d1 : SUCCESS in 54m 58s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/cc32960625d743fb9ad7d81408437134 : SUCCESS in 57m 53s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/85fdbf3daa994fdc85009fdf772b2a8f : SUCCESS in 1h 25m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9442db5bdb3544c2adbcf438ad1eb22c : SUCCESS in 1h 36m 32s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/8b452f6e40b047168d7757a62b714109 : SUCCESS in 1h 19m 51s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/921f29332fac4e209f63b751afb63ff9 : SUCCESS in 1h 01m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8332a53c2901ff1a4e5320bc0b6486fc368012fa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-30 07:41:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"660555d950250b4b33f983462468a7174f68a1fa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-30 07:42:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"960f8a3d5b04f976c217c3bbe6e52329d83f3050","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-30 07:44:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0abe9b8235ebc480d64fb89bb6cc5b29713e9224","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-30 08:22:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/09/679209/4/openstack-check/tripleo-build-containers-rhel-8-master/6e2a1c0/ : SUCCESS in 38m 00s","accounts_in_message":[],"_revision_number":4},{"id":"8a7d55d7556a81fe232deeed054b786125b42754","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-30 08:29:04.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"28492590195eacc7fa25709e131f2c5ed2675371","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-30 08:33:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d7788b95040c679cbb0114a35ede27b697a24025","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-30 09:27:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/09/679209/5/openstack-check/tripleo-build-containers-rhel-8-master/1a7acdc/ : SUCCESS in 53m 18s","accounts_in_message":[],"_revision_number":5},{"id":"86c6dfda25d47d10dc993f61d960bfb242c3e77c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 10:42:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f5d2e32bc49f4d648cbf999b9b095cec : SUCCESS in 42m 37s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/263ac969bb754843bcf85c1dadc4c4bd : SUCCESS in 56m 21s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/7e277fcfe1fe4c169442c8aa586ead4c : SUCCESS in 1h 31m 14s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/b388c6e6150340d6b1aa4408cb4bfd70 : SUCCESS in 1h 10m 03s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/c49d3b9afa95436f8e4b16dbef17828a : SUCCESS in 1h 35m 44s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/c68e650bda004fcfbc015f1b212dfbaa : SUCCESS in 55m 40s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/2a9c9f0647284f5998c34a308d2b282a : SUCCESS in 48m 03s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/459e45a16d244f15ad01cabd5bc47105 : SUCCESS in 1h 56m 28s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/5dd1ae833ecb43b19d306bb974b688ff : SUCCESS in 2h 00m 56s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/7bd9b4b4e9904ad687bcda0c34e61c12 : SUCCESS in 1h 29m 42s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ea3fb7043bf24ceba8d2a11568b8794e : SUCCESS in 1h 30m 47s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/fb486e5d84964de98fa527cd93e37df2 : SUCCESS in 1h 24m 43s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/62d809b283fe43f3b94bc5888da96df9 : SUCCESS in 24m 52s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36dba9e67bdc4fdaa93d0515e8369797 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3aa9e72202e48979baba6cf94061548 : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e05ce4838f794e2e8ad75376b03d82c7 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/524a2f81253246fdb6c78675bc9c6665 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/71548010080b43d08d313bfc06a16178 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6143dc2f8c545db9e1b70aed1e14fb1 : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/05b2d359633a4b3fbd2f7bc96548eb1a : SUCCESS in 25m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1c37b7517744227b1b47408ada2bc63 : SUCCESS in 30m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/10598a8aa388429e846701891872f08b : SUCCESS in 48m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d1efad8aad564fb185cadcdbcd78eb2e : SUCCESS in 48m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ab2200bab6347b29957701943b21e0c : SUCCESS in 51m 31s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/5b8af8fac23249abbf2ebcc2dff64a8d : SUCCESS in 1h 08m 55s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/7f85af31c2c147db92a59a71cf2c87a7 : SUCCESS in 1h 08m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/db13eb535654404d8b2748e67859e5a3 : SUCCESS in 1h 38m 44s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/de01c3b0606948449d5b112dde3e3bfe : SUCCESS in 1h 03m 05s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/5c7c61cebce24fdbb559bc3112bc8adb : SUCCESS in 1h 02m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5b8036565191021eab22631a80d793e0cf8450b2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-30 10:47:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f406488d3bd284c933cce3292ccc7c497f5192bf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-30 12:09:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b8f8fa6489a8c2d2e4af14358e3fc46ac4550e09","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-30 12:43:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4cf83e681b9fa70084d813d9917ac35cde843819","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-30 12:58:19.000000000","message":"Patch Set 7: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"8a75c7c741fdbca8145d6695f7a9edddf34f603e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-30 13:08:47.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"bc7185be295f5cf4f686770e1ef090c279fef63c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-30 13:09:25.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"fbe53e26fd62f8fa99685c1522d38d92e497bf39","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-08-30 13:28:00.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"5e827b90436440d4501416a4c52fc15404272759","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-30 13:47:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/09/679209/8/openstack-check/tripleo-build-containers-rhel-8-master/1e37f6e/ : SUCCESS in 37m 15s","accounts_in_message":[],"_revision_number":8},{"id":"315af2f1eeac9c9132d9633fdd350dc8f87df327","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 15:20:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/8c30090a006a4a46811d462c2a6143c3 : SUCCESS in 57m 36s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/f382ed13afb14b94842a9f32a29362b4 : SUCCESS in 1h 05m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3696d00b328340e8bf4c91f6313501e7 : SUCCESS in 1h 17m 00s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/a9121689edc24222bca2421c948da240 : SUCCESS in 1h 11m 30s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a4524ea98586449f97b1506799a5b63b : SUCCESS in 1h 28m 53s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/600cd213cf9c4200a83bbfa899aa7d7d : SUCCESS in 47m 57s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/5b75c96e113d485890f6139eb48dd6bb : SUCCESS in 46m 13s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/eb47c9b6989142b881c4d0f4f4094917 : SUCCESS in 1h 52m 23s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/10ad4c07748d4b93aa6665e27ae8d14f : SUCCESS in 2h 06m 53s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/acd328e8aec6485da819c17ba952fbea : SUCCESS in 1h 32m 22s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/33783d02ab254a68acb6bbc8c61cb8ef : SUCCESS in 1h 38m 28s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/ea586a5f2fc848abb9b78dc8e3d76a35 : SUCCESS in 1h 25m 08s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7229f6d7ee6d4ca2afe49f31f08afbea : SUCCESS in 30m 18s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6a6222053588467db119db86a1031819 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b152c64de4fc4a5b86fc8107fe022908 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/07cb0fb3fbea4450875ea86b66dff13c : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7700badb20b146b4b5a8ac27d0caac71 : SUCCESS in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cf1d233c6d2442a859f4bde7725dfd5 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab26c2b59cd441d82993937565c0353 : SUCCESS in 3m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7a9cd78fcab465ebfb6938edc14315a : SUCCESS in 4m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/eb1fd10128bd404aacb9f3bb8f6cf3b5 : SUCCESS in 38m 01s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b4ba03a0a2c04f00be7d1aa000e20952 : SUCCESS in 48m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ed9aa4a263c4186ad9b4ebe76740a19 : SUCCESS in 52m 09s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c6293f781d464bb791289a7fb7ab0308 : SUCCESS in 50m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2dc235bc66d345348f571c0d190fdef8 : SUCCESS in 57m 14s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/b2719d5be5354589b9f8421cb64d569d : SUCCESS in 1h 22m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e1f1f9f5016e4e6e8794b63caef9075b : SUCCESS in 1h 31m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/093c19bee1cc4ed4898c756bba08a5ef : SUCCESS in 1h 35m 07s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/c30e14d26ed948ad97324b49f4e69526 : SUCCESS in 1h 06m 48s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/0f678ba0cec54e33813e9d8b7d271582 : SUCCESS in 53m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c61fecdd3ffca6ca2012840d710be33e18d13214","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-08-30 15:24:15.000000000","message":"Patch Set 8: Review-Priority+2","accounts_in_message":[],"_revision_number":8},{"id":"f58772cb5747b9471e17cf8d1683091ce2fe4688","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-31 06:37:59.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"8463d8cf907bf1ac83ab7932807d392d11058475","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-31 06:38:57.000000000","message":"Patch Set 9:\n\ntrailing commas futureproofing","accounts_in_message":[],"_revision_number":9},{"id":"e61d9df6e2584de9efd12a436b7f6846fd832ce3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-31 06:40:57.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"d110bb39f934253ac3ae078aa0c344e02cb7be98","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-31 06:41:09.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"b0c3deeea0aa65256409920ea5b6d008b6af68b3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-31 06:54:09.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-build-containers-rhel-8-master http://logs.rdoproject.org/09/679209/9/openstack-check/tripleo-build-containers-rhel-8-master/d55a8c9/ : FAILURE in 14m 52s","accounts_in_message":[],"_revision_number":9},{"id":"a9381bd6ab9ddbbc37fde90115c73f7a5538e5fa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 08:39:19.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/f11b8b2feb6c42b0bcc7ad450d2e1857 : SUCCESS in 55m 00s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/bf9fcc3682444d3ea3bff5ba21500dd1 : SUCCESS in 57m 05s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/3485210997f447c7a30f814a4954282a : SUCCESS in 1h 23m 46s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/fad75d4b9c2744149b7b18df393a6a56 : SUCCESS in 1h 14m 20s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/154324ee74384adf872f545ece71a880 : SUCCESS in 1h 32m 32s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/1e1e8a2b26a241e69c3715dc0e9ba0f5 : SUCCESS in 55m 00s (non-voting)\n- kolla-ansible-bifrost-centos-source https://zuul.opendev.org/t/openstack/build/b4f1ad9bd6cc4c608a566f039c43e35b : SUCCESS in 45m 48s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/a451aa8abe2f477aad12019f9c672a10 : SUCCESS in 1h 38m 46s (non-voting)\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/e8ba3068e79a41739063e73ba81a370d : SUCCESS in 1h 54m 55s (non-voting)\n- kolla-ansible-centos-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/d31d8d592ec149d1a867e729603b3026 : SUCCESS in 1h 40m 01s (non-voting)\n- kolla-ansible-centos-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/077816f0cb15454eb19c1edc9c6316c9 : SUCCESS in 1h 42m 33s (non-voting)\n- kolla-ansible-centos-source-ironic https://zuul.opendev.org/t/openstack/build/99c87ecd9a04492296760f4e1354746a : SUCCESS in 1h 23m 36s (non-voting)\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a859939c811f4461ae53b39e85f0c938 : SUCCESS in 25m 45s (non-voting)\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/184ac13ca57149eb9027dc25213ff84a : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57de47bed83048c7b309cd3f5948c835 : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8947604038b04005a28801c86bb1fff0 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59a9552270054181821671cddf9c2408 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/063ec3d1fb3b42b9aa9331f4fd19bc01 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0075cf06f6d5411197c5d4918a6b244e : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d09362b0bfce497a941cec2ff3f39ea3 : SUCCESS in 5m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/76fc7d1e88a64065bc68f5892de2fa2d : SUCCESS in 23m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f8e7f78bc5248f3a9bc22b4e06dd915 : SUCCESS in 31m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca3a527858cf4668b8487deb7561ba21 : SUCCESS in 49m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c59716cae6dd4595a85e7c52b7613429 : SUCCESS in 53m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/232f224c32834f67a6ecdc22b3db9f1a : SUCCESS in 56m 46s\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8f6b05c3afae48baa1ec34a4ee2d2fb0 : SUCCESS in 1h 01m 52s (non-voting)\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/e6066bb344dd43999757b5a191ca2595 : SUCCESS in 1h 15m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/c94a3c8a811c41cea0e2425409ec764b : SUCCESS in 1h 30m 52s (non-voting)\n- kolla-ansible-ubuntu-source-cinder-lvm https://zuul.opendev.org/t/openstack/build/67709c8c8d224b608449b5fb2815153f : SUCCESS in 1h 03m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/8fb4af5c994a41e087d4e8e3843d66bb : SUCCESS in 58m 13s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"afa1b29a906508302c6eed6658b7ccdf8b3431ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 08:39:30.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"f62e7ce785f99c72f8239b63e80d1cafa20b18b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 10:03:54.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/87b0a95f4ac5425e9b95338ebb42b491 : SUCCESS in 40m 59s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/6cc64f3bec1b406f82727a273ca74017 : SUCCESS in 1h 10m 54s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/0ab0b4480ccd459da72112c23712d0a8 : SUCCESS in 1h 19m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d052c8ed771b4e7d9149deb99f38a015 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9629a94f3a76473cb9b6e6afc77e76f9 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6434648310eb44b0b53dcd0742202f30 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12105278ef6042e5a35e33f51e8cdc39 : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49b5bbb1a0034a65a371cc85d22c62c4 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93170f1e422143af93ef12fc559f27b5 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e01ea7f229d74d32840b3e8f3c32209d : SUCCESS in 5m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7e6d43c08e041ccbf7e7acb2d92b410 : SUCCESS in 23m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9cd9e23d3cd45a3b854feba78ada47a : SUCCESS in 50m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0808ec2602db45a59c372e9a1dd9e7eb : SUCCESS in 56m 04s","accounts_in_message":[],"_revision_number":9},{"id":"41f1b9d981ca618f22aba1808a587044f50dee60","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 10:03:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"24aea02cef731b26ce6ebdf42184fed98ead5d24","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-31 10:05:25.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea591d5dbf942a49a8afe1970939d52 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":9},{"id":"fa27deab83e3725903c8fef18979519cbe20f11d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-09-10 19:36:08.000000000","message":"Patch Set 9:\n\nbackground: https://etherpad.openstack.org/p/mariadb-upstream-yes-or-no\n\n(just cleaning open tabs)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"761d58717731dd72ce50f6dbb6cd8e2b5e608060","revisions":{"a3232cb6f75edb68c30a71b4a60fd3b04ce1c795":{"kind":"REWORK","_number":1,"created":"2019-08-29 09:07:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/679209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/679209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/679209/1"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-29 09:06:46.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-29 09:07:25.000000000","tz":0},"subject":"Move to MariaDB upstream repo on Ubuntu","message":"Move to MariaDB upstream repo on Ubuntu\n\nChange-Id: I16e4c40208912691e3015bfe2065ea0ce2358edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3232cb6f75edb68c30a71b4a60fd3b04ce1c795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3232cb6f75edb68c30a71b4a60fd3b04ce1c795"}]},"branch":"refs/heads/master"},"428590839a2799ccd626578333848ef4800f5dae":{"kind":"REWORK","_number":2,"created":"2019-08-29 10:25:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/679209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/679209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/679209/2"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-29 09:06:46.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-29 10:25:36.000000000","tz":0},"subject":"Move to MariaDB upstream repo on Ubuntu","message":"Move to MariaDB upstream repo on Ubuntu\n\nChange-Id: I16e4c40208912691e3015bfe2065ea0ce2358edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428590839a2799ccd626578333848ef4800f5dae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428590839a2799ccd626578333848ef4800f5dae"}]},"branch":"refs/heads/master"},"cc86457e2e7c72a15236387f3868862bbfe595d1":{"kind":"REWORK","_number":3,"created":"2019-08-30 07:41:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/679209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/679209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/679209/3"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-29 09:06:46.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-30 07:41:43.000000000","tz":0},"subject":"Move to MariaDB upstream repo on Ubuntu","message":"Move to MariaDB upstream repo on Ubuntu\n\nChange-Id: I16e4c40208912691e3015bfe2065ea0ce2358edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc86457e2e7c72a15236387f3868862bbfe595d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc86457e2e7c72a15236387f3868862bbfe595d1"}]},"branch":"refs/heads/master"},"3d681cf66d80e52facf61830955b3951a5fc7211":{"kind":"REWORK","_number":4,"created":"2019-08-30 07:44:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/679209/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/679209/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/679209/4"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-29 09:06:46.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-30 07:44:02.000000000","tz":0},"subject":"Move to MariaDB upstream repo on Ubuntu","message":"Move to MariaDB upstream repo on Ubuntu\n\nChange-Id: I16e4c40208912691e3015bfe2065ea0ce2358edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d681cf66d80e52facf61830955b3951a5fc7211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d681cf66d80e52facf61830955b3951a5fc7211"}]},"branch":"refs/heads/master"},"a65977c6c51aaae8abd7130b99647f9fa3ab5cf0":{"kind":"REWORK","_number":5,"created":"2019-08-30 08:33:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/679209/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/679209/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/679209/5"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-29 09:06:46.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-30 08:31:40.000000000","tz":0},"subject":"Move to MariaDB upstream repo on Ubuntu","message":"Move to MariaDB upstream repo on Ubuntu\n\nMariaDB in Ubuntu packages is backlevel (10.1).\nAll devstack gates are tested on MySQL 5.7 which is not compatible\nwith MariaDB 10.1, Neutron recently introduced a db migration\nthat breaks on MariaDB 10.1, but works on MariaDB 10.3.\n\nBumping up version to 10.3 on Ubuntu to match CentOS (RDO) version.\n\nChange-Id: I16e4c40208912691e3015bfe2065ea0ce2358edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a65977c6c51aaae8abd7130b99647f9fa3ab5cf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a65977c6c51aaae8abd7130b99647f9fa3ab5cf0"}]},"branch":"refs/heads/master"},"79056d7dced79e61e7d8896f319cf4e8d9b44feb":{"kind":"REWORK","_number":6,"created":"2019-08-30 12:09:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/679209/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/679209/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/679209/6"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-29 09:06:46.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-30 12:07:01.000000000","tz":0},"subject":"Move to MariaDB upstream repo and new version (10.3) on Ubuntu","message":"Move to MariaDB upstream repo and new version (10.3) on Ubuntu\n\nMariaDB in Ubuntu packages is backlevel (10.1).\nDevstack gates are tested on MySQL 5.7 which is not compatible\nwith MariaDB 10.1.\nNeutron recently introduced a db migration\nthat breaks on MariaDB 10.1, but works on MariaDB 10.3 [0].\n\nBumping up version to 10.3 on Ubuntu to match CentOS (RDO) version.\n\n[0]: https://review.opendev.org/#/c/677221/\n\nCloses-Bug: #1841907\nChange-Id: I16e4c40208912691e3015bfe2065ea0ce2358edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79056d7dced79e61e7d8896f319cf4e8d9b44feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79056d7dced79e61e7d8896f319cf4e8d9b44feb"}]},"branch":"refs/heads/master"},"face0cc99c4828084c427723f98fe0ece6e8708b":{"kind":"REWORK","_number":7,"created":"2019-08-30 12:43:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/679209/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/679209/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/679209/7"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-29 09:06:46.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-30 12:43:12.000000000","tz":0},"subject":"Move to MariaDB upstream repo and new version (10.3) on Ubuntu","message":"Move to MariaDB upstream repo and new version (10.3) on Ubuntu\n\nMariaDB in Ubuntu packages is backlevel (10.1).\nDevstack gates are tested on MySQL 5.7 which is not compatible\nwith MariaDB 10.1.\nNeutron recently introduced a db migration\nthat breaks on MariaDB 10.1, but works on MariaDB 10.3 [0].\n\nBumping up version to 10.3 on Ubuntu to match CentOS (RDO) version.\n\n[0]: https://review.opendev.org/#/c/677221/\n\nCloses-Bug: #1841907\nChange-Id: I16e4c40208912691e3015bfe2065ea0ce2358edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/face0cc99c4828084c427723f98fe0ece6e8708b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/face0cc99c4828084c427723f98fe0ece6e8708b"}]},"branch":"refs/heads/master"},"e7f9cc861b173c48624fc14355fbd54e4a618cb7":{"kind":"REWORK","_number":8,"created":"2019-08-30 13:09:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/09/679209/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/679209/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/679209/8"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-29 09:06:46.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-30 13:09:16.000000000","tz":0},"subject":"Move to MariaDB upstream repo and new version (10.3) on Ubuntu","message":"Move to MariaDB upstream repo and new version (10.3) on Ubuntu\n\nMariaDB in Ubuntu packages is backlevel (10.1).\nDevstack gates are tested on MySQL 5.7 which is not compatible\nwith MariaDB 10.1.\nNeutron recently introduced a db migration\nthat breaks on MariaDB 10.1, but works on MariaDB 10.3 [0].\n\nBumping up version to 10.3 on Ubuntu to match CentOS (RDO) version.\n\n[0]: https://review.opendev.org/#/c/677221/\n\nCloses-Bug: #1841907\nChange-Id: I16e4c40208912691e3015bfe2065ea0ce2358edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7f9cc861b173c48624fc14355fbd54e4a618cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7f9cc861b173c48624fc14355fbd54e4a618cb7"}]},"branch":"refs/heads/master"},"761d58717731dd72ce50f6dbb6cd8e2b5e608060":{"kind":"REWORK","_number":9,"created":"2019-08-31 06:37:59.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/09/679209/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/679209/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/679209/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/679209/9"}}},"commit":{"parents":[{"commit":"857cf3effcda2dcd178fd74c70d8384f0c22ca40","subject":"Merge \"Add fluentd_binary and fluentd_version labels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/857cf3effcda2dcd178fd74c70d8384f0c22ca40"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2019-08-29 09:06:46.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2019-08-31 06:35:43.000000000","tz":0},"subject":"Move to MariaDB upstream repo and new version (10.3) on Ubuntu","message":"Move to MariaDB upstream repo and new version (10.3) on Ubuntu\n\nMariaDB in Ubuntu packages is backlevel (10.1).\nDevstack gates are tested on MySQL 5.7 which is not compatible\nwith MariaDB 10.1.\nNeutron recently introduced a db migration\nthat breaks on MariaDB 10.1, but works on MariaDB 10.3 [0].\n\nBumping up version to 10.3 on Ubuntu to match CentOS (RDO) version.\n\n[0]: https://review.opendev.org/#/c/677221/\n\nCloses-Bug: #1841907\nChange-Id: I16e4c40208912691e3015bfe2065ea0ce2358edb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/761d58717731dd72ce50f6dbb6cd8e2b5e608060"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/761d58717731dd72ce50f6dbb6cd8e2b5e608060"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
